)]}'
{"id":"openstack%2Fneutron~993241","triplet_id":"openstack%2Fneutron~master~If0cbea75ae98cf153e01bcd5a7861b348cb3b248","project":"openstack/neutron","branch":"master","topic":"fix_vrf_del","attention_set":{},"removed_from_attention_set":{"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-06-23 12:52:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If0cbea75ae98cf153e01bcd5a7861b348cb3b248","subject":"[DNM] Run functional test jobs","status":"ABANDONED","created":"2026-06-13 15:05:51.000000000","updated":"2026-06-23 12:52:42.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"be6e1027267058dbcd2edf6f35e312175a546951","_number":993241,"virtual_id_number":993241,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-22 23:12:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-13 15:44:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ab19cc1f32f16a7ce0209387a58ded319385686","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-13 15:05:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b138d498433a798c438c2711d625c6fb9fb3fd00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 15:44:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a219d82fb8241dc978359f039888f58\n\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e2d4fd2beb564e36bbb5254e8c3b207b : FAILURE in 12m 59s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/53b595d0eb884206b39948b1bc286a44 : FAILURE in 28m 01s","accounts_in_message":[],"_revision_number":1},{"id":"545b67144a925e00d7f8af4c51bfb1142d03a596","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-14 18:10:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0581c4b9e6188be67c58117b8d75b0fcbef1fe32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 19:41:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/276612838b364e9bab169313d1433e62\n\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3ebc20c499124a96a7f1ae2649ae3c52 : SUCCESS in 1h 29m 15s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/e4be622c0af54389837464ae88bed9a5 : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":1},{"id":"44192bb970685eb135f2cc05e0906f04f1b7550d","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-16 01:17:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea84faed7275addf71b47dbf6c4dfa23b77fe452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 01:20:19.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/993240 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"a27e9f15ee54342c7741e2a1190a5e86eb586380","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-16 12:48:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"14d2b4513ea9badcacec141732f93a494d63eb5c","tag":"autogenerated:gerrit:restore","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-17 18:35:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"fdd1a08fece86bfc7d461fced93a95a8f65a2726","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-17 18:35:53.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5124d357cc30c0fe0340a17e2338657c0f437489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 19:26:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4844317de0c4dfdbe1e17910de67c33\n\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dc9c9aced9e24f05a6db323ca692f71b : SUCCESS in 46m 09s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/bd7aafb4a592466098bb3b78832614a4 : FAILURE in 42m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b3c53e039bfc8294775f3f8cacde7c15821bf54e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-17 22:31:41.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b2aac76265a0c8c8557f7f674499ef4b07c714d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 22:43:55.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/993854 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"8245d556bfdcc79c8fc422b8f07940c5d8068117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-17 23:06:25.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"109c303c7c06d064461cb6c8b63f14558e58438f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 00:34:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1954397d27dc410aa0b4aec40ffe4265\n\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8f3f8a2411264e80b489c03cded9d36b : SUCCESS in 1h 22m 28s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/b7577616f65140138d2e9b2fb4d08eee : FAILURE in 42m 47s","accounts_in_message":[],"_revision_number":4},{"id":"c6afd1ce60c2f0211f6a37ba332a59e76c653c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-18 01:29:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"0c0e7002ef55bfdc27f86cfa905ddf37a4950200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-18 01:30:29.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"9588962ae60f41ebecec270fad26c0f50c6ea0bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 02:23:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2c0b7ff22fa4b7fa6f9168b1af85a6d\n\n- neutron-functional https://zuul.opendev.org/t/openstack/build/10eebc6a3411494996a9f701f91f11f5 : SUCCESS in 46m 25s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/9613ab3e012d436c8c187021c456db8a : FAILURE in 25m 21s","accounts_in_message":[],"_revision_number":6},{"id":"f2ec69065117467d8e52bebe105eaa4afd239671","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-22 18:11:26.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"e90d2b603c43991427066e0371fafa309608e656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 19:20:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25182e3a5b6a4044aec2966da0ac8a88\n\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3c295f4b22d440f8ac3bac9e8bf74f37 : FAILURE in 1h 02m 13s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/93cd97e4c80e4b33be7f0a456efedf3a : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":7},{"id":"03d634f2ae11967acac308063fc4131e4d6ee78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-22 21:35:44.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"ee3055746807f66cba4a22cbe7cf3e04c72d3ec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 23:12:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/302ac926c11a4ab09e3a1da79bc6c614\n\n- neutron-functional https://zuul.opendev.org/t/openstack/build/60ef23c3893b46b2ab538a18286e0793 : SUCCESS in 1h 30m 49s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/45955ebb33164ed4af8fe99fdf20bde4 : SUCCESS in 55m 20s","accounts_in_message":[],"_revision_number":8},{"id":"be6e1027267058dbcd2edf6f35e312175a546951","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-23 12:52:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"746a90cf4549cad85a0c69c4523a3525bd3be0f6","revisions":{"c93ce6bd3f99c6229a3d81429d80526f7ba46397":{"kind":"REWORK","_number":1,"created":"2026-06-13 15:05:51.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/41/993241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/993241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/993241/1"}}},"commit":{"parents":[{"commit":"de087e8f12a916d46565a360b9a350356d8ed994","subject":"Merge \"ovn: pass gateway LRPs to _check_external_ips_changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de087e8f12a916d46565a360b9a350356d8ed994"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:04:30.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:04:30.000000000","tz":-240},"subject":"[DNM] Run functional test jobs","message":"[DNM] Run functional test jobs\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/993240\n\nChange-Id: If0cbea75ae98cf153e01bcd5a7861b348cb3b248\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c93ce6bd3f99c6229a3d81429d80526f7ba46397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c93ce6bd3f99c6229a3d81429d80526f7ba46397"}]},"branch":"refs/heads/master"},"94732112fe834b87abc31e22abe1e32ea8d44fe5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-17 18:35:53.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/41/993241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/993241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/993241/2"}}},"commit":{"parents":[{"commit":"de087e8f12a916d46565a360b9a350356d8ed994","subject":"Merge \"ovn: pass gateway LRPs to _check_external_ips_changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de087e8f12a916d46565a360b9a350356d8ed994"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:04:30.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-17 18:35:53.000000000","tz":0},"subject":"[DNM] Run functional test jobs","message":"[DNM] Run functional test jobs\n\n\n\nChange-Id: If0cbea75ae98cf153e01bcd5a7861b348cb3b248\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94732112fe834b87abc31e22abe1e32ea8d44fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94732112fe834b87abc31e22abe1e32ea8d44fe5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"653a11fc0820c06ff8bea083581d3956390a7883":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-17 22:31:41.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/41/993241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/993241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/993241/3"}}},"commit":{"parents":[{"commit":"de087e8f12a916d46565a360b9a350356d8ed994","subject":"Merge \"ovn: pass gateway LRPs to _check_external_ips_changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de087e8f12a916d46565a360b9a350356d8ed994"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:04:30.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-17 22:31:41.000000000","tz":0},"subject":"[DNM] Run functional test jobs","message":"[DNM] Run functional test jobs\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/993854\n\nChange-Id: If0cbea75ae98cf153e01bcd5a7861b348cb3b248\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/653a11fc0820c06ff8bea083581d3956390a7883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/653a11fc0820c06ff8bea083581d3956390a7883"}]},"branch":"refs/heads/master","description":"Edit commit message"},"21b80dfc407344c1c609526f36825b11779f6c31":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2026-06-17 23:06:25.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/41/993241/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/993241/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/993241/4"}}},"commit":{"parents":[{"commit":"319761f0697026227aa7b0a882bfdfddbae74f91","subject":"Merge \"Explicitly create log directories in the FrrFixture\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/319761f0697026227aa7b0a882bfdfddbae74f91"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:04:30.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-17 23:06:11.000000000","tz":-240},"subject":"[DNM] Run functional test jobs","message":"[DNM] Run functional test jobs\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/993240\n\nChange-Id: If0cbea75ae98cf153e01bcd5a7861b348cb3b248\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21b80dfc407344c1c609526f36825b11779f6c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21b80dfc407344c1c609526f36825b11779f6c31"}]},"branch":"refs/heads/master"},"bf0fcff15b10b3c59f47a19dbac9e7ab01f0e594":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2026-06-18 01:29:13.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/41/993241/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/993241/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/993241/5"}}},"commit":{"parents":[{"commit":"71f2feb57f5abe864950b6c5e3d13210c2a3cfdb","subject":"Improve _is_centos9_frr85 skip function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71f2feb57f5abe864950b6c5e3d13210c2a3cfdb"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:04:30.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-18 01:28:23.000000000","tz":-240},"subject":"[DNM] Run functional test jobs","message":"[DNM] Run functional test jobs\n\nCheck changes from\n https://review.opendev.org/c/openstack/neutron/+/993240\n\nChange-Id: If0cbea75ae98cf153e01bcd5a7861b348cb3b248\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf0fcff15b10b3c59f47a19dbac9e7ab01f0e594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf0fcff15b10b3c59f47a19dbac9e7ab01f0e594"}]},"branch":"refs/heads/master"},"72ac5d0c7b3c4a00134bd4897af5c9ccfbdc55a8":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-06-18 01:30:29.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/41/993241/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/993241/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/993241/6"}}},"commit":{"parents":[{"commit":"71f2feb57f5abe864950b6c5e3d13210c2a3cfdb","subject":"Improve _is_centos9_frr85 skip function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71f2feb57f5abe864950b6c5e3d13210c2a3cfdb"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:04:30.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-18 01:29:54.000000000","tz":-240},"subject":"[DNM] Run functional test jobs","message":"[DNM] Run functional test jobs\n\nCheck changes from\nhttps://review.opendev.org/c/openstack/neutron/+/993854\n\nChange-Id: If0cbea75ae98cf153e01bcd5a7861b348cb3b248\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72ac5d0c7b3c4a00134bd4897af5c9ccfbdc55a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72ac5d0c7b3c4a00134bd4897af5c9ccfbdc55a8"}]},"branch":"refs/heads/master"},"c519cf8ffa5211a93723a6fcc884f9495d3d0dc2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-22 18:11:26.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/41/993241/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/993241/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/993241/7"}}},"commit":{"parents":[{"commit":"5268f7c4fb4e0dae6dad520adab165a3e5c2d82f","subject":"Improve _is_centos9_frr85 skip function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5268f7c4fb4e0dae6dad520adab165a3e5c2d82f"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:04:30.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-22 18:10:42.000000000","tz":-240},"subject":"[DNM] Run functional test jobs","message":"[DNM] Run functional test jobs\n\nCheck changes from\nhttps://review.opendev.org/c/openstack/neutron/+/993854\n\nChange-Id: If0cbea75ae98cf153e01bcd5a7861b348cb3b248\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c519cf8ffa5211a93723a6fcc884f9495d3d0dc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c519cf8ffa5211a93723a6fcc884f9495d3d0dc2"}]},"branch":"refs/heads/master"},"746a90cf4549cad85a0c69c4523a3525bd3be0f6":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-06-22 21:35:44.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/41/993241/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/993241/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/993241/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/993241/8"}}},"commit":{"parents":[{"commit":"869fb25568fe0967cb9915b41f89a10383a0ee4f","subject":"Merge \"[CI] Move CentOS jobs to 10-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/869fb25568fe0967cb9915b41f89a10383a0ee4f"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:04:30.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-22 21:35:20.000000000","tz":-240},"subject":"[DNM] Run functional test jobs","message":"[DNM] Run functional test jobs\n\nCheck changes from\nhttps://review.opendev.org/c/openstack/neutron/+/993854\n\nChange-Id: If0cbea75ae98cf153e01bcd5a7861b348cb3b248\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/746a90cf4549cad85a0c69c4523a3525bd3be0f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/746a90cf4549cad85a0c69c4523a3525bd3be0f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
