)]}'
{"id":"openstack%2Fneutron~993245","triplet_id":"openstack%2Fneutron~master~I9b356da0bd62ab7fd17f6cbefb8d8c90f36ca82f","project":"openstack/neutron","branch":"master","topic":"rootwrap_filters","attention_set":{},"removed_from_attention_set":{"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-06-16 16:28:34.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-06-16 12:49:44.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"I9b356da0bd62ab7fd17f6cbefb8d8c90f36ca82f","subject":"Rootwrap filters for vtysh","status":"MERGED","created":"2026-06-13 15:34:54.000000000","updated":"2026-06-17 07:08:19.000000000","submitted":"2026-06-16 16:28:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"993245-rootwrap_filters","meta_rev_id":"70f191fd1e0ee33d909f15d8827c4d4779b7fb2d","_number":993245,"virtual_id_number":993245,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 16:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-16 06:17:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-16 09:27:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-16 12:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-16 12:49:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-13 15:35:03.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2026-06-13 17:46:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 06:17:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-16 09:27:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ff80c9d54b15b950eabb4b9377517534ea489311","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-13 15:34:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2d51989c546c0ec771eddb9cea8b339803b755a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-13 16:12:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6721e4716cc5385c1733d7c7dc1f7c1d3d527761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 17:46:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2546fd89cb9e4899af3a42fa85b86e0f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/53a6bfdaf58240b8a3285b44ddb8a988 : SUCCESS in 1h 27m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e5a2ed228784ddd9722c815b6b6b885 : SUCCESS in 1h 04m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e526a819a079447c81a947bab72a904a : SUCCESS in 1h 25m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/91c814b320234bb5a7398d49a9378a2a : SUCCESS in 2h 10m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d4faabb37ee6430d8d5ef92e73db4a76 : SUCCESS in 31m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b94aea4e85840e7885c91b72d2cd026 : SUCCESS in 1h 02m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f81748275ef244a3b424472d960744b5 : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/627a77b048d64126b7127c7b8d5bedb8 : SUCCESS in 28m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c81824cc3154f51ad21ce0e50591bbb : SUCCESS in 25m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d2ecfaa9e46f4e3f9330b9cbd7b81124 : SUCCESS in 36m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b400f5270c4becb7c9679253976509 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7145a274220c4c848ac0cb5813d89263 : SUCCESS in 20m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/79fb18b7dcb443d08e6574d1c816b6d5 : SUCCESS in 1h 29m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/32d1f4f8d20448569768a959966f6ca6 : SUCCESS in 40m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cfe42d45f0a48aca629d70ff5e59881 : SUCCESS in 56m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8d777c7058204a20997da9154a6435f9 : SUCCESS in 1h 11m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d1cff35ab314916865bce1d9cb910a8 : SUCCESS in 1h 42m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/302a3426d03f44309a4b56bba4faa3ca : SUCCESS in 1h 03m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cf31148932e14a1ea3613207ad3d3b42 : SUCCESS in 1h 11m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e202a134546d4226b296354b96fe1540 : SUCCESS in 1h 30m 48s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"eb8694c7def12ce6f0dee39c7cfe9c4d3af23c06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-15 15:52:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"73fa3bf198a50236f9e4f3804362af5460c16ec8","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-15 15:53:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36bffcdb8f14b379738291ff1e58f4ea100db32f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 18:07:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafd979984514cf686f24cf76152280f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7bc5998e5eaf44538cf8165815f5faf9 : SUCCESS in 1h 06m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3155daf5738f41479ae58c484145c2ae : SUCCESS in 1h 04m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/62a90fc27f8642699aa0cec80c1b711a : SUCCESS in 1h 38m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f7b89d572b6b4b718f905d358042cdab : SUCCESS in 2h 12m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8f74e7436b90478eb7c8e8569ef6da65 : SUCCESS in 36m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfe22a9d58464e48a9e45a62882253b4 : SUCCESS in 55m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b4e53c6da24cd1bff5bf61ed288299 : SUCCESS in 12m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5772097f0cbb41b9b6076e9d1d29b587 : SUCCESS in 36m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bc767e00408474f8e1b2161ab1dba3a : SUCCESS in 30m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c0f6fdb2b2149d286c32c64f4a1d152 : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148066790d7d4668bbb5e327126a26e9 : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f681cf68d4e4c268766164d6c109674 : SUCCESS in 23m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a5cba0e213a6496bb28136958a039604 : SUCCESS in 1h 30m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0baa4c354dd441ee947179c07ad10dc6 : SUCCESS in 1h 21m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/24e0411d26f6496b86df4d9e79fcd4fc : SUCCESS in 43m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7c2e1d7031ea4fb6a4542435da99be4a : SUCCESS in 2h 02m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eecb3b93d823483ba94a4082f0db76ac : SUCCESS in 1h 31m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5159ebd4e8674f37b8046c7086b1fc86 : SUCCESS in 1h 14m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6c59073c60864ab1b78834743944a69b : SUCCESS in 35m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5dc2121465bc4f0d9beddcd037765c1c : SUCCESS in 1h 05m 29s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"7d0587236225cf74a01084bc10dd1b1061898ea0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-16 06:17:39.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a55fea4d8639771d71d2b6db6e8daa465952ff8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-16 09:27:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ec116118f1834a51a4dca1977aea2cc7872d1520","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-16 12:49:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"829f9f94563193b454fe24f140e352305c96f00f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 12:50:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cafc4ca2df46c27a2b9e96ff10943b283975f7e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 16:28:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdaac643073c44389cd67600c284e2fd\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/56b2f4aadee942898826bf4c34229512 : SUCCESS in 1h 21m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6832e2dc193643bdbc9b5fa29534a8f9 : SUCCESS in 1h 06m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dd44959b02954c64a4878871e54c88c9 : SUCCESS in 2h 09m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1a39d32170294bdc940fab4575f8f7e3 : SUCCESS in 2h 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41951e66d60a46b5ba36d12644726bd3 : SUCCESS in 11m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e19af0b1ae744f779e8cd3a4b01f2406 : SUCCESS in 28m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7eac1698d5194a368e189da900ea734b : SUCCESS in 28m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8ce00334074e74b8ec835c5646f391 : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26aea79683d4458dbd831a1ee0114202 : SUCCESS in 15m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6158dc6d2fe0435e9fdcfa6f5e217e30 : SUCCESS in 1h 29m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/273ecdaa76064452a1ee88fc380d02be : SUCCESS in 1h 45m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5455623643ed4cfc88903286fc80063a : SUCCESS in 1h 24m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4a6a167332f445e864212602e37bb50 : SUCCESS in 1h 08m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0401787658cf4c378ab18a1f00953456 : SUCCESS in 1h 10m 20s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"e8abdd92bd10102e478462b67f2e48d8154c930a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 16:28:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"efdfacb45b5ee997031b1226eb8b245fce770417","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 16:30:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38b8f2042d7b4dbd8c16f67440572c88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3779102b20b34dd1aa18ec8af5048689 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2},{"id":"70f191fd1e0ee33d909f15d8827c4d4779b7fb2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 07:08:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eda38e62fe140a0862a5050f140b563\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/13959819e3d1418480df52f169674872 : SUCCESS in 48m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f587209209634aca9ebd167b10a007a6 : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8deb3c71ab4b44e3b74555ff789a5243 : SUCCESS in 38m 27s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e6863125283665182b5e906cbe09a917ccb1a51","revisions":{"f99bf76761d9ff44386a236c8c73b9162dbe03f9":{"kind":"REWORK","_number":1,"created":"2026-06-13 15:34:54.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/45/993245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/993245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/993245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/993245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/993245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/993245/1"}}},"commit":{"parents":[{"commit":"de087e8f12a916d46565a360b9a350356d8ed994","subject":"Merge \"ovn: pass gateway LRPs to _check_external_ips_changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de087e8f12a916d46565a360b9a350356d8ed994"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:31:40.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:31:40.000000000","tz":-240},"subject":"Rootwrap filters for vtysh","message":"Rootwrap filters for vtysh\n\nAdd rootwrap filters for executing frr_driver.py vtysh commands.\n\nThis is a stop gap solution until we add privsep support for vtysh\n\nChange-Id: I9b356da0bd62ab7fd17f6cbefb8d8c90f36ca82f\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99bf76761d9ff44386a236c8c73b9162dbe03f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99bf76761d9ff44386a236c8c73b9162dbe03f9"}]},"branch":"refs/heads/master"},"7e6863125283665182b5e906cbe09a917ccb1a51":{"kind":"REWORK","_number":2,"created":"2026-06-15 15:52:59.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/45/993245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/993245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/993245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/993245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/993245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/993245/2"}}},"commit":{"parents":[{"commit":"de087e8f12a916d46565a360b9a350356d8ed994","subject":"Merge \"ovn: pass gateway LRPs to _check_external_ips_changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de087e8f12a916d46565a360b9a350356d8ed994"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-13 15:31:40.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-15 15:52:27.000000000","tz":-240},"subject":"Rootwrap filters for vtysh","message":"Rootwrap filters for vtysh\n\nAdd rootwrap filters for executing frr_driver.py vtysh commands.\n\nThis is a stop gap solution until we add privsep support for vtysh\n\nChange-Id: I9b356da0bd62ab7fd17f6cbefb8d8c90f36ca82f\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e6863125283665182b5e906cbe09a917ccb1a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e6863125283665182b5e906cbe09a917ccb1a51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
