)]}'
{"id":"openstack%2Fneutron~993398","triplet_id":"openstack%2Fneutron~master~I34e2ecfe0f78f4cbc6326615745468761902e75e","project":"openstack/neutron","branch":"master","topic":"frr_vtysocket","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-06-18 10:41:24.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"38298":{"account":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"last_update":"2026-06-18 14:44:27.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-06-18 14:44:27.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-06-18 14:44:27.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-06-18 12:32:16.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"I34e2ecfe0f78f4cbc6326615745468761902e75e","subject":"FrrDriver Vtysh accepts --vty_socket argument","status":"MERGED","created":"2026-06-15 15:44:36.000000000","updated":"2026-06-18 14:46:15.000000000","submitted":"2026-06-18 14:44:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993398-frr_vtysocket","meta_rev_id":"817285c6b9847e6e567465ee56670856bfb99378","_number":993398,"virtual_id_number":993398,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2026-06-18 00:30:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-18 14:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-18 10:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":2,"date":"2026-06-18 12:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":1,"date":"2026-06-18 12:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-15 17:53:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-15 18:51:12.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 13:52:33.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-06-16 13:52:33.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"state":"REVIEWER"},{"updated":"2026-06-16 13:52:33.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2026-06-17 12:15:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"cc1ea1c65efeb96dd2737ede3b1a1433816f9e9a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-15 15:44:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f96a69b7ccaf91c89109398a1f21dbbdb938dfed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 17:53:13.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6cbb3813ea140d98f691ec324e26305\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4ba992e994e34e6fa5eeac2e08cbc84d : SUCCESS in 1h 17m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa15f90a2f43466cac2ddcced3e4043f : SUCCESS in 1h 17m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c0382ffec0dc4af08a4a54b9b9c301b5 : SUCCESS in 1h 45m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1698d3dbb1394c2f8fe9c4e103cb31e0 : SUCCESS in 2h 05m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7570bec38c64bb5829b6522c94a08a3 : SUCCESS in 37m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33cfc6583d27496b964eb107ed7d33be : SUCCESS in 1h 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aca0c18f3c3a4b92820e72f2bb62c084 : FAILURE in 10m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5dcb8ddd17440d092a257f881601856 : SUCCESS in 31m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3204113b95544c2b710c30e747366a0 : SUCCESS in 25m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a42b4c5128d040498cbbd192c4812431 : SUCCESS in 37m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff0cb7555b942a0b4759d953f9159bc : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec84f908adad45ada0bdbcc820c9a9db : SUCCESS in 22m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/df0959a5c5c24ce5b4691ebb65f0b8f5 : FAILURE in 54m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5db9cdae62224edbbf331240c9357327 : SUCCESS in 1h 13m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9fbda886d924c82a3f57e89f947589d : SUCCESS in 1h 13m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/076966e61d554320a4b7f34a47be68a8 : SUCCESS in 1h 45m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/581f187d9adf45efbfa741f33ab984d8 : FAILURE in 47m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eb9ecd907bb042cbad1a4cb90cb3802a : SUCCESS in 1h 07m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb4957f5e4834c9bbe9e1594ab108f3f : SUCCESS in 1h 18m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/990ff85d4540441ca310d86deacbb39e : SUCCESS in 1h 46m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ef7099e651ff1c6ca83672dc95d88485d70dedc0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-15 18:51:12.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/98/993398/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"5d2a5d5c736e9ff69833cb94414bb21ab29656fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-16 13:52:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3f2caf81f05ef1251d68d8cafdbaa403d47bfde1","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-16 13:52:33.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"fdfaac7453c7e005376397e312681c3e522a38b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-16 14:19:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c9e1d1b5a03b4af685373b1c9c23b8bda8c8455","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-16 14:20:22.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0637f4ea51ccaacd425de170711257dc9d07ffa3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-16 14:23:31.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e36e156e6bacda467a01f16684227d5503e1a5cf","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-16 14:35:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3bc2631556b25969cd14c7de1e745a9af0c35e7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-16 16:30:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a9d56df1559313a693b4ba2220ee6dd993ee7b28","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-16 16:31:26.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"62df5debdc131617779b885f5e265a589be28322","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 16:54:01.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/98/993398/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"56e895e077dd26271012f146cd127c444d7fdbce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 17:33:36.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/98/993398/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"18e0afe6288944e41481ca2cfffff8e46915200d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:57:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0081ef16dd4b455cab71a32ea1a1995e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aa8b160c781a46f1ab08c5b59958c193 : SUCCESS in 1h 21m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d0f9fabcdf84558a5312e678fec941f : SUCCESS in 1h 06m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/579c4a96627c4c8a8f54d95de7925227 : SUCCESS in 1h 19m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7ab43e34986b4538a62c299a658c5b91 : SUCCESS in 2h 24m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f4c1f4306ea14fffa6a5699ac214a76e : SUCCESS in 30m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10fe9701810d4f96a83f5d5bf013c26b : SUCCESS in 1h 06m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb39756e02e54a1fae237a62b714c9c0 : SUCCESS in 8m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45314903171744ffbc3e6b2fda6cbd5a : SUCCESS in 26m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7edf6cd06c3b44258d97c710a8339049 : SUCCESS in 27m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca0181dc44fc4fd78b550849ef3e8dcc : SUCCESS in 35m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41e2b79aabc1407988ada07143735e64 : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dd3d55950544b1198995c7f3ae3d86b : SUCCESS in 23m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b8a07b6247284aaba2ef97fc84e595ed : SUCCESS in 1h 31m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b2a23d970384a75a7d8d8ae08f17a50 : SUCCESS in 1h 24m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f427171541246c2b54cc4682a997640 : SUCCESS in 1h 17m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/428cf9f24d37456bbafbbfa275504ae6 : SUCCESS in 1h 44m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/893f14aede0f49039e9414c2436b0328 : SUCCESS in 1h 40m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9c089bd03f8c4c5b983ef7f0aa2605f6 : SUCCESS in 1h 08m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e5e20ae6e43a4ab0b4dd46ffe454484d : SUCCESS in 1h 09m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f934f93e0634632b865377c1f6143e7 : SUCCESS in 1h 19m 47s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"ae82fd864eadaa9bb9362702a307a875d471505f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 20:12:19.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/98/993398/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"cde1fbac6caf9a30766778109bf0ab2ba84e1e61","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-17 12:15:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"df4febaf36c8ae9effae2c323385a124a65d39ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:11:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f25a7db67748178688909941cfa541\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/48d90f6cbae7414a931e040fca805808 : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5fa72955e643432d997d1f26452f3932 : SUCCESS in 18m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dbc0080651754d0a87c8b469cc215a8a : SUCCESS in 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3380169de9ebbc1e3cdfb37c64b52f52e3a9e526","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-17 18:21:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"50ca168fce00367e8158c37a7218e4a0cfe91e5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-17 22:57:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9fec70be1ed8b1dde1171adf98ec25033d02d2f9","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-17 23:01:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f8392990cf65c54eca5f4f28e518f9a395dda915","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-18 00:30:43.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/98/993398/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"dac1c34222419759a6257a288a7e7d8df6453e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 01:02:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/200a888e983f4d558e591dc6ac0c8ed4\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3d26c3d0e28d4d86a3d51b73e27e9c36 : FAILURE in 58m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80a3e99cfb2e4f8e8fdcf3a4fa244d34 : SUCCESS in 37m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4100191d54574d9bb26dc632432cc2d4 : SUCCESS in 1h 20m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0daa573231eb4e84a2fb9c9f6c5f90fe : SUCCESS in 1h 47m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/118f4aa01faf4ebebd9ce9ae1649647a : SUCCESS in 31m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6916d1f620d49fcbc0d486e6a0416eb : SUCCESS in 43m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ac47595a5f4c19acce44a0028f696a : SUCCESS in 7m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f00813bbab644530980424f06fbc875f : SUCCESS in 21m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec242bca56db4c799dee1ce833bd0318 : SUCCESS in 25m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fb8854630032444884fc4db0dc8670bd : SUCCESS in 25m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/724cf19b6e394b258796063e38c50519 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bb2d338abab485cb985a1d454dbc97e : SUCCESS in 7m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/47ca56266494414b91f4a2b848ae00cf : SUCCESS in 45m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/870e0e4b4ded4158a7d849f62833975a : SUCCESS in 48m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfaf52a94b1240aba5d98a37d29ffb99 : SUCCESS in 1h 14m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b5b67af026d64bd883663430a65154d8 : SUCCESS in 1h 59m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/135bfe406ecb46e4bbbca12060cba94a : SUCCESS in 1h 48m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/80ea8b6a12d643e4b9acf5e2b41f683d : SUCCESS in 1h 07m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c684cdd4f4e34b56ad764f1c3401c1e4 : SUCCESS in 36m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/025102fe9b7b48a4b0858ef2b1fad487 : SUCCESS in 42m 46s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"b1df9eaeffe6c5d83a7195fe98460b582ee22d33","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-06-18 01:24:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f19e3cb404a1192731972037cb2e7867c035fc58","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-18 02:04:39.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/98/993398/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"c50ad125fd4eb582b95a62b4f1b1a0245a792bc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 03:31:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b156148fea924d259cbb2d6dad677986\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5524aff0d2674c5696ef665257b358f2 : SUCCESS in 1h 05m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1280adf5a9914d00b09c38fa88a23905 : SUCCESS in 1h 01m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2ac1ca6280534100b486830176f1cf69 : SUCCESS in 1h 29m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f30af27153e7470da6f9722ff5046398 : SUCCESS in 1h 57m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5775672beb3241609236a9ec141b16f7 : SUCCESS in 33m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed498cb8f3d64891a53d31f72442fdca : SUCCESS in 1h 02m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c15370f90b64b22a7b0642243d2608e : SUCCESS in 10m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7e0827f4ba842fa89fd09b7e9a8df35 : SUCCESS in 30m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27bc9417ff8d4fb8b407ca6cfd6688a5 : SUCCESS in 29m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/11325b795c064cb2b11f43a08a21e9bc : SUCCESS in 33m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e51662ffd56e4b688f71adad84a8a4e7 : SUCCESS in 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43e40e4cafef4e09bba2962e7775fea3 : SUCCESS in 22m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/472c4a76dfa2429486aec12e018dd1b7 : SUCCESS in 46m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/05e1b7ddd84b4ffdb6758293f6ad92f8 : SUCCESS in 1h 10m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/55832c3c220e4e44a582f52dcc5e2bfe : SUCCESS in 41m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/27286fee5f7848009b2209162e654650 : SUCCESS in 1h 58m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3a1fb3d17fe14cbf97903602762410b9 : SUCCESS in 1h 40m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fb55bccc3d794a90a7dae458d256cef1 : SUCCESS in 1h 05m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0a43e5a7e60c4187ab27ceea1344e231 : SUCCESS in 1h 13m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/77bb0d9aca4c407bb5c10643791c8d4b : SUCCESS in 42m 50s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"45a2261df3209d29855caf70b70364f8a600f9d1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-18 10:41:24.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"30bfc5b1c8e85c9501b67d8363de0a583601aca3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-18 12:32:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3fc910cb4268d1e8efa5d22230e19b2c8a0cafb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 12:32:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dc6bfd6bf2b6c58bd313151b920718fd9f10ca24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 14:44:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5966cd89ad9b486ebd8a9c8e7f6bed22\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/da20c10384b74abd9e4d5276c7bd00b5 : SUCCESS in 1h 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/043c31a674084ab89523372ca9c1d4af : SUCCESS in 37m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/196079480cb54348b37306fe995d0e08 : SUCCESS in 1h 48m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1bb66e36fabe4d00b7fa138c87d680c8 : SUCCESS in 2h 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49b46d8e35a4413ca9853e6d7d518a27 : SUCCESS in 12m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d16f58e4bfb4644b968392edd1571b2 : SUCCESS in 23m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2183e6b7c64f4727b3d7202523eccf4b : SUCCESS in 16m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44bece7c59734af1b91d14a1c353571e : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da487b6520a046769db6db173c7e7660 : SUCCESS in 8m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d5d5f4d3b0a441bcb339313c2f873416 : SUCCESS in 47m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ea2b3bab54d4531ace1474fb56514dc : SUCCESS in 1h 53m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b4eba5f161143e490994556230015ca : SUCCESS in 1h 23m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c9d74dec66441baa3f2c4821566ed19 : SUCCESS in 39m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ccfd599d399f4751b58ccfff34b88446 : SUCCESS in 1h 01m 55s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"3b9ae70fa98862640b91890b6e82e7f60e49c6d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 14:44:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"817285c6b9847e6e567465ee56670856bfb99378","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 14:46:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d086bfe9b4604c1f9b6a95712a9932a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2aee8b7f5a245deab29abdd9dca78a5 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"425f97dc8ce60a55f84b147c199b62437e4d378e","revisions":{"08c426feb511a71c8f0882e19f14812fa335f374":{"kind":"REWORK","_number":1,"created":"2026-06-15 15:44:36.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/98/993398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/993398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/993398/1"}}},"commit":{"parents":[{"commit":"de087e8f12a916d46565a360b9a350356d8ed994","subject":"Merge \"ovn: pass gateway LRPs to _check_external_ips_changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de087e8f12a916d46565a360b9a350356d8ed994"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-15 15:38:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-15 15:38:32.000000000","tz":-240},"subject":"FrrDriver Vtysh accepts --vty_socket argument","message":"FrrDriver Vtysh accepts --vty_socket argument\n\nVtysh already accepts passing different path to unix socket for\ncommunicating with frr via `--vty_socket` argument. This patch exposes\nthis argument in the frr_driver. The goal is to use this argument\nwhen frrdriver is run inside a container where the unix path might be\ndifferent than default.\n\nChange-Id: I34e2ecfe0f78f4cbc6326615745468761902e75e\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\nRelated-Bug: #2144617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08c426feb511a71c8f0882e19f14812fa335f374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08c426feb511a71c8f0882e19f14812fa335f374"}]},"branch":"refs/heads/master"},"df579b06f7103e1b037edd7b01c3070ae647f4c8":{"kind":"REWORK","_number":2,"created":"2026-06-16 13:52:16.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/98/993398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/993398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/993398/2"}}},"commit":{"parents":[{"commit":"8223b24310e636ff02e8db3f1da421a39b00dca0","subject":"Merge \"Remove NEUTRON_DEPLOY_MOD_WSGI from zuul jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8223b24310e636ff02e8db3f1da421a39b00dca0"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-15 15:38:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-16 13:51:42.000000000","tz":-240},"subject":"FrrDriver Vtysh accepts --vty_socket argument","message":"FrrDriver Vtysh accepts --vty_socket argument\n\nVtysh accepts passing different path to unix socket for\ncommunicating with frr via `--vty_socket` argument. This patch exposes\nthis argument in the frr_driver. The goal is to use this argument\nwhen frrdriver is run inside a container where the unix path might be\ndifferent than default.\n\nChange-Id: I34e2ecfe0f78f4cbc6326615745468761902e75e\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\nRelated-Bug: #2144617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df579b06f7103e1b037edd7b01c3070ae647f4c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df579b06f7103e1b037edd7b01c3070ae647f4c8"}]},"branch":"refs/heads/master"},"a8fc39d46f1abd900813e34c80a24a15b21c10a5":{"kind":"REWORK","_number":3,"created":"2026-06-16 14:19:27.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/98/993398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/993398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/993398/3"}}},"commit":{"parents":[{"commit":"8223b24310e636ff02e8db3f1da421a39b00dca0","subject":"Merge \"Remove NEUTRON_DEPLOY_MOD_WSGI from zuul jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8223b24310e636ff02e8db3f1da421a39b00dca0"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-15 15:38:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-16 14:17:21.000000000","tz":-240},"subject":"FrrDriver Vtysh accepts --vty_socket argument","message":"FrrDriver Vtysh accepts --vty_socket argument\n\nVtysh accepts passing different path to unix socket for\ncommunicating with frr via `--vty_socket` argument. This patch exposes\nthis argument in the frr_driver. The goal is to use this argument\nwhen frrdriver is run inside a container where the unix path might be\ndifferent than default.\n\nChange-Id: I34e2ecfe0f78f4cbc6326615745468761902e75e\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\nRelated-Bug: #2144617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8fc39d46f1abd900813e34c80a24a15b21c10a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8fc39d46f1abd900813e34c80a24a15b21c10a5"}]},"branch":"refs/heads/master"},"8df5beb298744123c5874df0e12e4cf33d1dc609":{"kind":"REWORK","_number":4,"created":"2026-06-16 16:30:39.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/98/993398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/993398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/993398/4"}}},"commit":{"parents":[{"commit":"4355bb2efe1940888809209572442c1ecda8655b","subject":"Merge \"Rootwrap filters for vtysh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4355bb2efe1940888809209572442c1ecda8655b"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-15 15:38:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-16 16:30:25.000000000","tz":-240},"subject":"FrrDriver Vtysh accepts --vty_socket argument","message":"FrrDriver Vtysh accepts --vty_socket argument\n\nVtysh accepts passing different path to unix socket for\ncommunicating with frr via `--vty_socket` argument. This patch exposes\nthis argument in the frr_driver. The goal is to use this argument\nwhen frrdriver is run inside a container where the unix path might be\ndifferent than default.\n\nRelated-Bug: #2144617\nAssisted-by: Opus 4.6 model\nChange-Id: I34e2ecfe0f78f4cbc6326615745468761902e75e\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8df5beb298744123c5874df0e12e4cf33d1dc609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8df5beb298744123c5874df0e12e4cf33d1dc609"}]},"branch":"refs/heads/master"},"425f97dc8ce60a55f84b147c199b62437e4d378e":{"kind":"REWORK","_number":5,"created":"2026-06-17 22:57:11.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/98/993398/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/993398/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/993398/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/993398/5"}}},"commit":{"parents":[{"commit":"4355bb2efe1940888809209572442c1ecda8655b","subject":"Merge \"Rootwrap filters for vtysh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4355bb2efe1940888809209572442c1ecda8655b"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-15 15:38:32.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-06-17 22:56:00.000000000","tz":-240},"subject":"FrrDriver Vtysh accepts --vty_socket argument","message":"FrrDriver Vtysh accepts --vty_socket argument\n\nVtysh accepts passing different path to unix socket for\ncommunicating with frr via `--vty_socket` argument. This patch exposes\nthis argument in the frr_driver. The goal is to use this argument\nwhen frrdriver is run inside a container where the unix path might be\ndifferent than default.\n\nRelated-Bug: #2144617\nAssisted-by: Opus 4.6 model\nChange-Id: I34e2ecfe0f78f4cbc6326615745468761902e75e\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/425f97dc8ce60a55f84b147c199b62437e4d378e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/425f97dc8ce60a55f84b147c199b62437e4d378e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
