)]}'
{"id":"openstack%2Fneutron~993487","triplet_id":"openstack%2Fneutron~master~I8fcd95fd8be0328277c794280900f8fb3d65ccbb","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-16 18:03:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8fcd95fd8be0328277c794280900f8fb3d65ccbb","subject":"Fix superfluous-parens pylint warnings","status":"MERGED","created":"2026-06-16 01:38:03.000000000","updated":"2026-06-17 10:05:12.000000000","submitted":"2026-06-16 18:03:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993487","meta_rev_id":"d590f913fba4b8d4b4cc84dd0413e481353362c4","_number":993487,"virtual_id_number":993487,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-16 03:56:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 18:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-16 06:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-16 14:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-16 14:11:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-16 03:44:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 03:56:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 06:10:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-16 14:11:12.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"172ad2f272b9651203b8a0157147aa5e114a2da8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-16 01:38:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"279f0df0b509f0349f090fdf1780e981dc63cbeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 03:44:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79fd25fddaa2436788a7395d5d58f30e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e3bc6c821a2a484e912ea3b8b86dc06a : SUCCESS in 1h 33m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5941ecefe17440179ac25e2a592544ac : SUCCESS in 1h 08m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6e2ee915d9e64082ab00c63fbb199f01 : SUCCESS in 1h 36m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6a2a6cf836a043ee8c09b1e13e106224 : SUCCESS in 1h 46m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec04460bb7794dfbbe0cc556e96d1530 : SUCCESS in 32m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aff9a2b13b2f4e19a80db6594458975d : SUCCESS in 1h 04m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b98badd63c4762accd9f30a61b281a : SUCCESS in 11m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c4efb1537444e769c0e17cdd2febd7b : SUCCESS in 21m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66f5917463c645d2a72ef4ddd3332739 : SUCCESS in 27m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/035c63b9cadd46b39e0cd13b2d583557 : SUCCESS in 24m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a376a74a7148beb6371f3224ed0547 : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c25276f4efb54edc93a6b7ba52d27ea1 : SUCCESS in 15m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ea119734b6bf406196eb652fc703f3e3 : SUCCESS in 1h 25m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f901206e6ac746618db943a0dcc0e92f : SUCCESS in 1h 10m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/01c6b33a66134d25859179f625f2e635 : SUCCESS in 1h 18m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f969875d5ac245e8893a8f7acdd95ea3 : SUCCESS in 2h 01m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/76e6989cb24a452a9753623d1a23a2cc : SUCCESS in 1h 52m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cac16f4f3bea48f39ac5d0dca85dae8c : SUCCESS in 54m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e5e5d83f8c414e6c8ca65fd0ad6ac7eb : SUCCESS in 1h 08m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9a87ac01d01b4b8e935cccef49e46a6c : SUCCESS in 1h 26m 47s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"d3b9dcf85aa54e96aeb85deb13ddd6d1e956bf6f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 03:56:52.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/87/993487/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c2ba8e82993c036003a4ec903c8519560e4f69cc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-16 06:10:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cd924ae884f40da0720ca2ab7db62efbb1940cfe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-16 14:11:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"676dcddeae4b361c1c6823121c69e0dc21c40c2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 14:17:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"310cf2f980cd36c7427771e8001bd74a70e5993b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:03:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b985f0874504cccbc59e34a5272f44b\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b3407fcc9b82404186d5a283a500a454 : SUCCESS in 1h 36m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eafdd988eb5d406e93fd6ca857302eab : SUCCESS in 1h 17m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1f542cb75076412a8b89e2e0cd1bd8f9 : SUCCESS in 1h 37m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3f15ed0c8fa44ed09fe5f6546aa16d47 : SUCCESS in 2h 38m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29432117aa14a418ca35d64bda25f1c : SUCCESS in 12m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec3db37e1fc04a7bb61c99f59c53d120 : SUCCESS in 28m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76d43fd92e41494bbec30516539a8b1f : SUCCESS in 28m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3088095af2bb4e8e9ac26e39a5addc4a : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/364ae1c7858944c0aa5c6a5ef41f5155 : SUCCESS in 21m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/de2b59bb8fa7496e9bce1425203c8799 : SUCCESS in 1h 29m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e78edfe96cca41bc957bfd7b00329e3a : SUCCESS in 1h 29m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/931dc65fb22340ae9207e3d529bfad95 : SUCCESS in 40m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2e925c0da974ef8951759607ff7de69 : SUCCESS in 54m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/de0ee5b761e547ce9e7b5a17957dea45 : SUCCESS in 1h 22m 32s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"7753d91f547ded826deaa4b4c68fccdb98a523f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:03:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e1d67267e4803241d520ab8479ec1582073a13af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:09:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f223a7c2099a42f6bbdb27560f47d2b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c63f63a1514c8c80f3ff139dfb49bc : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d590f913fba4b8d4b4cc84dd0413e481353362c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 10:05:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a97ba84eea34ad9a4742060c15ba9ce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2df0aabeabeb4f8f812862dda8b8dbb7 : SUCCESS in 20m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d00f10b56ba447b49c738d3a9a5ad0b1 : SUCCESS in 18m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3f13923cb0474a6b832e3bdff1306a0d : SUCCESS in 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b877e9b4bdf0910487eba0f4030e71d9349ede0c","revisions":{"b877e9b4bdf0910487eba0f4030e71d9349ede0c":{"kind":"REWORK","_number":1,"created":"2026-06-16 01:38:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/87/993487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/993487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/993487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/993487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/993487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/993487/1"}}},"commit":{"parents":[{"commit":"d8de4a7e54eee06d2bd6c042b0d1f89c7524d990","subject":"Merge \"Call static method ``RangeAllocator._make_params``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8de4a7e54eee06d2bd6c042b0d1f89c7524d990"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-16 01:35:23.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-16 01:36:19.000000000","tz":-240},"subject":"Fix superfluous-parens pylint warnings","message":"Fix superfluous-parens pylint warnings\n\nI had noticed a few of these had snuck-in over the\npast cycle, let\u0027s fix them and enable the pylint\ncheck.\n\nTrivialFix\n\nAssisted-by: Claude Sonnet 4.6\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I8fcd95fd8be0328277c794280900f8fb3d65ccbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b877e9b4bdf0910487eba0f4030e71d9349ede0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b877e9b4bdf0910487eba0f4030e71d9349ede0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
