)]}'
{"id":"openstack%2Fneutron~993489","triplet_id":"openstack%2Fneutron~master~I61cf99dab06642287a1f8c67f5f81b0b1bd8dede","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-06-22 11:08:39.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-06-22 09:03:42.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-22 11:08:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I61cf99dab06642287a1f8c67f5f81b0b1bd8dede","subject":"Fix unnecessary-pass pylint warnings","status":"MERGED","created":"2026-06-16 02:17:12.000000000","updated":"2026-06-22 11:10:33.000000000","submitted":"2026-06-22 11:08:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993489","meta_rev_id":"9627edeb6a663d92a7a9ddbb870a48ab49d412be","_number":993489,"virtual_id_number":993489,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-17 14:26:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-22 11:08:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-22 09:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-06-22 08:48:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-22 09:03:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-16 04:35:58.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 04:45:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-16 06:07:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-16 14:12:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-22 08:48:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"c17d34e870570f7672193716d25471151b062c17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-16 02:17:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e119e9c94c783d0e8d4b2ce9548884a721360b0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 04:35:58.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/89/993489/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9bb7b8fcc0eff650d5a13bf548d66b36cf14788d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 04:45:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f78e8e448bb64c288238b2cfeccfaa18\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1eafb037f19e4453a5fdff87fa83a66c : SUCCESS in 1h 23m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbee868fcf844b28841a2d1c458a9c0f : SUCCESS in 1h 07m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cf07a9cedc2e4f0e9ef7fff6ce59c463 : SUCCESS in 1h 36m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0d723bec2b8c4f6287a4b290e540f5fd : SUCCESS in 1h 59m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee4f3bf792c4438ba618ea9e7cb454cb : SUCCESS in 30m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14e360b34cdc438baa90427bfa8795d8 : SUCCESS in 46m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/724c5324d69f4e60a55f784d65c77478 : SUCCESS in 11m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0afe921954d413fbf2e992f621a7ae7 : SUCCESS in 31m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a9f884ca2d6406494a45be34f75bfd3 : SUCCESS in 26m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7beeece95f2e437e9a7ea924865bb045 : SUCCESS in 33m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0347f6b66c54e5194774595b0d8866c : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9d5d36aedbc47949e609b584ecb0736 : SUCCESS in 23m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f9a203b6cf724c1e9d71cd4355de1a83 : SUCCESS in 1h 25m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/85f968d854b44a90abda90edd5888ff3 : SUCCESS in 40m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/384bf1619720457fb547fba0195043d2 : SUCCESS in 1h 03m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2d03a4c153b54b20ac9ac877d51b0905 : FAILURE in 20m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1193e14d0c78489cbea440bfff8515f8 : FAILURE in 20m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2d80484d0d8e49189f8d19232d5040e0 : SUCCESS in 49m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ce9260995a5748f69889fb6d2f9d471b : SUCCESS in 1h 07m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c00e623294d40fea56030db3accd8ba : SUCCESS in 1h 44m 16s","accounts_in_message":[],"_revision_number":1},{"id":"bed6d75d2498d110585b00b655ff8c19fe6180c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-16 06:07:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f65961aab1d0e0847526e734eb2a1bc85376e10","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-16 06:47:21.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/89/993489/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"3a9d3366b6f1dfe73060df315036f83fea86fbf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 08:38:29.000000000","message":"Patch Set 1: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73fede23470e4396b72785a691787ae0\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b57acbca405a495592b0f98ccb3c8ee9 : SUCCESS in 1h 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4eb02c9b7be4d64897576da823a8dc8 : SUCCESS in 35m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4f0a63554d7647ec8379ce4b9dfe4c95 : SUCCESS in 1h 40m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6aec624023da43b4bb9234472be7a498 : SUCCESS in 2h 30m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6a1c732adfc44419d50fed66a04b163 : SUCCESS in 30m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88519ab9fc30471988f959baafa31226 : SUCCESS in 48m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3bd83e26e2e4edba83fa4eccb070790 : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ed30ca793d54d77ade1d972a5b9331d : SUCCESS in 25m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44b360f5651446c4acef16bbebd320c6 : SUCCESS in 16m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/973dfdaa478742f1b596ecf5a66f8973 : SUCCESS in 27m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b039ace36ca4af0b36113250c9fc15c : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ecb663a31144caf94b551b8882cfe3e : SUCCESS in 13m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7db8d3390a8b4ca18485543153bdb2a5 : SUCCESS in 52m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3123e05493d541cba598f454e7f86462 : SUCCESS in 56m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9514f039b6614ebe89fc3c710bba7a2c : SUCCESS in 1h 10m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4748a6eb645b4593a95465929bd06393 : SUCCESS in 1h 57m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6610a23ea175405589cca2a90ffde241 : SUCCESS in 1h 49m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7f5d9fa182a24368a5e7a179813a3050 : SUCCESS in 54m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/20b91ecda0e24499b860f8efa204064f : SUCCESS in 1h 09m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/76466589c03641dfb997134a5b98369e : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5f9d4aa4cda0f5e5294cfeab1e504f9ec8574c4f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-16 14:12:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05fe97680f85c0abc27ea9930c2570d8ca96d318","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:03:35.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 993489,1","accounts_in_message":[],"_revision_number":1},{"id":"608fd858a106917a6e39a156696874ad38e0c1ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 10:52:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a6489417484518a4df19e3dc7492ec\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ffc38b1e2d3449feb05a6642ea46d16e : SUCCESS in 26m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e8891ba4f6fb48458d1f191c4ed3544b : SUCCESS in 23m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/15642ed4d1cd463080cdfd04abec32a7 : SUCCESS in 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccadcfcbb555079d49c5ddfe04c4f1c3c5a273dd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-17 13:11:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d9b6ae7c6874762663aee859dec12495fc77227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-17 13:12:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9759ec923f2b63b1984f9c6189f69159e8bfcee7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-17 14:26:19.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/89/993489/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7cb0aade6f26d86286686015375b801e68dff3f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 15:34:40.000000000","message":"Patch Set 2: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04cea237d0ff4f04aa23ba0388663625\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0bea0cb29de9412aba1934d9e2d7923f : SUCCESS in 1h 06m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/846af1f413a34a75a0ce746771e16356 : SUCCESS in 41m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a39c2698798f40a583bff6f9e2fa2325 : SUCCESS in 1h 19m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d3a79c732d424741b866e74fa4bf4da9 : SUCCESS in 2h 12m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f17cccc89d34df787cd6e01e67e5a02 : SUCCESS in 31m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ace989bd4a564a2facf526b1384831a4 : SUCCESS in 53m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9785367453984df09423ec64d8044b8e : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4da0f8ec16e74939b4bc421801fe68a5 : SUCCESS in 25m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a92a30bee074cb8868acb7e35de0d6c : SUCCESS in 25m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d4d89eb912c04c359bdf2bd5ec1ac719 : SUCCESS in 41m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79df11fa0dee47658162452e2ade172f : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efb6ad09efe045c2a79a919887401556 : SUCCESS in 8m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/57c280040b36469db28be05eca4c436b : SUCCESS in 58m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d8961c3f2dd4000807385e89b22cf58 : SUCCESS in 1h 03m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a6e06c563634ac3b37342e74dfd0569 : SUCCESS in 57m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/df55d5884c9f43ff95e411eafa6823f1 : SUCCESS in 2h 11m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/75ee270c86ed4e569f453d5cf8400951 : SUCCESS in 2h 15m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a8ed050b2ac04de8937c1c2280713b1d : SUCCESS in 54m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/877f32c9153b4e75ba71f1d94444b57e : SUCCESS in 32m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/41788e7be7fd42c0b4ab4d84fed92b0c : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ff8c8e539954d00ca7a0f5b2dc8c33c29ecba8c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 00:33:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/589a96a1308645f597244d2fd098b1f4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1ed8856aa0514a4781bb603108bc9da2 : SUCCESS in 37m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9738acb9450f472796ca169ae326245f : SUCCESS in 52m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/338be13e767b41789b9858213e029727 : SUCCESS in 1h 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad2e380b10d87192b8ce6979da643086c34b084f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-22 08:48:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1a4a70b52be3ae3056d760176733c642c13ca103","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-22 09:03:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c3c432cc5e637ca3b7d3cf70ae42773f7c23bc7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 09:04:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"126ead2d38f49853a7db51a45a19b594fef90878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 11:08:39.000000000","message":"Patch Set 2: Verified+2\n\n(1 comment)\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/287df00c56104030a1aabe8076e8663e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/25350581de9d4eaf8f319fdc0838196e : SUCCESS in 1h 25m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b43a13ba87b43648e956da52024440e : SUCCESS in 55m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a39958be696427da09ed38c1db2abc8 : SUCCESS in 2h 03m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d9a3e14f61114efa954c136e5f3850c1 : SUCCESS in 1h 59m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7f04f46d9e48118f29a298059981ab : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef763d6c08ec4f1ab9dde2e1492a1c3c : SUCCESS in 21m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c1e47cec7e9465e8ec67635b08e1756 : SUCCESS in 23m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883ee6eeae5549bab8e648d7f57dbc44 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/559642f55ac74ec7aec64b792996695a : SUCCESS in 15m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/01ee72c65b674d8d8f56873c82c7430e : SUCCESS in 47m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/81cc05f4efc14a1fa33f9d44c6f52dbc : SUCCESS in 1h 55m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d24740b94ecc47c2bbba089713f22e51 : SUCCESS in 41m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84f659b99bde4f2b979311dacad58b84 : SUCCESS in 1h 10m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fad6dac3501b43c38a3bdaece5ff4da3 : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":2},{"id":"f8494872f7d5a44505aeba03a577b724cefeb915","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 11:08:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9627edeb6a663d92a7a9ddbb870a48ab49d412be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 11:10:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afe296b3ecde4952973e37e8e8ca8024\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1fa39969a045efb531567a3d0698b6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"718434b3ee800a6a845358de9806318c96da3a8a","revisions":{"9753865e37e619b977910f74a35af0208a6cc5cb":{"kind":"REWORK","_number":1,"created":"2026-06-16 02:17:12.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/89/993489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/993489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/993489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/993489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/993489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/993489/1"}}},"commit":{"parents":[{"commit":"d8de4a7e54eee06d2bd6c042b0d1f89c7524d990","subject":"Merge \"Call static method ``RangeAllocator._make_params``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8de4a7e54eee06d2bd6c042b0d1f89c7524d990"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-16 02:13:54.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-16 02:15:59.000000000","tz":-240},"subject":"Fix unnecessary-pass pylint warnings","message":"Fix unnecessary-pass pylint warnings\n\nNoticed one of these in an unrelated change, let\u0027s\nfix them and enable the pylint check.\n\nAlso start enforcing singleton-comparison as there\nare no warnings for it.\n\nTrivialFix\n\nChange-Id: I61cf99dab06642287a1f8c67f5f81b0b1bd8dede\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9753865e37e619b977910f74a35af0208a6cc5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9753865e37e619b977910f74a35af0208a6cc5cb"}]},"branch":"refs/heads/master"},"718434b3ee800a6a845358de9806318c96da3a8a":{"kind":"REWORK","_number":2,"created":"2026-06-17 13:12:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/89/993489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/993489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/993489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/993489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/993489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/993489/2"}}},"commit":{"parents":[{"commit":"319761f0697026227aa7b0a882bfdfddbae74f91","subject":"Merge \"Explicitly create log directories in the FrrFixture\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/319761f0697026227aa7b0a882bfdfddbae74f91"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-16 02:13:54.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-17 13:12:06.000000000","tz":-240},"subject":"Fix unnecessary-pass pylint warnings","message":"Fix unnecessary-pass pylint warnings\n\nNoticed one of these in an unrelated change, let\u0027s\nfix them and enable the pylint check.\n\nAlso start enforcing singleton-comparison as there\nare no warnings for it.\n\nTrivialFix\n\nChange-Id: I61cf99dab06642287a1f8c67f5f81b0b1bd8dede\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/718434b3ee800a6a845358de9806318c96da3a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/718434b3ee800a6a845358de9806318c96da3a8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
