)]}'
{"id":"openstack%2Fneutron~993767","triplet_id":"openstack%2Fneutron~master~I41eaa2e77ff6b907e376b6ab3250beb8659d45da","project":"openstack/neutron","branch":"master","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-06-18 08:34:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-18 08:34:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I41eaa2e77ff6b907e376b6ab3250beb8659d45da","subject":"Do not fail when api workers count can\u0027t be determined","status":"NEW","created":"2026-06-17 14:40:29.000000000","updated":"2026-06-18 08:54:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"37ade04fe173310a10fc2ca6c231b09828b678f8","_number":993767,"virtual_id_number":993767,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-18 04:14:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-18 05:43:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2026-06-18 06:41:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-17 15:44:38.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-17 16:11:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-06-17 17:02:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-18 06:41:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"27cfbed8c449679162cb67db7035818c738ebe43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-17 14:40:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d5816ba9b81e1deab696f16ad5de7b24ab59928","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-17 15:44:38.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/67/993767/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d70c9bda8efcf85e643d584da05aed6c4e4be62c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 17:02:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4c5d995a830496b874db6d648152d35\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/339be2fedf4a4c2d8a2175bbd98e6fbf : SUCCESS in 1h 41m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70f6dd93109b4a4aacd46cb62f7f454d : SUCCESS in 35m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4549c46f30e44dc1b0a8c0a8f9086012 : SUCCESS in 2h 19m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c06d4b500d6d46f48fcdc0e07764bad5 : SUCCESS in 42m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27d52954c7914299b1c6d3244654b9ab : SUCCESS in 1h 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c409e8c4e55d4a759379d5c59c95633a : FAILURE in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41177575b9d64a30825ebcce33f549aa : SUCCESS in 35m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d463de8afe94789ad7c0609b86a4150 : SUCCESS in 28m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb9add55537e46d7b75f6972ed036494 : SUCCESS in 34m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01915e26e98042c38fe772a0c7397403 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33278483014e416f8246f76b15085570 : SUCCESS in 24m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8d19cca615bd42e0a87fcf488ee6882b : SUCCESS in 1h 22m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/db43f64a94a242f9a48e19bb6288c093 : SUCCESS in 1h 16m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/efa22c71080e4626ace1d2f6ae37fbcb : SUCCESS in 1h 11m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/105de5155a4d4a08b4a0ed708bf2a1f6 : SUCCESS in 1h 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a4ecf4782b0f22ae380e869a31695c54ebbf5b4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 02:00:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be37935e0a37497ab4d7db03876f0c3c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e538694174064b7fb922e3d8cbaa0838 : SUCCESS in 1h 05m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d83b765c6f134390ab28ce4c3f846e26 : SUCCESS in 54m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/81c60d9cb1764387ade7808caf0b40e8 : SUCCESS in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b73549f35414c27dacf9405c61bd40b7848dff93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-18 03:35:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"68c23feda256624e59b214ec60627237764e502b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-06-18 04:14:49.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2026/Arista_Tempest_Testrefs/changes/67/993767/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b50ee7406a3e33f3ba91ec37b9ff23c336d3b46e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 05:43:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d69d0ce795c4b5ba28abeced7eef0f3\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5d3e3c400afa475e91d80637f4f4d8f1 : SUCCESS in 1h 30m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64c89dde633144a9af68bc75185c88a6 : SUCCESS in 1h 05m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/956022351d414106b05a7c0eb8a499b0 : SUCCESS in 2h 07m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7d52c977881b463aaba6ca199a73b94e : SUCCESS in 34m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cbea15046524b82868c592861500484 : SUCCESS in 37m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f17552fc294b2ab5dcb93d8ad04cfc : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa880b7497eb45839c52719a8429e6cb : SUCCESS in 33m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21d3994b93cf4ca89fb23cbdfee2edc6 : SUCCESS in 18m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3fd85b2ba6a46518e2b17bda7f0d6c2 : SUCCESS in 33m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9505166c2a7644e48ed159b9482bfaa1 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/603c683ede694164b068af37ca2a6698 : SUCCESS in 9m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/84cbcae53bd246979a788947091a978e : SUCCESS in 1h 29m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0ca7390a1d6415ea2202c044f1d17ce : SUCCESS in 43m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5bfc7b70f7e2405d9349b55c3095e066 : SUCCESS in 52m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf279a51b2134ad8a3501ad7020e12fd : SUCCESS in 36m 37s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"854ef0c8a9d4211353f6677c2addc7c62b02303e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-18 06:41:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"201959183481147b8b820b93d4e53d3a6dfcfe00","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-18 08:34:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37ade04fe173310a10fc2ca6c231b09828b678f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 08:54:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a58258b8506460981015dc087b527ee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/603eccbe044440799cc67b368571a5e1 : SUCCESS in 50m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/733fbbe886ef40b49d969cf148fdfeb6 : SUCCESS in 54m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/492ad8460eb54fa693ce52a54c40d42a : SUCCESS in 52m 26s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dbc88569638c62f90c36059eeb49ac7b5b38c9ae","revisions":{"3ed8df705f7768696402336cff4c769dbb28330c":{"kind":"REWORK","_number":1,"created":"2026-06-17 14:40:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/993767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/993767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/993767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/993767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/993767/1"}}},"commit":{"parents":[{"commit":"84a79cf3a4ada7f12a922988f8df8253f9664d16","subject":"Merge \"Remove the unused ``api_workers`` config knob\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84a79cf3a4ada7f12a922988f8df8253f9664d16"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-17 14:12:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-17 14:40:13.000000000","tz":540},"subject":"Do not fail when api workers count can\u0027t be determined","message":"Do not fail when api workers count can\u0027t be determined\n\nIn case neutron-api is run by a WSGI server different from uwsgi,\napi workers count can\u0027t be determined using the uwsgi library.\n\nIn case the count can\u0027t be determined, wait for the cache timeout\nperiod.\n\nRelated-Bug: #2155306\nChange-Id: I41eaa2e77ff6b907e376b6ab3250beb8659d45da\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ed8df705f7768696402336cff4c769dbb28330c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ed8df705f7768696402336cff4c769dbb28330c"}]},"branch":"refs/heads/master"},"dbc88569638c62f90c36059eeb49ac7b5b38c9ae":{"kind":"REWORK","_number":2,"created":"2026-06-18 03:35:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/993767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/993767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/993767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/993767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/993767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/993767/2"}}},"commit":{"parents":[{"commit":"84a79cf3a4ada7f12a922988f8df8253f9664d16","subject":"Merge \"Remove the unused ``api_workers`` config knob\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84a79cf3a4ada7f12a922988f8df8253f9664d16"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-17 14:12:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 03:35:09.000000000","tz":540},"subject":"Do not fail when api workers count can\u0027t be determined","message":"Do not fail when api workers count can\u0027t be determined\n\nIn case neutron-api is run by a WSGI server different from uwsgi,\napi workers count can\u0027t be determined using the uwsgi library.\n\nIn case the count can\u0027t be determined, wait for the cache timeout\nperiod.\n\nRelated-Bug: #2155306\nChange-Id: I41eaa2e77ff6b907e376b6ab3250beb8659d45da\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbc88569638c62f90c36059eeb49ac7b5b38c9ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbc88569638c62f90c36059eeb49ac7b5b38c9ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
