)]}'
{"id":"openstack%2Fneutron~994752","triplet_id":"openstack%2Fneutron~master~I5f9bdbb003dc158d8c7e9d371e0ac232cb0a8462","project":"openstack/neutron","branch":"master","topic":"bug/2069718","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-07-03 16:52:50.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-06-30 08:29:41.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"I5f9bdbb003dc158d8c7e9d371e0ac232cb0a8462","subject":"Make ``ovs_create_tap`` True by default","status":"MERGED","created":"2026-06-24 14:26:18.000000000","updated":"2026-07-03 16:54:29.000000000","submitted":"2026-07-03 16:52:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994752-bug/2069718","meta_rev_id":"17a012f1f745e63319fd41d9be802bdc400d2611","_number":994752,"virtual_id_number":994752,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-03 16:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-30 08:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-07-03 14:14:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-07-03 14:14:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-29 08:24:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 14:26:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2026-06-24 14:26:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-06-24 14:26:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-06-24 16:50:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 08:28:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-07-03 14:14:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"17467c9acbf069b1c300ef1358a8b5fdc8ef3e4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-24 14:26:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03fc031b59c7bf688cd04dc43d257bf0f27e0864","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-24 15:41:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"882252dff7acba3f80ba604fb07c4aef14fdb75f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 16:50:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09b87b0dd6034b2faf664f85a9d925ae\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/50913b11fb374843bc7a3e1ea0001546 : SUCCESS in 1h 09m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c87a6dbc07734710b4f7f99f729a9a6a : SUCCESS in 39m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/93d6bd1089c645aaa03ceb6e1a6e7fd4 : SUCCESS in 1h 32m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/851b4236dec2436fbc7e54b7b96ee84a : FAILURE in 2h 20m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb0c9f420c1743ab9495bea9f4de3d63 : SUCCESS in 35m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bab7949f838b4cada0782d43ee2f60f7 : SUCCESS in 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e9171ab30648c796f3242f4bd3270d : SUCCESS in 6m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54de5cd5aa11403f94d503e622db9cbf : SUCCESS in 29m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b52af1d956ea4de88ba2e9529ba7ccec : SUCCESS in 20m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/55e8fcd6452949948c196f411c76d315 : SUCCESS in 22m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a391143cb50e4249b71e373f0d8bfb9e : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b1a56735e14144a46df5ebe93404ef : SUCCESS in 9m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/971cc7025dd94840bdaa1c094d58c6e0 : FAILURE in 32m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3a2cbdb3bda4864915f478f392aaf60 : SUCCESS in 43m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eaef8fa42649427085de64b62bff153f : SUCCESS in 40m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2659c28ef16841248cfbdc2d41c5dad2 : SUCCESS in 1h 45m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3a62b50a953542ef9a098a0d1303431d : SUCCESS in 1h 07m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9aba76e7b8864b9795876d99498a6c37 : SUCCESS in 56m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7345b2c22c1c4c38a2d318397fc34aa9 : SUCCESS in 38m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5697b2a91e8f47e4bbdabb652d8eba1c : SUCCESS in 42m 32s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"289b2280dad008fd69c8e5cec897d821dd087c34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 18:17:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbad2fff5fe2468990d5f8da0688e1ca\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4a42eca50b8c46378f14946fe742b81e : SUCCESS in 56m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/54fc2025aaa94ef18838395de88617d8 : SUCCESS in 42m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6fb7eec1e3b84419b10cc71eeacaec39 : SUCCESS in 1h 07m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79945e63ac6ff57ae34219250247c3ba937a7379","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-25 07:45:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3f11125c57146951f0732e9f7adf628b60a18378","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 08:43:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7fdb6ebc5c34c22ac0c98cfffeefa24\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c75ac1d9bd104aa7ad89a26155b1c22e : SUCCESS in 54m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/573857add4c04a67ab3e34e9ee0b7ea4 : SUCCESS in 37m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bf2db475a9e046ddb4b63dd6747e563a : SUCCESS in 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ced328ba06e6ba9a9e347c3287f14ab3c4fe11a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 10:07:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a16e80814cbc408bb0194a9a8989e23a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ba2300ee338a47358b8446418f93a683 : SUCCESS in 1h 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/399c381f75004c739b3f01bda4be7af9 : SUCCESS in 1h 16m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9069f3d40ce24f46b8c02ccff1894238 : SUCCESS in 1h 29m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/679cb0fa7a9349f6b3d5b11032f9807e : FAILURE in 2h 13m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea1ce925ea46495ea7d199d5a445eba2 : SUCCESS in 32m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39a9e0bd9175450da990b9b2079f6b3b : SUCCESS in 45m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5937cc7420d8448fbba8eff81ff09e3d : SUCCESS in 12m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/804e3d3849dc4526bf9c357e73451edd : SUCCESS in 29m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c4ebd9c1060427f9a9b1f12a849934a : SUCCESS in 24m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bce88236d4b74f21830e9bb27f4add96 : SUCCESS in 31m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d6405b4b52426c92f8dc55da40081e : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4155826b3b4346be886c94cc970e77a0 : SUCCESS in 16m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/35221c5166414470bf9f286aa3222bc7 : SUCCESS in 1h 29m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9abca9bd9174ca9b4224dd753a6004c : SUCCESS in 1h 15m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd3d934caf9b4695a2e9c2ff8758b0c3 : SUCCESS in 1h 17m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c87b9b65e6bd4fd58d8bfa2ec9d37a98 : SUCCESS in 1h 45m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/22b910d1e743436f948400fc3e782ede : SUCCESS in 1h 42m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e5c2c97673ef4ce786bb8d2734fe4c92 : SUCCESS in 58m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4c93fb9d4e754e8aa7a2a259d1c30c66 : SUCCESS in 1h 02m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1b9bee798af34523baf83842b6cfab90 : SUCCESS in 1h 36m 11s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"51216e7c7b8fb36c1c59fd7929321e83d30c8922","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-26 08:28:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94d11050d6bd78959037245e8ef1bea3e7bba4df","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-26 13:29:59.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"140988e4c199c7cc937da3ff4ac478cf423236dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-29 08:24:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c19c2641f1a61d96d0592c9c72a1c8a36a2618c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-29 08:24:23.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"0ef5f4dc31f0b268d4669b5640716106d20da0ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 09:07:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d4600fe88944827be3674727afef25e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a77069af19043869adea63d6a3439ea : SUCCESS in 41m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b02b9b4acb594ee6aef30bddf47e100b : SUCCESS in 24m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d75cded0de5b457c926064dabb68d84c : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a930e1cc1b62f42d2d78b27cf22090391647d554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 10:40:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba65178fb033463ba3b47cca46a10f10\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1ad11631b7924543ad6040b5cf548807 : SUCCESS in 1h 34m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5322ec4d986a4129969a9016d7fff2bf : SUCCESS in 1h 12m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/89718f7ba0a14925bcd2d5c7e697cbf6 : SUCCESS in 1h 20m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d7926db53dc44aaab6da4f2dcbf8b79 : FAILURE in 2h 15m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04b4f72875cd40f48cb148540c95325a : SUCCESS in 30m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4b2a3dfa3414ff19d4c0a87b6cf8190 : SUCCESS in 1h 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd0f30e6686a40efa01c251cac2c5948 : SUCCESS in 11m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be60005fa4054e16afd4dd961dbd869c : SUCCESS in 20m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7be2ac4aa96f4df3894d2c57aa15529f : SUCCESS in 16m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0fe735f522ff4af99a81fbd6a74443a4 : SUCCESS in 24m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e480d695e2a14eb9b486059fbb033b27 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5505f48ebab4a17971538afad656d43 : SUCCESS in 22m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/493615d2ac754794b448e6c922d84f57 : SUCCESS in 1h 24m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3b9356f531b4c77b8c8a9c344434664 : SUCCESS in 42m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/553788b58c024d21ae5887a726617133 : SUCCESS in 1h 13m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4f2a46a4840a40c4bde80be90f375084 : SUCCESS in 2h 02m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/313bfeadf6e042f9a32ae3006a09f9c1 : FAILURE in 1h 07m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b79620d550694239976a6668500b4556 : SUCCESS in 1h 05m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f29da017377d4ee6813cf6b0f899972c : SUCCESS in 35m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ef0177044764a25a897812ec61ccfbb : SUCCESS in 42m 10s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"13203d775b5ce579efe31ff0b961c9439f2ea7a9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-29 12:14:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3772f58fc7c12f17fe1fbbcf056666e0ee8c3f83","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-30 08:29:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7c1f09914dcd03b2290118a1703880529f7f8cd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 10:02:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5859c2646d874bbb9b2dae92fdf5475a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e93ae1db45074e1a80b1e9133c5c5bd9 : SUCCESS in 44m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c22ff0b6e7a24713a6e6343ef28e5084 : SUCCESS in 46m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8607f476d72f468bbb0b56e3bca98e30 : SUCCESS in 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4f11eafea0546c1337e63845ac18b76f047e0858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 11:08:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75efa0ca3d7c4164bf03cfea5b14f987\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fd51116e994649babf8d71782d3d9a51 : SUCCESS in 1h 22m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88cdbe303e1c40fc9d1e8938a34f13d9 : SUCCESS in 1h 11m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/60afd3299df44775b4956396fb79d90a : SUCCESS in 1h 15m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a0afdb24fbfb4165916caed493f41f5b : SUCCESS in 2h 35m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07608ff9e43f4926bacbf6d805543e1e : SUCCESS in 30m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f31d2b5942a2412ead0b8a3b56542156 : SUCCESS in 1h 09m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d120da86f2ca44eabd60349db2b4ec94 : SUCCESS in 10m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2afb3fb3aeaf4797bd0a27acc53b23d3 : SUCCESS in 19m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f839424e3bc46fdad45f06a03a3b8fd : SUCCESS in 26m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c904e71dd544574865b766c77a666af : SUCCESS in 27m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d74da06e387440882694510d6a4a436 : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1dbe3c9701f4181a5407d25787f3a4b : SUCCESS in 10m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d0dd9f150b3848049856b71e0bd978d9 : SUCCESS in 1h 31m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6e953c652f24dbfa7fd35f21eb68405 : SUCCESS in 1h 19m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ef6b56857994434af8f3108d7eb26e5 : SUCCESS in 1h 17m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/43d1c8fdbc0545b4a67f275e3b71b272 : SUCCESS in 1h 56m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/52a3aaa93d674af88f4c89245e88b751 : SUCCESS in 1h 33m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0f6ccbe3841248f48e8d9dee934071c1 : SUCCESS in 1h 03m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/12d6e130155941e6bede1979906c1776 : SUCCESS in 1h 14m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/411756b56ba54b2c8a3363e4f24fc5e8 : SUCCESS in 1h 34m 34s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"f002ec5f7a168b82880d66882a6cab3ae6d82b4d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-30 11:14:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e2a692bd391e486803686be5170c7dc3f4577532","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 13:41:23.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46034bf87ad940a99d1ff36d8b14d668\n\n- openstack-tox-py313-with-oslo-master https://zuul.opendev.org/t/openstack/build/b7f6b739c6e041608f5337759e959ce9 : SUCCESS in 23m 07s\n- openstack-tox-py313-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a1bed66de345497a8067ac3f0a51c875 : SUCCESS in 17m 14s\n- openstack-tox-py313-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/97849b9178f44967b18387343c677ff1 : SUCCESS in 18m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/74028b47871944029243149339b9416d : SUCCESS in 2h 00m 40s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/58d39ade0b95413186b140dc202963f6 : SUCCESS in 1h 26m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/ba0aec0b493c4a0784a81bdc33052f1f : FAILURE in 1h 33m 30s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e6f83aa761ff4878b61e01c73dac4aef : SUCCESS in 2h 19m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e019521ad30a477c906e3e176fda9557 : SUCCESS in 1h 44m 37s\n- neutron-functional-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a8de0883d6ac446fb18b8e4f4b601709 : SUCCESS in 1h 28m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f6ad3bf29b3646f6baa013993d3f9a4f : SUCCESS in 1h 18m 52s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/fc536808ff4c4fc1ae9d1c89494d9f8a : SUCCESS in 1h 15m 09s\n- neutron-ovn-tempest-loki https://zuul.opendev.org/t/openstack/build/622ec7f847044e0485d25d7055c20f46 : SUCCESS in 1h 12m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/05448e7edbc84c2ea8ca292336a0c199 : SUCCESS in 1h 21m 20s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/e51e818087f34e40b8312259bba2eba6 : SUCCESS in 1h 07m 35s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/62829473ea64434e8dddeec7107f2849 : SUCCESS in 1h 37m 38s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/743258378ced471dac5c5926ef6b8bb7 : SUCCESS in 56m 12s\n- neutron-functional-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/4a59f010522d4577bee243759180f7cf : SUCCESS in 1h 23m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/95343192cd674e3ab0a8339db259092d : SUCCESS in 43m 23s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/113ee25fe3f042b8ae75ab3904de4001 : SUCCESS in 52m 24s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/747bc35caed34d11999be618d1784605 : SUCCESS in 2h 07m 17s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/5df187f6adab40deb09c2ce62950db6b : SUCCESS in 44m 55s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/3621a292377749e89d733e9547d0a395 : SUCCESS in 52m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/c8639ef527a744b1a0f6d9b2b7016688 : SUCCESS in 1h 15m 20s\n- neutron-ovn-tempest-ovs-master-centos-stream https://zuul.opendev.org/t/openstack/build/e12f5a2f77b846c1826473c9fbc5dc30 : SUCCESS in 1h 11m 06s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5bebe054e096477da856bb6e95a86137 : SUCCESS in 57m 08s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/f2fb1f951e3b47fb9248348d80b373d7 : SUCCESS in 59m 32s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/dc00d681710a4ffc86b69c888605b7d6 : SUCCESS in 1h 28m 14s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/343ab39c3d38421dbe82c606e5e1d6d1 : SUCCESS in 46m 49s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/5fb59895988e45fcae924a109510056a : FAILURE in 2h 24m 21s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/efefc1f5ef8d4c27999750362af600d6 : FAILURE in 2h 00m 48s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/55d61e6199ab40fe9d530c77d4483f6d : SUCCESS in 1h 09m 58s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/bc3aae0ff44242ca8a741ce0bb582842 : SUCCESS in 57m 05s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1806358c90f44dfa93a196d96943dae0 : SUCCESS in 48m 22s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/0c8180cb24e24aeb97752604255df866 : SUCCESS in 1h 16m 12s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/4c06a23301df400fae01d012bf299c39 : SUCCESS in 41m 51s\n- devstack-tobiko https://zuul.opendev.org/t/openstack/build/902969d3e4114e37a412437cddc0639f : SUCCESS in 47m 42s\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/7ce8897828b746479a6b4c25bac3c9c1 : SUCCESS in 1h 06m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0f8292d20e384b069f28f27b9f7f3cc3 : SUCCESS in 1h 33m 23s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/7a9927edd3eb4a479d1100bec2b8889a : SUCCESS in 25m 25s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/041c8da24d634a818058c453152b7ce8 : SUCCESS in 28m 16s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/8d876311dafd49d781253d1bcdb3e8f8 : SUCCESS in 50m 01s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/175b20c62e0747c48bb155e08554ff73 : SUCCESS in 1h 13m 08s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/7dca2501d7fa47e68c72dd3422c503ca : SUCCESS in 1h 04m 39s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/4757ff8659d047108600e50150ef8f0a : SUCCESS in 46m 43s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/a6113206f4ab4cbe81adebd326e0b7d7 : SUCCESS in 1h 14m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c92175e596714b6cb7a5b68f3fd5f88f : SUCCESS in 1h 04m 35s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/756d066965ab46bb938f81443bb9e37e : SUCCESS in 1h 18m 27s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"c4f6aeedcfdc9aa52a64eee5dff28f0b217afa26","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-03 14:14:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7303f14e57afbb1a61e779d1bfd2383ddebafdac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 14:14:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"51e55611c39ed2f6ff3b341f8ba3d81f3bc53593","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 16:52:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e54c3337954644509619906f4f97ceb7\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/caea233cfee9496facaed53476ae0ff3 : SUCCESS in 1h 40m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06b6c46efcae48258c6b5b210781d818 : SUCCESS in 1h 09m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/63f2c48de74b4a3c874fe66496c23a8f : SUCCESS in 1h 42m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dc815e1e26a347978c3b99aef54ca7ff : SUCCESS in 2h 07m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d769c8f67b45b1acc15df346d89a9f : SUCCESS in 13m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/253c095344144870b3ea01851557e9a3 : SUCCESS in 37m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d48bfd7e19834014967ec029eaebc1f2 : SUCCESS in 30m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba321033b23454eac3b69574faad66b : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d41ced59c22f4ab28c9d3ab06959f1e0 : SUCCESS in 22m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c2679b7fda6e4511bf111ded7f63fdac : SUCCESS in 1h 16m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7a7b7484b66e492697864e7798dcd21d : SUCCESS in 1h 57m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/450668e6a3c546debbfd0596c131f272 : SUCCESS in 1h 20m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c31058a05b5345db9081a44f03570444 : SUCCESS in 1h 15m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/636b8c536f7847c48c21174d59d5c9c7 : SUCCESS in 1h 07m 52s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"eef74de234cc357dac61c4425774dc5b64fc8ecf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 16:52:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"17a012f1f745e63319fd41d9be802bdc400d2611","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 16:54:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cea8f7c6efde4ea9added69bb5d55ef0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d84e850ee224b57832c0d17c91d8c8d : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1d22e94ec3f4220859cad54c1b27fb5 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2a41d493e022ad0b45a3266d6909b7b508a9b3c2","revisions":{"2b111a1663e9cf5e121b4a7aadf20a39a471e810":{"kind":"REWORK","_number":1,"created":"2026-06-24 14:26:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/994752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/994752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/994752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/994752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/994752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/994752/1"}}},"commit":{"parents":[{"commit":"1cab59daa0aee66d23150f7d1f92c82e0a960ab6","subject":"Merge \"Add configurable HAProxy timeouts for metadata proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1cab59daa0aee66d23150f7d1f92c82e0a960ab6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-24 14:20:39.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-24 14:20:39.000000000","tz":120},"subject":"Make ``ovs_create_tap`` True by default","message":"Make ``ovs_create_tap`` True by default\n\nThis flag was introduced in [1][2] in 2026.1 and was \"False\" by\ndefault. Now in 2026.2 all the needed code is in place in Nova,\nNeutron and os-vif. This patch is swapping it to \"True\" by default\nbecause:\n* Of the benefits this flag provides to the live migration in ML2/OVS\n  and ML2/OVN environments.\n* It will be deprecated in the next 2027.1 release assuming the \"True\"\n  value.\n\nNova is also updating the CI jobs to make \"ovs_create_tap\u003dtrue\" [3].\n\n[1]https://review.opendev.org/q/topic:%22bug/2069718%22\n[2]https://review.opendev.org/c/openstack/neutron/+/972039\n[3]https://review.opendev.org/c/openstack/nova/+/975500\n\nRelated-Bug: #2069718\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5f9bdbb003dc158d8c7e9d371e0ac232cb0a8462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b111a1663e9cf5e121b4a7aadf20a39a471e810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b111a1663e9cf5e121b4a7aadf20a39a471e810"}]},"branch":"refs/heads/master"},"a1320c5ff3e61540dfaa4b52ff3fa5c13a4cbca5":{"kind":"REWORK","_number":2,"created":"2026-06-25 07:45:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/994752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/994752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/994752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/994752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/994752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/994752/2"}}},"commit":{"parents":[{"commit":"a29365dc917d932a3fea8844da37e16179b9e969","subject":"Merge \"Fix PVLAN multinode support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a29365dc917d932a3fea8844da37e16179b9e969"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-24 14:20:39.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-25 07:44:17.000000000","tz":120},"subject":"Make ``ovs_create_tap`` True by default","message":"Make ``ovs_create_tap`` True by default\n\nThis flag was introduced in [1][2] in 2026.1 and was \"False\" by\ndefault. Now in 2026.2 all the needed code is in place in Nova,\nNeutron and os-vif. This patch is swapping it to \"True\" by default\nbecause:\n* Of the benefits this flag provides to the live migration in ML2/OVS\n  and ML2/OVN environments.\n* It will be deprecated in the next 2027.1 release assuming the \"True\"\n  value.\n\nNova is also updating the CI jobs to make \"ovs_create_tap\u003dtrue\" [3].\n\n[1]https://review.opendev.org/q/topic:%22bug/2069718%22\n[2]https://review.opendev.org/c/openstack/neutron/+/972039\n[3]https://review.opendev.org/c/openstack/nova/+/975500\n\nRelated-Bug: #2069718\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5f9bdbb003dc158d8c7e9d371e0ac232cb0a8462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1320c5ff3e61540dfaa4b52ff3fa5c13a4cbca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1320c5ff3e61540dfaa4b52ff3fa5c13a4cbca5"}]},"branch":"refs/heads/master"},"2a41d493e022ad0b45a3266d6909b7b508a9b3c2":{"kind":"REWORK","_number":3,"created":"2026-06-29 08:24:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/994752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/994752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/994752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/994752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/994752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/994752/3"}}},"commit":{"parents":[{"commit":"d0442f1c1a9be8746b1a7869a02d1b8b99622abe","subject":"Merge \"ovn: Handle missing ``neutron:network_name`` in sibling LRPs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0442f1c1a9be8746b1a7869a02d1b8b99622abe"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-24 14:20:39.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-29 08:24:01.000000000","tz":120},"subject":"Make ``ovs_create_tap`` True by default","message":"Make ``ovs_create_tap`` True by default\n\nThis flag was introduced in [1][2] in 2026.1 and was \"False\" by\ndefault. Now in 2026.2 all the needed code is in place in Nova,\nNeutron and os-vif. This patch is swapping it to \"True\" by default\nbecause:\n* Of the benefits this flag provides to the live migration in ML2/OVS\n  and ML2/OVN environments.\n* It will be deprecated in the next 2027.1 release assuming the \"True\"\n  value.\n\nNova is also updating the CI jobs to make \"ovs_create_tap\u003dtrue\" [3].\n\n[1]https://review.opendev.org/q/topic:%22bug/2069718%22\n[2]https://review.opendev.org/c/openstack/neutron/+/972039\n[3]https://review.opendev.org/c/openstack/nova/+/975500\n\nRelated-Bug: #2069718\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5f9bdbb003dc158d8c7e9d371e0ac232cb0a8462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a41d493e022ad0b45a3266d6909b7b508a9b3c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a41d493e022ad0b45a3266d6909b7b508a9b3c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
