)]}'
{"id":"openstack%2Fneutron~994936","triplet_id":"openstack%2Fneutron~master~I0bb7b02e9ff892b4b7001a4f653b18e3d0aa4781","project":"openstack/neutron","branch":"master","topic":"fix-network-context","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-08-10 15:50:30.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"18031":{"account":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"last_update":"2026-08-14 12:37:46.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"removed_from_attention_set":{"35928":{"account":{"_account_id":35928,"name":"Mohd Nisar Khan","email":"nisarkhancsengg@gmail.com","username":"nisarkhan"},"last_update":"2026-08-11 11:40:51.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-08-14 12:37:46.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I0bb7b02e9ff892b4b7001a4f653b18e3d0aa4781","subject":"Fix use of NetworkContext","status":"NEW","created":"2026-06-25 16:19:08.000000000","updated":"2026-08-14 12:37:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"5f8f252f5aa4ea09e7adcbfd69e407c58d873479","_number":994936,"virtual_id_number":994936,"owner":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":35928,"name":"Mohd Nisar Khan","email":"nisarkhancsengg@gmail.com","username":"nisarkhan"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-11 02:29:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":35928,"name":"Mohd Nisar Khan","email":"nisarkhancsengg@gmail.com","username":"nisarkhan"},"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-08-11 11:40:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35928,"name":"Mohd Nisar Khan","email":"nisarkhancsengg@gmail.com","username":"nisarkhan"},{"value":-1,"date":"2026-08-13 08:08:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":35928,"name":"Mohd Nisar Khan","email":"nisarkhancsengg@gmail.com","username":"nisarkhan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":35928,"name":"Mohd Nisar Khan","email":"nisarkhancsengg@gmail.com","username":"nisarkhan"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35928,"name":"Mohd Nisar Khan","email":"nisarkhancsengg@gmail.com","username":"nisarkhan"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-25 18:38:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-25 23:20:57.000000000","updated_by":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"reviewer":{"_account_id":35928,"name":"Mohd Nisar Khan","email":"nisarkhancsengg@gmail.com","username":"nisarkhan"},"state":"REVIEWER"},{"updated":"2026-08-03 21:51:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-08-04 12:54:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-08-04 12:54:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-08-07 07:56:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"CC"},{"updated":"2026-08-13 08:08:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"81ce986072c614ba8b7aaa9f5cb7ccd913a61c10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2026-06-25 16:19:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d34f5a627b4b9da38ffe99d1848a9862cabf85f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2026-06-25 16:27:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c70ab5a8f99210ebb329d070990cd49426589fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 18:38:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd1a97ec59ab40d7ad252292a526a22d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/15533f2872844841acc59176ee40970b : SUCCESS in 1h 29m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a8dcd83be2a496e9de367cb32c33f6c : SUCCESS in 59m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d74a4c829a1c4698a7d09aaeb80216b3 : SUCCESS in 1h 52m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/535596245cd4441b94873cadf27b9671 : SUCCESS in 2h 06m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/733b04fc440145689bba195412e058b6 : SUCCESS in 34m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdd961681845442eb8fe4b4afad78b4e : SUCCESS in 1h 00m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/162910d2a70f4712ae8ced853916ea46 : SUCCESS in 7m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a694a0fca8f64500a0c4020a8ccf7e94 : SUCCESS in 23m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3545ba926f04d65836f5bda4f762330 : SUCCESS in 26m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b84972251aee4600816d7eb7b8444924 : SUCCESS in 27m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad486fdd6d2847a7a9d0714517446741 : SUCCESS in 15m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d91fe35e78194509b983d8409bf6b7ab : SUCCESS in 20m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c899b6b1090d4a7c9a74703a23b19a21 : SUCCESS in 58m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7ca29473ca045a19afc887024761c66 : SUCCESS in 42m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/315ca4f556d64538a9957edda64787ae : SUCCESS in 1h 25m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/37cfd11d14a34ae2b5bc4934dcd3e841 : SUCCESS in 1h 05m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/adc7f42a079a444d813829a474d63692 : SUCCESS in 1h 22m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1816af34a1a3453f8adda40a88049206 : SUCCESS in 1h 11m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8de680f5d9a2413086290af6dab1ac86 : SUCCESS in 1h 10m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/215a3addcd5e4501909a5a8286ea489f : SUCCESS in 1h 20m 55s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"5e0a8bc4bf6938b9c65b9801e5e8ccf29e614c86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 21:15:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a5bcf219b6d4ac1863a61295205cdd9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45e33e73760144a3b9a6229491fa8f75 : SUCCESS in 40m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/830183f2e6c1409aafc111c432042e91 : SUCCESS in 23m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/34d48322c7a846d79165bbfac4c68815 : SUCCESS in 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02876428de62093f7b2d285a381eb58f89bb0514","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-04 12:54:18.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"96ddd224aa036b7af34f7279856ac5acca67542f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2026-08-04 13:44:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"6dc84dc261fba287f695dbdbe015a6e5516dfb1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 14:56:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1d7964b7eb4203bac5bb8d46584dae\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c3ef9fa2eedc41bdb33b5b5a8379da5c : SUCCESS in 45m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9d3cf531f9e14696a63f69495b55d052 : SUCCESS in 18m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6432c48a57ce473180314e11580202b4 : SUCCESS in 1h 08m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3929f97bcaea7332495acede3a7e5ddcc48fa319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 15:31:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d213d3d1c6b14fca8a3fbe8b76767f02\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3bb1fbc87ec94aa98eacd3905c11ee42 : SUCCESS in 1h 18m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd3204875b7144c6afb836af74485e97 : SUCCESS in 34m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bc6b736f6abb447d838c54e2f732d829 : SUCCESS in 1h 10m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d2151031614d40e58d703795677fef62 : SUCCESS in 1h 44m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1bece001b0db4d288205b2396f597f62 : SUCCESS in 31m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeb40925b76341f78bd31de052aadd63 : SUCCESS in 44m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74d0e794e1e14e69ae18d6a2e6f82c6b : SUCCESS in 12m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b49e6359d894c3c9c4d6cf3f08a50d5 : SUCCESS in 32m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/645e702be3984157850b84c23fe27efa : SUCCESS in 29m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a8ff77ca3bbd48ddb43d1a3b013a7a70 : SUCCESS in 28m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a788c3251194fd687fd1f7667325d88 : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3a2a345d499473a9c0c2980da592354 : SUCCESS in 26m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0d4cfbfb873547ff84bd30b2b3fec8f9 : SUCCESS in 1h 11m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f0ddfb78e69497db065bb888cda62c6 : SUCCESS in 1h 02m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6259a07f6a264e95839ce6bf3812b560 : SUCCESS in 1h 17m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/73eceb5087f7487b98dd2e79e64c5af8 : SUCCESS in 1h 33m 02s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d0af496a1c084c7184f28a6427a9ff15 : SUCCESS in 1h 26m 13s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/608b691a79de49b0b66055b6dedaa5a4 : SUCCESS in 57m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5e32085994164ff29194886a6ac89c39 : SUCCESS in 53m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3054df1515614b70bd9f7b15749bd0ba : SUCCESS in 1h 25m 06s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"f8ed0bbdef735022016d43693997a8017fb2f7c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-07 08:23:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"74959468f533fc5f518caa76b86688b286c846c8","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2026-08-07 12:20:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d66d09d327a245f0c5f286709063de50d368d862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2026-08-10 15:50:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"97e33c37a83421c1b42ad54fc2dfd54a951a4485","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2026-08-10 15:52:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be4c9eec6ecbfcce7a2e513aad9f3f81704cee96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 16:50:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bffd8916e540425ba84282be96008fdf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d758a7d97bc4e619c15b0470e510556 : SUCCESS in 43m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d4585c580e0a462eab62c7aadfbc4aed : SUCCESS in 42m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/68fec47acb86489599f749c4e1898b69 : SUCCESS in 56m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e227257cf4f66feee922a013de54e74ad4456dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 17:50:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b9be444dc4f4c14b25956bc78e30b09\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/012d4d9c018f4ca0a90569f738015eb8 : SUCCESS in 53m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e37e31782aff42ffafa9ca83136bcaa8 : FAILURE in 40m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/feb78dc1875247819ab58e2331568a12 : SUCCESS in 1h 00m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dbd68e350d774f48867653b71ea5d70a : SUCCESS in 1h 58m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e484764c0de94171b3b68cb97204d683 : SUCCESS in 30m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1785847a1c454e388d4402f46c4f49eb : SUCCESS in 1h 06m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e197b4c2724e4b22b1ef09285d558787 : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1e68c30f9e343b18026720b77d520d7 : SUCCESS in 25m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0bc12bb5dc9740e0890928d2626a92d2 : SUCCESS in 21m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/23c3fbc5fe3a45e1ad19bdd6b174d13f : SUCCESS in 27m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b36059e2868a497585f970a9ac8e51e3 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df3cf1af306c4055bb9efe6698eafdf6 : SUCCESS in 12m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/85fb04288cd14dee82811bf6fe228535 : SUCCESS in 52m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8594c12cfed640d4a29592bb5c3313ba : SUCCESS in 49m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a2cb906b0b84616aca74f21c68fe1c6 : SUCCESS in 40m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/43aa4d704679440f92e9ff4864f4c87e : FAILURE in 1h 14m 03s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/316fa8e2a4e3417185b979228b1d8535 : SUCCESS in 39m 54s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7ed645b49f4f47f2998cebe2d1a51e30 : SUCCESS in 39m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/348c89e27b9e4344a492661d6bd8afad : SUCCESS in 31m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/64747046cfd44cc7ac87f47a7934046c : SUCCESS in 1h 26m 12s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"dfc4969dcf8bfe793e765af65e9f7e8f1f3c18d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2026-08-11 00:47:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ab8c884e5604d1aee2b9663f74b4fd2e973b63a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 01:47:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3f2fad07623486bb9f4e4d3ebce657c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad54f99b47284d1d8468cef4afc308b1 : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4de7f228cca44f69cfcc6594d51f5cf : SUCCESS in 36m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f9f01129ee604528ad55bf461ac1783b : SUCCESS in 58m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f0a2c18be7d5c7a6005d7cf5c2c74001f29c5665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 02:29:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ed25a5c2ecd4c50a775d513e71ec615\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/812047bdfc8c44628fb189e3e97c906d : SUCCESS in 1h 23m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb19d4c0183c4096807585b931e06b1d : SUCCESS in 38m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5de3e0337a9c49d38ee346d60075c997 : SUCCESS in 1h 08m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cc05ea9797be493ba2f93e2b21a5d7ef : SUCCESS in 1h 39m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/350dd5b508664829ae21b042b7cf2c26 : SUCCESS in 29m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afbbe68a362e49fbb5e44d5394d5b203 : SUCCESS in 1h 07m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3928476d817b4116b81200febe608321 : SUCCESS in 12m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d9b2cd08dce43e5bdb61001ecbf032e : SUCCESS in 27m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ca8acc84207477ba2263dc784ce4f19 : SUCCESS in 25m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ad12f9b5a51e4376b492c2fd57ef6d27 : SUCCESS in 31m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a57889ee7a3149c1bcf4e9cb3a6c7e83 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc7c3e2d35714eb3b57725c8f1dbd70f : SUCCESS in 9m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4077809c041c4cc0beefa965faa1db02 : SUCCESS in 52m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef5a603314824c29badf5fd6e741ade2 : SUCCESS in 41m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef829eabbfd74fce8183d379a81e11da : SUCCESS in 56m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6b10506a7f7146e8ba7e5a3a8bf827e7 : SUCCESS in 45m 11s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d11a20c97574422a28c4672550a4a46 : SUCCESS in 1h 25m 46s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/edb423fbb76f40b1a26baa537ab8b8c3 : SUCCESS in 57m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/22612041aa684c9e87527356a81bb69a : SUCCESS in 1h 09m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56b9bba6626e48958b4a3b7f35ee669d : SUCCESS in 1h 15m 16s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":5},{"id":"7b3320a5a2a4568a85e335367d64b4fddb4b674b","author":{"_account_id":35928,"name":"Mohd Nisar Khan","email":"nisarkhancsengg@gmail.com","username":"nisarkhan"},"date":"2026-08-11 11:40:51.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d155c8bc122dc80353fffebd86881bbf5b7a757d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-13 08:08:13.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fc1e8ad6039452bc15aa1ebef899ad1668ad3988","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2026-08-13 15:45:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5f8f252f5aa4ea09e7adcbfd69e407c58d873479","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-14 12:37:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"147b0243443f264b194336320da295705f69ce02","revisions":{"d2d172c1208e3f8ba3602deb09393bcd3a3cea11":{"kind":"REWORK","_number":1,"created":"2026-06-25 16:19:08.000000000","uploader":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"ref":"refs/changes/36/994936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/994936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/994936/1"}}},"commit":{"parents":[{"commit":"c372acdcc19f4d3f95a2cd7d115e8655851e277c","subject":"Merge \"segments: Catch RP not found in ``_delete_nova_inventory``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c372acdcc19f4d3f95a2cd7d115e8655851e277c"}]}],"author":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-06-25 16:13:46.000000000","tz":0},"committer":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-06-25 16:18:55.000000000","tz":0},"subject":"[WIP]: Fix use of NetworkContext","message":"[WIP]: Fix use of NetworkContext\n\nThe patch in [0] was meant to fix the issue desribed in [1].\nHowever, the patch collapses the segments into a single level\n(level 0) when hierarchical port binding (HPB) is used.\n\nThis patch explores an alternative, where a NetworkContext\nis created and passed to the PortContext constructor instead\nof the network, where all segments (static and dyanmic) are\nused in the NetworkContext constructor.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/980637\n[1]: https://bugs.launchpad.net/neutron/+bug/2144501\n\nChange-Id: I0bb7b02e9ff892b4b7001a4f653b18e3d0aa4781\nSigned-off-by: Thomas Bachman \u003ctbachman@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d2d172c1208e3f8ba3602deb09393bcd3a3cea11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d2d172c1208e3f8ba3602deb09393bcd3a3cea11"}]},"branch":"refs/heads/master"},"dd769f536f6c608f0ba301b0139e25049f08e3a9":{"kind":"REWORK","_number":2,"created":"2026-06-25 16:27:28.000000000","uploader":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"ref":"refs/changes/36/994936/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/994936/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/994936/2"}}},"commit":{"parents":[{"commit":"c372acdcc19f4d3f95a2cd7d115e8655851e277c","subject":"Merge \"segments: Catch RP not found in ``_delete_nova_inventory``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c372acdcc19f4d3f95a2cd7d115e8655851e277c"}]}],"author":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-06-25 16:13:46.000000000","tz":0},"committer":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-06-25 16:26:56.000000000","tz":0},"subject":"[WIP]: Fix use of NetworkContext","message":"[WIP]: Fix use of NetworkContext\n\nThe patch in [0] was meant to fix the issue desribed in [1].\nHowever, the patch collapses the segments into a single level\n(level 0) when hierarchical port binding (HPB) is used.\n\nThis patch explores an alternative, where a NetworkContext\nis created and passed to the PortContext constructor instead\nof the network, where all segments (static and dyanmic) are\nused in the NetworkContext constructor.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/980637\n[1]: https://bugs.launchpad.net/neutron/+bug/2144501\n\nChange-Id: I0bb7b02e9ff892b4b7001a4f653b18e3d0aa4781\nSigned-off-by: Thomas Bachman \u003ctbachman@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd769f536f6c608f0ba301b0139e25049f08e3a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd769f536f6c608f0ba301b0139e25049f08e3a9"}]},"branch":"refs/heads/master"},"d997dd6a1c049e579b9d1e5666e30c18a3cd1628":{"kind":"REWORK","_number":3,"created":"2026-08-04 13:44:37.000000000","uploader":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"ref":"refs/changes/36/994936/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/994936/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/994936/3"}}},"commit":{"parents":[{"commit":"3cd91cd333b6b05bf335b724bfbb9889092d2897","subject":"Merge \"ml2: Add ``ovs_create_tap`` flag to ML2/OVS mechanism driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cd91cd333b6b05bf335b724bfbb9889092d2897"}]}],"author":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-06-25 16:13:46.000000000","tz":0},"committer":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-08-04 13:44:21.000000000","tz":0},"subject":"Fix use of NetworkContext","message":"Fix use of NetworkContext\n\nThe patch in [0] was meant to fix the issue desribed in [1].\nHowever, the patch collapses the segments into a single level\n(level 0) when hierarchical port binding (HPB) is used.\n\nThis patch explores an alternative, where a NetworkContext\nis created and passed to the PortContext constructor instead\nof the network, where all segments (static and dyanmic) are\nused in the NetworkContext constructor.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/980637\n[1]: https://bugs.launchpad.net/neutron/+bug/2144501\n\nChange-Id: I0bb7b02e9ff892b4b7001a4f653b18e3d0aa4781\nCloses-Bug: #2162752\nSigned-off-by: Thomas Bachman \u003ctbachman@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d997dd6a1c049e579b9d1e5666e30c18a3cd1628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d997dd6a1c049e579b9d1e5666e30c18a3cd1628"}]},"branch":"refs/heads/master"},"0c3c1146ab906069e6d7ee13c2639169c44c4135":{"kind":"REWORK","_number":4,"created":"2026-08-10 15:50:30.000000000","uploader":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"ref":"refs/changes/36/994936/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/994936/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/994936/4"}}},"commit":{"parents":[{"commit":"3cd91cd333b6b05bf335b724bfbb9889092d2897","subject":"Merge \"ml2: Add ``ovs_create_tap`` flag to ML2/OVS mechanism driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cd91cd333b6b05bf335b724bfbb9889092d2897"}]}],"author":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-06-25 16:13:46.000000000","tz":0},"committer":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-08-10 15:49:03.000000000","tz":0},"subject":"Fix use of NetworkContext","message":"Fix use of NetworkContext\n\nThe patch in [0] was meant to fix the issue described in [1].\nHowever, it results in both static and dynamic segments being\nreturned when binding at level 0, in effect collapsing all the\nsegments to a single HPB level. The mechanism manager\u0027s code to\nbind at each level expects to start with just the static segments.\nBinding at any level should return just the segments allocated\nfor the next level down, which are  used in the recursive call\nto _bind_port_level [3].\n\nThe problem with the patch in [0] is that once dynamic segments have\nbeen allocated, they will show up at level 0. This results in unintended\nbehavior: a mechanism driver that was only meant to bind a port at\na lower level (i.e. higher number) would instead bind the port to a\nsegment that isn\u0027t supposed to exist at that level.\n\nThe patch in [2] explores an alternative to [0], reverting this\nbehavior, and where a NetworkContext is created and passed to the PortContext\nconstructor instead of the network, where all segments (static and dyanmic)\nare used in the NetworkContext constructor.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/980637\n[1]: https://bugs.launchpad.net/neutron/+bug/2144501\n[2]: https://review.opendev.org/c/openstack/neutron/+/994936\n[3]: https://opendev.org/openstack/neutron/src/branch/master/neutron/plugins/ml2/managers.py#L919-L937\n\nChange-Id: I0bb7b02e9ff892b4b7001a4f653b18e3d0aa4781\nCloses-Bug: #2162752\nSigned-off-by: Thomas Bachman \u003ctbachman@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c3c1146ab906069e6d7ee13c2639169c44c4135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c3c1146ab906069e6d7ee13c2639169c44c4135"}]},"branch":"refs/heads/master"},"147b0243443f264b194336320da295705f69ce02":{"kind":"REWORK","_number":5,"created":"2026-08-11 00:47:26.000000000","uploader":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"ref":"refs/changes/36/994936/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/994936/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/994936/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/994936/5"}}},"commit":{"parents":[{"commit":"3cd91cd333b6b05bf335b724bfbb9889092d2897","subject":"Merge \"ml2: Add ``ovs_create_tap`` flag to ML2/OVS mechanism driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cd91cd333b6b05bf335b724bfbb9889092d2897"}]}],"author":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-06-25 16:13:46.000000000","tz":0},"committer":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2026-08-11 00:45:19.000000000","tz":0},"subject":"Fix use of NetworkContext","message":"Fix use of NetworkContext\n\nThe patch in [0] was meant to fix the issue described in [1].\n\nHowever, the initial commit for ML2 ([2], updated in [3]) expects\nthe \"original\" field to be relevant for update operations, and not\nfor create or delete operations.\n\nThe fix in [0] also breaks hierarchical port binding. The change causes both\nstatic and dynamic segments to be returned when binding at level 0, in effect\ncollapsing all the segments to the first HPB level. The mechanism manager\u0027s code\nto bind at each level expects to start with just the static segments. Binding at\nany level should return just the segments allocated for the next level, which\nare then used in the recursive call to _bind_port_level (see [4]).\n\nThe problem with the patch in [0] is that once dynamic segments have\nbeen allocated, they will show up at level 0. This results in unintended\nbehavior: a mechanism driver that was only meant to bind a port at\na lower level (i.e. higher number) would instead bind the port to a\nsegment that isn\u0027t supposed to exist at that level.\n\nThis patch reverts the change made in [0], and adds a unit test that\ncovers the bug described in [1].\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/980637\n[1]: https://bugs.launchpad.net/neutron/+bug/2144501\n[2]: https://review.opendev.org/c/openstack/neutron/+/33201\n[3]: https://review.opendev.org/c/openstack/neutron/+/182134\n[4]: https://opendev.org/openstack/neutron/src/branch/master/neutron/plugins/ml2/managers.py#L919-L937\n\nChange-Id: I0bb7b02e9ff892b4b7001a4f653b18e3d0aa4781\nCloses-Bug: #2162752\nAssisted-By: Codex (5.5)\nSigned-off-by: Thomas Bachman \u003ctbachman@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/147b0243443f264b194336320da295705f69ce02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/147b0243443f264b194336320da295705f69ce02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
