)]}'
{"id":"openstack%2Fneutron~995329","triplet_id":"openstack%2Fneutron~master~Ie674c8388fd415bf5b4e834e35311d273520f41b","project":"openstack/neutron","branch":"master","topic":"bug/2158628","attention_set":{},"removed_from_attention_set":{"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2026-06-30 12:36:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie674c8388fd415bf5b4e834e35311d273520f41b","subject":"Fix error handling in PVLAN","status":"MERGED","created":"2026-06-29 13:18:23.000000000","updated":"2026-06-30 12:37:23.000000000","submitted":"2026-06-30 12:36:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995329-bug/2158628","meta_rev_id":"3fc823b42d96fec6d3fe4757b189e73ee4898660","_number":995329,"virtual_id_number":995329,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-30 12:36:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-29 15:10:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-29 13:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-06-30 06:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-29 15:10:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-29 13:53:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-29 14:10:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-29 15:10:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-29 15:31:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-30 06:56:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"a9a9a7bbdb6e22fbe5977b405434c36974224c75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-06-29 13:18:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db4c897f99e4fbb8e94d5d6d57e626fbede6516f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2026-06-29 13:31:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9130d5b80fcdf64db8a4702d75c072bec0d0d980","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-29 13:53:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"caab3909ecc541fd9c57cb4612ac92add578fcb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 14:10:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7b0781cfc654ea5bf63702c1650bb84\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7f4f8b0d76ec4176bc6dc1b3a7fe0a85 : SUCCESS in 22m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7f60313fb77b4e8aa8756571e90fc1a6 : SUCCESS in 22m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dfea1ac6299c42748da3ba91e5545862 : SUCCESS in 37m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55937d8408087787f99494d433aadc45a61cf251","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-29 15:10:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b1c2bcebf72efa392781cebf5266550dfe1bb39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 15:31:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37678043597a488495b74aab717ac6d0\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7e93777067724153898b54338f2c0f02 : SUCCESS in 1h 26m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eac1ea530cbb44bfb49e2392e751eacd : SUCCESS in 1h 06m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/47fc9c176dbd444a97421ef1e60d2d6d : SUCCESS in 1h 39m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b82cf9c11c114a4eb5294f28ad48a103 : SUCCESS in 1h 57m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/17935d7b959146f2a87e3d83640264f0 : SUCCESS in 33m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc33f893529449fb94ff348c36047e3e : SUCCESS in 42m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4b12a6a02a4f6e97434b3002806759 : SUCCESS in 11m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6dfa398d0529428cbbb3630f5e4e15c2 : SUCCESS in 35m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5a0a0e92151436db3be7075e44a7b82 : SUCCESS in 26m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b9de29e2072432e9ddf521fea30ca75 : SUCCESS in 33m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f969a32ece554c45ad44643865cceada : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7bba6c76db74339bfc6fdd5f53185bd : SUCCESS in 25m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a810ad4c43ee4b22a2d633dd3b7d9e39 : SUCCESS in 1h 29m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9e86bb055b4421492bc09d0a9576965 : SUCCESS in 1h 10m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f189312201134c89adfcb4d6ab3b906d : SUCCESS in 1h 36m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/01a085d377fc4a0397247e746c523bb2 : SUCCESS in 1h 54m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eaad10ce160b4e05b9d1aabd7bd4cbd6 : FAILURE in 1h 29m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f7b4889f40cf4bf8baacf29799aa9ac4 : SUCCESS in 1h 06m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3371300406854249a06f281d4b696b65 : SUCCESS in 1h 09m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d4039bb8c753429d8e27095b2364439c : SUCCESS in 1h 30m 49s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"d320f639788a1eacc2b7d36f0585953a15803b68","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-29 16:24:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c2894f3fea3b10e850b5b3b4da9f64cf342b2d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 17:08:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21501df284d342e8868b813ff3ccee54\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c4af6ab8e684a0da097f6f4f21e533e : SUCCESS in 41m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f34bb5f4864548e08fc5b90e9cfd3bc9 : SUCCESS in 37m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cae9f4ff5f5f4ed49636ac72e442c9e2 : SUCCESS in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ffbadeb866aa97fdf072f8fe6344b689c52e19c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 18:23:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe211debc25840e6ac82f1bde64b7889\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3080ef8133d1452697a4f773094b42d7 : SUCCESS in 1h 28m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f90498d6d2664a8e887624a45d947772 : SUCCESS in 1h 04m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0ceed33789af4cf2b7f84e18487e52e9 : FAILURE in 1h 46m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/963f976621a84867b1800d1eea6e0290 : SUCCESS in 1h 54m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d46032e33f94371bd724740c3c9b04d : SUCCESS in 33m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db864a8c52cb465586acd4a5318ec627 : SUCCESS in 41m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448457efc61f45a285163f744a3addf0 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7adfa048cee643ffad3788390396574d : SUCCESS in 20m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4acb2fadb6484642babef3bf35292b27 : SUCCESS in 19m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c727ab6373d4f389d7f7ca59c2145b5 : SUCCESS in 32m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05af268d81a148fe986434193240e555 : SUCCESS in 14m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adb0de7f0813403ea8f31cb1fe126fe2 : SUCCESS in 22m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c387a02749724b08b2fb1f3832aac875 : SUCCESS in 1h 34m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c13cfb9d7bc64ee6a3971554b124941f : SUCCESS in 1h 08m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce64c862dac54b80a8e201be5b4251e3 : SUCCESS in 1h 11m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5d0ea7b84ec240be9578be96327105ab : SUCCESS in 1h 56m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5d49e1296a5446899b564b523126a715 : SUCCESS in 1h 32m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/20255f2b6f3d4e4a8091d3ccf21e72a5 : SUCCESS in 1h 06m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/716a1e3d35924b62b0e48ad9236e8051 : SUCCESS in 1h 06m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b70b2d61a40746dc93399c9261d5eab2 : SUCCESS in 1h 23m 28s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"6d8fc73014ce3870ac91b862e48c5ec543ceb6bd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-30 06:56:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"148a1e8285957dbbcdc6effb64495eec26a30980","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 07:44:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a28ed5aecab74b3d8098510370ec50ff\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d1124c1d87bc4e2aa9b8edd6a35ad6b6 : SUCCESS in 46m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/659f63cea5b34b538384cc4f50df42da : SUCCESS in 34m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7e15277472154820b9c762bdc993bf87 : SUCCESS in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f0158955a1a5e7fa87d4223358a065352e2fdb6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 09:29:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fadc2165ee24e3e8e302b27250766aa\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/de7c96d08fbd4f2382fb43450018c6af : SUCCESS in 1h 09m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34aa929bcd9243008d5626df9750d941 : SUCCESS in 34m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c2ffe13465354c559799b50fd00c88e3 : SUCCESS in 1h 36m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/19e3db2542334941ac23717b22891871 : SUCCESS in 2h 26m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/059377b7fb64473b80100aafc053a8a0 : SUCCESS in 37m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2963641f74344558b24d47c2bba132d : SUCCESS in 50m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d977e94f244414380090dba930c5531 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0be0aa952b964f62aec0eb8360697bb6 : SUCCESS in 22m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1129e74f81c44b308ed2aa476a0f0386 : SUCCESS in 26m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/128a0cf3a216406c955de8534bc03578 : SUCCESS in 31m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69b447ce0a2042f3b610be34456b20ac : SUCCESS in 13m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7ce49a408254de0a1f361fcfb6eb57c : SUCCESS in 9m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7d20f0628b3e4606badd0a5e7310d2eb : SUCCESS in 51m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1bf18b4b30724f26b745366ff47f3749 : SUCCESS in 54m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/91d9cf819e714b68a0636f58cc087dab : SUCCESS in 1h 15m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6c60458ff86f49718bbf172cfb142ce2 : SUCCESS in 1h 53m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0fb173c472364960b28232abd361771c : SUCCESS in 1h 30m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c13f07ba627d42a4ae83ddfed91a718d : SUCCESS in 52m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6508073e47404df891d3df86f1c6c7f1 : SUCCESS in 1h 07m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1ee8bfe28cbc41d88af1664b42f76bc0 : SUCCESS in 1h 21m 29s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"c79a58ce9bc0d763d5112e96628ea1834ec6f3eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 09:31:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e700673f49f3d09f911d7cea593734416826778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 12:36:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c806ed27f44ab7819e28057ee9313e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c42855b8d324776b11438c645b95253 : SUCCESS in 1h 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a7984051ff841a39c468f04786e2fb8 : SUCCESS in 37m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4a54d2e4d1c44ba7ac204c4a5b70db59 : SUCCESS in 1h 23m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a491dc6cbd4f44f5a1f6c9e9d00cab3f : SUCCESS in 2h 19m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bdaf7e58477406d96b69db4d36d8fce : SUCCESS in 11m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7157c0514a2f476f8fc0fc8adb59e3f8 : SUCCESS in 30m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1601e71fab57442ebc5839d1c1dea156 : SUCCESS in 20m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ab344ae3924e9faaa8b7b40b0d6def : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83d3d310f3d643819029861cbb377e11 : SUCCESS in 15m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/eb513dcab2b4407a9b0f5435947dd027 : SUCCESS in 51m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/61a214ecd8fd4eaea78d85eba6d768d9 : SUCCESS in 2h 20m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d5f706126e1453b8dbe911c92596f37 : SUCCESS in 1h 08m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/26f8cab87eb048859a0d2c5effabace6 : SUCCESS in 1h 10m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d85a65578c8949beb340e990a48af90a : SUCCESS in 56m 15s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"b8f7ac1224e12b7418abaeed21407b94dfa87280","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 12:36:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3fc823b42d96fec6d3fe4757b189e73ee4898660","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 12:37:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce2ebb67eeca4e9faf4c27b81832eccd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c346e04e88422497601f433024df30 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ca8c07443dc01ffb06506405687ff3d10432a2f","revisions":{"7369b467e2ac0194e7045e887e71e70d79b06ef1":{"kind":"REWORK","_number":1,"created":"2026-06-29 13:18:23.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/29/995329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/995329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/995329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/995329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/995329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/995329/1"}}},"commit":{"parents":[{"commit":"e4cc1b941b35b2c07a96508025f823f6390716aa","subject":"Merge \"[UT] Update base_mac config override\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4cc1b941b35b2c07a96508025f823f6390716aa"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-06-29 13:14:17.000000000","tz":120},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-06-29 13:18:10.000000000","tz":120},"subject":"Fix error handling in PVLAN","message":"Fix error handling in PVLAN\n\nPVLAN incorrectly handled 400 Bad Request errors as 500 Internal Server\nerrors. Replacing NeutronException with BadRequest exception fixes this\nissue.\n\nChange-Id: Ie674c8388fd415bf5b4e834e35311d273520f41b\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7369b467e2ac0194e7045e887e71e70d79b06ef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7369b467e2ac0194e7045e887e71e70d79b06ef1"}]},"branch":"refs/heads/master"},"5ca8c07443dc01ffb06506405687ff3d10432a2f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-29 13:31:08.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/29/995329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/995329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/995329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/995329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/995329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/995329/2"}}},"commit":{"parents":[{"commit":"e4cc1b941b35b2c07a96508025f823f6390716aa","subject":"Merge \"[UT] Update base_mac config override\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4cc1b941b35b2c07a96508025f823f6390716aa"}]}],"author":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-06-29 13:14:17.000000000","tz":120},"committer":{"name":"Elvira Garcia","email":"egarciar@redhat.com","date":"2026-06-29 13:30:42.000000000","tz":120},"subject":"Fix error handling in PVLAN","message":"Fix error handling in PVLAN\n\nPVLAN incorrectly handled 400 Bad Request errors as 500 Internal Server\nerrors. Replacing NeutronException with BadRequest exception fixes this\nissue.\n\nCloses-Bug: #2158628\n\nChange-Id: Ie674c8388fd415bf5b4e834e35311d273520f41b\nSigned-off-by: Elvira Garcia \u003cegarciar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ca8c07443dc01ffb06506405687ff3d10432a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ca8c07443dc01ffb06506405687ff3d10432a2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
