)]}'
{"id":"openstack%2Fneutron~995337","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I24f429adfcdbf7411663459998c36479aef92350","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2158374-stable/2026.1-stable/2025.2-stable/2025.1","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-06-30 11:19:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24f429adfcdbf7411663459998c36479aef92350","subject":"[UT] Update base_mac config override","status":"MERGED","created":"2026-06-29 14:08:52.000000000","updated":"2026-06-30 11:23:46.000000000","submitted":"2026-06-30 11:19:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995337-bug/2158374-stable/2026.1-stable/2025.2-stable/2025.1","cherry_pick_of_change":995336,"cherry_pick_of_patch_set":1,"meta_rev_id":"49b58c319c52cb925227c7ac4d6f58ed261bc457","_number":995337,"virtual_id_number":995337,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-30 11:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-29 14:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-30 08:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-30 08:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-29 14:09:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-29 16:00:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-30 08:36:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"312aee62e387c0e9d91aaa4829a977b8ef18e134","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-29 14:08:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"cc18977b64120b30cf3c28138e4a1d2bdac0e3b8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-29 14:09:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30896fa5ef3313a54f104d6efae82e45bf77dcb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 16:00:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fb73a9700e8414ba1ac2605be227820\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/787b80908fed40e2a973a8f1d710af4e : SUCCESS in 1h 13m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/25f204f265444b578623efdb24c1c62a : SUCCESS in 1h 50m 59s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/451d0d4363cd4b69b3c2fd2cb41a2919 : SUCCESS in 1h 38m 57s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/8a80f375a424488c9f16c326d5c92484 : SUCCESS in 31m 18s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/06b3d581621c4830ac2b86c838df8f13 : SUCCESS in 1h 28m 34s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/701d19f1fec74965af81d14085987137 : SUCCESS in 1h 14m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0edccfc226f443e8425f96df1f27287 : SUCCESS in 46m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a4a71cfe3f4c29a289d0b0c362fc1f : SUCCESS in 12m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9b4dbd7ca754f3299f8c366c477736f : SUCCESS in 19m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50d44ce605374f668d55de448ab1ff6b : SUCCESS in 20m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6456ec11bd842c79f77b9963871c64c : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13d495da2bd14defb052f4c0e5b85fc2 : SUCCESS in 21m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d99ad3d2d9c648848909fd772f510b02 : SUCCESS in 1h 04m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/48bedab041154fe7829e03e2b630b387 : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e2a4e6c681154c26f5bce52351670fa480756f6c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-30 08:36:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"792efc2d82be18da28d50bdbf41e3b0ac2c7764a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 08:36:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9deb920adf3aaf19152835ef8d7f572ffeae6d1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 11:19:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae69d0f3776468dae4af4edd9d13898\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/deb6c945af714010a3fac357ab2ac8de : SUCCESS in 2h 02m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca2e1dcdf21f48188155181d0d921b56 : SUCCESS in 11m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/714b5639826f495197baa8ff422a1477 : SUCCESS in 30m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/42fab8dc939a4453b7be1f657179a71d : SUCCESS in 14m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2db96905ae64e7c992b2a113f1c9a6c : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1572c2aecfcd445d8846f9f26a3db250 : SUCCESS in 16m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/80069ccbff0f4bba9fe6a0e8cde53a60 : SUCCESS in 1h 06m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/98c5d077f23e43e0b5c69c6f4a1f88ec : SUCCESS in 1h 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e944c7e9d5deec7fe34e412bd95e99ac429a9639","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 11:19:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"49b58c319c52cb925227c7ac4d6f58ed261bc457","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 11:23:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfee9226e8c6449abb0bcdf28719acab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1177060788e7476b91bcc57a257435bb : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2d28a7261d71c6e9794642d4dbd0a0bb5de4a71","revisions":{"f2d28a7261d71c6e9794642d4dbd0a0bb5de4a71":{"kind":"REWORK","_number":1,"created":"2026-06-29 14:08:52.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/37/995337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/995337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/995337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/995337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/995337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/995337/1"}}},"commit":{"parents":[{"commit":"68346f102f34be9345e8e590b68622101a9c4922","subject":"Increase ``allowed_network_downtime`` to 7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68346f102f34be9345e8e590b68622101a9c4922"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-06-26 10:54:49.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2026-06-29 14:08:52.000000000","tz":0},"subject":"[UT] Update base_mac config override","message":"[UT] Update base_mac config override\n\nSeeing random failures for MacAddressInUse in\nunit tests, extending the range to lower such\nconflicts.\n\nCloses-Bug: #2158374\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: I24f429adfcdbf7411663459998c36479aef92350\n(cherry picked from commit 2683e23aa99d89f42023fddec3b370a371cb6cd4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2d28a7261d71c6e9794642d4dbd0a0bb5de4a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2d28a7261d71c6e9794642d4dbd0a0bb5de4a71"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"6e846c1a2d5ad40f767e87b6c8ddb14b509ba5d7","ours":"68346f102f34be9345e8e590b68622101a9c4922","theirs":"54e749bf7f7910c4f8a8ebc117faf0d3658331a7","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
