)]}'
{"id":"openstack%2Fneutron~995387","triplet_id":"openstack%2Fneutron~master~Iec210d38629c737f8b2e03126d1bcc28beeb239d","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":["remove-enforce-scope-flag"],"change_id":"Iec210d38629c737f8b2e03126d1bcc28beeb239d","subject":"Remove setting of oslo.policy[enforce_scope] falg","status":"ABANDONED","created":"2026-06-29 17:27:11.000000000","updated":"2026-06-29 20:22:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f5a8acac93de992155ddb2be489cbabca3a24567","_number":995387,"virtual_id_number":995387,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-29 18:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-29 18:48:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3947008f56a625bae5f7dfc241783281300017b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-29 17:27:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bc3e9507e910c789d1db74f5db358ef041e0717","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-29 17:30:29.000000000","message":"Hashtag added: remove-enforce-scope-flag","accounts_in_message":[],"_revision_number":1},{"id":"1926765b5b95f4e29a56f842133aec873dfb1a4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 18:22:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/789a574c3d184d28ae402f05bf8f1f5c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de29447ed52d40beb8578330d01e1b50 : SUCCESS in 41m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a478e8d77d404cbdbbdae0cbba4a5926 : SUCCESS in 22m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3da8f321e67f4f7bb5916c4d87d26e55 : SUCCESS in 52m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edee6152492d9ac763a34fec003eb37741a2474a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 18:48:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3152475491bf4d01b867e6ce1a897088\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb9ecd1eafad43adaa5b40d654beb36c : SUCCESS in 1h 10m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce4e45221b349edb046dfbb0b050ec7 : SUCCESS in 1h 07m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b672e5a8d74b8da40b3cd8191583d6 : SUCCESS in 13m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19a5406552da40aea7dacfcf77bc456d : SUCCESS in 34m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2949b4b6d894378885c1c392a90dc26 : SUCCESS in 23m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06390c4500be471b9ef220462bdd48c7 : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc529d96cbb64fd39cbad25a0a2c6760 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d41f642beb5e4b8dbb24669c2c4a8715 : SUCCESS in 19m 32s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"541ce35fa942a5d40aff1865ccc717f055da1a81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-29 20:03:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f5a8acac93de992155ddb2be489cbabca3a24567","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-29 20:22:31.000000000","message":"Abandoned\n\nalready done in https://review.opendev.org/c/openstack/neutron/+/986729","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"802dc2135640f2b554c09aceae9d1579853b7621","revisions":{"12cb844d7b5b103191c14189beaade5fb0618a43":{"kind":"REWORK","_number":1,"created":"2026-06-29 17:27:11.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/995387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/995387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/995387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/995387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/995387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/995387/1"}}},"commit":{"parents":[{"commit":"ee192583e808687f92dc5e3df3ebab2f8e3e6e13","subject":"Add a release note for tenant_id deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee192583e808687f92dc5e3df3ebab2f8e3e6e13"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-29 16:51:05.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-29 16:51:05.000000000","tz":0},"subject":"Remove setting of oslo.policy[enforce_scope] falg","message":"Remove setting of oslo.policy[enforce_scope] falg\n\noslo.policy[enforce_scope] is enabled by default and setting\nit to True is not needed. Neutron set it to True in some of\nthe tests.\n\nChange-Id: Iec210d38629c737f8b2e03126d1bcc28beeb239d\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12cb844d7b5b103191c14189beaade5fb0618a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12cb844d7b5b103191c14189beaade5fb0618a43"}]},"branch":"refs/heads/master"},"802dc2135640f2b554c09aceae9d1579853b7621":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-29 20:03:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/995387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/995387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/995387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/995387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/995387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/995387/2"}}},"commit":{"parents":[{"commit":"a116142e714ef4ab0627bec6b71a696122ae3249","subject":"Merge \"Add PVLAN documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a116142e714ef4ab0627bec6b71a696122ae3249"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-29 16:51:05.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-29 20:03:58.000000000","tz":0},"subject":"Remove setting of oslo.policy[enforce_scope] falg","message":"Remove setting of oslo.policy[enforce_scope] falg\n\noslo.policy[enforce_scope] is enabled by default and setting\nit to True is not needed. Neutron set it to True in some of\nthe tests.\n\nChange-Id: Iec210d38629c737f8b2e03126d1bcc28beeb239d\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/802dc2135640f2b554c09aceae9d1579853b7621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/802dc2135640f2b554c09aceae9d1579853b7621"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ee192583e808687f92dc5e3df3ebab2f8e3e6e13","ours":"12cb844d7b5b103191c14189beaade5fb0618a43","theirs":"a116142e714ef4ab0627bec6b71a696122ae3249","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
