)]}'
{"id":"openstack%2Fneutron~996432","triplet_id":"openstack%2Fneutron~master~Ib128d88c15c33d987bc4bfddbdd90ddf14888b45","project":"openstack/neutron","branch":"master","topic":"multinodenontprole","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-07-14 07:32:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib128d88c15c33d987bc4bfddbdd90ddf14888b45","subject":"[DNM] Check jobs without multi-node-setup","status":"ABANDONED","created":"2026-07-08 09:21:26.000000000","updated":"2026-07-14 07:32:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f29eebe82448bbe168ab64a558a99108c2ad3b7","_number":996432,"virtual_id_number":996432,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-07-08 12:36:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 12:36:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7709cc3361b3c47b863eb8106e391678abc2ea0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-07-08 09:21:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"666f9b0005eb2c7cf44a28c35933e004e41f766e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 12:36:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5eabca0862b24ea4a6eac1467953db6f\n\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3ba5f71950ec42799788c2d11297595f : SUCCESS in 1h 15m 09s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/33ddc87dff814f3cb27ea9d72c741eea : TIMED_OUT in 3h 05m 15s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/b9286fd2fcd74ea0b7a96028c48c15b2 : SUCCESS in 1h 14m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bcba90dacb94eb19250cbb19300c891 : SUCCESS in 1h 03m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ca20c0879d14274934841473a358940 : SUCCESS in 1h 25m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2e17118b69ed4e2cb529fecdc06cf99e : SUCCESS in 1h 09m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/94fb69995fcb4778bf1506b0b7678666 : FAILURE in 1h 28m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6f29eebe82448bbe168ab64a558a99108c2ad3b7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-07-14 07:32:11.000000000","message":"Abandoned\n\nmoved to https://review.opendev.org/c/openstack/neutron/+/996388","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab4766bc38ec4e714829cd35afbdf80acd53ca07","revisions":{"ab4766bc38ec4e714829cd35afbdf80acd53ca07":{"kind":"REWORK","_number":1,"created":"2026-07-08 09:21:26.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/32/996432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/996432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/996432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/996432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/996432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/996432/1"}}},"commit":{"parents":[{"commit":"e4df77137e4624ffbf15edfa8e29f808a2d024cd","subject":"Merge \"Adapt to new testtools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4df77137e4624ffbf15edfa8e29f808a2d024cd"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-06-29 07:35:17.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-08 09:20:16.000000000","tz":330},"subject":"[DNM] Check jobs without multi-node-setup","message":"[DNM] Check jobs without multi-node-setup\n\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: Ib128d88c15c33d987bc4bfddbdd90ddf14888b45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab4766bc38ec4e714829cd35afbdf80acd53ca07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab4766bc38ec4e714829cd35afbdf80acd53ca07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
