)]}'
{"id":"openstack%2Fneutron~996485","triplet_id":"openstack%2Fneutron~unmaintained%2F2024.1~Ie038a896b1d3d779a04a1262b29791f996f40b15","project":"openstack/neutron","branch":"unmaintained/2024.1","topic":"ovs-rally-job-fix-osprofiler-unmaintained/2024.1","attention_set":{},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-07-08 18:59:51.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-07-08 18:59:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie038a896b1d3d779a04a1262b29791f996f40b15","subject":"[stable-only] Pin osprofiler to 4.3.0","status":"MERGED","created":"2026-07-08 14:18:25.000000000","updated":"2026-07-08 19:02:49.000000000","submitted":"2026-07-08 18:59:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996485-ovs-rally-job-fix-osprofiler-unmaintained/2024.1","cherry_pick_of_change":954615,"cherry_pick_of_patch_set":2,"meta_rev_id":"e92f545c80d0d7c7f5364bdbdf48d008e5badaf8","_number":996485,"virtual_id_number":996485,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-08 18:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-08 17:26:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-08 17:26:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 14:18:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-07-08 14:19:32.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-07-08 16:06:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-08 17:26:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5dd639fdcf25ffadb4aae37bc405f8653497ff72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-08 14:18:25.000000000","message":"Patch Set 1: Cherry Picked from branch unmaintained/2023.1.\n\nThe following files contain Git conflicts:\n* zuul.d/rally.yaml","accounts_in_message":[],"_revision_number":1},{"id":"386ea641c96e395398ba39484b26d397237b4cb9","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-08 14:19:32.000000000","message":"Patch Set 1: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch unmaintained/2024.1.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/neutron/zuul.d/rally.yaml@unmaintained/2024.1\", line 51, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/neutron/zuul.d/rally.yaml@unmaintained/2024.1\", line 52, column 13","accounts_in_message":[],"_revision_number":1},{"id":"0c795cb175b3ec9d7a570d4a25f5d88405d23a8a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-08 14:19:54.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6410c48bb3500e9a7ae38d797e15ddf6e42aea87","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-08 16:02:57.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"8de1474d7ac7de3750c37a4674b7825ef18de090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 16:06:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84ffeb3494d24af0a9c750b8de5c53d8\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/00687072e9f24c7395638605bb09708f : SUCCESS in 1h 14m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/f5dfc73b87514c0cb472f2470188f3dc : SUCCESS in 1h 29m 34s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/1cdedf9fc0784b55a6bbb7df1cc7a1d5 : SUCCESS in 1h 39m 27s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/6fd8506d58b442958d9c219286d2c2b7 : SUCCESS in 34m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/859f8f8da1634cff9eb4ca655089d3fb : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb36f80fe4a94d68b1a45d29d7aaeeee : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dfb63cd1b7a4be0a675b4735697a40d : SUCCESS in 21m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b8d9673d7aa343f69a04b9eee0cebecc : SUCCESS in 1h 29m 01s","accounts_in_message":[],"_revision_number":2},{"id":"bb83f862dfbec9609a5859af78a48797bb38f02d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-08 17:26:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3914e5e2a7492065acc3277676b527a940b60ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 17:28:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0dc606b151a0d0812b2cf943631a0182d54bf80b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 18:59:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb5403406b14a1fb2da01c5fe76538d\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/3889a71cbb1d4f03a746376b74c3b24a : SUCCESS in 1h 29m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35f2171dc5f041c78678e550d14bda82 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4927b28d6547c2b9c4bd741e3e56c8 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcbd67b1c4c24eb18a83522f08f27681 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e393daafd9611692ca7793e7cd9f289010928e85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 18:59:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e92f545c80d0d7c7f5364bdbdf48d008e5badaf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 19:02:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d86dfe33a7b4e3886962217e4019267\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f65f26b5f5724d7089c5293df6c17a57 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"42f442ec436d9ad9786687fb5d151e4ff2c4f7d9","revisions":{"5a7de681f91686f6f108b5e815a6efcb164c627f":{"kind":"REWORK","_number":1,"created":"2026-07-08 14:18:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/85/996485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/996485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/996485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/996485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/996485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/996485/1"}}},"commit":{"parents":[{"commit":"5610df195681062eb5bb6e6753588f11a21bd108","subject":"Merge \"Ignore designate NotFound errors during PTR delete\" into unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5610df195681062eb5bb6e6753588f11a21bd108"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-07-10 14:17:07.000000000","tz":120},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-08 14:18:25.000000000","tz":0},"subject":"[stable-only] Pin osprofiler to 4.3.0","message":"[stable-only] Pin osprofiler to 4.3.0\n\nneutron-ovs-rally-task job started to fail now that pyproject.toml was\nadded in osprofiler for package building.\nThis patch pins osprofiler to previous release that does not contain\nyet the new build method.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/955450\nChange-Id: Ie038a896b1d3d779a04a1262b29791f996f40b15\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n(cherry picked from commit 838bc618b1a67b5ec8d9b0e78e283e330ad67717)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a7de681f91686f6f108b5e815a6efcb164c627f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a7de681f91686f6f108b5e815a6efcb164c627f"}]},"branch":"refs/heads/unmaintained/2024.1","conflicts":{"base":"59c9ad47d876bb589d116c496d6a893c502d0e7e","ours":"5610df195681062eb5bb6e6753588f11a21bd108","theirs":"838bc618b1a67b5ec8d9b0e78e283e330ad67717","merge_strategy":"recursive","contains_conflicts":true}},"42f442ec436d9ad9786687fb5d151e4ff2c4f7d9":{"kind":"REWORK","_number":2,"created":"2026-07-08 14:19:54.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/85/996485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/996485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/996485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/996485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/996485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/996485/2"}}},"commit":{"parents":[{"commit":"5610df195681062eb5bb6e6753588f11a21bd108","subject":"Merge \"Ignore designate NotFound errors during PTR delete\" into unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5610df195681062eb5bb6e6753588f11a21bd108"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-07-10 14:17:07.000000000","tz":120},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-08 14:19:52.000000000","tz":0},"subject":"[stable-only] Pin osprofiler to 4.3.0","message":"[stable-only] Pin osprofiler to 4.3.0\n\nneutron-ovs-rally-task job started to fail now that pyproject.toml was\nadded in osprofiler for package building.\nThis patch pins osprofiler to previous release that does not contain\nyet the new build method.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/955450\nChange-Id: Ie038a896b1d3d779a04a1262b29791f996f40b15\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n(cherry picked from commit 838bc618b1a67b5ec8d9b0e78e283e330ad67717)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42f442ec436d9ad9786687fb5d151e4ff2c4f7d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42f442ec436d9ad9786687fb5d151e4ff2c4f7d9"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
