)]}'
{"id":"openstack%2Fneutron~996501","triplet_id":"openstack%2Fneutron~master~I65a3c0702433706e220229e0e5ddc609395b7f09","project":"openstack/neutron","branch":"master","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-08-11 21:21:19.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-08 20:33:26.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I65a3c0702433706e220229e0e5ddc609395b7f09","subject":"Start using Project* exceptions","status":"NEW","created":"2026-07-08 16:59:04.000000000","updated":"2026-08-11 21:21:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f3f43bbbac60b396efd61da58db0be1c09b8c361","_number":996501,"virtual_id_number":996501,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-08 19:02:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-07-09 12:14:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-08-08 20:33:26.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 19:02:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 12:14:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-08-11 21:21:19.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"4a96569d3b5be1de8b6fda5800baaeaa0b2c4795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-08 16:59:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6b8a249f882166a5eab411f8a694a81ff7f9dc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-08 17:02:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7af28e8d0af576a761bfb1564d130592e242dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 19:02:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c338082d61b4d33b1418fa5ee703b4f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/34a00a65f80443cba4878d7f77c3f426 : SUCCESS in 1h 04m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29623fd6a37c46379dd992a33ef991ad : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cd8126f2b10c471da3e68a5985d4adb5 : SUCCESS in 1h 19m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/61aed8c0740f420cb733ae8849f08385 : SUCCESS in 1h 55m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f331ba0678414f79adf5b74e807e4aae : SUCCESS in 35m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/214cc0ab3d6147d5b84f361587eab7b2 : SUCCESS in 1h 03m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151241faada64535901110ca492bcd73 : SUCCESS in 9m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/082da1c4b6a0408b9f2afeb420de75af : SUCCESS in 26m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42665fac4df54407be1a2f6439df9680 : SUCCESS in 21m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a07965af386f4c36a8326500eb90c7bd : SUCCESS in 38m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/debf2da5f18f4f7ead7ae3d7722a78e3 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/095f10cabc2d4ee1807acc8609286f88 : SUCCESS in 13m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cfbcfcadcd6347a98ce1196ff59fcc06 : SUCCESS in 1h 29m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5917a6707efb4ba8923bae5836848f35 : SUCCESS in 1h 21m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1e853647f8a4d6bbb78726b4b0d9171 : SUCCESS in 44m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/79d7f4415edd456c8f5da35a88193509 : SUCCESS in 1h 11m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b6669cdef0c948328e6307b7decd4565 : SUCCESS in 1h 50m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f3f4bb63f43c49909c0c90a669d813ca : SUCCESS in 1h 04m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/788c9beb130c463e9aed4d5b906e74cc : SUCCESS in 1h 13m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aadb9764a0234c06bb86f72a09722f97 : SUCCESS in 57m 51s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"614e1264b69de1048160e49894a4ed31574d02f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 21:23:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a778ebabc8f34181a648c58e57b7f4db\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce1aa0160c1e49ec918a449785647ed5 : SUCCESS in 53m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cc0a6df6868f4b2ba597bb0a1654812d : SUCCESS in 19m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4bed3370a8644a7f869161db286c1edc : SUCCESS in 1h 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ef921f6b69ee717361d83278060970fe37e09a8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-09 12:14:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"548a54d0e34d2e5834fb8615c9078dc10061ed86","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-08 20:33:26.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d60196ad75df0a000b890b20c7cb32e9247aee5c","revisions":{"7dcdfcfd697b7caf9d353d6a8a8e0bb811995f51":{"kind":"REWORK","_number":1,"created":"2026-07-08 16:59:04.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/01/996501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/996501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/996501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/996501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/996501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/996501/1"}}},"commit":{"parents":[{"commit":"04f6410f49bf85f7b63a156683eb0e69dd8a90f6","subject":"Merge \"trivial fix: Fix typo on PVLAN internals image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04f6410f49bf85f7b63a156683eb0e69dd8a90f6"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-08 16:57:24.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-08 16:57:24.000000000","tz":-240},"subject":"Start using ProjectQuotaNotFound exception","message":"Start using ProjectQuotaNotFound exception\n\nThis was added in order to move to strings using \u0027project\u0027\ninstead of \u0027tenant\u0027, but I forgot to change the neutron\nuser, do it now.\n\nTrivialFix\n\nChange-Id: I65a3c0702433706e220229e0e5ddc609395b7f09\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dcdfcfd697b7caf9d353d6a8a8e0bb811995f51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dcdfcfd697b7caf9d353d6a8a8e0bb811995f51"}]},"branch":"refs/heads/master"},"d60196ad75df0a000b890b20c7cb32e9247aee5c":{"kind":"REWORK","_number":2,"created":"2026-07-08 17:02:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/01/996501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/996501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/996501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/996501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/996501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/996501/2"}}},"commit":{"parents":[{"commit":"04f6410f49bf85f7b63a156683eb0e69dd8a90f6","subject":"Merge \"trivial fix: Fix typo on PVLAN internals image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04f6410f49bf85f7b63a156683eb0e69dd8a90f6"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-08 16:57:24.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-08 17:00:53.000000000","tz":-240},"subject":"Start using Project* exceptions","message":"Start using Project* exceptions\n\nProjectQuotaNotFound and DeviceIDNotOwnedByProject\nexceptions were added in order to move to strings\nusing \u0027project\u0027 instead of \u0027tenant\u0027, but I forgot\nto change the neutron users, do it now.\n\nTrivialFix\n\nChange-Id: I65a3c0702433706e220229e0e5ddc609395b7f09\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d60196ad75df0a000b890b20c7cb32e9247aee5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d60196ad75df0a000b890b20c7cb32e9247aee5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
