)]}'
{"id":"openstack%2Fneutron~997151","triplet_id":"openstack%2Fneutron~master~I6e3ae037a8575be137bad65f2fd0cf4c719033b3","project":"openstack/neutron","branch":"master","topic":"bug/2144617","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-07-14 15:32:48.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"38298":{"account":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"last_update":"2026-07-16 09:00:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e3ae037a8575be137bad65f2fd0cf4c719033b3","subject":"Save socket fd to be cleaned up in functional test","status":"MERGED","created":"2026-07-14 03:14:33.000000000","updated":"2026-07-16 09:01:41.000000000","submitted":"2026-07-16 09:00:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997151-bug/2144617","meta_rev_id":"e1ea5961617de20610d312758d441f452bf3a623","_number":997151,"virtual_id_number":997151,"owner":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-16 09:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-07-14 15:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"value":2,"date":"2026-07-16 07:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2026-07-16 01:10:54.000000000","_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"value":1,"date":"2026-07-16 07:06:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-07-14 15:33:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-14 03:52:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-14 05:19:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-14 13:37:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-07-14 15:32:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-07-14 15:33:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-07-14 15:33:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-07-14 15:33:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-07-16 07:06:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"64fc92109ccd4a6dfeccae290a379c0b83417e10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-07-14 03:14:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15406cc992be97b405175f469e13dc139611f4af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 03:52:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c022ac3ff2924bbe8c27bdc49c43406d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/25085e87dcb24806b5c448899026bf1d : NODE_FAILURE Node(set) request 31881a2f2fc844ed9bb0d41c4098aee7 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3e4a8472eb164ad089f4c9b003cf5b3b : NODE_FAILURE Node(set) request ecb42f0ff17a4804916f45daeaaa486f failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ea08db172c3d411c91842e9c94717bdf : NODE_FAILURE Node(set) request 4f07e064931a4f6ebd871b3f9a81d6b7 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e70a4f316cab1e23e2a30f4b8c9280949fcb403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 05:19:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/052a1830a9e94515bf3bc351c152c76a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/346681b322444350a96ff37e6917d7f2 : SUCCESS in 1h 23m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ae63199957147ba86be39e89c65305d : SUCCESS in 27m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fadd3d5befa54a03a622cb0282d92284 : SUCCESS in 1h 05m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/37563b9daa254f808acc625914bf975b : SUCCESS in 2h 00m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95f0f2ba8c2b4a1691ab42b8d9dfcca4 : SUCCESS in 1h 09m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa047bcb03a4f5bbc0f7dc13f387c8d : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1cdd0254c8a489791a523f6358ecfe9 : SUCCESS in 29m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28a763de51f749c1ba61c2902335fbdd : SUCCESS in 24m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa349904c45a4e81ba8077525d613344 : SUCCESS in 20m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3edfb97c7b074fd6b5222f472db6f663 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cc02a1ceb04496db107f8bab7a51bec : SUCCESS in 17m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3821ad3373ea496cb0f6e9a926dea894 : SUCCESS in 1h 01m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4393a242721840fa8158381ad612af6d : SUCCESS in 1h 09m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed5204598069448686a016d6bf40d1d2 : SUCCESS in 38m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/35fd10bf969f420487fdef3095d90d20 : SUCCESS in 1h 31m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5a78ec55738f4cd898c8d6d5a2a6248a : FAILURE in 1h 14m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cf24c444df6b429fb8e1cac149da4b59 : SUCCESS in 47m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ee86366a2eb64b83bd8915b4c0bab998 : SUCCESS in 54m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0075428ac14742a9abe8e921824004e5 : SUCCESS in 54m 04s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"ddda2eb1e71e593f4afd5a2be3eccde0ab02c30c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-07-14 13:30:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8f02d030ef650a35267dc89d49a6f6287136b5c1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-14 13:37:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13bc59aec06c2e5544e4caa6b2889714924cb42b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-14 13:37:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad6d934edff61cc652252f618b51a92b713c1f60","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-14 13:37:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a4de483712b0dbd16921e6a8c704ec624d9d0b7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-14 13:37:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fdd0fa380f9ec957254a7cea8d2a9777bee6ce80","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-14 13:38:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f597e35a77d7e562eec7b21576067e25ed469a5b","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-07-14 13:46:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"affa00660cfeeb7422e3bc2013522e0e68017437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-07-14 13:56:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"682882c01da676e3cef2e874f4652b5fb8a9f5e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-14 15:32:48.000000000","message":"Patch Set 3: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2c9e97f437c59b7a6ab87ed79f6f246dfd3146fc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-14 15:33:44.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"1911bf0015239cefc1d42bef7b71538567666b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 15:51:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e66e5495b3854f6cbed6f359c84d5f67\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2db3ba2b815743fa9b5ac2e1dc6c5ad9 : SUCCESS in 1h 09m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f4bb642268d49f4ac74240591a41346 : SUCCESS in 39m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bae9612e99434073b45c24447bdd5474 : SUCCESS in 1h 04m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/70d9d518380d4633afee5efb4ff3a711 : SUCCESS in 1h 34m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91a6a5cd474f43a19876d55886c5f06f : SUCCESS in 39m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f277a2f441f438095969f7d67385937 : SUCCESS in 10m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b48f151593184e9a8a606fcd8049bfc9 : SUCCESS in 29m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dabcaa0e3e4545108944a2f94a518812 : SUCCESS in 28m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/88fe1dfddee445b29fa6dadc50577ab3 : SUCCESS in 37m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4138ecf798e34ac882526113bb47cd45 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df650c8fd3af4e8d88a8ca5bf922ae63 : SUCCESS in 25m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0a524f61acdb48558b205dba93ba88d1 : SUCCESS in 55m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/11d2885f8e894e09bdf8b4a7b1afc627 : SUCCESS in 51m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f333d78a38e84652b141c9c876e97728 : SUCCESS in 1h 45m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/18eda1c9408a4f13b8f78aa3416d9e20 : SUCCESS in 39m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9408fd1cd3c24da7b6d339080564dec4 : SUCCESS in 40m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c5f6809f4ae64d53ad7626a66dc68a08 : SUCCESS in 1h 10m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/51a5522c6cea4e159e02f601f2bb4f77 : SUCCESS in 1h 15m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a764690ca184f378830e091fe84ee27 : SUCCESS in 1h 38m 13s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"e1badf398b954984db501bc7ffbf8416096ee7e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 16:24:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b2ae5de89a1456185e05e51418c4cd0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b70058c3cfa74d528c97a970dbe3f457 : SUCCESS in 1h 15m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4ae809d276c40c0b6ec25501191f8ef : SUCCESS in 1h 09m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ec0c7f3bddab49cfb5e7498ba1c1adcf : TIMED_OUT in 1h 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"53e3767647c574471011f2c2dad5bd9a7675dad6","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-07-15 01:51:36.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"47feccd9b630032ec732bfc6718917bf0ba63daa","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-07-16 01:10:54.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"2756da62ede20f14fc5dd3e8e28c1e3f5ac8a2ca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-16 07:06:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ea6e497c6842e39ed4a14cc7b5c210827d9724c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 07:07:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b09890b650d123df8984e34a94c47201656b81fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 09:00:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/638882086e5d4894b98628ba24963337\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fce8b6fea56c46a5bfd401f3dcdbf6f5 : SUCCESS in 1h 18m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa74ce75698a464f90dc137999f89d8e : SUCCESS in 1h 03m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/42900cb7bd294716827d594a632dc0cd : SUCCESS in 1h 01m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6de1a85a671f48e89b9a08019d4e26af : SUCCESS in 1h 52m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8669fefbbfa94b23b9adc5cdcf1a0c0e : SUCCESS in 10m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ab0fd528dfa467c8538bdc44a047c9b : SUCCESS in 28m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e00e1159ba34784b76aef72da9ad04d : SUCCESS in 16m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ae2f251f6d462885a2b075589b740d : SUCCESS in 14m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a96fcb4b9af4dd58ef9c3dcf0c818fa : SUCCESS in 27m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6b1e6db634f442cf91c7659f6d383d62 : SUCCESS in 1h 24m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4015a1821d3e4bc9ae401f8d297abe93 : SUCCESS in 1h 33m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/82d151ce96bd4b829da5609bb09c0445 : SUCCESS in 1h 19m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/653d64cd407c473b80c5a2b66b81a841 : SUCCESS in 56m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2e6752c4274f43f0bb7a865559baea89 : SUCCESS in 57m 27s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"36bd3d29034cdcdc1a7c444c9826635c0adda20e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 09:00:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e1ea5961617de20610d312758d441f452bf3a623","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 09:01:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9de93e76b3dd44ddb1a020cadc11d39c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36bf8b420cf74c4ea3958e84b06eaab3 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5ba251e192e6be62ddf11fe9268efa1f507e1d27","revisions":{"e46e259e49d455228ed0a03a0e43cd688740b25f":{"kind":"REWORK","_number":1,"created":"2026-07-14 03:14:33.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/51/997151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/997151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/997151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/997151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/997151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/997151/1"}}},"commit":{"parents":[{"commit":"6390bbd3faacbd7a13e189375e8c8f531c1ce4dc","subject":"Merge \"ovn: Advertise metadata IPv6 route via RA Route Information Option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6390bbd3faacbd7a13e189375e8c8f531c1ce4dc"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-07-13 16:44:07.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-07-13 16:44:07.000000000","tz":-240},"subject":"Save socket fd to be cleaned up in functional test","message":"Save socket fd to be cleaned up in functional test\n\nFunctional tests that create a netlink dispatcher thread should close\nthe netlink socket and terminate the netlink dispatcher thread.  This\nrequires the dispatcher thread to store the netlink socket\u0027s socket FD,\nsuch that a functional test case can close the socket to trigger the\nexisting code to terminate the dispatcher thread.\n\nRelated-Bug: #2144617\nAssisted-By: Claude Opus 4.6\nChange-Id: I6e3ae037a8575be137bad65f2fd0cf4c719033b3\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e46e259e49d455228ed0a03a0e43cd688740b25f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e46e259e49d455228ed0a03a0e43cd688740b25f"}]},"branch":"refs/heads/master"},"88d1440ca380dd4fc0d54a48f5c7058671122e45":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-14 13:30:31.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/51/997151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/997151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/997151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/997151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/997151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/997151/2"}}},"commit":{"parents":[{"commit":"32f99902dd62acf93b36e25ddc4173ed4bf16b6a","subject":"Merge \"[CI][FT] Tempoarary skip failing tests centos 10 stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32f99902dd62acf93b36e25ddc4173ed4bf16b6a"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-07-13 16:44:07.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-07-14 13:27:04.000000000","tz":-240},"subject":"Save socket fd to be cleaned up in functional test","message":"Save socket fd to be cleaned up in functional test\n\nFunctional tests that create a netlink dispatcher thread should close\nthe netlink socket and terminate the netlink dispatcher thread.  This\nrequires the dispatcher thread to store the netlink socket\u0027s socket FD,\nsuch that a functional test case can close the socket to trigger the\nexisting code to terminate the dispatcher thread.\n\nRelated-Bug: #2144617\nAssisted-By: Claude Opus 4.6\nChange-Id: I6e3ae037a8575be137bad65f2fd0cf4c719033b3\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88d1440ca380dd4fc0d54a48f5c7058671122e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88d1440ca380dd4fc0d54a48f5c7058671122e45"}]},"branch":"refs/heads/master"},"5ba251e192e6be62ddf11fe9268efa1f507e1d27":{"kind":"REWORK","_number":3,"created":"2026-07-14 13:56:33.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/51/997151/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/997151/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/997151/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/997151/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/997151/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/997151/3"}}},"commit":{"parents":[{"commit":"32f99902dd62acf93b36e25ddc4173ed4bf16b6a","subject":"Merge \"[CI][FT] Tempoarary skip failing tests centos 10 stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32f99902dd62acf93b36e25ddc4173ed4bf16b6a"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-07-13 16:44:07.000000000","tz":-240},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2026-07-14 13:53:12.000000000","tz":-240},"subject":"Save socket fd to be cleaned up in functional test","message":"Save socket fd to be cleaned up in functional test\n\nFunctional tests that create a netlink dispatcher thread should close\nthe netlink socket and terminate the netlink dispatcher thread.  This\nrequires the dispatcher thread to store the netlink socket\u0027s socket FD,\nsuch that a functional test case can close the socket to trigger the\nexisting code to terminate the dispatcher thread.\n\nRelated-Bug: #2144617\nAssisted-By: Claude Opus 4.6\nChange-Id: I6e3ae037a8575be137bad65f2fd0cf4c719033b3\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ba251e192e6be62ddf11fe9268efa1f507e1d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ba251e192e6be62ddf11fe9268efa1f507e1d27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
