)]}'
{"id":"openstack%2Fneutron~997550","triplet_id":"openstack%2Fneutron~stable%2F2026.1~Ib360dec982aff53e9e6abdab197de354899274d8","project":"openstack/neutron","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-15 14:33:29.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"39330":{"account":{"_account_id":39330,"name":"Amir Abbas","email":"am_882011@live.com","username":"amirabbas"},"last_update":"2026-09-15 18:53:59.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-09-15 18:53:59.000000000","reason":"Change was submitted"},"35499":{"account":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"last_update":"2026-09-15 18:53:59.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-09-15 18:53:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib360dec982aff53e9e6abdab197de354899274d8","subject":"bgp: Always assume just one provider switch","status":"MERGED","created":"2026-07-16 13:13:34.000000000","updated":"2026-09-15 18:55:37.000000000","submitted":"2026-09-15 18:53:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"997550","cherry_pick_of_change":988539,"cherry_pick_of_patch_set":3,"meta_rev_id":"9d9cc5aa1a3866af60277e061f4e2a16b1118861","_number":997550,"virtual_id_number":997550,"owner":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 18:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-09-15 14:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-09-15 16:20:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-09-15 16:20:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":39330,"name":"Amir Abbas","email":"am_882011@live.com","username":"amirabbas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-16 13:13:34.000000000","updated_by":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2026-07-16 13:17:33.000000000","updated_by":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"reviewer":{"_account_id":39330,"name":"Amir Abbas","email":"am_882011@live.com","username":"amirabbas"},"state":"CC"},{"updated":"2026-07-16 13:17:33.000000000","updated_by":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-07-16 13:17:33.000000000","updated_by":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-07-16 13:44:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-16 14:27:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-07-16 15:29:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 19:02:31.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2026-09-12 11:08:27.000000000","updated_by":{"_account_id":39330,"name":"Amir Abbas","email":"am_882011@live.com","username":"amirabbas"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2026-09-15 16:20:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"9cd43189a0eba9979dbcdf4737fba0e91596baf1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"date":"2026-07-16 13:13:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* neutron/services/bgp/commands.py\n* neutron/services/bgp/events.py\n* neutron/services/bgp/plugin.py\n* neutron/tests/functional/services/bgp/test_commands.py\n* neutron/tests/unit/services/bgp/test_plugin.py","accounts_in_message":[],"_revision_number":1},{"id":"b6dd95ae515f585864c5a0dc861e103e75b645cd","author":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"date":"2026-07-16 13:17:33.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"135cdcfefaebb7da124767261839953e0c3e8272","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 13:44:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d47c6422b04798858c16314c90fb90\n\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/128c0a4163bf44bf8580770cf8302836 : NODE_FAILURE Node(set) request 624981a9287b4f66bbe21d9d727bff82 failed in 0s\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9c20e7d1b6464b12bb8b5a2c34f0988f : NODE_FAILURE Node(set) request 7ef29d3f837c4012b6d3397bc995a34f failed in 0s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b244f56dfb04f2c84077a0bb76f8961 : NODE_FAILURE Node(set) request 32656952ee4b4f4e80fbd58843a0251a failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"697f829a5c1d7a34d67e33e6d85c44a601f82c1e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-16 14:27:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a58af2c32cb669f9e56dd4251dd8ec3dd1634829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 15:29:52.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9f96d8d2d0645f098341e7c8a6798a2\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5b11295b608646928369f32a1772b27e : SUCCESS in 1h 30m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50a8da6ad36641b9bb65ba79d64f8372 : SUCCESS in 1h 08m 27s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/f02415d612a148d386fa9c30aca8e76d : SUCCESS in 1h 22m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/ba4cb65aa77441b6bcfad801597972a7 : SUCCESS in 1h 52m 52s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/9d2ca2530e5d4e7ba5f8d96a1c5a5ba2 : SUCCESS in 2h 07m 37s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/d936509b75ca41bbb4edd9d30f0a6d1a : SUCCESS in 31m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1f0b4f65d8c4c068a5f99dd2009456d : FAILURE in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d55f6e583bd4a11b47c9fc246be62df : FAILURE in 8m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/605d330b722f40b9b29ca821144b92cb : FAILURE in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0de803ae94fe478580b33e84a0456b88 : FAILURE in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9e1233b2cd744db6a7017f475f3190d6 : FAILURE in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95f0d51f5416474a88134fe3eb501784 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56511375bc304683a7110c46b959a1b6 : SUCCESS in 21m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/64b31fa5fa9e48738fe404a794126766 : SUCCESS in 1h 14m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aa39b46336db4556bd412ff2e005bf23 : SUCCESS in 52m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/06157b6bb20c4443a9f1a27e52833bcd : FAILURE in 14m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c515e92d23d40bfbc5654bf780576dc : SUCCESS in 1h 01m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/77b46b2286a044bba145331b7d941f0a : SUCCESS in 1h 04m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4cba1ca85f6c4a0b9c4e44e70f047bdd : SUCCESS in 2h 09m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/13fa367dabdb47cabfab842f6fddc0d7 : SUCCESS in 1h 46m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/43154b3b52fa4af5831cc1d18882f8b4 : SUCCESS in 1h 13m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/360433080c4a4febbc4b6ec43735a0ff : SUCCESS in 1h 08m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d804ddd5e384b5d81b4460c51467d7b : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0b019a231434af935b418079e350baf5d77f450c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-09-03 18:55:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"589aea3d50c61bedf5fb826b2dfe98ff60e24921","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-09-03 19:02:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"287298f5b01e427a4e4a36401cac8a584c98b4fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 19:06:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/424c8d5a3b054e69b04600b09759aba1\n\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a65ce44d5a954a40bdead0c12a46addf : FAILURE in 8m 56s\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f8f9a15ccf17400899400e321e7f7d16 : FAILURE in 7m 32s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18287055b1e54ef7bb3d266483770a6c : FAILURE in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e40d53e566e5652cd34463a2ea69cc716d578560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 20:57:29.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3e7c505e04a4cd099ff9e682223d1df\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c564be39367d47829f5cf082e1a88589 : SUCCESS in 1h 20m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b0175fa742c428db4184ac02f8426ff : SUCCESS in 1h 04m 46s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/2916932b84dd4567aa859259249727c8 : SUCCESS in 1h 31m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/cd4e3cdf0e0449f49cec2b941dcfa962 : SUCCESS in 1h 46m 20s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/ac199c807f014ccf9e0b8bc1657af03f : SUCCESS in 1h 50m 45s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/a3a1be3b0440493190b55a14496fe183 : SUCCESS in 29m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/842b3abaee084c2e8051f0a95484c153 : FAILURE in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38dc3b095fb94af48eb3683e91e872c7 : FAILURE in 8m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0d3aeccf579471b90661e2e54bcd5a2 : FAILURE in 8m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aac759c3d8f740719253986fe1a97e00 : FAILURE in 4m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/252e1cd0d2604062acde52c57ea60471 : FAILURE in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4470b0c3438e4c90b57ad9db52c3aafc : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f921178b21d045618dd25af0e5883c8d : SUCCESS in 26m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8e72153e8fbb46afbc25c259d0adc58e : SUCCESS in 38m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c2ebcf3e621d4fcbb50240b07f6a00e4 : SUCCESS in 39m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b3c9cd6c4396463b92595df654b90e61 : FAILURE in 22m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4f64ce0d51d44cdbb713dacec2d3a7a : SUCCESS in 48m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7f4ca2ad93c487da775331e989f0a21 : SUCCESS in 1h 13m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/56a911f0e0504331aff3d9d3d55ed9d9 : SUCCESS in 1h 56m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c4bdc72e8533408bbcf05ebb3af9b320 : SUCCESS in 1h 47m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ccb1c8ea7c08414699d3094368f56290 : SUCCESS in 1h 07m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a2099425f43145149128d05607ac6e2a : SUCCESS in 1h 04m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/626e70594003466dafe3850f169b4b08 : SUCCESS in 1h 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9dc2514617274f510b14fd62a4269e3e98e612ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39330,"name":"Amir Abbas","email":"am_882011@live.com","username":"amirabbas"},"date":"2026-09-12 11:08:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"38bc6bf86cf56fec37ec1e6981aad63ea08a78f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 12:00:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/728def2745f94a92a2ec819e46249f5e\n\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce96383866674e25b08c65ae14a232a2 : SUCCESS in 32m 56s\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/01ce3f6aa752437c917f0775f7e7c14a : SUCCESS in 30m 43s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ca9b872044a40b886ca491997dfa2b6 : SUCCESS in 50m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9c193a795d1646d8f6397e504a082b6bee358a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:16:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/260b6e293d914ca7a31baab89b6b5f53\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5800742722544e24af43231246f8d7c4 : SUCCESS in 1h 01m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70b950b2b41a437da53441dd35c9bf3a : SUCCESS in 39m 02s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/83aab1bbcda644be8ddf92dec88503e1 : SUCCESS in 1h 33m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/458a27af20794390bcdc896c6b47365a : SUCCESS in 1h 30m 10s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/2e938a3cef25462781c9f43f7d6699f5 : SUCCESS in 2h 06m 30s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/fc3c82e3673c4241b66dac108f28b213 : SUCCESS in 29m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47db1d40227c40c3b33805de6b7add67 : SUCCESS in 45m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d739e7616efc4313b5893d96f011b2c9 : SUCCESS in 12m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41693ff2f10449a38f14c320e710978e : SUCCESS in 25m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a9d1acc93c24588a72fd306122d5dcc : SUCCESS in 20m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb2cbf3e88944c54b20b67bf022a69f0 : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f6abef12ce4fe8aeff10045276e585 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8c67f11dcd948758a86f39c68bf6a20 : SUCCESS in 15m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f090ed7a8d204365b881ac4a27d705b3 : SUCCESS in 40m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e10b0cd8a1a84c439dc373b180302056 : SUCCESS in 1h 16m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f0e39982a3aa4dbf9e3cf499523fb328 : SUCCESS in 1h 21m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b3be67aad50483eb8d0333c1c11e7bd : SUCCESS in 1h 11m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ffb3a6ed637a4bb384830e7548205af2 : SUCCESS in 1h 17m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c55f9f427f6442649d0d8c3c52f2d7fb : SUCCESS in 1h 13m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7bb532fc401b474dac4547f8754d324a : SUCCESS in 1h 42m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/012fe81c3f9140a79e26ecabf6c3d8ba : SUCCESS in 54m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7345260329a74baab16027142dea2beb : SUCCESS in 48m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/511e0efcf0c848fea0620ab131d01d56 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":2},{"id":"24875a45048d7f4fa7f84219b7e9c746da295636","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-15 14:33:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"634eb0423f497d9f1cf04e560ae91f9be67190cb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-15 16:20:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9065d3a17229d5f58cb10e5768b5e4eeb676c216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 16:21:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"939e407cd9f25b030fa76f2fa1361734a2a747c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:53:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/609f74ca404247219f536f5526dbe4c1\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d5d8b8279f0440aa98ae0084bab9e078 : SUCCESS in 1h 01m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75fca79a6be244069bd4300f76c7b5c2 : SUCCESS in 1h 04m 09s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/218c547d4d134c3fa0225b694265f918 : SUCCESS in 2h 31m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0301a6b1bd944214a0cd77bd6ad98386 : SUCCESS in 14m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d7f17e39e1e439ea278c2f2ede67147 : SUCCESS in 29m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b35022afa4b4b3695977cf19c4fb63f : SUCCESS in 23m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc65f657798f42a184f9f3f94f2ee621 : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72ee84ca3b3f4184b2463bf3e79ed3d2 : SUCCESS in 8m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/216474d70207414a9edbb4b184e8d8f3 : SUCCESS in 1h 04m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e083ed4ec24a44e88dd42a5dad69020c : SUCCESS in 1h 18m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfdfc3651b8244ab945f432be1e01021 : SUCCESS in 40m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9884d50269b54ea0a3be8e272736957a : SUCCESS in 54m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1c89c8dc3e804ab396469fe324998c71 : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f3e301f623e2c1cf3bac111352704fdb179d765a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:53:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9d9cc5aa1a3866af60277e061f4e2a16b1118861","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:55:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cac9b34b58b848db9380780deb8020ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfdc396559974fdd8a828697cb0e26b2 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"811e3cc799f7cccc6e5fc38f479f0a2072c0704b","revisions":{"eb34ccb5aea7424318211d7ff7ef4adb83536785":{"kind":"REWORK","_number":1,"created":"2026-07-16 13:13:34.000000000","uploader":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"ref":"refs/changes/50/997550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/997550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/997550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/997550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/997550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/997550/1"}}},"commit":{"parents":[{"commit":"fe5a33341209ba3fc657991b445bdfe459063ac1","subject":"Fix load_db_migration_drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe5a33341209ba3fc657991b445bdfe459063ac1"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-12 20:15:10.000000000","tz":-240},"committer":{"name":"alisafari","email":"alisafar1212@gmail.com","date":"2026-07-16 13:13:34.000000000","tz":0},"subject":"bgp: Always assume just one provider switch","message":"bgp: Always assume just one provider switch\n\nNow that we do not allow VLAN networks, it leaves us\nwith a single flat network for now. The vlan network\ncan be replaced in the future with evpn-vxlan so the\nadmin can have a provider network using the spine-n-leaf\nfabric.\n\nAssisted-By: Claude Opus 4.6\nChange-Id: Ib360dec982aff53e9e6abdab197de354899274d8\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n(cherry picked from commit 90ee661f9ebc503adbd2095afda2e3aaeb13f93b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb34ccb5aea7424318211d7ff7ef4adb83536785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb34ccb5aea7424318211d7ff7ef4adb83536785"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"5fd76f0a01206bed9bdf937cd87ccd551f078899","ours":"fe5a33341209ba3fc657991b445bdfe459063ac1","theirs":"90ee661f9ebc503adbd2095afda2e3aaeb13f93b","merge_strategy":"recursive","contains_conflicts":true}},"811e3cc799f7cccc6e5fc38f479f0a2072c0704b":{"kind":"REWORK","_number":2,"created":"2026-09-12 11:08:27.000000000","uploader":{"_account_id":39330,"name":"Amir Abbas","email":"am_882011@live.com","username":"amirabbas"},"ref":"refs/changes/50/997550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/997550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/997550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/997550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/997550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/997550/2"}}},"commit":{"parents":[{"commit":"fb082163a15385cf29245145e135e15f6139949d","subject":"[OVN] Fix shared admin_context in OVSDB monitor event handlers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb082163a15385cf29245145e135e15f6139949d"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-05-12 20:15:10.000000000","tz":-240},"committer":{"name":"Amir Abbas","email":"am_882011@live.com","date":"2026-09-12 10:47:55.000000000","tz":210},"subject":"bgp: Always assume just one provider switch","message":"bgp: Always assume just one provider switch\n\nNow that we do not allow VLAN networks, it leaves us\nwith a single flat network for now. The vlan network\ncan be replaced in the future with evpn-vxlan so the\nadmin can have a provider network using the spine-n-leaf\nfabric.\n\nAssisted-By: Claude Opus 4.6\nChange-Id: Ib360dec982aff53e9e6abdab197de354899274d8\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n(cherry picked from commit 90ee661f9ebc503adbd2095afda2e3aaeb13f93b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/811e3cc799f7cccc6e5fc38f479f0a2072c0704b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/811e3cc799f7cccc6e5fc38f479f0a2072c0704b"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
