)]}'
{"id":"openstack%2Fneutron~997729","triplet_id":"openstack%2Fneutron~master~I758414db87d1ec99e1fa700b6c6c37ad83168b26","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-07-24 10:30:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I758414db87d1ec99e1fa700b6c6c37ad83168b26","subject":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops","status":"ABANDONED","created":"2026-07-17 14:17:49.000000000","updated":"2026-07-24 10:30:10.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"aab5bbb30b04d4144d9c52eb29c519487ec68682","_number":997729,"virtual_id_number":997729,"owner":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-23 10:02:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"reviewer_updates":[{"updated":"2026-07-17 15:43:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 13:38:21.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"09e221938ef08f3e59b99c8223c41fa969a1371a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-17 14:17:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8ba06d17e8ee3dced86c634aa8c6719ceec7bb7","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-17 14:36:15.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"550029d5888dac191f5db2292c1c936fd5d294ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 15:43:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a2f82e3df2d4e19914d488c015efaff\n\n- neutron-ovn-evpn-tempest-multinode https://zuul.opendev.org/t/openstack/build/e43adb76e8364d2e8a7f3ca925fb778d : FAILURE in 34m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4e0b6deec903329892d1d56d296b710638ab7710","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-20 15:16:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ddcb47a8514d1d42764f85ff68a90518b4a7f867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:16:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dc66a74a5f14493a978e3f1b280b645\n\n- neutron-ovn-evpn-tempest-multinode https://zuul.opendev.org/t/openstack/build/757116fd022c46c4923ff252a43c7411 : FAILURE in 46m 49s","accounts_in_message":[],"_revision_number":2},{"id":"0c905666d64fb4e4b4c4802611c9e3686073b80a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-21 09:37:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"baadbd5c7d0394d99541cf62e86e29b030ae15f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-21 10:05:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"95ac5d29aeef47b4e3d6073db36443ae3ef5ce3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 12:37:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47edac5e32ad434a904dc5edb1841904\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/ac5c0e476efc4a80b7fb0f0970d22cbb : FAILURE in 2h 23m 59s\n- neutron-ovn-evpn-tempest-multinode https://zuul.opendev.org/t/openstack/build/92fb5347a5cd4d1aa9ab29f75defbaa6 : FAILURE in 55m 13s","accounts_in_message":[],"_revision_number":4},{"id":"72badd6d3850cee9731cf4ad70074c17f9fc5a69","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-21 13:38:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f6202d0100eabe6c9ba99fbd4c5e521f81d0cec8","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-07-21 13:38:21.000000000","message":"Patch Set 5: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  while parsing a quoted scalar\n    in \"openstack/neutron/zuul.d/tempest-multinode-bgp.yaml@master\", line 209, column 32\n  found unknown escape character\n    in \"openstack/neutron/zuul.d/tempest-multinode-bgp.yaml@master\", line 214, column 144","accounts_in_message":[],"_revision_number":5},{"id":"9e0943af8e61d44d8998d34d155eb2fb33cd577c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 13:39:01.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  while parsing a quoted scalar\n    in \"openstack/neutron/zuul.d/tempest-multinode-bgp.yaml@master\", line 209, column 32\n  found unknown escape character\n    in \"openstack/neutron/zuul.d/tempest-multinode-bgp.yaml@master\", line 214, column 144","accounts_in_message":[],"_revision_number":5},{"id":"104610ce723bed5f889af9bab88fa45a57b75bd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 13:39:11.000000000","message":"Patch Set 5:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  while parsing a quoted scalar\n    in \"openstack/neutron/zuul.d/tempest-multinode-bgp.yaml@master\", line 209, column 32\n  found unknown escape character\n    in \"openstack/neutron/zuul.d/tempest-multinode-bgp.yaml@master\", line 214, column 144","accounts_in_message":[],"_revision_number":5},{"id":"6a8804ae0dcecead47df20da1cdbd2744960b0f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-21 13:43:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"624d6e65a8f897b79cc8f7795b50b7726c9cfa2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 16:10:39.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"5e9c43bb1b2aaae62128532675f4d7819a7d515d","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-21 16:11:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"277ded29d1131e145b4a19f0056678de527c66dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 19:28:16.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/751fc9381aa549ee90699c31c103d348\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/fff255d4d0a64863beff8b15193ba869 : TIMED_OUT in 3h 10m 24s\n- neutron-ovn-evpn-tempest-multinode https://zuul.opendev.org/t/openstack/build/ce10be3526b54aebbb9974b47404664a : FAILURE in 49m 43s","accounts_in_message":[],"_revision_number":6},{"id":"c16fa38613fe0e3a5b23146990c46f4ab83f1947","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-22 10:45:45.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"f4e1c031796a8d02cda33e6c9e857f04ab7f5c26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 13:34:08.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"9d1e31c143766d4a3f170b0dec4ab16eccc3ecc7","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-22 13:34:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f23d9f833f239cb3373fa11b656e499bb8bcc3f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 16:53:15.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/270fab641274428db9db4d6e6a3423a1\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/be7d0dc66ac14c6ab8890506301af74a : TIMED_OUT in 3h 10m 54s\n- neutron-ovn-evpn-tempest-multinode https://zuul.opendev.org/t/openstack/build/71861a34421149769214cd5e74c78a21 : FAILURE in 1h 10m 30s","accounts_in_message":[],"_revision_number":7},{"id":"ba50d956b0204357d29a24992487db950d1ccef5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-23 09:22:15.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0e28d76ec6d23e6fca29b0f68c18705d1b4e190e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 10:02:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4783bd8b736343018ab53b0c1b356eef\n\n- neutron-ovn-evpn-tempest-multinode https://zuul.opendev.org/t/openstack/build/10a4149495b943e09f0e90b07b94fccc : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":8},{"id":"aab5bbb30b04d4144d9c52eb29c519487ec68682","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-07-24 10:30:10.000000000","message":"Abandoned\n\nabandoned in favor of https://review.opendev.org/c/openstack/neutron/+/994257","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e8fdeafb763b5f3cad081f605ae1a2cb4676e65b","revisions":{"b9cc54523f48531b7447b57981f31198c130ab43":{"kind":"REWORK","_number":1,"created":"2026-07-17 14:17:49.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/29/997729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/997729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/997729/1"}}},"commit":{"parents":[{"commit":"8e0bfed82a338212905e1e9b5afb127bda35cb2e","subject":"Add upstream EVPN multinode Zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e0bfed82a338212905e1e9b5afb127bda35cb2e"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-17 14:15:36.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-17 14:17:38.000000000","tz":120},"subject":"DNM - Run only EVPN multinode job to test advanced server ops","message":"DNM - Run only EVPN multinode job to test advanced server ops\n\nStrip the check pipeline to only neutron-ovn-evpn-tempest-multinode\nto validate the new EVPN network advanced server operations tests.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727\n\nAssisted-By: Claude Opus 4.6\nChange-Id: I758414db87d1ec99e1fa700b6c6c37ad83168b26\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9cc54523f48531b7447b57981f31198c130ab43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9cc54523f48531b7447b57981f31198c130ab43"}]},"branch":"refs/heads/master"},"12c454e5f57528e1764f1f630b079a840038d671":{"kind":"REWORK","_number":2,"created":"2026-07-20 15:16:40.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/29/997729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/997729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/997729/2"}}},"commit":{"parents":[{"commit":"8e0bfed82a338212905e1e9b5afb127bda35cb2e","subject":"Add upstream EVPN multinode Zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e0bfed82a338212905e1e9b5afb127bda35cb2e"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-17 14:15:36.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-20 15:08:25.000000000","tz":120},"subject":"DNM - Run only EVPN multinode job to test advanced server ops","message":"DNM - Run only EVPN multinode job to test advanced server ops\n\nStrip the check pipeline to only neutron-ovn-evpn-tempest-multinode\nto validate the new EVPN network advanced server operations tests.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727\n\nAssisted-By: Claude Opus 4.6\nChange-Id: I758414db87d1ec99e1fa700b6c6c37ad83168b26\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12c454e5f57528e1764f1f630b079a840038d671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12c454e5f57528e1764f1f630b079a840038d671"}]},"branch":"refs/heads/master"},"6184276b2616f6ce20a0ff0d30f4ce85f3acf6b4":{"kind":"REWORK","_number":3,"created":"2026-07-21 09:37:05.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/29/997729/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/997729/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/997729/3"}}},"commit":{"parents":[{"commit":"8e0bfed82a338212905e1e9b5afb127bda35cb2e","subject":"Add upstream EVPN multinode Zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e0bfed82a338212905e1e9b5afb127bda35cb2e"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-17 14:15:36.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-21 09:08:08.000000000","tz":120},"subject":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops","message":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops\n\nStrip the check pipeline to neutron-ovn-bgp-tempest-multinode and\nneutron-ovn-evpn-tempest-multinode to validate the new EVPN network\nadvanced server operations tests and check the effect of enabling\nlive migration in the base job.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727\n\nAssisted-By: Claude Opus 4.6\nChange-Id: I758414db87d1ec99e1fa700b6c6c37ad83168b26\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6184276b2616f6ce20a0ff0d30f4ce85f3acf6b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6184276b2616f6ce20a0ff0d30f4ce85f3acf6b4"}]},"branch":"refs/heads/master"},"2f8f9d783c1e79434b571d4e5e238f7e3c45cfa8":{"kind":"REWORK","_number":4,"created":"2026-07-21 10:05:47.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/29/997729/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/997729/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/997729/4"}}},"commit":{"parents":[{"commit":"8e0bfed82a338212905e1e9b5afb127bda35cb2e","subject":"Add upstream EVPN multinode Zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e0bfed82a338212905e1e9b5afb127bda35cb2e"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-17 14:15:36.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-21 10:05:23.000000000","tz":120},"subject":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops","message":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops\n\nStrip the check pipeline to neutron-ovn-bgp-tempest-multinode and\nneutron-ovn-evpn-tempest-multinode to validate the new EVPN network\nadvanced server operations tests and check the effect of enabling\nlive migration in the base job.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727\n\nAssisted-By: Claude Opus 4.6\nChange-Id: I758414db87d1ec99e1fa700b6c6c37ad83168b26\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f8f9d783c1e79434b571d4e5e238f7e3c45cfa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f8f9d783c1e79434b571d4e5e238f7e3c45cfa8"}]},"branch":"refs/heads/master"},"b91944109d73287b3a2b8388228ffe968178eb1e":{"kind":"REWORK","_number":5,"created":"2026-07-21 13:38:04.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/29/997729/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/997729/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/997729/5"}}},"commit":{"parents":[{"commit":"8e0bfed82a338212905e1e9b5afb127bda35cb2e","subject":"Add upstream EVPN multinode Zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e0bfed82a338212905e1e9b5afb127bda35cb2e"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-17 14:15:36.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-21 13:36:23.000000000","tz":120},"subject":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops","message":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops\n\nStrip the check pipeline to neutron-ovn-bgp-tempest-multinode and\nneutron-ovn-evpn-tempest-multinode to validate the new EVPN network\nadvanced server operations tests and check the effect of enabling\nlive migration in the base job.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/998149\n\nAssisted-By: Claude Opus 4.6\nChange-Id: I758414db87d1ec99e1fa700b6c6c37ad83168b26\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b91944109d73287b3a2b8388228ffe968178eb1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b91944109d73287b3a2b8388228ffe968178eb1e"}]},"branch":"refs/heads/master"},"379d737e3cb71415becd0bb270b7fb4b083b4d6d":{"kind":"REWORK","_number":6,"created":"2026-07-21 13:43:52.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/29/997729/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/997729/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/997729/6"}}},"commit":{"parents":[{"commit":"8e0bfed82a338212905e1e9b5afb127bda35cb2e","subject":"Add upstream EVPN multinode Zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e0bfed82a338212905e1e9b5afb127bda35cb2e"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-17 14:15:36.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-21 13:43:32.000000000","tz":120},"subject":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops","message":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops\n\nStrip the check pipeline to neutron-ovn-bgp-tempest-multinode and\nneutron-ovn-evpn-tempest-multinode to validate the new EVPN network\nadvanced server operations tests and check the effect of enabling\nlive migration in the base job.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/998149\n\nAssisted-By: Claude Opus 4.6\nChange-Id: I758414db87d1ec99e1fa700b6c6c37ad83168b26\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/379d737e3cb71415becd0bb270b7fb4b083b4d6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/379d737e3cb71415becd0bb270b7fb4b083b4d6d"}]},"branch":"refs/heads/master"},"25cca4566004c98a366a2d1519ec4dfeb31d9aa9":{"kind":"REWORK","_number":7,"created":"2026-07-22 10:45:45.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/29/997729/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/997729/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/997729/7"}}},"commit":{"parents":[{"commit":"8e0bfed82a338212905e1e9b5afb127bda35cb2e","subject":"Add upstream EVPN multinode Zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e0bfed82a338212905e1e9b5afb127bda35cb2e"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-17 14:15:36.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-22 10:44:55.000000000","tz":120},"subject":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops","message":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops\n\nStrip the check pipeline to neutron-ovn-bgp-tempest-multinode and\nneutron-ovn-evpn-tempest-multinode to validate the new EVPN network\nadvanced server operations tests and check the effect of enabling\nlive migration in the base job.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/998149\n\nAssisted-By: Claude Opus 4.6\nChange-Id: I758414db87d1ec99e1fa700b6c6c37ad83168b26\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25cca4566004c98a366a2d1519ec4dfeb31d9aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25cca4566004c98a366a2d1519ec4dfeb31d9aa9"}]},"branch":"refs/heads/master"},"e8fdeafb763b5f3cad081f605ae1a2cb4676e65b":{"kind":"REWORK","_number":8,"created":"2026-07-23 09:22:15.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/29/997729/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/997729/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/997729/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/997729/8"}}},"commit":{"parents":[{"commit":"4f8b96e0275589b9f5f8b7826d11b5dac3083e7b","subject":"Add upstream EVPN multinode Zuul job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f8b96e0275589b9f5f8b7826d11b5dac3083e7b"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-17 14:15:36.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-07-23 09:22:06.000000000","tz":120},"subject":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops","message":"DNM - Run BGP and EVPN multinode jobs to test advanced server ops\n\nStrip the check pipeline to neutron-ovn-bgp-tempest-multinode and\nneutron-ovn-evpn-tempest-multinode to validate the new EVPN network\nadvanced server operations tests and check the effect of enabling\nlive migration in the base job.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/997727\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/998149\n\nAssisted-By: Claude Opus 4.6\nChange-Id: I758414db87d1ec99e1fa700b6c6c37ad83168b26\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8fdeafb763b5f3cad081f605ae1a2cb4676e65b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8fdeafb763b5f3cad081f605ae1a2cb4676e65b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
