)]}'
{"id":"openstack%2Fneutron~997993","triplet_id":"openstack%2Fneutron~master~I9f663a5ae8267f8c508f70bd8b0672e8460475d4","project":"openstack/neutron","branch":"master","topic":"bug/2161095","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-07-20 21:27:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9f663a5ae8267f8c508f70bd8b0672e8460475d4","subject":"[OVN] Set external_mac for bound distributed Floating IPs","status":"MERGED","created":"2026-07-20 13:49:46.000000000","updated":"2026-07-20 21:29:11.000000000","submitted":"2026-07-20 21:27:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"997993-bug/2161095","meta_rev_id":"a4ba0f08c032007a47342e5306c7701db9071a17","_number":997993,"virtual_id_number":997993,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-20 21:27:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-07-20 14:53:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-07-20 19:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-07-20 19:31:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-07-20 14:53:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 14:53:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-07-20 14:53:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-07-20 14:53:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-07-20 15:41:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-20 15:44:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-20 19:31:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"ee5abd9dcad423409e0a1a8537b47ac35f7a4316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-07-20 13:49:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac9276f01960b211de5da7d498fe8645e6d96f58","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-20 14:53:15.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e83fb35bc8c31dd1259dbd5da0766998aa3aa8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 15:41:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4b5b834aa884abd8f3abf97136a31be\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/98800054ef124d77a7b4cd5ce59f2ad6 : SUCCESS in 24m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cce337bbef6f4d2782e8f350a5f72414 : SUCCESS in 49m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/305790a331af49a7a0fc92362c25dedd : SUCCESS in 59m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49aedce4b75ec046d47b0f70e9e47cc0108c72f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 15:44:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6ab1679b077411e8210672a7cc13df9\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a0605dfc41774e40b191d66fa1bf07c3 : SUCCESS in 1h 07m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22e4fcfb795040f890297b5603f43a81 : SUCCESS in 1h 22m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0db6278949d54ec4bff2fc57d651d7e5 : SUCCESS in 1h 54m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ecca806554c45b5b041c68b8a5e5497 : SUCCESS in 1h 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ab9097cf364d2c92e9af858bbe8d9d : SUCCESS in 11m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa0732d8e8a940e38b6bdb530f8cefae : SUCCESS in 24m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/938c66b8ea6b4be19207a1f532edee5d : SUCCESS in 24m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00e36bb2ad384a42a2fd240efcd195d0 : SUCCESS in 32m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/347c983dc4c1467da4dce8ba74c9e9fd : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1931806eaa8344f08fc240de56d340cf : SUCCESS in 21m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f8d0e23e32114189809a396dd6daef49 : SUCCESS in 1h 34m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/86d41eb308de4f05bfa8b4b0d64271f9 : SUCCESS in 1h 20m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/71ccb6f8d6db41f891f322e07dd4068e : SUCCESS in 1h 03m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/efe51524c46b47cbbeb8f765472fb717 : SUCCESS in 1h 30m 59s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"113ede1ca8d2724c8f0fd8490a93df842c7fd478","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-20 19:31:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"03882ca5a7830dcaf19bae5db8dde3cc21cca955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 19:32:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e63a91e28809e2bea957ef9854f3ab90dbd2fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 21:27:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ece38d2a9fc84db6898afde06d2bc00d\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/757dd5081ef04930ba6cc6a977f76694 : SUCCESS in 1h 27m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/099fe695b985445891e224203bc55b0e : SUCCESS in 1h 08m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ba587a2c825647eeba6dac6f38ab5c98 : SUCCESS in 1h 52m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a1bc8d9e78405289f2dc36b2337b5b : SUCCESS in 9m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3631928f883f4013b3f454771b95ea29 : SUCCESS in 29m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c332e7e95d624514baabf02cd0711736 : SUCCESS in 26m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec1cb0ee4224579aa8713db28e3c2f3 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a9f420a763e4721a784c6f050df9784 : SUCCESS in 21m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d68041316f0c436da5d11c6a28260338 : SUCCESS in 1h 28m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/122f4fc97cd6408ea3e43b689d109423 : SUCCESS in 1h 18m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e09a7acc3f0444bca1ccf574f5d79f43 : SUCCESS in 38m 50s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"934999f5dfc7d114b7bb0b05bcdeb0baeaa61620","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 21:27:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4ba0f08c032007a47342e5306c7701db9071a17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 21:29:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/116a1181d88243b586c84de0197579bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87af395f578c4fa1890f5c6f6502211b : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80ac8d44af4d4da8808e8a99d8a90be9a88eaac9","revisions":{"80ac8d44af4d4da8808e8a99d8a90be9a88eaac9":{"kind":"REWORK","_number":1,"created":"2026-07-20 13:49:46.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/93/997993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/997993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/997993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/997993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/997993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/997993/1"}}},"commit":{"parents":[{"commit":"199191b5e1d9b9abda8e7689ec0133c4adad7ac4","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/199191b5e1d9b9abda8e7689ec0133c4adad7ac4"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-20 13:34:19.000000000","tz":0},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-07-20 13:49:40.000000000","tz":0},"subject":"[OVN] Set external_mac for bound distributed Floating IPs","message":"[OVN] Set external_mac for bound distributed Floating IPs\n\nWhile rebuilding ovn dbs using neutron-ovn-db-sync-util\nexternal_mac was not getting set when floating ips\nare distributed and bound, this patch handles it.\n\nport up event handles this already as part of\nsync_routers_and_rports but that only works when\novn controller is connected to the ovn sb db and\nclaims the port and mark it up. But for the scenarios\nwhere ovn dbs are rebuild from scratch using neutron\ndb ovn controller would not reconnect to the cluster SB db\nuntil those are restarted or \"ovn-appctl sb-cluster-state-reset\"\nis ran.\n\nAssisted-By: Claude\nCloses-Bug: #2161095\nChange-Id: I9f663a5ae8267f8c508f70bd8b0672e8460475d4\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80ac8d44af4d4da8808e8a99d8a90be9a88eaac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80ac8d44af4d4da8808e8a99d8a90be9a88eaac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
