)]}'
{"id":"openstack%2Fneutron~998053","triplet_id":"openstack%2Fneutron~master~Icafb28437a3de1981315abf91ecdf7da49d2ff64","project":"openstack/neutron","branch":"master","topic":"evpn_standalone","hashtags":[],"change_id":"Icafb28437a3de1981315abf91ecdf7da49d2ff64","subject":"[DNM] POC standalone tool for creating evpn router","status":"ABANDONED","created":"2026-07-20 19:19:50.000000000","updated":"2026-08-10 12:54:10.000000000","total_comment_count":1,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"1b03e4b2e967770749ddb9a79c4a282d8f4270ce","_number":998053,"virtual_id_number":998053,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-07-21 20:53:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-07-27 19:24:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"reviewer_updates":[{"updated":"2026-07-20 20:53:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-22 09:17:37.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"CC"}],"messages":[{"id":"8898aa4b41d0353f4bb2f696fa0fa0fcf8b63d9c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-07-20 19:19:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d6617f09cf307ce4f024f7a8b0a6fd8ff71beac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 20:53:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f5aeaa3831a4997b90da8637c9a4637\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a389f72daa4a45b796ff01374bdd2d11 : SUCCESS in 1h 00m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7fcfa73fec0407eb011cdb4ce800b0f : FAILURE in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1fb9977a8b48d1b891fe88981ba952 : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb61a91d863f44c298211f63cc6161fe : SUCCESS in 20m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/36369e4585d9401a92fa43980aa68063 : SUCCESS in 1h 28m 30s","accounts_in_message":[],"_revision_number":1},{"id":"03b8d0aa4b73fcc8183bcbdb131399c158dded3c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-07-21 19:23:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"295ee17f2ab0592c5cd3319e9eac1cf117f3af31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 20:53:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15dcf250c4d24af5ac7030ba2459a0d8\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af4ff467dcf045a498a2c1eefabc545d : SUCCESS in 1h 03m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a805829cc26d4915880aa63a7eaf3d69 : FAILURE in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556f0fb79acc40ccafa398ced0872a8b : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85e6df22ea744cdc8962248ad43b09df : SUCCESS in 21m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6da99cacb45d4fe98251357abd479a59 : SUCCESS in 1h 24m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7810a18c02f2cd974c7616cd584000d45d8c92b4","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-07-27 19:24:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b03e4b2e967770749ddb9a79c4a282d8f4270ce","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-08-10 12:54:10.000000000","message":"Abandoned\n\nSimilar tool will go into the tempest neutron directly https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/998604","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df1298e2fe70ba6eb7e3b0bc66d687b0b821aac3","revisions":{"cc8966f678d595b73edf6900b63e91678478830f":{"kind":"REWORK","_number":1,"created":"2026-07-20 19:19:50.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/53/998053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/998053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/998053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/998053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/998053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/998053/1"}}},"commit":{"parents":[{"commit":"199191b5e1d9b9abda8e7689ec0133c4adad7ac4","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/199191b5e1d9b9abda8e7689ec0133c4adad7ac4"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-07-20 19:18:44.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-07-20 19:18:44.000000000","tz":-240},"subject":"[DNM] POC standalone tool for creating evpn router","message":"[DNM] POC standalone tool for creating evpn router\n\nChange-Id: Icafb28437a3de1981315abf91ecdf7da49d2ff64\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc8966f678d595b73edf6900b63e91678478830f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc8966f678d595b73edf6900b63e91678478830f"}]},"branch":"refs/heads/master"},"df1298e2fe70ba6eb7e3b0bc66d687b0b821aac3":{"kind":"REWORK","_number":2,"created":"2026-07-21 19:23:32.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/53/998053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/998053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/998053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/998053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/998053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/998053/2"}}},"commit":{"parents":[{"commit":"199191b5e1d9b9abda8e7689ec0133c4adad7ac4","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/199191b5e1d9b9abda8e7689ec0133c4adad7ac4"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-07-20 19:18:44.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2026-07-21 19:23:11.000000000","tz":-240},"subject":"[DNM] POC standalone tool for creating evpn router","message":"[DNM] POC standalone tool for creating evpn router\n\nChange-Id: Icafb28437a3de1981315abf91ecdf7da49d2ff64\nSigned-off-by: Miro Tomaska \u003cmtomaska@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df1298e2fe70ba6eb7e3b0bc66d687b0b821aac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df1298e2fe70ba6eb7e3b0bc66d687b0b821aac3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
