)]}'
{"id":"openstack%2Fneutron~999075","triplet_id":"openstack%2Fneutron~master~I901c07a69470bee9c2db5033b9a3c04b96dcc649","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-29 17:48:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I901c07a69470bee9c2db5033b9a3c04b96dcc649","subject":"Enable hacking plugins from neutron-lib","status":"MERGED","created":"2026-07-29 09:22:27.000000000","updated":"2026-07-29 22:26:33.000000000","submitted":"2026-07-29 17:48:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999075","meta_rev_id":"1eb81f134eb0fe37c663ba1968d8506a5c9e8d1f","_number":999075,"virtual_id_number":999075,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-29 17:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-07-29 09:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-07-29 10:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-07-29 10:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-29 09:43:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-07-29 10:31:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-07-29 11:01:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21e3d76b372bf8c5d5bda3f6f6182a6ffae56ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-29 09:22:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56ed83db480b087fa97539db557fdbcafdb5b401","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-29 09:43:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5f10a8317551c0f6039cc3006f4687c152db714","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-29 10:31:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc6b61b1f3c8df2b5c95bbf427c569b71114babf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 11:01:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c5da92f8db64586b5b1d1ecff35d82a\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1037db9cb07347a1b9cd24c41324f6a8 : SUCCESS in 1h 06m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94a3732c0cd14f038922b18b4c4fb9b7 : SUCCESS in 1h 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351e1ac1aa51492c9fc4157ea33c8a1f : SUCCESS in 11m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b6d1a66aaed455988962435269270bc : SUCCESS in 30m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a3b57cf630f4a57bdc7fd2b04eaa9c4 : SUCCESS in 28m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/99ae24ee6bef4e95b19de1a5a2b1fc76 : SUCCESS in 35m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2102642c36d74da2a049e856d3923823 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84dbea3f23f74a55a15c1621b9f002c6 : SUCCESS in 25m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/21f6201786d34366b27b1e745164f68e : SUCCESS in 1h 33m 33s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"f1cb456ffd3c998b6ac3d2d907e22f60b4844748","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 11:03:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"350f726930f222f4a0210ac6142bb33b64a12bd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 12:41:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dd64ffdba4d42e3958ceef7a3f08ec7\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68374bfcef4c4845b23b9440254b6316 : SUCCESS in 44m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201d2773be184f21ba35e72cad7a3f03 : SUCCESS in 12m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4b587456dd2407bb2bd89a5bad9f1bd : SUCCESS in 29m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6372877a3aeb47159a2ff8c18b5c6c8d : SUCCESS in 24m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/687cef6e6b394b6f8e697763fcf0b911 : SUCCESS in 14m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50d1476d8b504e40b0bd17d20b671cba : SUCCESS in 24m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/838ad2d9606941feb858faba91e97574 : FAILURE in 50m 36s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"e2a916a377655d70b3d35b0f63797c219fe3169e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 13:14:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/432680b93f384630a099783477508eae\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b026971ab164b7cac2bc8ef9580be61 : SUCCESS in 43m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c30eb27792ae49b09e0c3674268eec7d : SUCCESS in 1h 18m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/90f234ae0bb64fc7883ac5f73effadf4 : TIMED_OUT in 1h 21m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e3e7fabfdb70eb56ecd7abaf1018cb132349184","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-29 14:52:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d14e9ce9cd9ae5dc98763b5f8b9632c3e274a804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 16:13:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ea4f40d24b14134835bbe5933d68f5b\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f95f43b432074c52946d20135036c4ab : SUCCESS in 1h 04m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40fcde1113cc4b9581299041233900f9 : SUCCESS in 43m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786a974678774d7988ce3d67243c6723 : SUCCESS in 13m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99102b6e7ca54f17adaf47a6431e2255 : SUCCESS in 20m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1f69dde7d94413a9e78723bd2148441 : SUCCESS in 17m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4df5e03d2aac4e69b151436bd2c3638c : SUCCESS in 40m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6040f89a00a448cbddc82e3c831d31c : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abc5fde5a53a4b888431d5cf0bf0cc0f : SUCCESS in 23m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f6bf9ad47c34b1283dd0a7133b93b51 : SUCCESS in 1h 13m 49s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"95f23a610f67d485132381b0be0901172a956cca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 16:14:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59d7c331b661d41cb53f8c2bc66e13b4184d1ee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:48:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d30b7bd050cb4762ab68b1f233d0c5df\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a16f0e1e6ee34df3801955aca4b1e673 : SUCCESS in 1h 00m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3a13041ea70442f87af6433a33c7377 : SUCCESS in 10m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/793b186809854c86a0077b16badf9d19 : SUCCESS in 27m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa06ab44ee984b48b26fa1af86de186a : SUCCESS in 22m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d532722f8fc64225a42f0c183ed3333e : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec03b7aa02864f7cb80926d325296ece : SUCCESS in 18m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4963dc67889f4f9eaa2d9bf490f4f964 : SUCCESS in 1h 29m 25s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":1},{"id":"e31c37388c8d41a59e466f3d48527d606e19604d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:48:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd8e953faa43e6d734d208745edbbec2655feceb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:51:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fa35c72fd1c488b8066f18267e9b8db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5491fa56badc406a92ac5ee263ffb8bf : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ff1368b7bc4475f85aac46551d4816b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1},{"id":"1eb81f134eb0fe37c663ba1968d8506a5c9e8d1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 22:26:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90ef14c8ee624b37badb30e46b1f57f2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/514a5712cacf4fbdad8aba0c587d287e : TIMED_OUT in 1h 21m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dabbdaad7d324a1ca1a5ad48a24a7d9e : SUCCESS in 37m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f43552527bc54a9f9087f36e14d0aeed : SUCCESS in 58m 18s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79ebae9880368de93e80de05f2e3f41d030d7274","revisions":{"79ebae9880368de93e80de05f2e3f41d030d7274":{"kind":"REWORK","_number":1,"created":"2026-07-29 09:22:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/999075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/999075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/999075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/999075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/999075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/999075/1"}}},"commit":{"parents":[{"commit":"92553f76934dc1d32dbbc1fdebe37074937f187c","subject":"Merge \"db: Remove LBaaS dead migration code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92553f76934dc1d32dbbc1fdebe37074937f187c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-29 09:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-29 09:22:01.000000000","tz":540},"subject":"Enable hacking plugins from neutron-lib","message":"Enable hacking plugins from neutron-lib\n\nSince d1f6c22cb07f76a2ce1a847824dcb62b5a22e2e3 was merged, these\nplugins are disabled by default.\n\nChange-Id: I901c07a69470bee9c2db5033b9a3c04b96dcc649\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79ebae9880368de93e80de05f2e3f41d030d7274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79ebae9880368de93e80de05f2e3f41d030d7274"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
