)]}'
{"id":"openstack%2Fneutron~999732","triplet_id":"openstack%2Fneutron~master~I507a22040079401ba82ba832154b2c975c372ab2","project":"openstack/neutron","branch":"master","topic":"cs10_swap_bump","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-08-11 13:28:04.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-08-11 17:12:52.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-10 15:14:57.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I507a22040079401ba82ba832154b2c975c372ab2","subject":"[CI] Bump swap cs10 fips job","status":"MERGED","created":"2026-08-04 14:41:03.000000000","updated":"2026-08-11 17:15:13.000000000","submitted":"2026-08-11 17:12:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999732-cs10_swap_bump","meta_rev_id":"b0caa316c928d4756c567d402b704087adc1d9f0","_number":999732,"virtual_id_number":999732,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-11 17:12:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-08-10 14:20:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-08-10 15:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-08-10 15:14:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2026-08-10 14:28:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-04 15:34:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-08-04 17:32:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-10 14:20:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-08-10 14:28:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-08-10 14:28:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a03695b44f9819bba340540430fd9dc79b3c3e2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-04 14:41:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f323d391b57701915ea1d5d2bfbf505b659a67c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 17:32:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e29d305dc1243798ce5470b206001f1\n\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/0e6ee14ff3fd49e0be76971f94229be8 : FAILURE in 2h 41m 12s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"a04551240e73655247ef94b9186629364a929772","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-05 04:39:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14c80d8e1f1d948daff5211e005a9405ff0013ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 06:42:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94bbf4d6076849cbaa52972cf75dd3d5\n\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/8c834a9ee6334195b509cb5837568837 : FAILURE in 1h 59m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"147913aa28c9716f8318721775578d4a23ae3666","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-10 10:11:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abb7ed6ca63c0aa92f9644ddcafdbe6291976162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 12:20:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a25a2da99316459f90985527d3e1a3ef\n\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/3b88afbf5be04aa88ae01c5e1e339b57 : SUCCESS in 2h 07m 22s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job neutron-ovs-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"89bbd3c7443f2b6eddf62d43150e9ea8663d26a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-10 12:54:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"46b02f7fa538623503437fb28d11f68617527e64","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-10 14:20:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4261dfc1b6074af2a125be26ff6fc8442a8f4dac","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-10 14:28:29.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"2d1258127dbd72939990574c280929ad7cfc39d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 15:13:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b9a7f4d328547abb9a19ba8fb99aeaf\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5db8d57caad4fa087e40146119083de : SUCCESS in 54m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4cfc8e13927c4c8a819606bd5f5789c9 : SUCCESS in 1h 08m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/68c8b69dd0e541e982b07be4ffacc04f : SUCCESS in 2h 10m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9ad955bab1e4d1fa5d381d4e7eb47e3 : SUCCESS in 32m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/184472eb62c0491b95beea0e979eebfb : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559ea2face29468ba155d8666bcbe801 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02bbfb01cdfe40a888ad97c1e40b303b : SUCCESS in 19m 32s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"0ab435c767e912c829cb688f9479cbdb247680ae","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-10 15:14:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"852044ffe3badcdb4abb5ae6308a81205f942a74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 15:15:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4a9d847659d765d6e32185412eb3db39d3a3d2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 16:54:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8a7f52d7d03405f897b13c439b5e25c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b68d8296f171402d8ee8e21b58485be8 : FAILURE in 1h 00m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7b427c32d424669a1b3060fc91ce3c2 : SUCCESS in 1h 23m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/01aeeab89e7540b7841c5bd50404b73b : SUCCESS in 1h 37m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e6c59c7c4249f296744af200367d76 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddf19c3c208e45f38cade201cbc55229 : SUCCESS in 13m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9d7bd1f19764181ae24334be0aaab56 : SUCCESS in 21m 56s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"b6b446b03723506bb425e547f602516542c5b189","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-11 06:30:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29510c76e6daaaff7ed7a4dcdfe5cd0ed1e6a5ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 08:17:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/196f546765a54db7a3e3ebda84ef11c2\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f17fe848cf084699aa9cc0fcbf7133dc : SUCCESS in 45m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b712b5121fcc49b1bc4bca0dac991c5a : SUCCESS in 1h 06m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/abdc501deab54902bdbff0455312a35e : SUCCESS in 1h 46m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e0b33a7132624d9790ac3f25b4d55528 : SUCCESS in 30m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11336cec75bf40bb82aef390a210b830 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542056514dbc40fb9d19b5f96cb2c413 : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b86bcaa3034a4c5fa814db12dc0197b8 : SUCCESS in 18m 42s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"e5ee89fa98fd216b7bfab5c356e88cc8fe7f7ed8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 08:18:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d33f1fad99e82c4a47d5458b18f5a685802a8dc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 09:57:52.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a43370da08f242d29e40afa51082d473\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60221c2cee7840a3aa96e5916b9b6af6 : SUCCESS in 34m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8103248a7020429c8e0cec7152a2cf27 : FAILURE in 1h 05m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bab8a21bfbfc4053ac40b0dc242b40d0 : SUCCESS in 1h 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f811d22ac648f2be796f7a200d6f13 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0facc95011284ce19814781b2fe92ed4 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5905271443341ecadc111631dbbf745 : SUCCESS in 23m 49s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"847d1cd59533c84fdac51827ec1db746e3034a21","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-08-11 13:28:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ee282efffd87c137a59b2e25668fb692cd5fb05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 15:29:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b16c4e3d9a4b4db7eb090c816e92a0\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b6ee6d011e54571ade971bc919a33e8 : SUCCESS in 55m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9401bb2d782341f7a5d7a3339ed4f949 : SUCCESS in 1h 18m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fc1178b53b6645a49c325766bef8dc08 : SUCCESS in 1h 59m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c6eef8be0d44e5d825296d3c8e226eb : SUCCESS in 34m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/285ea748049042779dbf34b6ddecd9aa : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e75596159746e6a7dce6293dac6b18 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dde390c49ee472886e2af468a31e0a6 : SUCCESS in 17m 07s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"a9c4a5999ef1096f412e2da92fa0730868e17c61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 15:30:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"31af89a2437593f61707083cb361ef75fcd73079","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 17:12:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a097f0ddfe7e4a18b704b03b5b5f58fb\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e56bbb2a77e4a51853c270105efd7c8 : SUCCESS in 27m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fa58e01633f448e38c0255c44f432cbd : SUCCESS in 1h 07m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5b70d4bc16ef45e582fb88c645ef041e : SUCCESS in 1h 42m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/999648507fe64121abba16c797dd9d78 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebef5dc2d76499aa147c36934cf32be : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e5e05626099498f958b32d25d685d8e : SUCCESS in 23m 25s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":2},{"id":"7d4b649cf9b799cc665701b3266fc41c1923345f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 17:12:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b0caa316c928d4756c567d402b704087adc1d9f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 17:15:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b5c295e0cb249f98eff97ef1749e517\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3023e179c534a0fbc62f98be1fe641a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4074aa8aaf627bbf569e224952b2489809e83394","revisions":{"89580b5138b1457a5a4640c0bd2ddbc22a2db324":{"kind":"REWORK","_number":1,"created":"2026-08-04 14:41:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/32/999732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/999732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/999732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/999732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/999732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/999732/1"}}},"commit":{"parents":[{"commit":"3cd91cd333b6b05bf335b724bfbb9889092d2897","subject":"Merge \"ml2: Add ``ovs_create_tap`` flag to ML2/OVS mechanism driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cd91cd333b6b05bf335b724bfbb9889092d2897"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-08-04 14:40:27.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-08-04 14:40:29.000000000","tz":330},"subject":"[WIP] Bump swap cs10 fips job","message":"[WIP] Bump swap cs10 fips job\n\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: I507a22040079401ba82ba832154b2c975c372ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89580b5138b1457a5a4640c0bd2ddbc22a2db324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89580b5138b1457a5a4640c0bd2ddbc22a2db324"}]},"branch":"refs/heads/master"},"4074aa8aaf627bbf569e224952b2489809e83394":{"kind":"REWORK","_number":2,"created":"2026-08-10 12:54:04.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/32/999732/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/999732/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/999732/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/999732/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/999732/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/999732/2"}}},"commit":{"parents":[{"commit":"3cd91cd333b6b05bf335b724bfbb9889092d2897","subject":"Merge \"ml2: Add ``ovs_create_tap`` flag to ML2/OVS mechanism driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cd91cd333b6b05bf335b724bfbb9889092d2897"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-08-04 14:40:27.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-08-10 12:52:56.000000000","tz":330},"subject":"[CI] Bump swap cs10 fips job","message":"[CI] Bump swap cs10 fips job\n\nThis fips centos 10 stream job is running on node\nwith memory less than 7 GB and fails randomly due to\noom kills. Bumping swap to give some room and avoid\noom kills in this job.\n\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: I507a22040079401ba82ba832154b2c975c372ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4074aa8aaf627bbf569e224952b2489809e83394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4074aa8aaf627bbf569e224952b2489809e83394"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
