)]}'
{"id":"openstack%2Fneutron~999830","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I9b6ebb928b6f440bd06f8a976f6f83268f29609a","project":"openstack/neutron","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-08 01:34:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b6ebb928b6f440bd06f8a976f6f83268f29609a","subject":"Fix accidental removal of router external gateway info","status":"MERGED","created":"2026-08-05 13:19:42.000000000","updated":"2026-08-08 01:35:37.000000000","submitted":"2026-08-08 01:34:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999830","cherry_pick_of_change":999776,"cherry_pick_of_patch_set":1,"meta_rev_id":"ddf02861c6ba9d1fde7aac38565e1c4566fcc9fd","_number":999830,"virtual_id_number":999830,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-08 01:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-08-05 13:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-08-07 23:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-08-07 23:11:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-08-05 13:23:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-05 13:23:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-08-05 15:32:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-07 23:11:00.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"07b36a1682742258226d3c3c1722757fea02d23e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-05 13:19:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b9f23ca83d353a5c2fe6bc2b8c0d468daf567b6c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-05 13:23:02.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"c5e40093d77cb82c8003d529b71f86976a5da4ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 15:32:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/547d4440bd5c456cb4b5618fe68de5dd\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/963389b34a774a81849ce4b3a40545bc : SUCCESS in 1h 24m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/6bfe38eacfe64d54bbf86f0cdffe1e62 : SUCCESS in 1h 42m 06s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/19b63287e39d403bb1d8dea7d59b4038 : SUCCESS in 1h 40m 20s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/5c108917ed6a4142be9b1c78232dc281 : SUCCESS in 34m 08s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/063b0e889b494bae97a69c28ab17a017 : SUCCESS in 1h 37m 20s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/716922b785664407b215afd8f8cb4f41 : SUCCESS in 1h 22m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b52e5a04cf3c43348e91d5eb72785141 : SUCCESS in 44m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77e962487ff64d1cabdecc9976fcbf2e : SUCCESS in 15m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a109169852ff466b9265216a02f7079f : SUCCESS in 23m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae96108f2d8d426982d9e3a2fdeb4c97 : SUCCESS in 18m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882634a45be14dca9aaefc526c12efd3 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5047e3c1d674e6f84f333754a529f9e : SUCCESS in 26m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3d2ef324f1b640f0b35cc7cb6cc771d1 : SUCCESS in 1h 29m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d32c0377b04947489c9ca3900a857ddb : SUCCESS in 1h 53m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bd2bb1112eda4f449456d854a3bc4bb4 : SUCCESS in 1h 56m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f49ddc2605f246aa9b5d9d9774bb3dd7 : SUCCESS in 1h 41m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b558be9b10584094a918a0d64af92dc1 : SUCCESS in 1h 02m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ef42025214c44cfb859285db8fe3e091 : SUCCESS in 1h 11m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d3299af555814809a7ebd34583bb0556 : SUCCESS in 1h 17m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/01cd50e06bbe4024867edd9d5c7788e1 : SUCCESS in 1h 34m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fcc442ac252f71dd46e902de6b2ab98b31e8c71a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2026-08-07 23:11:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41ca896c540088d68630336cb456d74d10c14805","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 23:11:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52c410ab7a7c167e071271ddb66a09c29f56f5a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 01:34:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c2c4ce3c5a94795817faeba4db8c6ca\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/65ae8f52197b4a2aaa1b78446ce2ec8e : SUCCESS in 1h 50m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef9481c552a4a88b2c6313dcf54bac1 : SUCCESS in 11m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d2cb2389e16496097cef2b20bacc089 : SUCCESS in 21m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6dc8b926c4b64fcaa084ea88d950ee97 : SUCCESS in 21m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec1d6bf71f194be794e94bf61e844853 : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a510c5053504e1baf4649a4baf03382 : SUCCESS in 25m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9889a934fd07438088b555a86a14d966 : SUCCESS in 1h 30m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0128ee3eaabf4d689de1f1445a6a4367 : SUCCESS in 1h 33m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/091b7c7d285f42d6bc25b46e88b216ee : SUCCESS in 1h 53m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/11b9a2a0aa7a4ad09f7c2557e104158a : SUCCESS in 1h 07m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b408015aafdf4c959c3cc92bc129add8 : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e1a87cdd75b63659e4d4bdf43d3be2538d79d1f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 01:34:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ddf02861c6ba9d1fde7aac38565e1c4566fcc9fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 01:35:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df5e2e804cf642e889a7f2683f8b8456\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a3c6b2657574ac3aab130db6a63d6b2 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e0b342e52e4b9a0c4ea99a0a179b5a1a513dd43","revisions":{"6e0b342e52e4b9a0c4ea99a0a179b5a1a513dd43":{"kind":"REWORK","_number":1,"created":"2026-08-05 13:19:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/30/999830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/999830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/999830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/999830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/999830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/999830/1"}}},"commit":{"parents":[{"commit":"b778a6464f36fa5580e7cd4f503fdc7181abf406","subject":"Prevent cross-project subnet onboard on shared networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b778a6464f36fa5580e7cd4f503fdc7181abf406"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-08-05 01:23:22.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-08-05 13:19:42.000000000","tz":0},"subject":"Fix accidental removal of router external gateway info","message":"Fix accidental removal of router external gateway info\n\nMove \"router in use\" check earlier in delete_router()\nso that BEFORE_DELETE callbacks are not triggered\naccidentally, which could result in removal of the\nexternal gateway information.\n\nChange-Id: I9b6ebb928b6f440bd06f8a976f6f83268f29609a\nCloses-bug: #2162868\nAssisted-by: Claude Sonnet 4.6\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n(cherry picked from commit 19d7151a3c9169a90afee3be25ea3fce5cf4ff86)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0b342e52e4b9a0c4ea99a0a179b5a1a513dd43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0b342e52e4b9a0c4ea99a0a179b5a1a513dd43"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"0f353dc98bde24938392c9127e75933b275f9ce3","ours":"b778a6464f36fa5580e7cd4f503fdc7181abf406","theirs":"19d7151a3c9169a90afee3be25ea3fce5cf4ff86","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
