)]}'
{"id":"openstack%2Fnova-powervm~669090","triplet_id":"openstack%2Fnova-powervm~master~I3fdc567cf05bba9bd210f2403a74ab6c2eb12c34","project":"openstack/nova-powervm","branch":"master","hashtags":[],"change_id":"I3fdc567cf05bba9bd210f2403a74ab6c2eb12c34","subject":"Merged changes : Add image type capability flags: \"https://github.com/openstack/nova/commit/ aafd873e2176076609ff17d63137dd6f84ddf8d3# diff-98d1a06e0feb7e5ff43a21433afd9f34\"","status":"ABANDONED","created":"2019-07-04 07:57:57.000000000","updated":"2019-07-09 18:07:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a0ec23ea10cde64d9e2aedd1bc68b21d8116d407","_number":669090,"virtual_id_number":669090,"owner":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-07-04 08:21:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 08:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48466b47b3b02fd6743de19b127113bd9f9a89ba","author":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"date":"2019-07-04 07:57:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c4006f8b672dce1191f6e0ffa161e7581983018","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 08:21:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/90/669090/1/check/openstack-tox-docs/b20a061/html/ : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/669090/1/check/openstack-tox-lower-constraints/49f7f16/ : FAILURE in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/90/669090/1/check/openstack-tox-pep8/fd372ac/ : SUCCESS in 7m 50s\n- openstack-tox-py27 http://logs.openstack.org/90/669090/1/check/openstack-tox-py27/e331b4e/ : SUCCESS in 7m 31s\n- openstack-tox-py36 http://logs.openstack.org/90/669090/1/check/openstack-tox-py36/51781c4/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a0ec23ea10cde64d9e2aedd1bc68b21d8116d407","author":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"date":"2019-07-09 18:07:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d89ab0e358a25b8872ed48c183cd724af9dc7fa","revisions":{"3d89ab0e358a25b8872ed48c183cd724af9dc7fa":{"kind":"REWORK","_number":1,"created":"2019-07-04 07:57:57.000000000","uploader":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"ref":"refs/changes/90/669090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-powervm","ref":"refs/changes/90/669090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/90/669090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/90/669090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/90/669090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-powervm refs/changes/90/669090/1"}}},"commit":{"parents":[{"commit":"bf68706a837af97452bd5caf388cb61bed197ad5","subject":"Merge \"Add Debug Log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/bf68706a837af97452bd5caf388cb61bed197ad5"}]}],"author":{"name":"VedaAnnayappa","email":"va00121481@techmahindra.com","date":"2019-07-04 07:57:35.000000000","tz":-300},"committer":{"name":"VedaAnnayappa","email":"va00121481@techmahindra.com","date":"2019-07-04 07:57:35.000000000","tz":-300},"subject":"Merged changes : Add image type capability flags: \"https://github.com/openstack/nova/commit/ aafd873e2176076609ff17d63137dd6f84ddf8d3# diff-98d1a06e0feb7e5ff43a21433afd9f34\"","message":"Merged changes :\nAdd image type capability flags:\n\"https://github.com/openstack/nova/commit/\naafd873e2176076609ff17d63137dd6f84ddf8d3#\ndiff-98d1a06e0feb7e5ff43a21433afd9f34\"\n\nUse new ``initial_xxx_allocation_ratio`` CONF\n\"https://github.com/openstack/nova/commit/\n08f3ae960623c94bdd997cacb3e81f04b4bbba69#\ndiff-46d49d172cae6b899c1be3628dede5d1\"\n\nChange-Id: I3fdc567cf05bba9bd210f2403a74ab6c2eb12c34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/3d89ab0e358a25b8872ed48c183cd724af9dc7fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/3d89ab0e358a25b8872ed48c183cd724af9dc7fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
