)]}'
{"id":"openstack%2Fnova-powervm~687003","triplet_id":"openstack%2Fnova-powervm~master~I7291077407c818c712fb5c98952a15020191a2cc","project":"openstack/nova-powervm","branch":"master","hashtags":[],"change_id":"I7291077407c818c712fb5c98952a15020191a2cc","subject":"Fixing Unit Test Issues","status":"MERGED","created":"2019-10-07 11:10:35.000000000","updated":"2019-11-21 17:30:34.000000000","submitted":"2019-10-09 11:38:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687003-1570621093717-e8c316f0","meta_rev_id":"015185386d3cd421eb93ea8212c74e84d97d0ecd","_number":687003,"virtual_id_number":687003,"owner":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 11:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"all":[{"value":2,"date":"2019-10-09 11:25:48.000000000","_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"value":2,"date":"2019-10-09 08:33:28.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"all":[{"value":1,"date":"2019-10-09 11:25:48.000000000","_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 13:17:10.000000000","updated_by":{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},"reviewer":{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},"state":"REVIEWER"},{"updated":"2019-10-09 08:33:28.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-09 11:25:48.000000000","updated_by":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"reviewer":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"state":"REVIEWER"},{"updated":"2019-10-09 11:38:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74bc00b1d76335c22d88acbc85cd45e8f6296927","author":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"date":"2019-10-07 11:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3041a13e1525f78899376dc3a2c6ae67b03590b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 11:17:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdee7ffc325945f59c0e56a92dc193ca : SUCCESS in 4m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/957fc0cf73ae4a5ab0225a831fbd65aa : SUCCESS in 2m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dce5f86d7f814323a0ab1cb2515bb95a : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2d3c3492bb841e39685f76adeaa20db : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8cf4773a5c94accbce1bda1eeb9579b : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25c357f240514306930c1161bd3476ba : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"230f642ae4912f559492b7a1da1bffc1d7ba6695","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-07 18:37:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nGood that tests pass, but please fix the commit message.","accounts_in_message":[],"_revision_number":1},{"id":"5cf2763354071b528d6b9d845995064b0a997e18","author":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"date":"2019-10-09 04:08:17.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"146bd28ad7ffefe5519ebc5d9bef98ad5873aed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 04:18:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6baaf3d0c8bd459badb261e31f3f69d5 : SUCCESS in 3m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/171210b6a5b84c209856465b1921576e : SUCCESS in 1m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f8bcfa8ae2d4d3f963337829d08ec9b : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b3a5b11d753425fb5a5541ed32f98fa : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66c923eb4ebc4b97a87ac1ec4e50f568 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10897d90ca6a4773a703789350f48924 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4552a7266645df8e119a787ee56deef59533af22","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-09 08:33:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b4ce1f0c59dee4d70cef6421c2e77c84fe2413bf","author":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"date":"2019-10-09 11:25:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4d670291de98172a873e8812eecbca3446933f99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 11:25:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18a9fbade7006f5530e07cf3b6d2bb76f369f030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 11:38:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e6fbf190944f899f33c798f43cd247 : SUCCESS in 4m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9fb569c6fb9d4f7497513ada73ed6f3d : SUCCESS in 2m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d407b6b82724ca38cd31869ce1614bd : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe834038df24064b423cf7e3ca58a3e : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc247740d6704cad8b219efe635213a2 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a8c114edcc34a28a14238d4c89a489a : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6a50eae39c0bbe628900199f0d0a1c0da1164ccf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 11:38:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"015185386d3cd421eb93ea8212c74e84d97d0ecd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-11-21 17:30:34.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a0c045487ebb5e4fb548c7f679200ef7814be31b","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1fe310f4be1f6b8f9d2ab1e3a6294566af0cc5e1","revisions":{"83eb6559daf5f64c6666cb4680ce501955d6d5f8":{"kind":"REWORK","_number":1,"created":"2019-10-07 11:10:35.000000000","uploader":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"ref":"refs/changes/03/687003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-powervm","ref":"refs/changes/03/687003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/03/687003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/03/687003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/03/687003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-powervm refs/changes/03/687003/1"}}},"commit":{"parents":[{"commit":"5de7a5ebdca34ba56407c9e8cece6953278ce27f","subject":"Updated Libraries:","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/5de7a5ebdca34ba56407c9e8cece6953278ce27f"}]}],"author":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-10-07 11:09:39.000000000","tz":-240},"committer":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-10-07 11:09:39.000000000","tz":-240},"subject":"Fixing UT issues which were created by the latest nova changes","message":"Fixing UT issues which were created by the latest nova changes\n\nChange-Id: I7291077407c818c712fb5c98952a15020191a2cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/83eb6559daf5f64c6666cb4680ce501955d6d5f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/83eb6559daf5f64c6666cb4680ce501955d6d5f8"}]},"branch":"refs/heads/master"},"1fe310f4be1f6b8f9d2ab1e3a6294566af0cc5e1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-09 04:08:17.000000000","uploader":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"ref":"refs/changes/03/687003/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-powervm","ref":"refs/changes/03/687003/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/03/687003/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/03/687003/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/03/687003/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-powervm refs/changes/03/687003/2"}}},"commit":{"parents":[{"commit":"5de7a5ebdca34ba56407c9e8cece6953278ce27f","subject":"Updated Libraries:","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/5de7a5ebdca34ba56407c9e8cece6953278ce27f"}]}],"author":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-10-07 11:09:39.000000000","tz":-240},"committer":{"name":"VedaAnnayappa","email":"va00121481@techmahindra.com","date":"2019-10-09 04:07:56.000000000","tz":0},"subject":"Fixing Unit Test Issues","message":"Fixing Unit Test Issues\n\nUpdating to 1.1.23 version of pypowervm and Fixing UT issues accordingly\n\nChange-Id: I7291077407c818c712fb5c98952a15020191a2cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/1fe310f4be1f6b8f9d2ab1e3a6294566af0cc5e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/1fe310f4be1f6b8f9d2ab1e3a6294566af0cc5e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
