)]}'
{"id":"openstack%2Fnova-powervm~700683","triplet_id":"openstack%2Fnova-powervm~master~I5e06e582795bb6a4fc252f7bd8c57993289e4ef5","project":"openstack/nova-powervm","branch":"master","hashtags":[],"change_id":"I5e06e582795bb6a4fc252f7bd8c57993289e4ef5","subject":"Drop python 2.7 support and testing","status":"NEW","created":"2019-12-27 07:50:48.000000000","updated":"2020-04-25 23:59:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b325bd0dde050314536b12b40853a59f69458875","_number":700683,"virtual_id_number":700683,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"date":"2020-04-06 12:15:46.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"date":"2020-04-18 18:38:37.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-02 11:52:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2020-04-06 12:15:37.000000000","_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2020-04-06 12:15:37.000000000","_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"date":"2020-04-06 12:15:37.000000000","_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"date":"2020-04-06 12:15:37.000000000","_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"date":"2020-04-09 07:27:39.000000000","_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-06 12:15:37.000000000","_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 22:55:10.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-02-02 11:52:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-06 12:15:37.000000000","updated_by":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"reviewer":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"state":"REVIEWER"},{"updated":"2020-04-06 12:15:37.000000000","updated_by":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"reviewer":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"state":"REVIEWER"},{"updated":"2020-04-06 12:15:37.000000000","updated_by":{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},"reviewer":{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},"state":"REVIEWER"},{"updated":"2020-04-06 12:15:37.000000000","updated_by":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"reviewer":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"state":"REVIEWER"},{"updated":"2020-04-06 12:15:37.000000000","updated_by":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"reviewer":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"state":"REVIEWER"},{"updated":"2020-04-06 12:15:46.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-09 07:27:39.000000000","updated_by":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"reviewer":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"state":"REVIEWER"},{"updated":"2020-04-18 18:38:37.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"9ce6f15fcf988d90a954d51a1373ac99f81d0c58","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-12-27 07:50:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71542cfe8b417520a1b059e13d13bbf0adaf5f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 07:56:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1171a7d244b4c04bffca5ef3e260ed5 : SUCCESS in 3m 24s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9c1a9a23009546bd8177fe1202dede74 : SUCCESS in 2m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7a0c3ccbb1d4f63a09827a55ac9c39b : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1646aead9ed14d5bb51c2c176592d022 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f72e23eb662f18bffa389f0bae372ebd279eea34","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-06 22:55:10.000000000","message":"Patch Set 1: Code-Review-1\n\nThis looks right to me.\n\nHowever:\n\n- There should be a release note.\n- We should show a successful PowerVM CI run under py3 before we move ahead on this.","accounts_in_message":[],"_revision_number":1},{"id":"77a4847c9e66249777f661d64eff74be8f4bf7d0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-22 01:43:53.000000000","message":"Patch Set 1: Code-Review-1\n\nalong with review comments, you need to cleanup requirements also.\n\nDo all things mentioned in https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#completion-criteria","accounts_in_message":[],"_revision_number":1},{"id":"dd969ea0a4b8d1cf1024cfec699f6aa06b7982f3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-02 11:23:00.000000000","message":"Patch Set 1:\n\nEric, this repo has no releasenotes.","accounts_in_message":[],"_revision_number":1},{"id":"4019d8873f0db901585aab99e22f133d746fd44d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-02 11:27:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f3c476ee9fdd5bea185f74a07c6fcede00c077a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 11:27:16.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/nova-powervm on branch master.  The error was:\n\n  The project template \"openstack-python-ussuri-jobs\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-lower-constraints-jobs\n        - openstack-python-ussuri-jobs\n        - periodic-stable-jobs\n\n  in \"openstack/nova-powervm/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"748a9ef59f7ca437c6a8e7158d01f054d0e1e320","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-02 11:29:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aaf4df11c9a13ddfe45ba9a94b467a932c7f69d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 11:36:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c5c28c673b4ae19088733ccfc36f06 : SUCCESS in 3m 22s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/556aa458acc94cf7974374c60c6fd1e1 : SUCCESS in 2m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/99450e4f91ab4f7f892280693ddc8157 : SUCCESS in 2m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8633ce52a9fc47c4856ad54c7bff5daa : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77e73d73acf843b8bfc5d7f3cddcdf1f : FAILURE in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/765aa40a8ad44182b244ea11ff424f12 : FAILURE in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca6a367565414d659223cd08a0d01fa9 : FAILURE in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ecd7928f9944f89b66896e16aedb0ab : FAILURE in 2m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb13d061c6862c85d49d9f5d093b2685f52d4458","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-02 11:46:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6cc111d88b979c08ff3d9fadbc5a27a4a327fdb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 11:52:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad89945fbf6d48ea9f57ca66e4b51ee8 : SUCCESS in 3m 37s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac15b812906048bfa5d20f54e19ed8e4 : SUCCESS in 2m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0816192d31dc4b63a73c5369c83d837c : SUCCESS in 2m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c06efd8a24a49d78f7b939765fa5303 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef728e73839c4b019e406d610b1a9510 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d9b00ec49264026a9a8aad2352e22fd : FAILURE in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ea225fe851c4f128703f68e7fcab2a3 : FAILURE in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4831acac9ba34e2b96a822083d163f99 : FAILURE in 2m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1648fa7a396cb361a323025819968811d6464eb8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-15 19:01:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6f75485701c17f1c95fbbec6da7b47fd78c65ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 19:09:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b888b8db68a84dcb94ed42a35d390262 : SUCCESS in 3m 03s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/77faaa9513564e4abe8f91b1c8698b47 : SUCCESS in 1m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/063af934f1c14a399e64f0ccd5fa9e65 : SUCCESS in 2m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7c156d51e4949a296dd411d0283800d : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c8ad33924c4071a14285a0bc321c05 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58f8ddcc2b9c4976ac69ebb459778d96 : FAILURE in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/760a5ae3a55849308f4992e560dd1604 : FAILURE in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22eff5d619f546f882ecab3e5268fa7e : FAILURE in 2m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"423c6994abbde713cebe5fd0e2b815e00e3c0fbf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-06 12:15:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c912ef96001620ec8e9fccc0486bc38aed2fbc51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 12:27:50.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e148593ecf4ac28d804e2fc1b0f9ac : SUCCESS in 4m 55s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a7714b5a12d14af69ffec391126ac217 : SUCCESS in 2m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8de1592eae9f4686bb818bcc16696a51 : SUCCESS in 2m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6cbf9cdf1b24e1fabb78111f22eb0d0 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee1afc87e404a019f2c3a116c1e1758 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b95a3808a1114d1a9f09e81d7cbd6b5b : FAILURE in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5449f565ebc9493180c9054240a1a599 : FAILURE in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/964202cc695d42ddb1ddbf42b5d965bc : FAILURE in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e66f9764c6e7ea0541300b07275b17c78b86095b","author":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"date":"2020-04-09 07:27:39.000000000","message":"Patch Set 4:\n\nThe UTs are failing as the master isn\u0027t in sync with the changes made in nova/master. We need to fix that before this can be merged. Will get that done as soon as possible.","accounts_in_message":[],"_revision_number":4},{"id":"a4a6c356a29d02da1964fbac30ae3e1a13cec107","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-18 18:38:37.000000000","message":"Patch Set 4:\n\n\u003e The UTs are failing as the master isn\u0027t in sync with the changes\n \u003e made in nova/master. We need to fix that before this can be merged.\n \u003e Will get that done as soon as possible.\n\nThanks, Please note deadline was m-1 and we are going to complete the py2 drop goal by next week. All the pending patches will be left over from tracking.","accounts_in_message":[],"_revision_number":4},{"id":"b325bd0dde050314536b12b40853a59f69458875","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-25 23:59:50.000000000","message":"Topic drop-py27-support removed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6d74e71d45c7145c86943560c8d2b1e0956f0269","revisions":{"eb186af93829f679e5a542caf5ccb8ac1ceb38ce":{"kind":"REWORK","_number":1,"created":"2019-12-27 07:50:48.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/83/700683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-powervm","ref":"refs/changes/83/700683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/1"}}},"commit":{"parents":[{"commit":"376d9493e2a10313068508daf9054d7ecf6d121f","subject":"Merge \"Adding LPAR Detailed Settings section.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/376d9493e2a10313068508daf9054d7ecf6d121f"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-12-27 07:51:18.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-12-27 07:51:18.000000000","tz":480},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I5e06e582795bb6a4fc252f7bd8c57993289e4ef5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/eb186af93829f679e5a542caf5ccb8ac1ceb38ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/eb186af93829f679e5a542caf5ccb8ac1ceb38ce"}]},"branch":"refs/heads/master"},"5a723bc7a91434ffdb2c81ffda1053ccaf06bc0e":{"kind":"REWORK","_number":2,"created":"2020-02-02 11:27:02.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/83/700683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-powervm","ref":"refs/changes/83/700683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/2"}}},"commit":{"parents":[{"commit":"376d9493e2a10313068508daf9054d7ecf6d121f","subject":"Merge \"Adding LPAR Detailed Settings section.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/376d9493e2a10313068508daf9054d7ecf6d121f"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-12-27 07:51:18.000000000","tz":480},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-02 11:26:53.000000000","tz":60},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nSwitch jobs to use ussuri template, otherwise pep8 is not run.\n\nNote: This repo has no releasenotes, so releasenotes are not added.\n\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nChange-Id: I5e06e582795bb6a4fc252f7bd8c57993289e4ef5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/5a723bc7a91434ffdb2c81ffda1053ccaf06bc0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/5a723bc7a91434ffdb2c81ffda1053ccaf06bc0e"}]},"branch":"refs/heads/master"},"ed402f1b8b8a245e9dccb6df11c158b528fd0d2a":{"kind":"REWORK","_number":3,"created":"2020-02-02 11:29:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/83/700683/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-powervm","ref":"refs/changes/83/700683/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/3"}}},"commit":{"parents":[{"commit":"376d9493e2a10313068508daf9054d7ecf6d121f","subject":"Merge \"Adding LPAR Detailed Settings section.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/376d9493e2a10313068508daf9054d7ecf6d121f"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-12-27 07:51:18.000000000","tz":480},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-02 11:29:29.000000000","tz":60},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nSwitch jobs to use ussuri template, otherwise pep8 is not run.\n\nNote: This repo has no releasenotes, so releasenotes are not added.\n\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nChange-Id: I5e06e582795bb6a4fc252f7bd8c57993289e4ef5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/ed402f1b8b8a245e9dccb6df11c158b528fd0d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/ed402f1b8b8a245e9dccb6df11c158b528fd0d2a"}]},"branch":"refs/heads/master"},"6d74e71d45c7145c86943560c8d2b1e0956f0269":{"kind":"REWORK","_number":4,"created":"2020-02-02 11:46:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/83/700683/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-powervm","ref":"refs/changes/83/700683/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-powervm refs/changes/83/700683/4"}}},"commit":{"parents":[{"commit":"376d9493e2a10313068508daf9054d7ecf6d121f","subject":"Merge \"Adding LPAR Detailed Settings section.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/376d9493e2a10313068508daf9054d7ecf6d121f"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-12-27 07:51:18.000000000","tz":480},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-02 11:45:57.000000000","tz":60},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nSwitch jobs to use ussuri template, otherwise pep8 is not run.\n\nNote: This repo has no releasenotes, so releasenotes are not added.\n\nRemove hacking, newer pip complains about duplicate requirements.\n\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nChange-Id: I5e06e582795bb6a4fc252f7bd8c57993289e4ef5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/6d74e71d45c7145c86943560c8d2b1e0956f0269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-powervm/commit/6d74e71d45c7145c86943560c8d2b1e0956f0269"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
