)]}'
{"id":"openstack%2Fnova-specs~1002901","triplet_id":"openstack%2Fnova-specs~master~If325a63614978f060f7b07e7b52aa637936acb87","project":"openstack/nova-specs","branch":"master","topic":"bp/availability-zone-scoped-quotas","attention_set":{},"removed_from_attention_set":{"35086":{"account":{"_account_id":35086,"name":"Khoi","email":"nguyenhuukhoinw@gmail.com","username":"nhk"},"last_update":"2026-09-07 23:28:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If325a63614978f060f7b07e7b52aa637936acb87","subject":"Add availability-zone-scoped quota spec","status":"ABANDONED","created":"2026-08-30 00:12:38.000000000","updated":"2026-09-07 23:28:03.000000000","total_comment_count":4,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"d049bb4fe69742162df0c8846b5b776f1530fcca","_number":1002901,"virtual_id_number":1002901,"owner":{"_account_id":35086,"name":"Khoi","email":"nguyenhuukhoinw@gmail.com","username":"nhk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-30 00:25:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2026-08-31 08:49:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 00:25:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 08:49:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"6d1c93b1f95c662e8d2998b53d0b8c22eac5d57f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35086,"name":"Khoi","email":"nguyenhuukhoinw@gmail.com","username":"nhk"},"date":"2026-08-30 00:12:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f866d556c3b57a8275e5634ffedcf11331f01e72","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35086,"name":"Khoi","email":"nguyenhuukhoinw@gmail.com","username":"nhk"},"date":"2026-08-30 00:21:43.000000000","message":"Topic set to bp/availability-zone-scoped-quotas","accounts_in_message":[],"_revision_number":1},{"id":"4365b4ae999c879f34b0fbc748abea0371cf3497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 00:25:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef7601e94a1f4c868612d00dd1f2d1d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc61b5071e89447eb6fdebf4ee024a47 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001c957f076540a485a7f7dddc48ad87 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8f1d1e0d228bd95051218013fa6a6ff296fcfd54","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-31 08:49:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d049bb4fe69742162df0c8846b5b776f1530fcca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35086,"name":"Khoi","email":"nguyenhuukhoinw@gmail.com","username":"nhk"},"date":"2026-09-07 23:28:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1280a4f89564eae14be3ee270cfce9c83eba3b0a","revisions":{"1280a4f89564eae14be3ee270cfce9c83eba3b0a":{"kind":"REWORK","_number":1,"created":"2026-08-30 00:12:38.000000000","uploader":{"_account_id":35086,"name":"Khoi","email":"nguyenhuukhoinw@gmail.com","username":"nhk"},"ref":"refs/changes/01/1002901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/1002901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/1002901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/1002901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/1002901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/1002901/1"}}},"commit":{"parents":[{"commit":"23297f93da97a7011eb7a19399c144f7ee45ccc6","subject":"Create specs directory for 2027.1 Indri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/23297f93da97a7011eb7a19399c144f7ee45ccc6"}]}],"author":{"name":"nhk","email":"nguyenhuukhoinw@gmail.com","date":"2026-08-30 00:06:54.000000000","tz":420},"committer":{"name":"nhk","email":"nguyenhuukhoinw@gmail.com","date":"2026-08-30 00:06:54.000000000","tz":420},"subject":"Add availability-zone-scoped quota spec","message":"Add availability-zone-scoped quota spec\n\nAdd a design for optional project quotas scoped to a compute availability zone for instances, cores, and RAM. Scoped limits are enforced in addition to the existing project-wide limits and remain opt-in.\n\nChange-Id: If325a63614978f060f7b07e7b52aa637936acb87\nSigned-off-by: nhk \u003cnguyenhuukhoinw@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1280a4f89564eae14be3ee270cfce9c83eba3b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1280a4f89564eae14be3ee270cfce9c83eba3b0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
