)]}'
{"id":"openstack%2Fnova-specs~130005","triplet_id":"openstack%2Fnova-specs~master~I58de5b61b1db05b1392b6a5ea54e37d43e2b867e","project":"openstack/nova-specs","branch":"master","topic":"bp/s","hashtags":[],"change_id":"I58de5b61b1db05b1392b6a5ea54e37d43e2b867e","subject":"Nova specs doc for dynamic server groups","status":"ABANDONED","created":"2014-10-21 18:55:27.000000000","updated":"2014-11-21 17:09:52.000000000","total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"050c7ee229c7bd1a6cb041b137e013dc959bafff","_number":130005,"virtual_id_number":130005,"owner":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2014-11-07 01:14:40.000000000","_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},{"value":1,"date":"2014-11-05 23:09:27.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2014-11-05 23:11:29.000000000","_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"value":-1,"date":"2014-11-14 23:05:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":-1,"date":"2014-11-07 09:49:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-10-21 19:08:02.000000000","updated_by":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"reviewer":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"state":"REVIEWER"},{"updated":"2014-10-27 21:58:08.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"},{"updated":"2014-11-05 23:09:27.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2014-11-07 01:14:40.000000000","updated_by":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"reviewer":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"state":"REVIEWER"},{"updated":"2014-11-07 09:49:19.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2014-11-14 23:05:54.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"e382dac42bdbdc75fbe15c9417d45a76238c109f","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-10-21 18:55:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e377ec4b35676835a10c1015a304cd773e96851f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-21 19:00:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/05/130005/1/check/gate-nova-specs-docs/4af64af/doc/build/html/ : SUCCESS in 1m 40s\n- gate-nova-specs-python27 http://logs.openstack.org/05/130005/1/check/gate-nova-specs-python27/4b89d3a : FAILURE in 2m 02s","accounts_in_message":[],"_revision_number":1},{"id":"51247d0c72bc280675b026ebff56ac6bfe1b1776","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2014-10-21 19:08:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4ae112e481132150a64b0e163ff9d1749b87237","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-10-21 19:26:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)\n\nI think a ton more detail needs to be in this spec before we can even consider it, knowing the dragons that lie ahead.\n\nTo be honest, I\u0027d rather we didn\u0027t go down this path at all. I don\u0027t think there is much to be gained, and the potential for further complicating our boot process (not to mention getting this implementation wrong) is high, IMHO. In a post-apocalyptic world where we have a separate scheduler, we may have an option to do better here. However, for now, my opinion vote on whether this is worth the trouble is -2.","accounts_in_message":[],"_revision_number":1},{"id":"c152cbde234f35bf3ecb4295f3ad9eb2d39258a2","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-10-21 21:55:37.000000000","message":"Patch Set 1:\n\nThank you for the feedback. I will address the review comments and have a new version out for review in a few days.","accounts_in_message":[],"_revision_number":1},{"id":"6b24d013521a0b94e36700b40d794590ab465938","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-10-22 00:32:59.000000000","message":"Patch Set 1:\n\nI echo Jennifer\u0027s \"thanks\" for the review comments.  In the next round of reviews, the various race conditions will be clearly enumerated with a proposed solution.\n\nBoth Jennifer and I will be in Paris as well, so we\u0027d love to chat with folks as well on this topic as I\u0027m sure review will still be ongoing.  ;-)","accounts_in_message":[],"_revision_number":1},{"id":"df2afeb43196e9937938c7220b2d915d40be7e45","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-10-31 19:22:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de7fae549e9aad53486a6746435e560b4da5da04","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-10-31 19:22:57.000000000","message":"Topic changed from bp/s to bp/dynamic-server-groups","accounts_in_message":[],"_revision_number":2},{"id":"45ac7c3ee5818612b5901de3e34e1ce7927fb220","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-31 19:25:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/05/130005/2/check/gate-nova-specs-docs/3d6739d/doc/build/html/ : SUCCESS in 3m 00s\n- gate-nova-specs-python27 http://logs.openstack.org/05/130005/2/check/gate-nova-specs-python27/81d873d : FAILURE in 1m 26s","accounts_in_message":[],"_revision_number":2},{"id":"478eff707a9f278a8bb3fc16eeaae7785616e43d","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-10-31 19:29:39.000000000","message":"Patch Set 1:\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c83780da1f8908229463a2623b0b1a1570921ea7","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-05 23:06:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"18e853ccc4ebfebf2034685d571943ce1011482e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-05 23:09:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/05/130005/3/check/gate-nova-specs-docs/49ed908/doc/build/html/ : SUCCESS in 2m 57s\n- gate-nova-specs-python27 http://logs.openstack.org/05/130005/3/check/gate-nova-specs-python27/79aabe2 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":3},{"id":"54a7287c04024ec1e1f98fdb483405b1ae505d20","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-05 23:11:29.000000000","message":"Patch Set 3:\n\nUploaded a new version since the table had format issues in version 2. This table is too big to be in ASCII text and still be 79 characters per line. Instead, I saved it as a PDF and linked to it in the specification.","accounts_in_message":[],"_revision_number":3},{"id":"d82c2b26053f63933fe0b3230b9d73b47b74ab07","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-11-07 01:14:40.000000000","message":"Patch Set 3:\n\nComments are welcome, but as we discussed in the summit\u0027s \"Nova Unconference\" session on 11/6 with Joe Gordon, John and Sylvain, we will split this into two blueprints... one for \"add VM to group\" and another for \"remove VM from group\", particularly since the latter case is extremely trivial since no policy checking need be done and requires minimal code changes, yet still adds value.","accounts_in_message":[],"_revision_number":3},{"id":"21a4b0c861d0f01013e9f7087a8a649a527e8485","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2014-11-07 09:49:19.000000000","message":"Patch Set 3: Code-Review-1\n\nPlease understand my -1 as just a formal blocking vote on this current version until new specs are submitted based on the unconference feedback.","accounts_in_message":[],"_revision_number":3},{"id":"bab9b4c29fc64684a3b994d7d6db983607f9d223","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-11-14 23:05:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nAttempting to highly lock down the times during which add/remove operations can succeed sounds to me like it will make this less useful, and also probably increase the amount of work we have to do during group-based placement, potentially impacting performance.\n\nI still prefer we not do this.","accounts_in_message":[],"_revision_number":3},{"id":"dfb1377b627c1836dc4dd03cfaec9eb23a08f044","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-11-14 23:20:38.000000000","message":"Patch Set 3:\n\nThanks Dan.  Wanted to address a comment:\n\n\u003e \"Attempting to highly lock down the times during which add/remove operations can succeed sounds to me like it will make this less useful\"\n\nAdding/removing servers to a group is a fairly infrequent operation relative to other operations... it\u0027s really just for flexibility since admins aren\u0027t going to always know their groups at boot time, so I don\u0027t think it\u0027s really any less useful due to this restriction (to alleviate the races).","accounts_in_message":[],"_revision_number":3},{"id":"35d5317e98f5f956413ae16490ba94a35b394cfd","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-21 17:09:40.000000000","message":"Patch Set 3:\n\nAbandoning this changest to break this up into two changesets as discussed in the \"Nova Unconference\" session on 11/6. There will be a changeset for removing a server from a server group and another for adding a server to a server group.","accounts_in_message":[],"_revision_number":3},{"id":"ca8453a656bd4e9a0d9730c3a70b137beb81fb74","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-21 17:09:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c3993dd23d6a37eb2d4bd0e59c7fac92516413ae","revisions":{"4035c57ef28ca83016d89d1b33a5223db1e62538":{"kind":"REWORK","_number":1,"created":"2014-10-21 18:55:27.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/05/130005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/130005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/1"}}},"commit":{"parents":[{"commit":"2e3f8a525418e1748877eca6bf8898a13934656f","subject":"(Re)Propose: make resource tracker use objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2e3f8a525418e1748877eca6bf8898a13934656f"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-10-21 18:50:27.000000000","tz":-300},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-10-21 18:55:11.000000000","tz":-300},"subject":"Nova specs doc for dynamic server groups","message":"Nova specs doc for dynamic server groups\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups\n\nChange-Id: I58de5b61b1db05b1392b6a5ea54e37d43e2b867e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4035c57ef28ca83016d89d1b33a5223db1e62538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4035c57ef28ca83016d89d1b33a5223db1e62538"}]},"branch":"refs/heads/master"},"76decc6e794bfc94f82529c8e43f0b78c0633b12":{"kind":"REWORK","_number":2,"created":"2014-10-31 19:22:38.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/05/130005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/130005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/2"}}},"commit":{"parents":[{"commit":"f3f289b1eef76a2571dda6976016ab309abad7e5","subject":"Nova Pagination Sorting Enhancements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f3f289b1eef76a2571dda6976016ab309abad7e5"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-10-21 18:50:27.000000000","tz":-300},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-10-31 19:21:55.000000000","tz":-300},"subject":"Nova specs doc for dynamic server groups","message":"Nova specs doc for dynamic server groups\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups\n\nChange-Id: I58de5b61b1db05b1392b6a5ea54e37d43e2b867e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/76decc6e794bfc94f82529c8e43f0b78c0633b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/76decc6e794bfc94f82529c8e43f0b78c0633b12"}]},"branch":"refs/heads/master"},"c3993dd23d6a37eb2d4bd0e59c7fac92516413ae":{"kind":"REWORK","_number":3,"created":"2014-11-05 23:06:27.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/05/130005/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/130005/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/130005/3"}}},"commit":{"parents":[{"commit":"f3f289b1eef76a2571dda6976016ab309abad7e5","subject":"Nova Pagination Sorting Enhancements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f3f289b1eef76a2571dda6976016ab309abad7e5"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-10-21 18:50:27.000000000","tz":-300},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-05 23:04:13.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups","message":"Nova specs doc for dynamic server groups\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups\n\nChange-Id: I58de5b61b1db05b1392b6a5ea54e37d43e2b867e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c3993dd23d6a37eb2d4bd0e59c7fac92516413ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c3993dd23d6a37eb2d4bd0e59c7fac92516413ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
