)]}'
{"id":"openstack%2Fnova-specs~136487","triplet_id":"openstack%2Fnova-specs~master~I1728e3b347b4ae0364785d167ac530bf4109f3bf","project":"openstack/nova-specs","branch":"master","topic":"bp/dynamic-server-groups-remove","hashtags":[],"change_id":"I1728e3b347b4ae0364785d167ac530bf4109f3bf","subject":"Nova specs doc for dynamic server groups - remove","status":"ABANDONED","created":"2014-11-21 20:58:13.000000000","updated":"2015-04-27 21:51:16.000000000","total_comment_count":41,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"88ce89c2f8bae95718acad99d324f6aa3025a695","_number":136487,"virtual_id_number":136487,"owner":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2014-12-18 21:30:21.000000000","_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":1,"date":"2014-12-17 05:12:44.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"date":"2015-01-11 08:39:19.000000000","_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"date":"2014-12-17 05:11:53.000000000","_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":-1,"date":"2014-12-17 18:22:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":-1,"date":"2015-04-14 17:49:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"date":"2014-12-18 21:40:18.000000000","_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"date":"2014-12-18 21:27:37.000000000","_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-11-21 21:08:24.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2014-11-21 21:17:43.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2014-11-23 22:46:35.000000000","updated_by":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"reviewer":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"state":"REVIEWER"},{"updated":"2014-11-24 20:41:52.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2014-12-02 05:05:07.000000000","updated_by":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"reviewer":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"state":"REVIEWER"},{"updated":"2014-12-02 21:36:19.000000000","updated_by":{"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"},"reviewer":{"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"},"state":"REVIEWER"},{"updated":"2014-12-11 20:42:38.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2014-12-12 14:21:15.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2014-12-17 04:10:16.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2014-12-17 05:12:44.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2014-12-17 18:22:50.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2014-12-18 21:27:37.000000000","updated_by":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"reviewer":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"state":"REVIEWER"},{"updated":"2014-12-18 21:30:21.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2014-12-18 21:40:18.000000000","updated_by":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"reviewer":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"state":"REVIEWER"},{"updated":"2015-01-11 08:39:19.000000000","updated_by":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"reviewer":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"state":"REVIEWER"},{"updated":"2015-04-14 17:49:05.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"}],"messages":[{"id":"a0f0f7bc759f896f5958f94b77a9fbcfb76b728d","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-21 20:58:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59f807da62299770dcfcc25a7549ce31a9dd81f3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-21 21:04:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/87/136487/1/check/gate-nova-specs-docs/16f515f/doc/build/html/ : SUCCESS in 1m 56s\n- gate-nova-specs-python27 http://logs.openstack.org/87/136487/1/check/gate-nova-specs-python27/9aad698 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1},{"id":"555268d32c033ae85a6b2fc38d6abe1d28569fda","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-21 21:09:52.000000000","message":"Patch Set 1:\n\nThis is the nova spec for the trivial new \u0027remove a server from a server group\u0027 API. This was discussed in the Nova Unconference Session by Joe Cropper from IBM on Nov 6.","accounts_in_message":[],"_revision_number":1},{"id":"92bf812a9a00c3716edec08f3f1a55532e3fa3fe","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2014-11-22 07:23:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d8915146d40402d7d52562ce20a76065f8c4d82f","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-11-22 22:50:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa8c4157b6676f20483ea995c73bd6413c459f74","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-23 00:59:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41825e801821e39c8b1b647d65c57052eba5b92a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-23 01:01:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/87/136487/2/check/gate-nova-specs-docs/f4d2348/doc/build/html/ : SUCCESS in 1m 48s\n- gate-nova-specs-python27 http://logs.openstack.org/87/136487/2/check/gate-nova-specs-python27/d48fbfd : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b8477327c11a119ac2a66f2ec1539f3f55b96c9f","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-23 01:05:40.000000000","message":"Topic changed from bp/s to bp/dynamic-server-groups-remove","accounts_in_message":[],"_revision_number":2},{"id":"57e5f5944f8782c8e2f58d39b1c6fbd70962061c","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-23 01:05:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cbbc25cc159b83d91933bde2fa2ecc621012022","author":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"date":"2014-11-23 22:46:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"28a9579d342374a4ce05134da4064d13a09269cb","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2014-11-26 17:58:41.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nto be clear, I\u0027m giving -1 because I\u0027m not sure it shouldn\u0027t be covered by Tempest","accounts_in_message":[],"_revision_number":2},{"id":"1393cf5ba4a14d5bb85371421fbbd0917e050256","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-26 18:13:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"20584578ae744210ac2a91362b13f7ad94258f08","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-26 18:13:49.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a2fb708089f3f15caed170f29f4dcd18f6d0aed7","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-11-26 18:14:19.000000000","message":"Patch Set 3:\n\nUploaded new patch to address Sylvain\u0027s comments.","accounts_in_message":[],"_revision_number":3},{"id":"c3e19383e393690cc808a0855c3690517791d8f3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-26 18:15:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/87/136487/3/check/gate-nova-specs-docs/efb4e0a/doc/build/html/ : SUCCESS in 1m 38s\n- gate-nova-specs-python27 http://logs.openstack.org/87/136487/3/check/gate-nova-specs-python27/50d7466 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":3},{"id":"be0e1c443d17cfca23d84722f44af911d1459619","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2014-11-26 19:39:33.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3688920f33a0ab83d677c8bf908b18d24dfe12c2","author":{"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"},"date":"2014-11-27 02:04:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nI have included feedback on the suggested API changes, from a discussion in the API-WG meeting we had earlier today.","accounts_in_message":[],"_revision_number":3},{"id":"28e64f76c10929effbb53c769c79919b6c2f4c82","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-02 01:13:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"82a86961292e89687d26a52aa82d89f1fabf5283","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-02 01:14:34.000000000","message":"Patch Set 4:\n\nAddressed comments from Miguel.","accounts_in_message":[],"_revision_number":4},{"id":"e3c2344818194cd159e752064122e1998e08fe3f","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-02 01:15:04.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"09af5ac39c46fb4140e92d8d1f78db827076b741","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-02 01:20:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/87/136487/4/check/gate-nova-specs-docs/5c834aa/doc/build/html/ : SUCCESS in 3m 02s\n- gate-nova-specs-python27 http://logs.openstack.org/87/136487/4/check/gate-nova-specs-python27/bace852 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":4},{"id":"0ecb52fdfc72f32b0ba665f9bd16c4a13bcafb38","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-02 01:24:04.000000000","message":"Topic changed from bp/s to bp/dynamic-server-groups-remove","accounts_in_message":[],"_revision_number":4},{"id":"472a32d443750b2de1172db92948008e7ddb5345","author":{"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"},"date":"2014-12-02 01:26:17.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nLooks much better now, thanks! I have one small improvement to add a hyperlink to the server membership resource (see inline comment).","accounts_in_message":[],"_revision_number":4},{"id":"be74fcfa470c8567b0ca927cb43d48d3bdf99948","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2014-12-02 05:05:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nComments inline, but overall looks great","accounts_in_message":[],"_revision_number":4},{"id":"d1778ed6aabfe49ff715eb7bf02d70ded330da7a","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-12-02 05:08:53.000000000","message":"Patch Set 4:\n\nThanks for the feedback thus far.\n\nQuick question... is there general consensus that tempest tests are needed for this API, or can we sufficiently do this with unit tests as the logic is fairly trivial and not really subject to races?","accounts_in_message":[],"_revision_number":4},{"id":"cb85d5ff9868a1f1368f2ce98f8bf9bea50cbe84","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-12-02 05:17:56.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4b842ffeb780387ef399f0c7279e17b898ff56de","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2014-12-02 06:28:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c1d018cbec7334eddf6419271b215374d55cdd5","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-02 20:40:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"410a693ab567fca20b12a6f5f77ff475fc80a8dd","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-02 20:40:23.000000000","message":"Topic changed from bp/s to bp/dynamic-server-groups-remove","accounts_in_message":[],"_revision_number":5},{"id":"dcf55233d2fa175cf2272c4ebfc19d5f6d2733f3","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-02 20:40:32.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f93d5900be44fdf3db17c81bbe80136e8f9a66f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-02 20:52:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/87/136487/5/check/gate-nova-specs-docs/b263247/doc/build/html/ : SUCCESS in 2m 55s\n- gate-nova-specs-python27 http://logs.openstack.org/87/136487/5/check/gate-nova-specs-python27/393a571 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":5},{"id":"3aa6ecb0c8ae653dbf28f908b7ef70633f3a04d6","author":{"_account_id":12606,"name":"Miguel Grinberg","email":"miguel.grinberg@gmail.com","username":"miguelgrinberg"},"date":"2014-12-02 21:36:19.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM on the API front. Nice that you figured out how to keep the hyperlink to the server in the representation. Thanks!","accounts_in_message":[],"_revision_number":5},{"id":"5228fb65d47186aa4c009b3ad2ac7f26c5c7d20d","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-02 22:03:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"253449559c94594ee3da2837d6abafabc4ec2ec0","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2014-12-03 08:56:21.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7b0bed8b0ef823a4d13bc85516a31e314d4b7bba","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-12-03 18:24:58.000000000","message":"Patch Set 5: Code-Review-1\n\nI have a meta question. I think this was split from the original add/remove case because add is hard. The question is: if we decide not to do add, is remove really useful on its own? Is it going to be more confusing to users if they can remove things from a group that are still running, but can\u0027t add them? Right now, it\u0027s pretty symmetric. You get into a group at birth, and leave at death.\n\nCould we at least document a use case where it makes sense to do this without add, if add never gets implemented/approved? Apologies if it has already been discussed somewhere else and I missed it.","accounts_in_message":[],"_revision_number":5},{"id":"afd9d489b2d235859e6951e48f05966b9f84a23b","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-12-03 18:37:12.000000000","message":"Patch Set 5:\n\nHey Dan, great question.  This was discussed a bit, albeit mostly verbally.  While it would be nice to have both remove and add, remove on its own can still be useful (and allows us to stage this to keep it simple).\n\nAn example use case would be:\n- Admin deploys VM-1, VM-2, VM-3 in SG-A and specifies \u0027affinity\u0027 policy for performance reasons.\n- Company buys an uber-fast 10GB switch, alleviating their initial performance concerns.\n- Admin can now remove VM-3 from SG-A and put it on another machine to help free some resource capacity (this is better than having to delete VM-3 to remove it and re-deploy/configure the VM).","accounts_in_message":[],"_revision_number":5},{"id":"e8740dc33ff38eb31735a0b0ae01d90c57431864","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-12-03 18:52:16.000000000","message":"Patch Set 5:\n\nRight, so the converse situation \"justifies\" add:\n\n- Admin deploys VM-1 and VM-2\n- Customer realizes that performance between these is bad, wants them co-located\n- Admin adds them to an affinity group and hits migrate on one so they land together\n\nI think that once you disassociate the group membership from the lifetime of a VM, then the user is going to be seriously confused about why they can remove but not add. That\u0027s why I think that the fate of remove and add should probably be tied together.","accounts_in_message":[],"_revision_number":5},{"id":"cc5a6bccd6d1898d3f116891eb5b4214ebb43a9f","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-12-03 19:06:25.000000000","message":"Patch Set 5:\n\nYep, it would be nice to have symmetry - and that\u0027s why the \u0027add\u0027 spec will be posted by Jennifer in the next few days.  That said, based on some discussions I had with jogo, johnthetubaguy and bauzas at the summit - we felt getting \u0027remove\u0027 done in Kilo was a good way to stage this, and still provide value-add over what an admin can do today.\n\nI don\u0027t think the confusion the admin will have about why they can\u0027t add will outweigh not being able to remove at all, but that\u0027s just my $0.02 after discussion with several cloud admins.  :-)","accounts_in_message":[],"_revision_number":5},{"id":"ee247de758b90f37689cb5e734420a44b2747658","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-12-04 13:11:23.000000000","message":"Patch Set 5:\n\nI agree with Dan that having remove and not add can be confusing. I also think that remove is significantly easier to implement then add. So perhaps we should think out how add would work and make sure its feasible (target post K), before approving this spec.\n\nI am not too concerned about having remove and not add for a single cycle. But if we end up with just remove, that will be confusing.","accounts_in_message":[],"_revision_number":5},{"id":"5ac03a6ceb2d4c4540d6fb7d2c09cc53fbef5272","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-12-04 13:53:29.000000000","message":"Patch Set 5:\n\nThanks, jogo.  We\u0027ll be posting \u0027add\u0027 spec likely in the next day or so.  It should be feasible, but will likely take some iterating since there is more complexity as we discussed.  We will circle back with you once that\u0027s posted.  Thanks.","accounts_in_message":[],"_revision_number":5},{"id":"d10cd51a488812b97016cb2c432f2d67a97fec62","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-12-04 16:00:16.000000000","message":"Patch Set 5:\n\nJoe, right, my point is that I think add is going to be really hard/contentious. I don\u0027t want to split these up because remove is easy and we want to schedule it for kilo, only to find out that we can\u0027t agree on an add implementation.\n\nI also think that it\u0027s entirely possible that we could settle on a plan for add and find out that the implementation is not as straightforward as we expected. Thus, I would rather see them tied together from a planning POV. The patches to add the two functions can be separate, but I would hesitate to merge remove until we have settled on an add implementation.\n\nAnyway, just MHO.","accounts_in_message":[],"_revision_number":5},{"id":"35a418f9d4992be601a1cd51088f69aeeddbe616","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-05 00:11:02.000000000","message":"Patch Set 5:\n\nI have submitted the add spec for review:\nhttps://review.openstack.org/139272","accounts_in_message":[],"_revision_number":5},{"id":"1fadd60ba78fbae4966bc48bdeb6013a5b4e7f16","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2014-12-05 08:18:14.000000000","message":"Patch Set 5:\n\nCorrect me if I\u0027m wrong, but at the moment adding an instance to a group is implicit by asking for a group UUID when booting.\n\nThe main problem that operators would face is that they can rollback their previous decisions and leave written in the marble (eh, stone ;-)) their current topologies. Here, IMHO this spec, even if simplier to implement (because of no race conditions to manage) is still very valuable.\n\nWaiting for another spec which would just extend the API for managing those sounds quite unfair, as it would delay the main feature that operators are wanting.\n\nAnyway, as grealy said Dan, that\u0027s just MHO.","accounts_in_message":[],"_revision_number":5},{"id":"56e37feaa5884c98533a7cbb9bbcfa293ed0388b","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-12-09 03:20:10.000000000","message":"Patch Set 5:\n\nAs Joe Gordon requested and previously mentioned by Jennifer, the spec for add has been posted.  While we understand add may not be accepted for Kilo (although we\u0027d welcome it :), I hope you can see sufficient thought and detail has been posted in the spec such that we can minimally proceed with this (far simpler) remove spec for Kilo to make forward progress.\n\nThanks again!","accounts_in_message":[],"_revision_number":5},{"id":"9eec5dfeadcf2b81cc3e042b3e2f6276c743d5a8","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-11 16:04:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f93a9d04c51afe0d1d4977106c53e11c0a2787aa","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-11 16:05:58.000000000","message":"Patch Set 6:\n\nUploaded new patch to link to the corresponding add spec for completeness.","accounts_in_message":[],"_revision_number":6},{"id":"262ec8e9baedd024f92a695857927e969517ce92","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-11 16:06:20.000000000","message":"Topic changed from bp/s to bp/dynamic-server-groups-add","accounts_in_message":[],"_revision_number":6},{"id":"ba678f493541c176cb0436c36924b50c5fce49ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-11 16:08:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/87/136487/6/check/gate-nova-specs-docs/179a3dd/doc/build/html/ : SUCCESS in 2m 48s\n- gate-nova-specs-python27 http://logs.openstack.org/87/136487/6/check/gate-nova-specs-python27/d617099 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":6},{"id":"74c3a3543aeb4761894fb6ea1030325f5ca4484a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2014-12-11 20:42:38.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"61a6555fa35824fe14b2cfdcabfd6f7309b9fad6","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-12-12 14:21:15.000000000","message":"Patch Set 6: Code-Review-1\n\nI think this needs to only come in on v2.1 as a micro version","accounts_in_message":[],"_revision_number":6},{"id":"c94eb1ca87591fdc1e7524032550a3bd95e577a3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-12-12 15:24:35.000000000","message":"Patch Set 6: Code-Review-1\n\nJust re-asserting my -1 that add and remove should be the same spec.","accounts_in_message":[],"_revision_number":6},{"id":"f98af577c3cc74e20c570f9a9127f0d607ac7c65","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-12 18:56:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8484f077ff06ebe6e2df1f0fa74a348cf9ef0e9b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-12-12 19:00:57.000000000","message":"Patch Set 7: Code-Review-1\n\nJust re-asserting my -1 that add and remove should be the same spec.","accounts_in_message":[],"_revision_number":7},{"id":"831d28ae83dcbec8f767cdd3d99fb128baaaed61","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-12 19:06:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/87/136487/7/check/gate-nova-specs-docs/f0d9edb/doc/build/html/ : SUCCESS in 3m 03s\n- gate-nova-specs-python27 http://logs.openstack.org/87/136487/7/check/gate-nova-specs-python27/156241a : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":7},{"id":"d729fa56442285099e4783d12e7b07113bf85708","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-12 21:41:34.000000000","message":"Topic changed from bp/s to bp/dynamic-server-groups-remove","accounts_in_message":[],"_revision_number":7},{"id":"9a879077ccc3a6ada6fa4369f01eb47c2fe8b0fb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-12-16 22:48:26.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)\n\n-1 on a few smaller issues.\n\nRegarding blocking this until an implementation is agreed on for the add function, I can see Dan\u0027s point, if we can\u0027t get add in it\u0027s going to be a bit weird to just have remove, and once the remove group member API is in we\u0027re stuck with it given the API contract.\n\nI honestly don\u0027t know all of the ins and outs of the server group stuff to speak to it to make an uber informed vote, and haven\u0027t yet gone through the add member spec.\n\njohnthetubaguy was involved in the discussion at the summit so I\u0027d like his input here.","accounts_in_message":[],"_revision_number":7},{"id":"85c2c17308997fe96106aa0a2a886d54e2dc013e","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-12-16 22:59:31.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f3e059c6efcdc05f1d8078bb03d7c4668fb846b6","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-16 23:18:20.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"2f7e7c555dfae187aba615f035808a5e6548663f","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-16 23:21:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6270f3c69053ce069f81a0958834640bc3f3dc43","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-16 23:21:34.000000000","message":"Topic changed from bp/s to bp/dynamic-server-groups-remove","accounts_in_message":[],"_revision_number":8},{"id":"ef23d3434aaf26e7623b5de601210f06f5efb1bd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-12-16 23:39:06.000000000","message":"Patch Set 8:\n\nJust re-asserting my -1 that add and remove should be the same spec.","accounts_in_message":[],"_revision_number":8},{"id":"16359bedc7cbf25e757988d50a922211e178a82f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-12-16 23:39:35.000000000","message":"Patch Set 8: Code-Review-1","accounts_in_message":[],"_revision_number":8},{"id":"22a3fb7a839c44744978c64bbbd4478b8a3945bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-17 00:06:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/87/136487/8/check//gate-nova-specs-docs/bd112c3//doc/build/html/ : SUCCESS in 1m 48s\n- gate-nova-specs-python27 http://logs.openstack.org/87/136487/8/check//gate-nova-specs-python27/86cccc2/ : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":8},{"id":"031a6dfc68ea58407e8dc55d0605510daad36997","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-12-17 00:13:38.000000000","message":"Patch Set 8:\n\nDan - https://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-add is the add spec too.\n\nHow about we approve both of these (still naturally subject to code reviews) and make them conditional on each other?  It will allow the work to be split up nicely as well as advised by jogo and johnthetubaguy.\n\nOtherwise, as I mentioned previously, I\u0027m not sure why we are so worried about an admin getting \"confused\" about why you can remove, but not add.  I don\u0027t think the confusion will be any less than an admin wondering why s/he can\u0027t do either of these anyway and having to delete their VM.  Additionally, we do have the add spec (mentioned previously) with a lot of detail about how to implement it.\n\nThanks again for the comments -- hopefully we can reach some concurrence on this.  :)","accounts_in_message":[],"_revision_number":8},{"id":"4dab463b5d2c6d709f9a66d0abdb3c9f7896f402","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-12-17 04:09:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5306bc0f8def868cb8cbc5021db3dbdf21eb1c73","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-12-17 04:10:16.000000000","message":"Patch Set 8: Code-Review-1\n\nYou missed my comment about adding the APIImpact tag to the commit message.\n\nSorry Dan, you\u0027re gonna need to re-apply that -1. :)","accounts_in_message":[],"_revision_number":8},{"id":"9647ab1471d29dca7cc9a277cd33adca5787becc","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-12-17 04:12:09.000000000","message":"Patch Set 8:\n\nBTW, I added this to the nova meeting agenda for Thursday to get some more eyes on it.","accounts_in_message":[],"_revision_number":8},{"id":"6b11397a687bed10ef96230b9637457f1a7467e0","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2014-12-17 04:54:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"82bf600569a4e4be1c4afbedf46973a0cfc496a3","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-17 05:10:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c0c0ce20d4a0abb606eab69f408f5230c0f38c18","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-17 05:11:53.000000000","message":"Patch Set 9:\n\nfixed the commit message.","accounts_in_message":[],"_revision_number":9},{"id":"d4c4c3877053ffb775346779937f89c5a8f25eba","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-17 05:11:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f740c4a6c37a526cd9e8785146c17f3f9d5cd67c","author":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"date":"2014-12-17 05:12:10.000000000","message":"Topic changed from bp/s to bp/dynamic-server-groups-remove","accounts_in_message":[],"_revision_number":9},{"id":"5f82052badc8ac96ae8c55e090a4e323662732eb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-17 05:12:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs http://docs-draft.openstack.org/87/136487/9/check//gate-nova-specs-docs/e74b5cc//doc/build/html/ : SUCCESS in 2m 33s\n- gate-nova-specs-python27 http://logs.openstack.org/87/136487/9/check//gate-nova-specs-python27/43ecf53/ : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":9},{"id":"b113c9a6fa07653335f129395e6b39a59d2dacb2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-12-17 18:22:50.000000000","message":"Patch Set 9: Code-Review-1\n\nJust re-asserting my -1 that add and remove should be the same spec.","accounts_in_message":[],"_revision_number":9},{"id":"e6438300c9d491360c330f55e1bd7437cbb5c601","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-12-18 21:28:28.000000000","message":"Patch Set 9: Code-Review-2\n\nThere isn\u0027t clear consensus on this one right now. We are at kilo feature proposal freeze. This is low enough down the priorities list that I don\u0027t really think its worth hashing out right now.  server groups themselves is a huge can of worms","accounts_in_message":[],"_revision_number":9},{"id":"2b900827f8497ee644c5f9de6f6ac337c5b5b48a","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2014-12-18 21:30:21.000000000","message":"Patch Set 9:\n\nI agree that we want add/remove approved together.  I\u0027m less concerned that they\u0027re in the same spec, but don\u0027t think we should approve one until both are ready.  Although putting them in one spec makes that easier :)","accounts_in_message":[],"_revision_number":9},{"id":"8c63eea23993e008884b4f4b63261bc23c0a32b0","author":{"_account_id":7664,"name":"Joe Cropper","email":"jwcroppe@us.ibm.com","username":"jwcroppe"},"date":"2015-01-11 08:39:19.000000000","message":"Patch Set 9:\n\nThanks for the reviews everyone.  Please check out the spec for add - https://review.openstack.org/#/c/139272/ - we\u0027d love to get feedback so we can land it for \"L\" then.  :-)\n\nThanks!","accounts_in_message":[],"_revision_number":9},{"id":"882ae5d67711a9effa1b4c927303a05b7592ef31","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2015-04-14 17:49:05.000000000","message":"Patch Set 9: Code-Review-1\n\nLiberty is open","accounts_in_message":[],"_revision_number":9},{"id":"88ce89c2f8bae95718acad99d324f6aa3025a695","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2015-04-27 21:51:16.000000000","message":"Abandoned\n\nAbandoning this because its targeted to kilo. Please reopen it if you\u0027re interested in pursuing this work in liberty.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"bc74499eba15726a398c5bd66caa57f4021adb74","revisions":{"d0a6997a600c41bf068a5461b71162d0f47a2cb9":{"kind":"REWORK","_number":1,"created":"2014-11-21 20:58:13.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/87/136487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/136487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/1"}}},"commit":{"parents":[{"commit":"a75867126f6a6e8b0c0c39f6cbdedd727ef3c620","subject":"Merge \"Validate database migrations and model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a75867126f6a6e8b0c0c39f6cbdedd727ef3c620"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:52:03.000000000","tz":-360},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:57:53.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups - remove","message":"Nova specs doc for dynamic server groups - remove\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-remove\n\nChange-Id: I1728e3b347b4ae0364785d167ac530bf4109f3bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0a6997a600c41bf068a5461b71162d0f47a2cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0a6997a600c41bf068a5461b71162d0f47a2cb9"}]},"branch":"refs/heads/master"},"53ac0ccaea7d48d9de9d0b21129760b77aa449bc":{"kind":"REWORK","_number":2,"created":"2014-11-23 00:59:12.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/87/136487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/136487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/2"}}},"commit":{"parents":[{"commit":"a75867126f6a6e8b0c0c39f6cbdedd727ef3c620","subject":"Merge \"Validate database migrations and model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a75867126f6a6e8b0c0c39f6cbdedd727ef3c620"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:52:03.000000000","tz":-360},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-23 00:58:17.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups - remove","message":"Nova specs doc for dynamic server groups - remove\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-remove\n\nChange-Id: I1728e3b347b4ae0364785d167ac530bf4109f3bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/53ac0ccaea7d48d9de9d0b21129760b77aa449bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/53ac0ccaea7d48d9de9d0b21129760b77aa449bc"}]},"branch":"refs/heads/master"},"ed14ac86abfe8a4a4a09d1aeff5f0d4c7cc4b813":{"kind":"REWORK","_number":3,"created":"2014-11-26 18:13:24.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/87/136487/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/136487/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/3"}}},"commit":{"parents":[{"commit":"59b883548e415293569ccc92ffd5ceefe7a07d89","subject":"Merge \"Support iSCSI live migration for different iSCSI target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/59b883548e415293569ccc92ffd5ceefe7a07d89"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:52:03.000000000","tz":-360},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-26 18:13:15.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups - remove","message":"Nova specs doc for dynamic server groups - remove\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-remove\n\nChange-Id: I1728e3b347b4ae0364785d167ac530bf4109f3bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed14ac86abfe8a4a4a09d1aeff5f0d4c7cc4b813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed14ac86abfe8a4a4a09d1aeff5f0d4c7cc4b813"}]},"branch":"refs/heads/master"},"9496517d92fcc8e4ca04c35aa17adfe81afbbdc3":{"kind":"REWORK","_number":4,"created":"2014-12-02 01:13:55.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/87/136487/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/136487/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/4"}}},"commit":{"parents":[{"commit":"d013356b5dfd83af8401e825b6f93a073b741d85","subject":"Merge \"Fill out kilo priorities as discussed at the summit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d013356b5dfd83af8401e825b6f93a073b741d85"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:52:03.000000000","tz":-360},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-12-02 01:12:54.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups - remove","message":"Nova specs doc for dynamic server groups - remove\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-remove\n\nChange-Id: I1728e3b347b4ae0364785d167ac530bf4109f3bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9496517d92fcc8e4ca04c35aa17adfe81afbbdc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9496517d92fcc8e4ca04c35aa17adfe81afbbdc3"}]},"branch":"refs/heads/master"},"d8dccf684cbde6f98b2468e5300587852670fff2":{"kind":"REWORK","_number":5,"created":"2014-12-02 20:40:00.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/87/136487/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/136487/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/5"}}},"commit":{"parents":[{"commit":"82ecbaf7e31f3bb52108a827a4a7f8d1b6e93748","subject":"Merge \"Use configdrive with Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/82ecbaf7e31f3bb52108a827a4a7f8d1b6e93748"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:52:03.000000000","tz":-360},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-12-02 20:39:32.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups - remove","message":"Nova specs doc for dynamic server groups - remove\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-remove\n\nChange-Id: I1728e3b347b4ae0364785d167ac530bf4109f3bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d8dccf684cbde6f98b2468e5300587852670fff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d8dccf684cbde6f98b2468e5300587852670fff2"}]},"branch":"refs/heads/master"},"26b0d3adc69e0bac65a36ec929e6b1df01157b70":{"kind":"REWORK","_number":6,"created":"2014-12-11 16:04:32.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/87/136487/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/136487/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/6"}}},"commit":{"parents":[{"commit":"c7c16f33a5a4d6be750b7f5c8229b86717b71145","subject":"cleanup: api-microversions spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c7c16f33a5a4d6be750b7f5c8229b86717b71145"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:52:03.000000000","tz":-360},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-12-11 16:04:18.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups - remove","message":"Nova specs doc for dynamic server groups - remove\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-remove\n\nChange-Id: I1728e3b347b4ae0364785d167ac530bf4109f3bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/26b0d3adc69e0bac65a36ec929e6b1df01157b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/26b0d3adc69e0bac65a36ec929e6b1df01157b70"}]},"branch":"refs/heads/master"},"512bf71cc1d0332a560010fe5ce50f898ba3adc2":{"kind":"REWORK","_number":7,"created":"2014-12-12 18:56:07.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/87/136487/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/136487/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/7"}}},"commit":{"parents":[{"commit":"788d82436134c832a391cddbe5869c77d5d86334","subject":"Merge \"Cells v2 mapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/788d82436134c832a391cddbe5869c77d5d86334"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:52:03.000000000","tz":-360},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-12-12 18:55:55.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups - remove","message":"Nova specs doc for dynamic server groups - remove\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-remove\n\nChange-Id: I1728e3b347b4ae0364785d167ac530bf4109f3bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/512bf71cc1d0332a560010fe5ce50f898ba3adc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/512bf71cc1d0332a560010fe5ce50f898ba3adc2"}]},"branch":"refs/heads/master"},"4f1cb699fb7e0c220896f56d608065de67a6cef4":{"kind":"REWORK","_number":8,"created":"2014-12-16 23:21:10.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/87/136487/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/136487/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/8"}}},"commit":{"parents":[{"commit":"074106210ec1bee9c6be08d748c16c81f31be361","subject":"Adds lock-free-quota-management spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/074106210ec1bee9c6be08d748c16c81f31be361"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:52:03.000000000","tz":-360},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-12-16 23:21:14.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups - remove","message":"Nova specs doc for dynamic server groups - remove\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-remove\n\nChange-Id: I1728e3b347b4ae0364785d167ac530bf4109f3bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f1cb699fb7e0c220896f56d608065de67a6cef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f1cb699fb7e0c220896f56d608065de67a6cef4"}]},"branch":"refs/heads/master"},"bc74499eba15726a398c5bd66caa57f4021adb74":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2014-12-17 05:10:07.000000000","uploader":{"_account_id":7221,"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","username":"jmulsow"},"ref":"refs/changes/87/136487/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/87/136487/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/87/136487/9"}}},"commit":{"parents":[{"commit":"5c6b1afb03d68bbd71e7ea774b4bf4883357c673","subject":"Merge \"Consolidate the APIs for getting consoles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5c6b1afb03d68bbd71e7ea774b4bf4883357c673"}]}],"author":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-11-21 20:52:03.000000000","tz":-360},"committer":{"name":"Jennifer Mulsow","email":"jmulsow@us.ibm.com","date":"2014-12-17 05:04:24.000000000","tz":-360},"subject":"Nova specs doc for dynamic server groups - remove","message":"Nova specs doc for dynamic server groups - remove\n\nThis is the documentation for\nhttps://blueprints.launchpad.net/nova/+spec/dynamic-server-groups-remove\n\nAPIImpact\nChange-Id: I1728e3b347b4ae0364785d167ac530bf4109f3bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bc74499eba15726a398c5bd66caa57f4021adb74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bc74499eba15726a398c5bd66caa57f4021adb74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
