)]}'
{"id":"openstack%2Fnova-specs~390116","triplet_id":"openstack%2Fnova-specs~master~I5cfc88b89f87b07dbceca031718c36c5ba0bada2","project":"openstack/nova-specs","branch":"master","topic":"sched-error-reporting","hashtags":[],"change_id":"I5cfc88b89f87b07dbceca031718c36c5ba0bada2","subject":"Improve Scheduler Logging","status":"ABANDONED","created":"2016-10-23 17:39:58.000000000","updated":"2017-04-04 10:13:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f6e4b5b643b91663ed828e5b612304e1467286a5","_number":390116,"virtual_id_number":390116,"owner":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2016-10-23 17:42:18.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-11-10 14:54:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"value":1,"date":"2016-11-01 12:17:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-10-23 17:42:18.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-11-01 12:17:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2016-11-10 14:54:47.000000000","updated_by":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"reviewer":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"state":"REVIEWER"}],"messages":[{"id":"9c32d640a2fc7e023688dd93280c10006a6820e3","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2016-10-23 17:39:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"276fb0a4f9c5a8db259de37f4fb8749d0a3c01ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-23 17:42:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/16/390116/1/check/gate-nova-specs-docs-ubuntu-xenial/b0fbace//doc/build/html/ : SUCCESS in 2m 08s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/16/390116/1/check/gate-nova-specs-pep8-ubuntu-xenial/775eb66/ : SUCCESS in 1m 29s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/16/390116/1/check/gate-nova-specs-python27-ubuntu-xenial/b299812/ : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0477d9cb027849b797a5b53863ae005ff41e19c7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-11-01 12:17:06.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks like a trivial reapproval as the idea is still valid. I wonder if the \u00272\u0027 should be dropped from the topic, though?","accounts_in_message":[],"_revision_number":1},{"id":"4d38e77e3f6a9deb0256a9c134c39736e27237cb","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2016-11-01 15:32:46.000000000","message":"Topic changed from sched-error-reporting2 to sched-error-reporting","accounts_in_message":[],"_revision_number":1},{"id":"a8cee15e1d789876052669f16587d8e674bed7c7","author":{"_account_id":6772,"name":"Stephen Gordon","email":"sgordon@redhat.com","username":"sgordon"},"date":"2016-11-10 14:54:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f6e4b5b643b91663ed828e5b612304e1467286a5","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2017-04-04 10:13:44.000000000","message":"Abandoned\n\nLets restore and move this to pike if there are things we still need.\n\nI thought we got a lot of this merged already without this spec, but I may have lost track.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72cb93d2aad09c5e47fd463961089a5d54501778","revisions":{"72cb93d2aad09c5e47fd463961089a5d54501778":{"kind":"REWORK","_number":1,"created":"2016-10-23 17:39:58.000000000","uploader":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"ref":"refs/changes/16/390116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/16/390116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/16/390116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/16/390116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/16/390116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/16/390116/1"}}},"commit":{"parents":[{"commit":"462ba34be936c077a4dcc295fe84d81edffcce3c","subject":"Merge \"Detach and attach boot volumes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/462ba34be936c077a4dcc295fe84d81edffcce3c"}]}],"author":{"name":"Chris Friesen","email":"chris.friesen@windriver.com","date":"2016-10-23 17:37:42.000000000","tz":-360},"committer":{"name":"Chris Friesen","email":"chris.friesen@windriver.com","date":"2016-10-23 17:37:42.000000000","tz":-360},"subject":"Improve Scheduler Logging","message":"Improve Scheduler Logging\n\nThe nova scheduler includes a number of very complicated filters with\nnon-obvious failure modes (the NUMATopologyFilter in particular comes\nto mind). It is possible to have a situation where a given instances\nfails to schedule, and it is not immediately apparent as to what\nexactly caused the failure.\n\nAccordingly, it is proposed that we allow for optional detailed\nmessages about precisely *why* a scheduler filter is failing.\n\n(This was originally approved for Newton, but the implementation\nwas not merged in time.)\n\nChange-Id: I5cfc88b89f87b07dbceca031718c36c5ba0bada2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/72cb93d2aad09c5e47fd463961089a5d54501778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/72cb93d2aad09c5e47fd463961089a5d54501778"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
