)]}'
{"id":"openstack%2Fnova-specs~423872","triplet_id":"openstack%2Fnova-specs~master~Ia9d18ea7bba36424b145a397c37bf750f914c647","project":"openstack/nova-specs","branch":"master","topic":"bp/placement-notifications","hashtags":[],"change_id":"Ia9d18ea7bba36424b145a397c37bf750f914c647","subject":"Add placement nofications using versionedObjects","status":"ABANDONED","created":"2017-01-23 00:51:24.000000000","updated":"2019-02-23 10:55:01.000000000","total_comment_count":27,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a888d7e9bf0c210d0eac1012201704e0aa3b076d","_number":423872,"virtual_id_number":423872,"owner":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2017-01-23 02:23:07.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2017-01-30 13:45:07.000000000","_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"value":-1,"date":"2017-01-30 16:04:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":-1,"date":"2017-01-23 15:39:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"date":"2017-01-23 13:17:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-23 02:23:07.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-01-23 13:17:29.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2017-01-30 16:04:28.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"}],"messages":[{"id":"2d8e9a13606557aad86318ca9d46b2c617167a7e","author":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"date":"2017-01-23 00:51:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bf4fe0e4d39b99ae564f5fa8cf0f0fc995121af","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-23 00:53:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/72/423872/1/check/gate-nova-specs-docs-ubuntu-xenial/dff77ab//doc/build/html/ : SUCCESS in 2m 12s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/72/423872/1/check/gate-nova-specs-pep8-ubuntu-xenial/cf9dc3f/ : SUCCESS in 1m 33s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/72/423872/1/check/gate-nova-specs-python27-ubuntu-xenial/ba76692/ : FAILURE in 2m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2875a607772c5b46f0d6dd92c7d52e8e29106622","author":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"date":"2017-01-23 02:13:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6629c36203c1b21074252d531891cc599e7ea18d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-23 02:16:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/72/423872/2/check/gate-nova-specs-docs-ubuntu-xenial/34d417b//doc/build/html/ : SUCCESS in 2m 19s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/72/423872/2/check/gate-nova-specs-pep8-ubuntu-xenial/c21b429/ : SUCCESS in 1m 48s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/72/423872/2/check/gate-nova-specs-python27-ubuntu-xenial/ac01ec4/ : FAILURE in 1m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6ddf82da0156df57cae67edc10a2cd8645c8bc73","author":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"date":"2017-01-23 02:19:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"81e159c8922b15f78c9b37e81d1b98e03cb8d0ab","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-23 02:23:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/72/423872/3/check/gate-nova-specs-docs-ubuntu-xenial/c253ee1//doc/build/html/ : SUCCESS in 2m 15s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/72/423872/3/check/gate-nova-specs-pep8-ubuntu-xenial/f19a4c0/ : SUCCESS in 3m 18s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/72/423872/3/check/gate-nova-specs-python27-ubuntu-xenial/1d59da8/ : FAILURE in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"e5641e723fb19c2f6f59cbd0c20c4beb44e5d948","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-01-23 15:39:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(11 comments)\n\nI think we needs some details here and there but it is a good start.\nSee inline.\nCheers,\ngibi","accounts_in_message":[],"_revision_number":3},{"id":"f1403b7310ee46d368b3d008993a7ced8217f3c5","author":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"date":"2017-01-30 13:45:07.000000000","message":"Patch Set 3:\n\n(11 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6dac4a875ef06f8ba5560426d740724e681d07f1","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-01-30 16:04:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)\n\nI agree this could do with more detail on the form of the notifications being made, as they will likely be quite different from action to action.\n\nAlso, some consideration should be made about how much information should be provided about a resource provider when an inventory or aggregate is updated.","accounts_in_message":[],"_revision_number":3},{"id":"bf927cfe1b0efd4b92ee51c9aec71888e154e900","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2017-04-04 11:09:51.000000000","message":"Abandoned\n\nSeems like this has been left for many weeks with an unaddressed -ve comments.\n\nPlease restore if you want to keep going with this one.","accounts_in_message":[],"_revision_number":3},{"id":"99ea85ee46b7c84bd9ec1102f0689a5b36f66706","author":{"_account_id":29983,"name":"Sudhakar Reddy","email":"sudredram@yahoo.com"},"date":"2019-02-23 05:10:31.000000000","message":"Patch Set 3:\n\nAny progress on this currently we have corner cases where placement does not delete the allocation in time and nova already deleted the instance ID and Ironic made node available after clean up.When new reservation comes for same BM node nova accepts the reservation and placement fails it.","accounts_in_message":[],"_revision_number":3},{"id":"22abfad9f00a6350eb7e74967c31216ec91b52b9","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-02-23 10:55:01.000000000","message":"Patch Set 3:\n\n\u003e Any progress on this currently we have corner cases where placement\n \u003e does not delete the allocation in time and nova already deleted the\n \u003e instance ID and Ironic made node available after clean up.When new\n \u003e reservation comes for same BM node nova accepts the reservation and\n \u003e placement fails it.\n\nThere are no plans at this time to implement notifications in placement in the near future. How do you envision notifications helping in your scenario?\n\nIt might make sense to make a posting to the openstack-discuss list or to create a bug if you have a minimal test case that demonstrates the issue you\u0027re seeing.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6376762cf2810e323b62e74cb7f8585e0a5e047d","revisions":{"b01cd984dfde944e5a71dade1484875fd85b6084":{"kind":"REWORK","_number":1,"created":"2017-01-23 00:51:24.000000000","uploader":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"ref":"refs/changes/72/423872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/72/423872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/1"}}},"commit":{"parents":[{"commit":"45a7f0df14e6f63694a76d9fcfb91ad5469519e6","subject":"Merge \"Update the generic resource pools spec to reflect reality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45a7f0df14e6f63694a76d9fcfb91ad5469519e6"}]}],"author":{"name":"digambar15","email":"digambarpatil15@yahoo.co.in","date":"2017-01-22 04:27:09.000000000","tz":330},"committer":{"name":"digambar15","email":"digambarpatil15@yahoo.co.in","date":"2017-01-22 04:29:22.000000000","tz":330},"subject":"Add placement nofications using versionedObjects.","message":"Add placement nofications using versionedObjects.\n\nFollowing notification will be added -\n\n* placement.resource_provider.create\n* placement.resource_provider.delete\n* placement.resource_provider.aggregate.associate\n* placement.resource_provider.aggregate.disassociate\n* placement.resource_provider.inventory.update\n* placement.allocation.create\n* placement.allocation.delete\n\nChange-Id: Ia9d18ea7bba36424b145a397c37bf750f914c647\nImplements: blueprint placement-notifications\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b01cd984dfde944e5a71dade1484875fd85b6084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b01cd984dfde944e5a71dade1484875fd85b6084"}]},"branch":"refs/heads/master"},"9d253a1d6a22d5c8ca51160cde4f0e640433b1a8":{"kind":"REWORK","_number":2,"created":"2017-01-23 02:13:30.000000000","uploader":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"ref":"refs/changes/72/423872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/72/423872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/2"}}},"commit":{"parents":[{"commit":"45a7f0df14e6f63694a76d9fcfb91ad5469519e6","subject":"Merge \"Update the generic resource pools spec to reflect reality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45a7f0df14e6f63694a76d9fcfb91ad5469519e6"}]}],"author":{"name":"digambar15","email":"digambarpatil15@yahoo.co.in","date":"2017-01-22 05:50:42.000000000","tz":330},"committer":{"name":"digambar15","email":"digambarpatil15@yahoo.co.in","date":"2017-01-22 05:50:42.000000000","tz":330},"subject":"Add placement nofications using versionedObjects","message":"Add placement nofications using versionedObjects\n\nFollowing notification will be added -\n\n* placement.resource_provider.create\n* placement.resource_provider.delete\n* placement.resource_provider.aggregate.associate\n* placement.resource_provider.aggregate.disassociate\n* placement.resource_provider.inventory.update\n* placement.allocation.create\n* placement.allocation.delete\n\nChange-Id: Ia9d18ea7bba36424b145a397c37bf750f914c647\nImplements: blueprint placement-notifications\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d253a1d6a22d5c8ca51160cde4f0e640433b1a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9d253a1d6a22d5c8ca51160cde4f0e640433b1a8"}]},"branch":"refs/heads/master"},"6376762cf2810e323b62e74cb7f8585e0a5e047d":{"kind":"REWORK","_number":3,"created":"2017-01-23 02:19:37.000000000","uploader":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"ref":"refs/changes/72/423872/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/72/423872/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/72/423872/3"}}},"commit":{"parents":[{"commit":"45a7f0df14e6f63694a76d9fcfb91ad5469519e6","subject":"Merge \"Update the generic resource pools spec to reflect reality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45a7f0df14e6f63694a76d9fcfb91ad5469519e6"}]}],"author":{"name":"digambar15","email":"digambarpatil15@yahoo.co.in","date":"2017-01-22 05:57:12.000000000","tz":330},"committer":{"name":"digambar15","email":"digambarpatil15@yahoo.co.in","date":"2017-01-22 05:57:12.000000000","tz":330},"subject":"Add placement nofications using versionedObjects","message":"Add placement nofications using versionedObjects\n\nFollowing notification will be added -\n\n* placement.resource_provider.create\n* placement.resource_provider.delete\n* placement.resource_provider.aggregate.associate\n* placement.resource_provider.aggregate.disassociate\n* placement.resource_provider.inventory.update\n* placement.allocation.create\n* placement.allocation.delete\n\nChange-Id: Ia9d18ea7bba36424b145a397c37bf750f914c647\nImplements: blueprint placement-notifications\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6376762cf2810e323b62e74cb7f8585e0a5e047d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6376762cf2810e323b62e74cb7f8585e0a5e047d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
