)]}'
{"id":"openstack%2Fnova-specs~502306","triplet_id":"openstack%2Fnova-specs~master~Ie7be551f4f03957ade9beb64457736f400560486","project":"openstack/nova-specs","branch":"master","topic":"bp/bandwidth-resource-provider","hashtags":[],"change_id":"Ie7be551f4f03957ade9beb64457736f400560486","subject":"Network bandwidth resource provider","status":"MERGED","created":"2017-09-10 09:20:16.000000000","updated":"2018-05-08 17:30:33.000000000","submitted":"2018-05-08 17:30:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":839,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"502306-1525800633682-3ebd00cc","meta_rev_id":"c0d79c74a5b84e5616aaed5054786042544afb8c","_number":502306,"virtual_id_number":502306,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":2,"date":"2018-05-08 17:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1689,"name":"Robert Kukura","email":"rk@theep.net","username":"rkukura"},{"value":0,"_account_id":7776,"name":"Paul Carver","email":"pcarver@att.com","username":"pcarver"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2018-05-07 17:41:49.000000000","_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"date":"2018-05-08 16:11:36.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2018-05-08 00:07:03.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":1,"date":"2018-05-08 07:27:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":1,"date":"2018-05-08 07:10:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2018-05-08 17:19:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1689,"name":"Robert Kukura","email":"rk@theep.net","username":"rkukura"},{"value":0,"_account_id":7776,"name":"Paul Carver","email":"pcarver@att.com","username":"pcarver"},{"value":1,"date":"2018-05-08 16:01:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":1,"date":"2018-05-07 15:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2018-05-08 17:19:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1689,"name":"Robert Kukura","email":"rk@theep.net","username":"rkukura"},{"value":0,"_account_id":7776,"name":"Paul Carver","email":"pcarver@att.com","username":"pcarver"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1689,"name":"Robert Kukura","email":"rk@theep.net","username":"rkukura"},{"value":0,"_account_id":7776,"name":"Paul Carver","email":"pcarver@att.com","username":"pcarver"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":1689,"name":"Robert Kukura","email":"rk@theep.net","username":"rkukura"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":7776,"name":"Paul Carver","email":"pcarver@att.com","username":"pcarver"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-12 20:00:23.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2017-09-13 00:56:42.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2017-09-29 06:48:14.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2017-10-14 17:18:24.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2017-10-15 19:05:47.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-10-19 14:28:47.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2017-10-22 14:00:30.000000000","updated_by":{"_account_id":1689,"name":"Robert Kukura","email":"rk@theep.net","username":"rkukura"},"reviewer":{"_account_id":1689,"name":"Robert Kukura","email":"rk@theep.net","username":"rkukura"},"state":"REVIEWER"},{"updated":"2018-03-06 09:58:46.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2018-03-07 03:17:09.000000000","updated_by":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"reviewer":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"state":"REVIEWER"},{"updated":"2018-03-18 07:44:50.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2018-03-21 23:56:13.000000000","updated_by":{"_account_id":7776,"name":"Paul Carver","email":"pcarver@att.com","username":"pcarver"},"reviewer":{"_account_id":7776,"name":"Paul Carver","email":"pcarver@att.com","username":"pcarver"},"state":"REVIEWER"},{"updated":"2018-04-04 01:19:24.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2018-04-04 02:43:18.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2018-04-25 14:10:10.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2018-04-27 07:53:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2018-05-03 22:19:17.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2018-05-07 15:55:17.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2018-05-07 17:41:49.000000000","updated_by":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"reviewer":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"state":"REVIEWER"},{"updated":"2018-05-08 00:07:03.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2018-05-08 07:10:20.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2018-05-08 07:27:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2018-05-08 16:01:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2018-05-08 16:11:36.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2018-05-08 17:19:37.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-05-08 17:30:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba2c3a05c391bb3a43a44bdbb9760a6e3b025c82","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-09-10 09:20:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59dffe017795eba80535a45c7f1787a9ae1ff674","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-10 09:25:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-specs-docs-ubuntu-xenial http://logs.openstack.org/06/502306/1/check/gate-nova-specs-docs-ubuntu-xenial/dd771e0/ : FAILURE in 3m 36s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/1/check/gate-nova-specs-pep8-ubuntu-xenial/2de1e7a/ : SUCCESS in 4m 52s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/1/check/gate-nova-specs-python27-ubuntu-xenial/c4ca9cf/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d23f92043b1ef0e54da129db23fd16c6163538c6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-09-18 09:24:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44205ae7ecd86bdf56e6b9af3a0b5ff1bc2ebf64","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-18 09:29:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/06/502306/2/check/gate-nova-specs-docs-ubuntu-xenial/ffe1707//doc/build/html/ : SUCCESS in 4m 33s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/2/check/gate-nova-specs-pep8-ubuntu-xenial/2af1675/ : SUCCESS in 3m 42s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/2/check/gate-nova-specs-python27-ubuntu-xenial/95435f5/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7be677df2da78810b7a7c4f1f0e70b3b32ba09e6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2017-09-19 21:17:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1c5f3d984457ce56b14d27ab5c1dcebcea400aa5","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2017-10-02 00:14:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(12 comments)\n\nPlease see in-line comments?","accounts_in_message":[],"_revision_number":2},{"id":"17df6a0d9c56c017a86bc9e8594408c9a0d7c30d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-10-02 20:25:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f2217a006d3e710583e7357ba4f4c414524a1638","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-10-03 11:43:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"647429a045b43df88d1e3b94a2712cd6702ec0ec","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-03 14:40:09.000000000","message":"Patch Set 2:\n\n(32 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b8d2700e48cbae226207d8aab4ffcdd3786aebd0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-03 14:40:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b0082227b2045d80eae26ffe8ed2d2b2d04a07dd","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-10-03 14:57:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nConceptually this seems okay, but there\u0027s some confusion on how resource provider, inventory, and resource class interact that may be important to clear up.\n\nI\u0027m not certain that my concerns are correct, or maybe instead I\u0027m missing a detail about the implementation that requires the strange (to me) setup.","accounts_in_message":[],"_revision_number":3},{"id":"b825ca77f57e9fd6a2a7488bc99c5799ae58ae5e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-03 15:11:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)\n\nI agree with cdent.  The model should probably be more like:\n\n Compute host RP\n      |\n      +-\u003e Network Resource RP {\n      |       traits: [CUSTOM_PHYSNET_1]\n      |       inventory: {\n      |           NET_BANDWIDTH_INGRESS_BITS_SEC \u003d 1000000000,\n      |           NET_BANDWIDTH_EGRESS_BITS_SEC \u003d 1000000000\n      |       }\n      |   }\n      |\n      +-\u003e Network Resource RP {\n              traits: [CUSTOM_PHYSNET_2]\n              inventory: {\n                  NET_BANDWIDTH_INGRESS_BITS_SEC \u003d 1000000000,\n                  NET_BANDWIDTH_EGRESS_BITS_SEC \u003d 1000000000\n              }\n          }","accounts_in_message":[],"_revision_number":3},{"id":"c549156107c6a95145384ed764638c7c17c585c5","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-10-03 15:34:32.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"29738662fea46628362837149a4b05100d11a067","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-10-03 15:49:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(15 comments)\n\nGetting close, Rodolfo! :)","accounts_in_message":[],"_revision_number":3},{"id":"ef3436e3330b49f89f159f6a27e2ff1c1a6c91ea","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-10-03 16:13:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"19d6393e231951f509eff695313c55fbf90abfd9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-03 17:02:41.000000000","message":"Patch Set 3: Workflow-1\n\n(12 comments)\n\nWIP: I didn\u0027t finish the review. I still need to address all the comments.","accounts_in_message":[],"_revision_number":3},{"id":"34e35bf29c580b71ff05fa6e2957ee2054adae45","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-04 17:16:18.000000000","message":"Patch Set 3: -Workflow\n\n(12 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9551d8b64d767dd55a7ac840014c8955abc8d399","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-04 17:18:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2cdd8e65981b1e2229542a79eecdae64f66bcb8d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-04 19:03:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/06/502306/4/check/gate-nova-specs-docs-ubuntu-xenial/05b1802//doc/build/html/ : SUCCESS in 4m 21s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/4/check/gate-nova-specs-pep8-ubuntu-xenial/debf228/ : SUCCESS in 2m 52s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/4/check/gate-nova-specs-python27-ubuntu-xenial/59637ca/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"f8c5ceaebe9f5361128fe32e464f382f08b536cc","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-10-04 20:29:45.000000000","message":"Patch Set 4: Code-Review-1\n\n(9 comments)\n\nFew more things, Rodolfo, but we\u0027re making good progress here.\n\n-jay","accounts_in_message":[],"_revision_number":4},{"id":"781c7b406d5fba20635e600d0edd18ac78f303d6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-04 21:22:21.000000000","message":"Patch Set 4: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":4},{"id":"adaa4ab9d36052dc16125bc480385d6a0a9a0a4f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-05 15:32:19.000000000","message":"Patch Set 4:\n\n(18 comments)\n\nI think I\u0027m addressing all comments made.\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"2701da913ed61f5e89c3ac95642275856c001e83","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-05 15:33:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"26f72771a53cafce9eadc880d2ad20dec4ffb6d9","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-10-05 15:35:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dcd7802780bb36fa1badcf14e8572c46c8cd74af","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-05 15:40:27.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"740880d7cfe00dcdaa70f2df2776f1c4cec3feae","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-05 16:32:42.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)\n\nContinued discussion in PS4","accounts_in_message":[],"_revision_number":6},{"id":"fbb9d8a7f0d4bbaf638931443a458897f07635ec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-05 17:56:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/06/502306/6/check/gate-nova-specs-docs-ubuntu-xenial/2d53596//doc/build/html/ : SUCCESS in 4m 37s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/6/check/gate-nova-specs-pep8-ubuntu-xenial/c9be267/ : SUCCESS in 2m 42s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/6/check/gate-nova-specs-python27-ubuntu-xenial/9201445/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":6},{"id":"778ad28e30d73fdfe4dff6a72857d2e791c02e33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-06 05:13:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/6/check/build-openstack-sphinx-docs/60af208/html/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/6/check/openstack-tox-pep8/d546c88/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":6},{"id":"28b3922e8f2884dd9a820519d4725f3ff22ecab7","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2017-10-07 00:08:44.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nA small typo in-line. Please also note that you have unaddressed comments from Eric Fried (left in PS4) that pertain to PS6","accounts_in_message":[],"_revision_number":6},{"id":"3a0e51ea31db58a52b330dfaee0a0bed7086ec29","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-10 08:05:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"487dfdf410b4bfdb73601a4934121a461597c820","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-10 08:17:39.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"121dad1a8824314e63db9b1dd6d9d3d9a030fed7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-10 09:05:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"45d5c366a21697965c736d0dfeda9f202239e6a7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-10 09:40:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/06/502306/7/check/gate-nova-specs-docs-ubuntu-xenial/dbe404e//doc/build/html/ : SUCCESS in 4m 28s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/7/check/gate-nova-specs-pep8-ubuntu-xenial/a168059/ : SUCCESS in 2m 48s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/7/check/gate-nova-specs-python27-ubuntu-xenial/5c0ca0c/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":7},{"id":"7d20d80605a5bfb081e5834f140c77ecd1a078f1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2017-10-10 11:44:33.000000000","message":"Patch Set 4:\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a93d47baaf822c514748ff812f2e4bc5f86cfd33","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-11 16:36:43.000000000","message":"Patch Set 4:\n\n(9 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6718251c36cf02d72663d52fd90fddb7d6c1dab7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-11 16:38:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f427e25c5f4fb5d55919f0c19c8a65494e00c866","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-11 16:41:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7685d5fbb3efe498273ee62d5fdaf51b918e4796","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-11 17:35:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4652572d3dc3c98ef37fced5c3bbf8c96a372b52","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-11 19:39:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-specs-docs-ubuntu-xenial http://logs.openstack.org/06/502306/9/check/gate-nova-specs-docs-ubuntu-xenial/0624865/ : FAILURE in 56s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/9/check/gate-nova-specs-pep8-ubuntu-xenial/270b479/ : FAILURE in 1m 57s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/9/check/gate-nova-specs-python27-ubuntu-xenial/753856e/ : FAILURE in 1m 23s","accounts_in_message":[],"_revision_number":9},{"id":"4467a02c24efef84b13badf3e8d4b0475f16cb53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-11 20:58:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/9/check/build-openstack-sphinx-docs/ebcfb19/html/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 finger://ze07.openstack.org/b5b2ccd649ce4977b12f2dc39378e57c : RETRY_LIMIT in 1m 35s","accounts_in_message":[],"_revision_number":9},{"id":"74b86010344312ffe92e0975229b032a1fa11a1b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-12 11:11:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"20283972b5e04be079cd6974645bf29f7af0c475","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-12 11:14:19.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"d7935c82543cea7728d7d1f6bab7b0ae263a930c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-12 11:15:33.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"63a5e6e7964352b3aee41ec546b2c853ac707891","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-12 11:18:45.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-specs-docs-ubuntu-xenial http://logs.openstack.org/06/502306/10/check/gate-nova-specs-docs-ubuntu-xenial/e9f94d9/ : FAILURE in 1m 56s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/10/check/gate-nova-specs-pep8-ubuntu-xenial/38f7b7e/ : FAILURE in 1m 22s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/10/check/gate-nova-specs-python27-ubuntu-xenial/94e2445/ : FAILURE in 1m 11s","accounts_in_message":[],"_revision_number":10},{"id":"621d0a88ddb504c07edeb0c5dfed7b9177443ebe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-12 12:29:52.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/10/check/build-openstack-sphinx-docs/d09b5d4/html/ : SUCCESS in 5m 23s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/10/check/openstack-tox-pep8/19983e8/ : RETRY_LIMIT in 1m 37s","accounts_in_message":[],"_revision_number":10},{"id":"8f828b9ad2d61986600218fa1be034f7f501f2e1","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-12 12:31:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c7ff2995c09f2f0cdd8d0141ad2857d25b76d632","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-12 12:35:24.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"24a5208f6ae23e1010e3960cc8f7fc9f8620c816","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-12 12:39:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/06/502306/10/check/gate-nova-specs-docs-ubuntu-xenial/5ea1055//doc/build/html/ : SUCCESS in 3m 35s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/10/check/gate-nova-specs-pep8-ubuntu-xenial/da70241/ : SUCCESS in 2m 23s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/10/check/gate-nova-specs-python27-ubuntu-xenial/3c3ba11/ : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":10},{"id":"29d360d47d29bea06d1f2a5da1b5e37b1f3c2470","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-12 13:07:49.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/10/check/build-openstack-sphinx-docs/f4cdd8b/ : RETRY_LIMIT in 1m 50s\n- openstack-tox-pep8 finger://ze01.openstack.org/03e7ff89a750494fb8f8d21701c99eb7 : RETRY_LIMIT in 1m 35s","accounts_in_message":[],"_revision_number":10},{"id":"6b32cd1a30fd9dea277e3a2ab7179a5ef5b3c4b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-13 08:37:58.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"2fa932f09eec9c6f75e01ab4a7e8cfe5eaac63cb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-13 09:21:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/06/502306/10/check/gate-nova-specs-docs-ubuntu-xenial/b93b105//doc/build/html/ : SUCCESS in 4m 17s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/10/check/gate-nova-specs-pep8-ubuntu-xenial/4f5aed9/ : SUCCESS in 3m 34s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/10/check/gate-nova-specs-python27-ubuntu-xenial/0612d9a/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":10},{"id":"2c5a04998d8a2474286fd4aa23cd53a0a761bcd0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-15 19:01:50.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"768fc02ee7670e127d671fd68a3d2a218eabc029","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-15 19:05:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/06/502306/11/check/gate-nova-specs-docs-ubuntu-xenial/443a512//doc/build/html/ : SUCCESS in 3m 45s\n- gate-nova-specs-pep8-ubuntu-xenial http://logs.openstack.org/06/502306/11/check/gate-nova-specs-pep8-ubuntu-xenial/3499240/ : SUCCESS in 2m 31s\n- gate-nova-specs-python27-ubuntu-xenial http://logs.openstack.org/06/502306/11/check/gate-nova-specs-python27-ubuntu-xenial/3aab403/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":11},{"id":"b296737c4d79b7a07337f8e4c63be1fda9be21a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-15 19:08:26.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/11/check/build-openstack-sphinx-docs/c215810/html/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/11/check/openstack-tox-pep8/e1b4d96/ : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":11},{"id":"8d36820d5fc03749c0bdfe9f94adf3632cd3ad1f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-10-16 09:47:28.000000000","message":"Patch Set 11: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":11},{"id":"94e009872a70dbdbb944e3914b04817e541152fd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-16 14:33:40.000000000","message":"Patch Set 11:\n\n(7 comments)","accounts_in_message":[],"_revision_number":11},{"id":"6ec36c57689e86b567745b2026a053334506cdbd","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2017-10-16 14:46:36.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nnit about clarity","accounts_in_message":[],"_revision_number":11},{"id":"d53d00798c827aacbb65ef68795e7444adc8a0a2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-16 14:56:29.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5277a6e9f5556548a222966828a3b46056b52f0a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-16 15:30:25.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3108ed294922cb54fe25330249f12ecf53431815","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-16 15:44:48.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/12/check/build-openstack-sphinx-docs/0adb804/html/ : SUCCESS in 6m 48s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/12/check/openstack-tox-pep8/351dc08/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":12},{"id":"46a9da62a0388a9a3e770f8a15f3e63715c97d9b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2017-10-17 13:34:30.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"4d88f52ee5895b35aee9c825d4146eeb0c7bcca9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-17 13:57:46.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"971aeee0696fe059958d0a27cd52b498dfe03723","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-10-17 16:33:43.000000000","message":"Patch Set 12: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":12},{"id":"cae0f704d33ab6e38244513dccbd4c57257da3ef","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-10-17 16:45:55.000000000","message":"Patch Set 12:\n\nThis spec also doesn\u0027t go into much detail about what happens during a move operation for an instance with ports that have bandwidth allocations against a given network resource provider under the source compute host provider. When nova moves the instance, allocations will have to be created against the target host, and eventually cleaned up from the source host. Where does the allocation cleanup for the source happen once the instance has successfully moved?","accounts_in_message":[],"_revision_number":12},{"id":"575e49849afce2b1d883a74bfe974977220231f5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-17 19:37:06.000000000","message":"Patch Set 12: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":12},{"id":"ff000212d725e5d66d7b5162a2d47e43ec82805b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2017-10-18 06:27:34.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"e852e50cdbdbd43dfc66c6cc6bf41f1ef243d961","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2017-10-18 07:03:27.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a6a91d1b0007c4baeab8946900d9e8ba37404173","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-18 16:17:09.000000000","message":"Patch Set 12:\n\n(24 comments)","accounts_in_message":[],"_revision_number":12},{"id":"fe75d1ed17c161c416d8c18a9dbcfd0b2f2f3016","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-18 16:18:10.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ff2ebc611ddb86f60d72d117a1c5ad32bf84d3a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-18 16:40:34.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/13/check/build-openstack-sphinx-docs/a17178f/html/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/13/check/openstack-tox-pep8/9217b65/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":13},{"id":"956667f4738971003ee25eb3e11760d54cb6cf29","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-10-18 20:20:13.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"356ad2ec6cdac5997c2b79e5702938abe7178a42","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-10-18 20:28:17.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"9cbf9a796cd26cfe36b18c46b64810c44f39388f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-10-18 20:33:12.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5fd79c9333e2d436d1158a0549f089c0e673d4aa","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-10-18 20:52:31.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)\n\nI don\u0027t think nova should be responsible for managing network bw provider port allocations during server create and move operations, I think neutron should be the source of truth for managing the inventory and allocations of these types of resources. Changes to the allocations can be triggered when nova binds and unbinds a port, but neutron should be handling the resource tracking with placement.","accounts_in_message":[],"_revision_number":13},{"id":"06427df64bdc4e6c93db94a8dc2c0f85b8e191dd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-18 21:01:21.000000000","message":"Patch Set 12:\n\n(2 comments)\n\nTo restate what Matt has said: it needs to be made crystal clear:\n- Who is the \u0027consumer\u0027 of the bandwidth (and VF/ovs-port) inventory?  Is it the port or is it the instance?\n- Who is responsible for creating that allocation?  Is it the nova scheduler?  The neutron agent?  Neutron itself?  And when?  Host selection?  Port binding?  Port activation?","accounts_in_message":[],"_revision_number":12},{"id":"541cc2543e33acd38645a786ab8c4f78e9e7df22","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-18 21:04:21.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"658f839f4b0754a0d005cbb1815bae9bb932d95b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2017-10-18 21:25:08.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4e119dce98e16b6933bfb750f008ec2dab24395a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-19 11:43:18.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"24cadb26739bc9af634db4fb1bc6b79a106df28b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-19 12:18:40.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"a6310c100d4fd183fad1a290497524c8ed946114","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-19 13:25:03.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"3df7dc98d9db35c693354168cf218be3be3485ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 13:34:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/14/check/build-openstack-sphinx-docs/9461977/html/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/14/check/openstack-tox-pep8/6b8c9e9/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":14},{"id":"d6c6a80652078bc65e03c14b88460a7218a45ba2","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-10-19 14:26:09.000000000","message":"Patch Set 13:\n\n(3 comments)\n\nNew version came in while I was reviewing, will see if my comments still apply, but generally I think a bit more detail on allocation handling and rp generation is in order.","accounts_in_message":[],"_revision_number":13},{"id":"d2a309f553c2bbfccfe9bcefca6d56b69cb7093a","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-10-19 14:28:47.000000000","message":"Patch Set 14: Code-Review-1\n\nYeah, at least two of the comments from ps13 still apply in terms of detail on rp and allocation handling.","accounts_in_message":[],"_revision_number":14},{"id":"22f9b578ff9e6b4af6fb5276b21af0b2802d6262","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-10-20 08:20:22.000000000","message":"Patch Set 13:\n\n(2 comments)\n\nI would like to have an agreement on who is in charge of making the allocation record: Nova conductor or Neutron agent.\n\n- Chris, Matt and Miguel asked why this allocation is not created by the Neutron agent.\n- Sean Mooney (and Jay as far as I know) think the allocation should be created as fast as possible in the conductor to avoid delays between the guest allocation and the allocation record creation, avoiding possible clashes and resource oversubscriptions.","accounts_in_message":[],"_revision_number":13},{"id":"b3460cef2e4653e66d98a35b439f4a09c203ff24","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2017-11-14 08:26:13.000000000","message":"Abandoned\n\nNot approved for Queens","accounts_in_message":[],"_revision_number":14},{"id":"4e84b3f983b78cd2421f3298cdab2a19732cbbf4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2018-01-31 16:38:56.000000000","message":"Restored\n\nRequested by Balazs Gibizer","accounts_in_message":[],"_revision_number":14},{"id":"b47e26bf8b83bd182554859ca595816d5347c3f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 17:22:25.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/14/check/build-openstack-sphinx-docs/f3105f2/html/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/14/check/openstack-tox-pep8/1d0855b/ : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":14},{"id":"619f5a011c221758b2c5721d75de9bae4b525d70","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-02-15 16:07:52.000000000","message":"Patch Set 14:\n\nWhile I tried to prepare myself to the PTG I needed to summarize where we are with this spec.\n\nSeem to be agreed in the spec:\n* neutron is responsible to create RPs under the compute RP for network related resources like bandwidth\n* neutron is responsible to report inventory for the network RPs\n* the requested minimum bandwidth for a neutron port will be defined in / will be accessible through the neutron API\n* The port will be the resource consumer.  The port_id will be the consumer_id in placement for the bandwidth allocations\n\n\nNeeds to be agreed:\n* How the requested minimum bandwidth will be incorporated into the GET allocation_candidate request sent to the placement API during server create/move operation?\n** today nova-scheduler sends the GET allocation_candidate placement API during the scheduling\n** the requested resources are calculated from the flavor fields and flavor extra_spec\n** suggestion: change nova to use the requested minimum bandwidth from the neutron port when calculating the resources that needs to be requested in the GET allocation_candidate placement call\n* Do we need to move the neutron port creation from the nova-compute to the conductor?\n** eventually yes, but not necessarily a hard dependency for minimum bandwidth feature\n** suggestion: support minimum bandwidth first only for servers that are booted with pre-created ports provided in the server create request. When nova finished moving the port creation to the conductor then the minimum bandwidth feature can be extended to support servers booted without pre-created ports.\n* Who will claim the bandwidth resource in placement?\n** nova uses the allocation candidates returned from GET allocation_candidates placement call for further filtering and then to POST allocations of the selected candidate to the placement to claim the resources. As the GET allocation_candidate request already contains the requested bandwidth the returned response already contains possible bandwidth allocation requests to be sent to the placement. So nova has the information to claim the bandwidth resource.\n** If there are two separate claims one from nova (for cpu, mem, disk) and one from neutron (for the bandwidth) then the claim becomes non atomic and opens up the possibility of race condition.\n* Is there somebody form Intel to push this forward? \n** If not then I can take this space over","accounts_in_message":[],"_revision_number":14},{"id":"5e1bd28db5f50cddb7008445b5392c4184e859ad","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-02-16 13:26:48.000000000","message":"Patch Set 14:\n\n\u003e While I tried to prepare myself to the PTG I needed to summarize\n \u003e where we are with this spec.\n \u003e \n \u003e Seem to be agreed in the spec:\n \u003e * neutron is responsible to create RPs under the compute RP for\n \u003e network related resources like bandwidth\n \u003e * neutron is responsible to report inventory for the network RPs\n \u003e * the requested minimum bandwidth for a neutron port will be\n \u003e defined in / will be accessible through the neutron API\n \u003e * The port will be the resource consumer.  The port_id will be the\n \u003e consumer_id in placement for the bandwidth allocations\n \u003e \n \u003e \n \u003e Needs to be agreed:\n \u003e * How the requested minimum bandwidth will be incorporated into the\n \u003e GET allocation_candidate request sent to the placement API during\n \u003e server create/move operation?\n \u003e ** today nova-scheduler sends the GET allocation_candidate\n \u003e placement API during the scheduling\n \u003e ** the requested resources are calculated from the flavor fields\n \u003e and flavor extra_spec\n \u003e ** suggestion: change nova to use the requested minimum bandwidth\n \u003e from the neutron port when calculating the resources that needs to\n \u003e be requested in the GET allocation_candidate placement call\n\nThis can be done with the newly proposed placement request filter feature of nova: https://review.openstack.org/#/c/544585/6/specs/rocky/approved/placement-req-filter.rst\n\n \u003e * Do we need to move the neutron port creation from the\n \u003e nova-compute to the conductor?\n \u003e ** eventually yes, but not necessarily a hard dependency for\n \u003e minimum bandwidth feature\n \u003e ** suggestion: support minimum bandwidth first only for servers\n \u003e that are booted with pre-created ports provided in the server\n \u003e create request. When nova finished moving the port creation to the\n \u003e conductor then the minimum bandwidth feature can be extended to\n \u003e support servers booted without pre-created ports.\n \u003e * Who will claim the bandwidth resource in placement?\n \u003e ** nova uses the allocation candidates returned from GET\n \u003e allocation_candidates placement call for further filtering and then\n \u003e to POST allocations of the selected candidate to the placement to\n \u003e claim the resources. As the GET allocation_candidate request\n \u003e already contains the requested bandwidth the returned response\n \u003e already contains possible bandwidth allocation requests to be sent\n \u003e to the placement. So nova has the information to claim the\n \u003e bandwidth resource.\n \u003e ** If there are two separate claims one from nova (for cpu, mem,\n \u003e disk) and one from neutron (for the bandwidth) then the claim\n \u003e becomes non atomic and opens up the possibility of race condition.\n \u003e * Is there somebody form Intel to push this forward?\n \u003e ** If not then I can take this space over","accounts_in_message":[],"_revision_number":14},{"id":"df9fed6e87536d59c41bdeb4a7dd167c9d8b5687","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-02-16 13:27:56.000000000","message":"Patch Set 14:\n\nSome more to agree on:\n* Where will be in the neutron API the minimum bandwidth request available?\n** This spec mention the port binding API as a possible place but that seems to exist only too late when the target compute host is already selected. Nova needs this information to select the compute host.\n** In the neutron spec it is defined this way \"The required minimum network bandwidth needed for a port is modeled defining a QoS policy along with a QoS minimum bandwidth rule\".\n** Reading this from nova needs multiple API round trip so I suggest to expose a read-only copy of this information directly in the neutron port field.\n** If the format of the new port field follows the placement API allocation_candidate request format i.e. \"resources: {NET_BANDWIDTH_INGRESS_BITS_SEC: 10000, NET_BANDWIDTH_EGRESS_BITS_SEC: 10000}\" then that would allow nova to handle this resource request (and any further port resource request) in a generic way.\n\n* This spec try to handle both OVS and SRIOV minimum bandwidth rules. However for the SRIOV case we have to agree first how to model PFs and VFs in placement. So I suggest to phase the feature. First focus on only the OVS minimum bandwidht rules the modelling seems to be easier there. Later when the SRIOV modelling is done in placement then this feature can be extended to support minimum bandwidth rules for SRIOV as well.","accounts_in_message":[],"_revision_number":14},{"id":"3b0b62543175709d9548c4042397689625efc5cc","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-02-16 14:15:14.000000000","message":"Patch Set 14:\n\n(5 comments)\n\nA few more comments before the ptg.","accounts_in_message":[],"_revision_number":14},{"id":"9c965c8f254eb9e8ff2f96c2813d523a87663a2b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-02-22 23:44:11.000000000","message":"Patch Set 14:\n\n(7 comments)\n\nMy thoughts on Balazs\u0027s comments:\n\n* Do we need to move the neutron port creation from the nova-compute to the conductor? \n\nI agree that it is not necessary\n\n* suggestion: support minimum bandwidth first only for servers that are booted with pre-created ports provided in the server create request. When nova finished moving the port creation to the conductor then the minimum bandwidth feature can be extended to support servers booted without pre-created ports.\n\nThe way you specify the requirement the bandwidth requirement for a port is by assigning to it a QoS policy with a minimum bandwidth rule. That can be done explicitly or by inheriting the port\u0027s network QoS policy. I can see how Nova will be able to create ports that inherit a QoS policy from its network, but how do you propose to manage NOva creating a port and assigning a QoS policy to it?\n\n* If there are two separate claims one from nova (for cpu, mem, disk) and one from neutron (for the bandwidth) then the claim becomes non atomic and opens up the possibility of race condition.\n\nMy understanding is that Nova should manage the claims. Neutron only provides the data in Resource providers to enable the claims. Am I missing something?\n\n* Is there somebody form Intel to push this forward?\n\nNo, there is not. Please take this work over :-)\n\n* Where will be in the neutron API the minimum bandwidth request available? Reading this from nova needs multiple API round trip so I suggest to expose a read-only copy of this information directly in the neutron port field.\n\nYes, you are right. There is an already approved Neutron spec that deals with this. Specifically, we are going to add a read-only \"resources\" attribute to ports (https://review.openstack.org/#/c/396297/7/specs/pike/strict-minimum-bandwidth-support.rst@68). This way, all Nova has to do is GET port and will receive the necessary data. Nova won\u0027t need to mess with Neutron QoS calls or the multiple port binding extension","accounts_in_message":[],"_revision_number":14},{"id":"678b690844b9b79a762e1c81b250105186fa2a5a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-02-22 23:44:57.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d44c0c4a94a3adbe2bd158eeb83d1d7403ecc4e6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-02-23 08:56:11.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"193d9fd603b3d9a5f30cb0c8adefd5f997e4b466","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-02-23 13:42:19.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"513a2eed938f3d1526b65e33b90f7a779517ffbd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-02-26 16:56:19.000000000","message":"Patch Set 14:\n\n\u003e \n \u003e * suggestion: support minimum bandwidth first only for servers that\n \u003e are booted with pre-created ports provided in the server create\n \u003e request. When nova finished moving the port creation to the\n \u003e conductor then the minimum bandwidth feature can be extended to\n \u003e support servers booted without pre-created ports.\n \u003e \n \u003e The way you specify the requirement the bandwidth requirement for a\n \u003e port is by assigning to it a QoS policy with a minimum bandwidth\n \u003e rule. That can be done explicitly or by inheriting the port\u0027s\n \u003e network QoS policy. I can see how Nova will be able to create ports\n \u003e that inherit a QoS policy from its network, but how do you propose\n \u003e to manage NOva creating a port and assigning a QoS policy to it?\n \u003e\n\nThe server create either called with a net_id or port_id. When it is called with port_id then nova needs does not need to create ports in neutron but just use the port that is provided. When server create is called with net_id then nova will go and create ports in neutron. In this case nova will not attach any QoS policy to the port it creates so it will inherit the QoS policy from the network if any. I think this is a meaningful limitation of the use case when the server is created with the net_id provided.\n\n\n\n \u003e \n \u003e * If there are two separate claims one from nova (for cpu, mem,\n \u003e disk) and one from neutron (for the bandwidth) then the claim\n \u003e becomes non atomic and opens up the possibility of race condition.\n \u003e \n \u003e My understanding is that Nova should manage the claims. Neutron\n \u003e only provides the data in Resource providers to enable the claims.\n \u003e Am I missing something?\n\nThere was some conflicting ideas to let neutron claim the bandwidth resource in placement. But I agree with you that we should only have a single place from where we claim resources for a nova instance.\n\n \u003e \n \u003e * Is there somebody form Intel to push this forward?\n \u003e \n \u003e No, there is not. Please take this work over :-)\n\nMeanwhile alex_xu told me that he will also try to help from intel. \n\n \u003e \n \u003e * Where will be in the neutron API the minimum bandwidth request\n \u003e available? Reading this from nova needs multiple API round trip so\n \u003e I suggest to expose a read-only copy of this information directly\n \u003e in the neutron port field.\n \u003e \n \u003e Yes, you are right. There is an already approved Neutron spec that\n \u003e deals with this. Specifically, we are going to add a read-only\n \u003e \"resources\" attribute to ports (https://review.openstack.org/#/c/396297/7/specs/pike/strict-minimum-bandwidth-support.rst@68).\n \u003e This way, all Nova has to do is GET port and will receive the\n \u003e necessary data. Nova won\u0027t need to mess with Neutron QoS calls or\n \u003e the multiple port binding extension\n\nSounds good, I missed that. BTW, in the linked spec has the example \"NIC_BW_EGRESS.ext-net\": 1000 . Does the .ext_net part of the resource name is the name of the network the port is attached to? I think it would be nice to have only the resource class name there as placement does not care about the network name.","accounts_in_message":[],"_revision_number":14},{"id":"70c8af98d266974802320089f6aa55683fe2b60a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-02-26 20:31:03.000000000","message":"Patch Set 14:\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e5758638870b723ca6bc9ec5a41846e189089984","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-02-27 21:19:50.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d667d9ecd4913c6b1e685166acf05e1bdac971fb","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-02-28 20:39:10.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5d566b1d5189df47d3a9ba1d12ee3212f75de1f9","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-02-28 20:42:56.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4bde15577922bc54422a4e268fc1876b3e4c343d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-02 12:08:57.000000000","message":"Patch Set 14:\n\nThis was brought up at the PTG on Thursday.  It\u0027s a limitation in our ability to do unambiguous modeling/requests in the general case when the bandwidth inventory comes from sharing providers.  Consider the following as a counterexample:\n\n         [the rest of the cn rp tree]\n                  /      \\\n +-----+      +--+--+  +--+--+      +-----+\n |SHBW1|......|PF1  |  |PF2  |......|SHBW2|\n |BW: 8| agg1 |VF: 4|  |VF: 4| agg2 |BW: 8|\n +-----+      +-----+  +-----+      +-----+\n\nPF1 and PF2 are providers representing network physical functions.  They exist somewhere in the compute node\u0027s provider tree (the remainder of which isn\u0027t important).  They provide VF inventory.\n\nSHBW1 and SHBW2 are sharing providers (e.g. TOR switches) providing bandwidth inventory.  (8 what?  Don\u0027t know, don\u0027t care - not the point.)\n\nSHBW1 is associated via aggregate `agg1` with PF1.\nSHBW2 is associated via aggregate `agg2` with PF2.\n\nAssume the PFs are effectively identical (e.g. connect to the same physnet, etc.).  Assume the bandwidth providers are effectively identical.\n\nNow we want to express a request for e.g. `VF:1,BW:2`, BUT if the VF is provided by PF1, the BW must be provided by SHBW1; and likewise, if the VF is provided by PF2, the BW must be provided by SHBW2.\n\nThe issue is that we have no generic way of expressing this request with this restriction.  Under the current implemented/conceived architecture, the only options entail explicitly naming things.  For example, requesting allocation candidates with member_of:agg1.  Or marking PF1 and SHBW1 with the same trait and using ?required\u003d\u003cthat trait\u003e.","accounts_in_message":[],"_revision_number":14},{"id":"542f8473fdbac8d3db3ab0184b6add00abee4b8e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-04 14:48:30.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"8439480c9c9b3a0e9b41fd6b8eaf008c55b88209","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-04 14:49:42.000000000","message":"Patch Set 15:\n\n(2 comments)\n\nSpec has been reworked based on the PTG description.","accounts_in_message":[],"_revision_number":15},{"id":"67f976a8adbc71f2ae062ba8f47c7695dc7ddcdf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-04 14:50:15.000000000","message":"Patch Set 15:\n\n\u003e This was brought up at the PTG on Thursday.  It\u0027s a limitation in\n \u003e our ability to do unambiguous modeling/requests in the general case\n \u003e when the bandwidth inventory comes from sharing providers. \n \u003e Consider the following as a counterexample:\n \u003e \n \u003e [the rest of the cn rp tree]\n \u003e /      \\\n \u003e +-----+      +--+--+  +--+--+      +-----+\n \u003e |SHBW1|......|PF1  |  |PF2  |......|SHBW2|\n \u003e |BW: 8| agg1 |VF: 4|  |VF: 4| agg2 |BW: 8|\n \u003e +-----+      +-----+  +-----+      +-----+\n \u003e \n \u003e PF1 and PF2 are providers representing network physical functions. \n \u003e They exist somewhere in the compute node\u0027s provider tree (the\n \u003e remainder of which isn\u0027t important).  They provide VF inventory.\n \u003e \n \u003e SHBW1 and SHBW2 are sharing providers (e.g. TOR switches) providing\n \u003e bandwidth inventory.  (8 what?  Don\u0027t know, don\u0027t care - not the\n \u003e point.)\n \u003e \n \u003e SHBW1 is associated via aggregate `agg1` with PF1.\n \u003e SHBW2 is associated via aggregate `agg2` with PF2.\n \u003e \n \u003e Assume the PFs are effectively identical (e.g. connect to the same\n \u003e physnet, etc.).  Assume the bandwidth providers are effectively\n \u003e identical.\n \u003e \n \u003e Now we want to express a request for e.g. `VF:1,BW:2`, BUT if the\n \u003e VF is provided by PF1, the BW must be provided by SHBW1; and\n \u003e likewise, if the VF is provided by PF2, the BW must be provided by\n \u003e SHBW2.\n \u003e \n \u003e The issue is that we have no generic way of expressing this request\n \u003e with this restriction.  Under the current implemented/conceived\n \u003e architecture, the only options entail explicitly naming things. \n \u003e For example, requesting allocation candidates with member_of:agg1. \n \u003e Or marking PF1 and SHBW1 with the same trait and using\n \u003e ?required\u003d\u003cthat trait\u003e.\n\nThanks Eric. I incorporated this problem description into the latest version of the spec.","accounts_in_message":[],"_revision_number":15},{"id":"3f1ac3846980d9fa3d5e60005c9230dd428388dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-04 14:58:35.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/15/check/build-openstack-sphinx-docs/a9b172b/html/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/15/check/openstack-tox-pep8/e938ce3/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":15},{"id":"f036b251c7f3d9b672f115b69db1b82ba1a185f5","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-03-06 09:39:51.000000000","message":"Patch Set 15: Code-Review-1\n\n(7 comments)\n\nfew comments here.","accounts_in_message":[],"_revision_number":15},{"id":"8bcf9fef01451f2e40308d9f776554238620b63a","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-03-06 09:58:46.000000000","message":"Patch Set 15: Code-Review-1\n\n(3 comments)\n\nSome comments inline, still reviewing","accounts_in_message":[],"_revision_number":15},{"id":"4e8b033298591463c0f4108e8d39b6b6a5ad814c","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-03-06 11:53:20.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"106f8ab371ec74f5a7459889ba505933075358ac","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-03-07 01:45:10.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c43cf974183894719491894e88ed85fec2b3e094","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-03-07 02:53:37.000000000","message":"Patch Set 15:\n\n(7 comments)","accounts_in_message":[],"_revision_number":15},{"id":"09ce9a5feface3287e54c5a73eb12fa534ca7320","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2018-03-07 03:17:09.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"a6f3da4c0ba8da95befa26d60d3f84477adc8e62","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-03-07 06:48:53.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"50e530914e17b96e6f2c0045f8d96a4877e6d36b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-07 17:01:46.000000000","message":"Patch Set 15:\n\n(13 comments)\n\nThanks for the reviews so far. I tried to answer some of your questions inline. I will update the spec later this week to gather some more comments to fix.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":15},{"id":"3e9692992085d73a53e9f5c1fbe1ef753a96d8d7","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-03-08 05:47:53.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"c31179167d2651f96403352e26dd4d215289a918","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-03-08 05:57:30.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"953db5b3cb8e18c477fdd8b7e063c6326847faf9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-08 10:19:33.000000000","message":"Patch Set 15:\n\n(2 comments)\n\n@Alex: some more answers inline.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":15},{"id":"1a75f26b9b7635080c09e69acc7f19450ac83640","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2018-03-08 11:32:10.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"64840e2fdc1bbc954f17e98c6c016a636a538325","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-03-08 15:21:08.000000000","message":"Patch Set 15:\n\n(6 comments)","accounts_in_message":[],"_revision_number":15},{"id":"f2e1007a42fa233269ac6bfbca2941722b3c6598","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-03-09 08:37:05.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b36210100cc29e9abb0188f994182eec449a2ba0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-09 08:58:03.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"6f63c39cb9a55fcbb531f6cff5983f9685732c65","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-09 16:01:07.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"5d418d6165246cb74a62ae04d39237f19c0d069f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-09 16:03:26.000000000","message":"Patch Set 15:\n\n(17 comments)\n\nI\u0027ve updated the spec based on the comments so far.\nCheers,\ngibi","accounts_in_message":[],"_revision_number":15},{"id":"8cacf05375d1e9b8f90aaa36716267e912ca6637","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-09 16:09:36.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/16/check/build-openstack-sphinx-docs/169faef/html/ : SUCCESS in 5m 41s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/16/check/openstack-tox-pep8/5017070/ : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":16},{"id":"e70ff715fd516dd4b84739bceea4be645a427d6b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-09 16:18:19.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"ec0226d72b84da53a0e3d23b19560d4fd477634f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-09 16:24:53.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/17/check/build-openstack-sphinx-docs/a64d9ee/html/ : SUCCESS in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/17/check/openstack-tox-pep8/69a7db4/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":17},{"id":"8468a83ad996a840b85524e9c9f059fab16cc41e","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-09 20:31:01.000000000","message":"Patch Set 15:\n\n(17 comments)","accounts_in_message":[],"_revision_number":15},{"id":"a25488bfc7cec9809cc4d26b8e984bad2cbcac87","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-09 20:32:21.000000000","message":"Patch Set 17: Code-Review+2\n\nLeft my grammar/typos/notes on the previous revision. I\u0027m actually good with this spec as-is. It hits all the right points, IMHO, and works with the placement API the way I expect it to.\n\n-jay","accounts_in_message":[],"_revision_number":17},{"id":"5411dee96937510d9e59317eef0dcf5590d1da04","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-03-11 14:21:44.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ca2eee19a4fd1c6054223cdbd9277d0431429ea7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-12 11:27:38.000000000","message":"Patch Set 15:\n\n(8 comments)\n\n@Jay: thanks for the review. I left some answers inline. I\u0027m planning to do an update on this spec around Wednesday to gather a bit more feedback.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":15},{"id":"22f0eef08c1597bbc6302a47ebc90fe7f284076d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-14 14:26:42.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"ad70c75b2631031a28d75661e32da1e777f60378","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-14 14:30:38.000000000","message":"Patch Set 18:\n\n(9 comments)\n\nI will not be available until Monday so I will only check the incoming comments next week.\nCheers,\ngibi","accounts_in_message":[],"_revision_number":18},{"id":"ab2565f9336559210934276e81e3f13943c09387","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-14 14:32:48.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"711f03144f6cfadb264c8bcddf16337f2eb09dd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 14:34:03.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/18/check/build-openstack-sphinx-docs/27dc07b/html/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/18/check/openstack-tox-pep8/3e39534/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":18},{"id":"c1114ddcd949c714dae325b6c3b7f7a52615f88d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-03-14 15:21:57.000000000","message":"Patch Set 17:\n\n(24 comments)","accounts_in_message":[],"_revision_number":17},{"id":"41ec0ee7fd523c772fb887dc866bd5cc4e5dccab","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-03-14 15:24:12.000000000","message":"Patch Set 18: Code-Review-1\n\nLots of comments and questions in PS17 while PS18 was being pushed.\n\nI don\u0027t think we need a new compute API microversion for this, so -1 for that. Otherwise I want to make sure my comments/questions are addressed so I fully understand what is being proposed, and that we\u0027re aware of all of the various caveats.","accounts_in_message":[],"_revision_number":18},{"id":"dc7d49c5cd92f094339f9c52ddb46ed649dbb8cb","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-03-18 00:45:55.000000000","message":"Patch Set 17:\n\n(26 comments)","accounts_in_message":[],"_revision_number":17},{"id":"27747e535161f36500fd1486d826cc97fc05716d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-03-18 00:46:31.000000000","message":"Patch Set 18:\n\nLeft my comments in PS17, as Matt did","accounts_in_message":[],"_revision_number":18},{"id":"ab78e5592501b56f2daa22a948acb297abe73a3c","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-03-18 06:34:21.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"555f6ff387b6afbf91fe04c14b4ded296efcb18e","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-03-18 07:44:50.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"1a3221ced6b79f684f9d2385566c23223296bd8c","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-03-19 15:04:44.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"d47b4fd58da5091a1d0cdea2e362dc3a0993af4a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-19 16:54:32.000000000","message":"Patch Set 17:\n\n(21 comments)\n\nThanks Matt, Miguel, Alex for the comments. I tried to answer the questions inline. I see two major things to be settled:\n\n* Do we need microversion bump or not to signal the API behavior change in case of server create with network having QoS bandwidth rule and during interface attach with QoS aware port?\n\n* I have to clarify how will the resource request from the neutron port will end up in the query parameters of the allocation candidate GET request. \n\nI will try to update the spec on Wednesday.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":17},{"id":"d6ab60537fb1676a153bbf3030a390a97d8d3f1c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-20 10:27:32.000000000","message":"Patch Set 18:\n\n(3 comments)\n\n@Bence: thanks for the review. I will incorporate your feedback to the next revision of the spec.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":18},{"id":"1aece32759da81a2e48479fd07eeb6c43750ad97","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-20 15:55:05.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"ac7610154d10585c159eaf6fc8ddfe06baafa3d8","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-03-20 16:22:09.000000000","message":"Patch Set 18:\n\nThe neutron-side spec was updated (ps4) and is now in sync with this: Ib995837f6161bcceb09735a5601d8b79a25a7354","accounts_in_message":[],"_revision_number":18},{"id":"8066e284a24921feced7bf7c65f34c2a7fa2b38d","author":{"_account_id":7776,"name":"Paul Carver","email":"pcarver@att.com","username":"pcarver"},"date":"2018-03-21 23:56:13.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"954c3da5ee871eb090580920b390e71d5a7bad59","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-22 08:04:39.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"24c6bf6fa75c3f9ee21acd937e946ff0219c3a36","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-03-22 14:11:02.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a5d43c71cd7401e704446600b556ac63fbd3bf37","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-22 18:39:25.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"a2d5d3a14e7b50e49e49d132a0a4da489dbece94","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-22 18:41:21.000000000","message":"Patch Set 19:\n\n(34 comments)\n\nI fixed every comments and did promised extensions to the spec except Bence\u0027s last comment about SRIOV physical device mapping. I need to think about that first.\nCheers,\ngibi","accounts_in_message":[],"_revision_number":19},{"id":"6e9a35a1c8e0bb701227521c596121ed79f207a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-22 19:03:25.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/19/check/build-openstack-sphinx-docs/0db1928/html/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/19/check/openstack-tox-pep8/a1821fe/ : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":19},{"id":"d496bc177a69d90158b25808a983b7ebef89e8d3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-23 10:14:23.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"5f9831ba884e5cf9d0c8136f46b814a69a780a1d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-03-23 16:06:51.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"db30f9851fcfab626e3dfa428b25b585cdb4af3f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-23 16:36:52.000000000","message":"Patch Set 17:\n\n(8 comments)","accounts_in_message":[],"_revision_number":17},{"id":"69a5cc5d770865cc853a885360b7c316e3ff57ef","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-23 17:32:08.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"3f36cfb55a129ab988f7153f93febb4331d216cd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-23 17:33:01.000000000","message":"Patch Set 19:\n\n(10 comments)\n\nMade some modification based on the recent comments and added two new subsection with newly discovered issues.","accounts_in_message":[],"_revision_number":19},{"id":"7a1d3a544d77c47f4a8a9941670008fa7570a62e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-23 17:44:03.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"8c51f8a12b0c659e7686900ae6048c96795e2dfd","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-23 18:05:13.000000000","message":"Patch Set 20: Code-Review-1\n\n(12 comments)\n\nthis is really close. my only concern is that I do not think we need the NET_VNIC_TYPE_XXX traits -- and I would prefer to leave those out of the spec.","accounts_in_message":[],"_revision_number":20},{"id":"d47cc1f1d739faf094052adff8f9465b4fa376a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-23 18:42:55.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/20/check/build-openstack-sphinx-docs/fc04fe8/html/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/20/check/openstack-tox-pep8/07a0e42/ : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":20},{"id":"7cf5cd61aab784dce3814fa45e9b732396b15bde","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-25 19:39:00.000000000","message":"Patch Set 20: Code-Review-1\n\n(24 comments)\n\nI didn\u0027t quite get to finish reviewing this today.  Will finish up tomorrow.  There\u0027s at least one -1-worthy technical error so far though (L490).","accounts_in_message":[],"_revision_number":20},{"id":"ddc62109cfbb63b2a54c3e5f05006a4e3723fa4e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-26 11:59:41.000000000","message":"Patch Set 20:\n\n(13 comments)\n\n@Jay: Thanks for the feedback. I tried to add more reasoning about the need for the VNIC_TYPE\n@Eric: Thanks for the feedback. I tried to answer your concern about the numbered groups.","accounts_in_message":[],"_revision_number":20},{"id":"f7e207855c76c5023271a30ce7c2da62940999af","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-26 12:48:39.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"c42c07c1779cf95ca4177d022672b15c1a83effa","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-26 14:01:52.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"65ccd1fab6d1aa1665534e259bb0125b20feb0b6","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-26 14:03:08.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"22b23123491fd6eac8913c33b7a3cebe416afea5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-26 15:20:28.000000000","message":"Patch Set 20:\n\n(2 comments)\n\n@Jay: responded inline","accounts_in_message":[],"_revision_number":20},{"id":"7f36f8d68923a2d1f61ea6678c48dddbbf8de500","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-27 14:30:55.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"7f1ac26a79e012c9a54c1c1166b3482ec21fd3d9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-27 14:32:00.000000000","message":"Patch Set 21:\n\n(27 comments)","accounts_in_message":[],"_revision_number":21},{"id":"ad758b22ed1ce9a77cfd0588ccc5daca6b8eb910","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-27 14:42:22.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/21/check/build-openstack-sphinx-docs/95af57b/html/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/21/check/openstack-tox-pep8/875f9ad/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":21},{"id":"5ff9db361389c67d50cc1c56cb254c7b14f94afa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2018-03-27 14:54:33.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"c71618b13e088e5369dfc9a9ad2cd3bbe0c56efd","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-27 17:03:07.000000000","message":"Patch Set 21: Code-Review-1\n\nOK, summarizing requested changes from the etherpad where me, gibi and mlavalle discussed this spec (https://etherpad.openstack.org/p/X0RboWOe7C):\n\n1) Remove any mention of NET_VNIC_TYPE_XXX or NET_BACKEND_XXX traits. The only traits that need to be mentioned in the spec are the CUSTOM_PHYSNET_XXX traits that indicate the physical network the NIC is attached to.\n\n2) Reduce the number of use case scenarios to two. The first use case is two 10G NICs attached to different physical networks. The second use case is two 10G NICs attached to the same physical network:\n\nUSE CASE 1: Two separate 10G physical NICs that are attached to two difference physical networks and bandwidth needs to be tracked for each of the physical networks.\n\ncompute_node_provider:\n    physical_network_0_provider:\n        traits: CUSTOM_PHYSNET_0\n        inventories:\n            NET_BANDWIDTH_INGRESS_KILOBITS_PER_SECOND: 1000000\n    physical_network_1_provider:\n        traits: CUSTOM_PHYSNET_1\n        inventories:\n            NET_BANDWIDTH_INGRESS_KILOBITS_PER_SECOND: 1000000\n\nUSE CASE 2: Two separate 10G physical NICs that are both supplying bandwidth to the same physical network and need to have 20G of bandwidth available for instances to consume but a single instance cannot consume more than the bandwidth provided by a single physical NIC:\n\ncompute_node_provider:\n    physical_network_0_provider_0:\n        traits: CUSTOM_PHYSNET_0\n        inventories:\n            NET_BANDWIDTH_INGRESS_KILOBITS_PER_SECOND: 1000000, max_unit: 1000000\n    physical_network_0_provider_1:\n        traits: CUSTOM_PHYSNET_0\n        inventories:\n            NET_BANDWIDTH_INGRESS_KILOBITS_PER_SECOND: 1000000, max_unit: 1000000\n\n3) Change the name of the resource classes to NET_BANDWIDTH_INGRESS_KILOBITS_PER_SECOND and NET_BANDWIDTH_EGRESS_KILOBITS_PER_SECOND","accounts_in_message":[],"_revision_number":21},{"id":"286f8dcf94851bbeee223cdc71a8441d67449d4d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-27 17:21:19.000000000","message":"Patch Set 21: Code-Review-1\n\n(20 comments)","accounts_in_message":[],"_revision_number":21},{"id":"f5d5f77e309c95971fd06b1f55bf653aad705cb5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-27 17:27:26.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"a3fb42ae9e70ff9f2b4add7048c551713f822a76","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-27 18:30:53.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"05dc512ac4916a222eed9299f56d866ebaf384a4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-27 18:32:58.000000000","message":"Patch Set 22:\n\n(17 comments)\n\nI updated the spec after the discussion with Jay and Miguel. Also I fixed some of the comments from Eric, but not all of them. I have to look at the comments in the section\n\"Mapping between physical resource consumption and claimed resources\" again tomorrow with a fresh brain.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":22},{"id":"ffd5f25eea941179d803780e7443e1db5f206b29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-27 18:37:43.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/22/check/build-openstack-sphinx-docs/6e64e55/html/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/22/check/openstack-tox-pep8/a370efc/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":22},{"id":"52b8052e53966a7101ae549f72ca65577a76830b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-03-27 23:31:59.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"dbb6ba30f3d8c30103abf3be227981191b8f0a2e","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-03-28 08:15:04.000000000","message":"Patch Set 22: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"7cfc60dd2eef817f0bf7785f672080078744ae2b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-03-28 09:15:52.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"31c91fe5d69d4c77d96f00196c459f2d47c5901f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-29 08:56:56.000000000","message":"Patch Set 22: Code-Review-1\n\n(3 comments)\n\nWe still big open items in the spec:\n* Do we need vnic_type in the model. See comments in PS22 L420\n* We cannot use physnet trait in the port\u0027s resource request. See comments in PS22 L450\n* Who and how does the mapping of a port\u0027s resource request and the subset of the allocations in the selected allocation candidate. See Eric\u0027s comments in PS21 L664 in the section \u0027Where the mapping will be calculated\u0027 and the IRC discussion in http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2018-03-27.log.html#t2018-03-27T17:06:10","accounts_in_message":[],"_revision_number":22},{"id":"590dfbffd3bd364addd2d1e65703c73148d63b29","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-29 15:27:10.000000000","message":"Patch Set 22:\n\n\u003e (3 comments)\n \u003e \n \u003e We still big open items in the spec:\n \u003e * Do we need vnic_type in the model. See comments in PS22 L420\n \u003e * We cannot use physnet trait in the port\u0027s resource request. See\n \u003e comments in PS22 L450\n\nMore discussion happened on IRC about it http://eavesdrop.openstack.org/irclogs/%23openstack-neutron/%23openstack-neutron.2018-03-29.log.html#t2018-03-29T14:03:22\n\nAlso further details has been added to the etherpad after L80 https://etherpad.openstack.org/p/X0RboWOe7C\n\n \u003e * Who and how does the mapping of a port\u0027s resource request and the\n \u003e subset of the allocations in the selected allocation candidate. See\n \u003e Eric\u0027s comments in PS21 L664 in the section \u0027Where the mapping will\n \u003e be calculated\u0027 and the IRC discussion in http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2018-03-27.log.html#t2018-03-27T17:06:10","accounts_in_message":[],"_revision_number":22},{"id":"07a612cfd488f1319750c023a44aeadedff9174a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-03-29 22:02:39.000000000","message":"Patch Set 22: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4c380635319c772ed5f78b211708827e646d3600","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-04-03 12:16:31.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"12cc2cdc4dbb22d803833bb3b715db8def73371c","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-04-03 16:07:02.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"c069648df9d2b04a38ec11f7f0c4afd401211bc8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-04-04 02:43:18.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"fc8fec2940bbb5c86310cc0c9ec9b63dcc95964a","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-04-04 08:40:52.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"f8f5b9c4924244dc3f9a1de89fb0aae97c9f7df8","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-04-05 13:22:43.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"e5a77450a44fa5dae5233879b643db52dfdda1a8","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-06 19:01:06.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"126b4c564c60ac40ca30af4d8d8bdbecba243187","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-04-09 14:54:20.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"c1a71df2aec17c71b385dbb4d34f65fbbd3a2e7a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-04-19 15:52:36.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"540cf190db05a69a24880a38d70d61d7479e698e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-04-19 15:53:14.000000000","message":"Patch Set 22: -Code-Review\n\n(9 comments)\n\nI\u0027ve update the spec based on the last two weeks discussion.","accounts_in_message":[],"_revision_number":22},{"id":"e37910be2d0ce1d121aa70cea1382044918b2619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-19 17:38:18.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/23/check/build-openstack-sphinx-docs/9f1f3c9/html/ : SUCCESS in 7m 11s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/23/check/openstack-tox-pep8/b60b51e/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":23},{"id":"1920ab182b65a24988cc95da8b4ef1296b769d3c","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-20 16:11:10.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"90c3bfa05e97a6e44f8c23c4d570cbdf7443d994","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-20 17:28:01.000000000","message":"Patch Set 23: Code-Review-1\n\nI\u0027ve added my thoughts on how to accomplish this without any vnic type traits:\n\nhttps://etherpad.openstack.org/p/net-bw-providers","accounts_in_message":[],"_revision_number":23},{"id":"df5430cfca8a3a0ee1a9d57cb7acd2647599595e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-04-21 15:13:13.000000000","message":"Patch Set 23:\n\n\u003e I\u0027ve added my thoughts on how to accomplish this without any vnic\n \u003e type traits:\n \u003e \n \u003e https://etherpad.openstack.org/p/net-bw-providers\n\nI\u0027ve responded to in the etherpad","accounts_in_message":[],"_revision_number":23},{"id":"f9cf120128023ee8131fce4f43d934592a2f4e4b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-04-24 09:32:30.000000000","message":"Patch Set 23:\n\nWe had a hangout with Jay and Bence yesterday. I will rework the spec based on the agreement we made there.\n\nLet\u0027s try to limit the scope of the specs to:\n* Neutron provides RP tree and inventory (with physnet and vnic_type traits)\n* Neutron provides the resource_request in the port as an optional API extension\n* Nova takes the resource_request from the port (if exists) and blindly include it into the GET allocation_candidate request\n* Nova will include the whole allocation (it made in placement) to the port binding. \n\nWe will not solve the following problems in the current iteration:\n* mapping the pieces of the allocation back to the pieces of the resource_request coming from individual ports. Neutron can fail the port binding for now for complicated cases\n* trying to model the granular resource group proximity parameter for the Neutron ports. If the proximity is defined in the flavor extra_spec then that policy will be applied for the ports\u0027 resource_request as well for now.\n* implementing the neutron backend preference order with a scheduler weigher in nova-scheduler.","accounts_in_message":[],"_revision_number":23},{"id":"b6f2ff31cff7d15db18a44c7c3cb86bb14ce3226","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-04-24 17:47:28.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"cd2a7b3838a07bbe38bd64b45c4b3762f882f43d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-04-24 17:48:35.000000000","message":"Patch Set 24:\n\nI cut back on the scope (an on the size of the spec) based on the latest discussion. The connected neutron spec will also be updated accordingly in the coming days.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":24},{"id":"7d57417ce8cbd0569a9bf403f4941e02179a55cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-24 19:01:35.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/24/check/build-openstack-sphinx-docs/a567915/html/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/24/check/openstack-tox-pep8/44b1bd1/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":24},{"id":"b7d390269e2ec3a32933b87e2a5d13126656b7c8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-04-25 08:29:07.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"0bdef8d711421c4bddf1b9ce479647bbe13acf5d","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-04-25 14:10:10.000000000","message":"Patch Set 24:\n\n\u003e The connected neutron spec will also be updated\n \u003e accordingly in the coming days.\n\nDone. Please see ps8 of Ib995837f6161bcceb09735a5601d8b79a25a7354.","accounts_in_message":[],"_revision_number":24},{"id":"7f250fdb01a118358827b671af2b70914f687e4c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2018-04-27 07:53:11.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"200d72bb2ab254d4c3b5e7cc567f576f125e17bd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-04-27 11:03:11.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"2a8221b0325a9c1917f5cb1e4926030160da9bdc","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-27 18:00:18.000000000","message":"Patch Set 24:\n\n(9 comments)","accounts_in_message":[],"_revision_number":24},{"id":"3bdbc32aa714762328d352f1bd40fc8ceb3c5d5a","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-27 18:04:56.000000000","message":"Patch Set 24: Code-Review+2\n\nI\u0027m +2 on the ideas in the spec and the proposed nova-side changes. The scope of this spec was limited to essentially the following:\n\n1) nova-conductor calls to neutron to retrieve the set of requested resources and traits for a port when the new Neutron API extension is present and the user supplied port information in the nova boot call.\n\n2) nova-conductor writes the requested resources and traits to the RequestSpec\n\n3) nova-scheduler includes the network-specific requested resources in its call to GET /allocation_candidates as separate numbered request groups\n\n4) nova-compute passes allocation dicts, serialized into a field of the port binding dict, to neutron during the setup_networks_on_host() activity\n\nAll other steps are done in Neutron.\n\n-jay","accounts_in_message":[],"_revision_number":24},{"id":"eef37202b1f5a18ab90afb49db3004a6504674ee","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-05-01 02:49:11.000000000","message":"Patch Set 24: Code-Review+1\n\n(4 comments)\n\nI am happy with this spec. Also some nits in case you roll another revision","accounts_in_message":[],"_revision_number":24},{"id":"92ecbe8da3a9982b3e6f9a8c7a3e68e5d5a9e593","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-05-02 00:12:14.000000000","message":"Patch Set 24: Code-Review+1","accounts_in_message":[],"_revision_number":24},{"id":"d65b9f4be8a93a0a4741828c897eae2c9c2307ba","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-05-02 08:55:25.000000000","message":"Patch Set 24: Code-Review+1","accounts_in_message":[],"_revision_number":24},{"id":"f8c125548bae82c0016196f4221ac6f9ebe74a44","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-05-02 13:43:19.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"ca3d8b8a9bc67eb5537156ef05f037fbda28571d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-05-02 13:44:07.000000000","message":"Patch Set 24:\n\n(10 comments)\n\nThanks for the reviews. I fixed the nits and added references to two new placement specs aiming to support any(traits) type of queries.\n\ncheers,\ngibi","accounts_in_message":[],"_revision_number":24},{"id":"ad13649db35f6c01a2f935c076e8bde89e80a295","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 13:53:50.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/25/check/build-openstack-sphinx-docs/e83d42b/html/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/25/check/openstack-tox-pep8/ca5803a/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":25},{"id":"b257e9147a3be885a2e8479ce27767b6e8af71c0","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-05-02 14:48:35.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"7f3fa07930ba5ff913ed8fafd70a31288544c077","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-05-02 15:14:55.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"44b1a159b7f0ca4d5e37acb278ca3bdf01dfa380","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-05-02 18:10:28.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"25367ab9ee0e38fe8f0c1a075ddf36889edc3571","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-05-02 22:26:42.000000000","message":"Patch Set 25: Code-Review-1\n\n(5 comments)\n\nThis is super close.\n\nMost of my comments really center around the same issue.  I think I\u0027ll be satisfied if you show me specifically what the network resource request looks like, and how that gets mapped to granular request groups in the placement querystring.\n\nI would also like it clarified that Neutron (or is it the virt driver?  Or the plugin/agent?) will need to use the original request in concert with the allocation to figure out how to create/assign the ports appropriately.","accounts_in_message":[],"_revision_number":25},{"id":"408cc9eec99f458ee55866055cdf63235092b7bd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-05-03 15:39:33.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"5ed8c41d28a0016fb6e4fc4e834cdd1d638019ed","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-05-03 15:40:20.000000000","message":"Patch Set 25:\n\n(5 comments)\n\nI followed Jay\u0027s suggestion to re-use the RequestGroup object as internal storage for the port\u0027s resource request in Nova.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":25},{"id":"76d90f821dab55d8460872579cc7335734fc446a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 15:47:02.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/26/check/build-openstack-sphinx-docs/87c9485/html/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/26/check/openstack-tox-pep8/cc963ca/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":26},{"id":"13ed6fce97b503ec8ce935ec5791ae9b9672b0c4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-05-03 15:48:58.000000000","message":"Patch Set 26: Code-Review+1\n\n(1 comment)\n\ntypo if you respin","accounts_in_message":[],"_revision_number":26},{"id":"703c1646b1a1e90bffc230aaaa85d939485b81a1","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-05-03 22:19:17.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"ed338ef4a7c24d3e92db16e71087e35e45ec7dbd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-05-04 15:40:56.000000000","message":"Patch Set 26:\n\n(25 comments)","accounts_in_message":[],"_revision_number":26},{"id":"142434339bb841af4ce0bf2b45916e45659f287a","author":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"date":"2018-05-04 17:58:10.000000000","message":"Patch Set 26:\n\nUp front, I kinda skimmed through this as I don\u0027t fully understand it all. But, from an Ironic perspective:\n\nAre we going to include ironic in this? I think it can be done, but would take some work, specifically:\n\n* ensuring NICs for ironic nodes are represented in placement \n* somehow making sure all NICs for a node are allocated to the instance, as we can\u0027t split them up between multiple instances. Alternatively, just don\u0027t let them be allocated to another instance if not all are taken.\n* ensuring an ironic instance consumes the entire NIC, for the same reason.\n\nI\u0027m trying to remember if Ironic has a way of mapping ports to a specific NIC, this may need to be thought out.\n\nAlso, sounds like Neutron will be putting the physical NIC information in placement? Will ironic need to push that information into Neutron? Today, ironic is the source of truth for physical NIC info.\n\nI could be completely misunderstanding these, please do let me know if I am :)\n\nThe other option is ignoring ironic (for now?) and not being able to use network resources in a flavor.","accounts_in_message":[],"_revision_number":26},{"id":"81e1b93f1f55cc4bcc5233d10dc593d40bf3f9d3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-05-07 14:41:32.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"8340ae7aa792bf8fad184951b25afc4fd4bc5ad4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-05-07 14:47:14.000000000","message":"Patch Set 26:\n\n(26 comments)\n\n\u003e Up front, I kinda skimmed through this as I don\u0027t fully understand\n \u003e it all. But, from an Ironic perspective:\n \u003e \n \u003e Are we going to include ironic in this? I think it can be done, but\n \u003e would take some work, specifically:\n\nI suggest not to try to include Ironic in the first iteration. Let\u0027s plan Ironic support for a later iteration.\n\n \u003e \n \u003e * ensuring NICs for ironic nodes are represented in placement\n\nThat could be done by Ironic when reporting the RP tree from the virt driver.\n\n \u003e * somehow making sure all NICs for a node are allocated to the\n \u003e instance, as we can\u0027t split them up between multiple instances.\n \u003e Alternatively, just don\u0027t let them be allocated to another instance\n \u003e if not all are taken.\n \u003e * ensuring an ironic instance consumes the entire NIC, for the same\n \u003e reason.\n\nFor Ironic a NIC\u0027s bandwidth is not a resource but a only a capability. I think this can be handled with traits. \n\n \u003e \n \u003e I\u0027m trying to remember if Ironic has a way of mapping ports to a\n \u003e specific NIC, this may need to be thought out.\n \u003e \n \u003e Also, sounds like Neutron will be putting the physical NIC\n \u003e information in placement? Will ironic need to push that information\n\nYes, Neutron reports the bandwidth inventory of a NIC on the compute.\n\n \u003e into Neutron? Today, ironic is the source of truth for physical NIC\n \u003e info.\n\nThis is one of the reason I don\u0027t want to include Ironic in the scope right now. The current spec handles NICs that are owned by Neutron. Ironic is different in this sense and I think it can be solved separately.\n\n \u003e \n \u003e I could be completely misunderstanding these, please do let me know\n \u003e if I am :)\n \u003e \n \u003e The other option is ignoring ironic (for now?) and not being able\n \u003e to use network resources in a flavor.","accounts_in_message":[],"_revision_number":26},{"id":"94ea0cff6605b77eb4865b045217228bcb688941","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-07 14:53:00.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/27/check/build-openstack-sphinx-docs/809245c/html/ : SUCCESS in 6m 03s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/27/check/openstack-tox-pep8/98c8eec/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":27},{"id":"80b236991c9864dd82bfcea2b62ed3551a9dfa71","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-05-07 15:55:17.000000000","message":"Patch Set 27: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"832f63fafd69e38a71e3f8aff9ffde0db1a70470","author":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"date":"2018-05-07 17:41:49.000000000","message":"Patch Set 27:\n\nOkay, I\u0027m fine leaving ironic out of this for now. Thanks gibi :)","accounts_in_message":[],"_revision_number":27},{"id":"5849040ee62ed5c26d6869dbafa89e5ea1080f93","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-05-08 00:07:03.000000000","message":"Patch Set 27: Code-Review+2\n\nI\u0027m happy with this.","accounts_in_message":[],"_revision_number":27},{"id":"d0711fa3ca2b1bb32335f18130ca2f47cf698c47","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-05-08 07:10:20.000000000","message":"Patch Set 27: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":27},{"id":"9993ecbe3317fafab589433743acf1c49bd8e121","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-05-08 07:27:39.000000000","message":"Patch Set 27: Code-Review+1","accounts_in_message":[],"_revision_number":27},{"id":"7f4bdae94d9841d4d0f167d10ac551b8eeb081ba","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-05-08 16:01:43.000000000","message":"Patch Set 27: Code-Review+1\n\n(6 comments)\n\nCouple of nits in there but this is taking shape. At this point I really just want to see the code :)","accounts_in_message":[],"_revision_number":27},{"id":"c29fb1dead5f69ae950ce009248eb4fd620118ea","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-05-08 16:11:36.000000000","message":"Patch Set 27:\n\n\u003e (6 comments)\n \u003e \n \u003e Couple of nits in there but this is taking shape. At this point I\n \u003e really just want to see the code :)\nMe too :)\nThanks for the review.","accounts_in_message":[],"_revision_number":27},{"id":"a2887cd8614f004226633f5665b5f40e1227975a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-05-08 17:19:37.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1\n\n(7 comments)\n\nA few nits inline but nothing worth holding this up, and the implementation details can be hashed out during implementation, which needs to get started.","accounts_in_message":[],"_revision_number":27},{"id":"2107726cdafb8ce9b731a816c350bfbad5bed750","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 17:19:51.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"60a1d1986e97822a38742a18c5d71f316432b061","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 17:30:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":27},{"id":"934a0376a2fb33d8217ada489030eb81da8689dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 17:30:33.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/502306/27/gate/build-openstack-sphinx-docs/3aa5694/html/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/06/502306/27/gate/openstack-tox-pep8/59fc0eb/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"7d29f3410260747eb0d4fc3d9610b491d4af398f","revisions":{"1d3d1189ba694ef5f20aa7d30d2d9b1d88d01b71":{"kind":"REWORK","_number":1,"created":"2017-09-10 09:20:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/1"}}},"commit":{"parents":[{"commit":"f49397be06404d310483f49b2c062c8b3dd70767","subject":"Uncap sphinx version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f49397be06404d310483f49b2c062c8b3dd70767"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:16:11.000000000","tz":60},"subject":"Network bandwitdh resource provider","message":"Network bandwitdh resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d3d1189ba694ef5f20aa7d30d2d9b1d88d01b71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d3d1189ba694ef5f20aa7d30d2d9b1d88d01b71"}]},"branch":"refs/heads/master"},"2c2e34517c27b1282044357c309fd9094508d3d5":{"kind":"REWORK","_number":2,"created":"2017-09-18 09:24:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/2"}}},"commit":{"parents":[{"commit":"f49397be06404d310483f49b2c062c8b3dd70767","subject":"Uncap sphinx version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f49397be06404d310483f49b2c062c8b3dd70767"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-18 09:23:04.000000000","tz":60},"subject":"Network bandwitdh resource provider","message":"Network bandwitdh resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2c2e34517c27b1282044357c309fd9094508d3d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2c2e34517c27b1282044357c309fd9094508d3d5"}]},"branch":"refs/heads/master"},"b437ef271e8523d42f39e893c194e1558669b72b":{"kind":"REWORK","_number":3,"created":"2017-10-03 14:40:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/3"}}},"commit":{"parents":[{"commit":"f49397be06404d310483f49b2c062c8b3dd70767","subject":"Uncap sphinx version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f49397be06404d310483f49b2c062c8b3dd70767"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-03 14:39:06.000000000","tz":60},"subject":"Network bandwitdh resource provider","message":"Network bandwitdh resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b437ef271e8523d42f39e893c194e1558669b72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b437ef271e8523d42f39e893c194e1558669b72b"}]},"branch":"refs/heads/master"},"ed2bfbbb460f41a4df5759bfac4b3253ffa7083a":{"kind":"REWORK","_number":4,"created":"2017-10-04 17:18:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/4"}}},"commit":{"parents":[{"commit":"f49397be06404d310483f49b2c062c8b3dd70767","subject":"Uncap sphinx version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f49397be06404d310483f49b2c062c8b3dd70767"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-04 17:15:39.000000000","tz":60},"subject":"Network bandwitdh resource provider","message":"Network bandwitdh resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed2bfbbb460f41a4df5759bfac4b3253ffa7083a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed2bfbbb460f41a4df5759bfac4b3253ffa7083a"}]},"branch":"refs/heads/master"},"086481d1a6a824965d8f30842d911cdf719a32be":{"kind":"REWORK","_number":5,"created":"2017-10-05 15:33:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/5"}}},"commit":{"parents":[{"commit":"f49397be06404d310483f49b2c062c8b3dd70767","subject":"Uncap sphinx version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f49397be06404d310483f49b2c062c8b3dd70767"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-05 15:32:04.000000000","tz":60},"subject":"Network bandwitdh resource provider","message":"Network bandwitdh resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/086481d1a6a824965d8f30842d911cdf719a32be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/086481d1a6a824965d8f30842d911cdf719a32be"}]},"branch":"refs/heads/master"},"c5f2752bd4f0dc60d2a2707377e639fdf12dc217":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2017-10-05 15:40:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/6"}}},"commit":{"parents":[{"commit":"f49397be06404d310483f49b2c062c8b3dd70767","subject":"Uncap sphinx version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f49397be06404d310483f49b2c062c8b3dd70767"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-05 15:39:54.000000000","tz":0},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c5f2752bd4f0dc60d2a2707377e639fdf12dc217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c5f2752bd4f0dc60d2a2707377e639fdf12dc217"}]},"branch":"refs/heads/master"},"2242d77a18b7a3ae221699a7075e57b568382fda":{"kind":"REWORK","_number":7,"created":"2017-10-10 09:05:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/7"}}},"commit":{"parents":[{"commit":"f49397be06404d310483f49b2c062c8b3dd70767","subject":"Uncap sphinx version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f49397be06404d310483f49b2c062c8b3dd70767"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-10 08:34:39.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2242d77a18b7a3ae221699a7075e57b568382fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2242d77a18b7a3ae221699a7075e57b568382fda"}]},"branch":"refs/heads/master"},"4f43782d8c4aadfd3a59d449a88ae7525aa9309f":{"kind":"REWORK","_number":8,"created":"2017-10-11 16:38:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/8"}}},"commit":{"parents":[{"commit":"db7819b5d5c611d6700d74329687cbd22d84a116","subject":"Merge \"Add trait support in the allocation candidates API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/db7819b5d5c611d6700d74329687cbd22d84a116"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-11 16:37:41.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f43782d8c4aadfd3a59d449a88ae7525aa9309f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f43782d8c4aadfd3a59d449a88ae7525aa9309f"}]},"branch":"refs/heads/master"},"25a73625fd083942dcc71fe845f7c0752d5ef420":{"kind":"REWORK","_number":9,"created":"2017-10-11 16:41:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/9"}}},"commit":{"parents":[{"commit":"db7819b5d5c611d6700d74329687cbd22d84a116","subject":"Merge \"Add trait support in the allocation candidates API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/db7819b5d5c611d6700d74329687cbd22d84a116"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-11 16:40:28.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/25a73625fd083942dcc71fe845f7c0752d5ef420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/25a73625fd083942dcc71fe845f7c0752d5ef420"}]},"branch":"refs/heads/master"},"8314c33346e0d1633c0935e0d88181e797d7eae3":{"kind":"NO_CHANGE","_number":10,"created":"2017-10-12 11:14:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/10"}}},"commit":{"parents":[{"commit":"db7819b5d5c611d6700d74329687cbd22d84a116","subject":"Merge \"Add trait support in the allocation candidates API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/db7819b5d5c611d6700d74329687cbd22d84a116"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-12 11:11:04.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8314c33346e0d1633c0935e0d88181e797d7eae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8314c33346e0d1633c0935e0d88181e797d7eae3"}]},"branch":"refs/heads/master"},"dbbf7163fd97a57757de2553237fcd03991e2b7f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2017-10-15 19:01:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/11"}}},"commit":{"parents":[{"commit":"3a5d54a83ad283e793a1161ad62906664262fe7e","subject":"Merge \"Reset the instance keypair while rebuilding (spec)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3a5d54a83ad283e793a1161ad62906664262fe7e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-15 19:01:50.000000000","tz":0},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dbbf7163fd97a57757de2553237fcd03991e2b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dbbf7163fd97a57757de2553237fcd03991e2b7f"}]},"branch":"refs/heads/master"},"9fd254d95b535eed045c65dcff8d0b7bd6457cc9":{"kind":"REWORK","_number":12,"created":"2017-10-16 15:30:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/12"}}},"commit":{"parents":[{"commit":"3a5d54a83ad283e793a1161ad62906664262fe7e","subject":"Merge \"Reset the instance keypair while rebuilding (spec)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3a5d54a83ad283e793a1161ad62906664262fe7e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-16 15:29:15.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9fd254d95b535eed045c65dcff8d0b7bd6457cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9fd254d95b535eed045c65dcff8d0b7bd6457cc9"}]},"branch":"refs/heads/master"},"6cda1d709e0587d499b4c33ed3ce2bc92675763a":{"kind":"REWORK","_number":13,"created":"2017-10-18 16:18:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/13"}}},"commit":{"parents":[{"commit":"3a5d54a83ad283e793a1161ad62906664262fe7e","subject":"Merge \"Reset the instance keypair while rebuilding (spec)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3a5d54a83ad283e793a1161ad62906664262fe7e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-18 16:17:00.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cda1d709e0587d499b4c33ed3ce2bc92675763a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6cda1d709e0587d499b4c33ed3ce2bc92675763a"}]},"branch":"refs/heads/master"},"0f692ac856de45cde082bcdd6452861f628b241c":{"kind":"REWORK","_number":14,"created":"2017-10-19 13:25:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/502306/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/14"}}},"commit":{"parents":[{"commit":"3a5d54a83ad283e793a1161ad62906664262fe7e","subject":"Merge \"Reset the instance keypair while rebuilding (spec)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3a5d54a83ad283e793a1161ad62906664262fe7e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-09-10 09:14:17.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"rodolfo.alonso.hernandez@intel.com","date":"2017-10-19 12:20:38.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis resource provider will model the total amount of bandwidth\navailable in a hardware device or a user define value, which will\noverwrite the value read from the compute agent driver.\nThis available bandwidth will change depending on every new port\ncreated/deleted/modified with a minimum bandwidth QoS network policy\ndefined.\n\nblueprint bandwidth-resource-provider\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0f692ac856de45cde082bcdd6452861f628b241c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0f692ac856de45cde082bcdd6452861f628b241c"}]},"branch":"refs/heads/master"},"df6ddaad629f4ef9d169eafba21cf087e32f3491":{"kind":"REWORK","_number":15,"created":"2018-03-04 14:48:30.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/15"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:47:46.000000000","tz":0},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/df6ddaad629f4ef9d169eafba21cf087e32f3491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/df6ddaad629f4ef9d169eafba21cf087e32f3491"}]},"branch":"refs/heads/master"},"8acaed70674edcb7a91a94308c3a8cd5dde2a979":{"kind":"REWORK","_number":16,"created":"2018-03-09 16:01:07.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/16"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-09 15:59:51.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8acaed70674edcb7a91a94308c3a8cd5dde2a979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8acaed70674edcb7a91a94308c3a8cd5dde2a979"}]},"branch":"refs/heads/master"},"6550dcda755c345c23b78f6fae96228d8df5c433":{"kind":"REWORK","_number":17,"created":"2018-03-09 16:18:19.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/17"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-09 16:18:11.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6550dcda755c345c23b78f6fae96228d8df5c433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6550dcda755c345c23b78f6fae96228d8df5c433"}]},"branch":"refs/heads/master"},"95a5884be69f339d943e7c65b65b7b29627d3ac3":{"kind":"REWORK","_number":18,"created":"2018-03-14 14:26:42.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/18"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-14 14:26:27.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/95a5884be69f339d943e7c65b65b7b29627d3ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/95a5884be69f339d943e7c65b65b7b29627d3ac3"}]},"branch":"refs/heads/master"},"782c3e7e5bc23e26ca3819878796856cf1844c9c":{"kind":"REWORK","_number":19,"created":"2018-03-22 18:39:25.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/19"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-22 18:38:21.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\nCo-Authored-By: Bence Romsics \u003cbence.romsics@ericsson.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/782c3e7e5bc23e26ca3819878796856cf1844c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/782c3e7e5bc23e26ca3819878796856cf1844c9c"}]},"branch":"refs/heads/master"},"714d572a073232a288f69f71458575901fca9819":{"kind":"REWORK","_number":20,"created":"2018-03-23 17:32:08.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/20"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-23 17:31:38.000000000","tz":60},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\nCo-Authored-By: Bence Romsics \u003cbence.romsics@ericsson.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/714d572a073232a288f69f71458575901fca9819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/714d572a073232a288f69f71458575901fca9819"}]},"branch":"refs/heads/master"},"3d16a2ab0ff2ada5838cb49ae7093e48925d1b2b":{"kind":"REWORK","_number":21,"created":"2018-03-27 14:30:55.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/21"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-27 14:30:47.000000000","tz":120},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\nCo-Authored-By: Bence Romsics \u003cbence.romsics@ericsson.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3d16a2ab0ff2ada5838cb49ae7093e48925d1b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3d16a2ab0ff2ada5838cb49ae7093e48925d1b2b"}]},"branch":"refs/heads/master"},"333c7ade2ae0ac44657e7c23191522e0b996d1aa":{"kind":"REWORK","_number":22,"created":"2018-03-27 18:30:53.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/22"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-27 18:30:04.000000000","tz":120},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\nCo-Authored-By: Bence Romsics \u003cbence.romsics@ericsson.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/333c7ade2ae0ac44657e7c23191522e0b996d1aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/333c7ade2ae0ac44657e7c23191522e0b996d1aa"}]},"branch":"refs/heads/master"},"57eadebff2d70d7a38ba18fd84ba59320c2d9ab0":{"kind":"REWORK","_number":23,"created":"2018-04-19 15:52:36.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/23"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-04-19 15:52:06.000000000","tz":120},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\nCo-Authored-By: Bence Romsics \u003cbence.romsics@ericsson.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/57eadebff2d70d7a38ba18fd84ba59320c2d9ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/57eadebff2d70d7a38ba18fd84ba59320c2d9ab0"}]},"branch":"refs/heads/master"},"b1b8eede78b77aa2549a3523ccd31d32a2bff2f9":{"kind":"REWORK","_number":24,"created":"2018-04-24 17:47:28.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/24"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-04-24 17:46:34.000000000","tz":120},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources in Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\nCo-Authored-By: Bence Romsics \u003cbence.romsics@ericsson.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b1b8eede78b77aa2549a3523ccd31d32a2bff2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b1b8eede78b77aa2549a3523ccd31d32a2bff2f9"}]},"branch":"refs/heads/master"},"3c2d204c572b4ffb801213cbcc9b76fafb2f239a":{"kind":"REWORK","_number":25,"created":"2018-05-02 13:43:19.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/25"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-02 13:42:33.000000000","tz":120},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources\nin Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\nCo-Authored-By: Bence Romsics \u003cbence.romsics@ericsson.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3c2d204c572b4ffb801213cbcc9b76fafb2f239a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3c2d204c572b4ffb801213cbcc9b76fafb2f239a"}]},"branch":"refs/heads/master"},"dab08add2456463d1aa5b610b041965e537a9d01":{"kind":"REWORK","_number":26,"created":"2018-05-03 15:39:33.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/26"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-03 15:39:22.000000000","tz":120},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources\nin Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\nCo-Authored-By: Bence Romsics \u003cbence.romsics@ericsson.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dab08add2456463d1aa5b610b041965e537a9d01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dab08add2456463d1aa5b610b041965e537a9d01"}]},"branch":"refs/heads/master"},"7d29f3410260747eb0d4fc3d9610b491d4af398f":{"kind":"REWORK","_number":27,"created":"2018-05-07 14:41:32.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/502306/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/06/502306/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/06/502306/27"}}},"commit":{"parents":[{"commit":"8b692c5264694c9d33f4dadfbd7a863b4acf0bd7","subject":"Merge \"Update Provider Tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8b692c5264694c9d33f4dadfbd7a863b4acf0bd7"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-04 14:04:34.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-07 14:41:24.000000000","tz":120},"subject":"Network bandwidth resource provider","message":"Network bandwidth resource provider\n\nThis spec proposes adding new resource classes representing network\nbandwidth and modeling network backends as resource providers in\nplacement. As well as adding scheduling support for the new resources\nin Nova.\n\nblueprint bandwidth-resource-provider\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003crodolfo.alonso.hernandez@intel.com\u003e\nCo-Authored-By: Bence Romsics \u003cbence.romsics@ericsson.com\u003e\n\nChange-Id: Ie7be551f4f03957ade9beb64457736f400560486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d29f3410260747eb0d4fc3d9610b491d4af398f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d29f3410260747eb0d4fc3d9610b491d4af398f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
