)]}'
{"id":"openstack%2Fnova-specs~547850","triplet_id":"openstack%2Fnova-specs~master~Ie45afa4711a7a25e5f24ca0799eee6e1533b0481","project":"openstack/nova-specs","branch":"master","topic":"bp/policy-defaults-refresh","hashtags":[],"change_id":"Ie45afa4711a7a25e5f24ca0799eee6e1533b0481","subject":"Policy Default Refresh spec","status":"MERGED","created":"2018-02-25 17:48:32.000000000","updated":"2019-07-04 14:07:46.000000000","submitted":"2019-07-02 22:22:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":237,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"547850-1562106179881-5fda7c2d","meta_rev_id":"6af54bee649ae2a9d051c9a791c78d4567174bb1","_number":547850,"virtual_id_number":547850,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"date":"2019-06-28 10:29:24.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2019-07-02 22:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"date":"2019-06-27 18:24:15.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-07-02 08:47:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-27 16:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":2,"date":"2019-07-02 21:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":1,"date":"2019-07-02 21:57:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-27 12:27:21.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2018-04-19 20:54:48.000000000","updated_by":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"reviewer":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"state":"REVIEWER"},{"updated":"2018-06-29 15:23:58.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2018-11-15 14:03:39.000000000","updated_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"reviewer":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"state":"REVIEWER"},{"updated":"2019-03-28 21:34:46.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-04-19 01:16:05.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2019-05-03 21:28:22.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-05-11 00:57:10.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-05-16 16:36:50.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-06-07 05:46:59.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-06-27 16:35:54.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-06-27 18:24:15.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-07-02 08:47:38.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-07-02 21:57:23.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-07-02 22:22:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e57709e5a770b4e8d759c9ea07f76310ab54a41","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-02-25 17:48:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25e67a15fedddb2b9bc659c18d9ecf86d34e2e89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-25 17:55:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/50/547850/1/check/build-openstack-sphinx-docs/f9c20e4/html/ : SUCCESS in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/1/check/openstack-tox-pep8/f6a70bf/ : FAILURE in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"14712aced66e6c4265a0a4d72f4da90a5a36f9af","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-02-25 18:04:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a0d129a94424440d0366729d02ec3672d79b891","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-25 18:15:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/50/547850/2/check/build-openstack-sphinx-docs/8054512/html/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/2/check/openstack-tox-pep8/442a587/ : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"3e0d9a508497ecafb56580c7eacdf5d43f574d11","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-02-26 11:04:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5d1240b727786b7eeabb47d60fe94d39be50a43f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-26 11:16:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/50/547850/3/check/build-openstack-sphinx-docs/c3c51cd/html/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/3/check/openstack-tox-pep8/bd95b1f/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ef6cd08ebceed1961621e295699ab6f33c61bde2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-03-02 13:30:36.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"26bcdb3353b54e02771c7543af423b61a0d4255b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-02 13:48:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/50/547850/4/check/build-openstack-sphinx-docs/a59e1d7/html/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/4/check/openstack-tox-pep8/1998b7a/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":4},{"id":"153a83105b381937e1977776485fc169a2941a44","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-13 13:33:29.000000000","message":"Patch Set 4: Code-Review+1\n\n(7 comments)\n\nI only have nits inline but the proposed change looks good to me.\nCheers,\ngibi","accounts_in_message":[],"_revision_number":4},{"id":"ef56956ec17784fd05dd29aa2eaaadaddf14551e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-03-26 07:54:30.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9b54153a5225eca906a3e9adc07f748e43272fc7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-03-26 07:54:39.000000000","message":"Patch Set 5:\n\nthanks done.","accounts_in_message":[],"_revision_number":5},{"id":"94c1e9880916466544616f565a214a0d5fc7223c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-26 08:02:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/50/547850/5/check/build-openstack-sphinx-docs/38d33a6/html/ : SUCCESS in 5m 23s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/5/check/openstack-tox-pep8/aca1a29/ : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":5},{"id":"cadd452b0a83c225921acce940cb702f2dc56402","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-03-27 12:27:21.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)\n\nSome nits.\n\nIn general, I think that granular policy is a pain, so it is critical that whatever this does, it does it in a way that people who are already happy or don\u0027t care don\u0027t need to do anything. The spec seems to cover that, so good.","accounts_in_message":[],"_revision_number":5},{"id":"00e18bfa745b35d1da9a1298da59de29c3dda1e3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-03-28 03:10:13.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"ec3558c62893d030ec20d02b304367e56c3d0eac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 03:19:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/50/547850/6/check/build-openstack-sphinx-docs/d156fae/html/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/6/check/openstack-tox-pep8/b820979/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":6},{"id":"0bad8a09eec48658fa0ab2cc71fdbb905dfa83f8","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2018-04-04 10:04:36.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nSo scope creep alert here, but I am thinking about only having policy churn once for the users.\n\nI think we should instead look at adding \"global\" scope vs not global and Admin vs Writer vs Reader roles. That would force us into adding this level of granularity, but actually make them usable for operators out the box.","accounts_in_message":[],"_revision_number":6},{"id":"e5e79faa664731130a0b3f3f97f143c13ed7905f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-05-31 03:02:00.000000000","message":"Patch Set 6: Workflow-1\n\n(2 comments)\n\n\u003e (1 comment)\n \u003e \n \u003e So scope creep alert here, but I am thinking about only having\n \u003e policy churn once for the users.\n \u003e \n \u003e I think we should instead look at adding \"global\" scope vs not\n \u003e global and Admin vs Writer vs Reader roles. That would force us\n \u003e into adding this level of granularity, but actually make them\n \u003e usable for operators out the box.\n\nYeah, that is good idea and better way to move forward with less overhead on operator as well as on developers. \ni will wait for default roles spec [1] before this. Let\u0027s put this as WIP till than. \n\n..1 https://review.openstack.org/#/c/245629/8","accounts_in_message":[],"_revision_number":6},{"id":"502c22f83c6759678cdbf298ceb8899da3cba1e3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-05-31 05:00:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bf17d4e92b82feb1e6760e959123e0d743bec1cc","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2018-06-29 15:23:58.000000000","message":"Patch Set 6: Code-Review-1\n\n(22 comments)\n\nI think the spec should try to pick establish role terminology so as to not confuse people.\n\nCurrently Keystone is using:\n\n- admin\n- member\n- reader\n\nMember was recently renamed to \u0027member\u0027. https://review.openstack.org/#/c/576203/\nAuditor was recently renamed to \u0027reader\u0027. https://review.openstack.org/#/c/570990/\n\nThis should help to disseminate the same default roles understanding across the community.","accounts_in_message":[],"_revision_number":6},{"id":"4c1fce9987aeb021569993ec0d0d9fe4a002591e","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2018-06-29 16:45:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"edda7899c0f5bc19257023c077ae8409f27443ad","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-07-02 01:02:03.000000000","message":"Patch Set 6:\n\n\u003e (1 comment)\n\nThat is good question. We have 2 options to make this change \n\nOption 1:  \n- Introduce the new rules with new default roles \n- Keep enforcing the old rules with same default for deprecation period.\n- Give warning for user about switching to old rules to new rules\n- Note we will enforce the old rules as it is till deprecation period and new rule as soft enforcement only as it should not effect any user for deprecation period. \n- after deprecation period. we will remove the old rules and keep enforcing the new rules only. \n\nOption 2 :\n- Introduce new rule with same default as old rules\n- warn user and ask them to switch to new rules if they have overridden the rules.\n- If old rule is overridden then enforce old rules otherwise enforce new rules. \n- after deprecation period, remove the old rule completely and keep enforcing new rules only.  \n- Then change the defaults of new rules to appropriate one.\n- Before changing the default value, warn operator about new default value change. \n- After another deprecation period, change the default of new rule. \n\nWe can compare the both way here in term of how many times we are asking operator to update their policy to consume our new rules with new defaults:\n \nOption 1: ONCE, In this way user will take care of upgrade of policy once where they will switch to new rules with new defaults. \n \nOption 2: - Twice. This way we are asking operator to make policy change twice, one to switch old rule to new rule and then switch to new default value. \n\noption 1 is better IMO.\n\nI will update this spec with those details and also what default will go with what rule.","accounts_in_message":[],"_revision_number":6},{"id":"6b2974fc6520cf28656c65d219102bd2c16f6b99","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-07-02 01:04:21.000000000","message":"Patch Set 6:\n\n\u003e (22 comments)\n \u003e \n \u003e I think the spec should try to pick establish role terminology so\n \u003e as to not confuse people.\n \u003e \n \u003e Currently Keystone is using:\n \u003e \n \u003e - admin\n \u003e - member\n \u003e - reader\n \u003e \n \u003e Member was recently renamed to \u0027member\u0027. https://review.openstack.org/#/c/576203/\n \u003e Auditor was recently renamed to \u0027reader\u0027. https://review.openstack.org/#/c/570990/\n \u003e \n \u003e This should help to disseminate the same default roles\n \u003e understanding across the community.\n\nThanks. i will take care of your comment in next PS.","accounts_in_message":[],"_revision_number":6},{"id":"9c19cc81610ae99d47b3348e0798433f5ead1277","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2018-07-09 19:20:14.000000000","message":"Patch Set 6:\n\nKeystone member rename is actually here: https://review.openstack.org/#/c/572243/\n\nI provided the wrong link before.","accounts_in_message":[],"_revision_number":6},{"id":"ccdbcc525bbb172b580684cafc482cbe26ad3cce","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-09-19 15:08:29.000000000","message":"Patch Set 6:\n\nI\u0027m all for seeing something like this! I think this has a few relatively large initiatives lumped into it though. I did start a thread last week to see if we can find a convention that we all can move towards (per operator feedback) [0].\n\nI\u0027m curious to know if you have feedback on that convention. Ideally, it would be nice to establish a convention before you start any of this work. I\u0027d hate to have something like that come down the pipe after you\u0027ve started this, only to change it again. We do have some documentation in oslo.policy for renaming policies [1].\n\n[0] http://lists.openstack.org/pipermail/openstack-dev/2018-September/134597.html\n[1] https://docs.openstack.org/oslo.policy/latest/reference/api/oslo_policy.policy.html#default-rule","accounts_in_message":[],"_revision_number":6},{"id":"d46e9c91692a5743dfb323aee7cbda8322598ed4","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2018-09-20 10:47:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"de7b0ac5400428c634bfb85e8c57e12b372d26d0","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2018-09-20 10:48:04.000000000","message":"Patch Set 7: Workflow-1\n\nAdded a new spec with some ideas in, clearly its all a WIP right now.","accounts_in_message":[],"_revision_number":7},{"id":"33c76d33de5cc22a682018071e704302d06306cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-20 13:11:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/7/check/openstack-tox-docs/204fbb2/html/ : SUCCESS in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/7/check/openstack-tox-pep8/ed9d474/ : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":7},{"id":"9595ccda56d69ba5dcc1a0e0e9f20ccec4575cf2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-09-20 13:19:40.000000000","message":"Patch Set 7:\n\n(13 comments)\n\nI\u0027m happy to see this discussion happening. I left some comments inline, but let me know if you need me to clarify anything or chip in and help out.","accounts_in_message":[],"_revision_number":7},{"id":"d8c6b7aab1d16f738b93c16ae3480123f8f9333e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2018-09-21 13:22:37.000000000","message":"Patch Set 7:\n\n(10 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9f4d56bc1f0829bb4acae7daac603eee6a989a6f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-18 02:16:17.000000000","message":"Patch Set 7:\n\n(7 comments)\n\nLet me merge the 3 updates together and have some PoC ready to reflect how these 3 updates looks like.","accounts_in_message":[],"_revision_number":7},{"id":"e52f1f9dcdff02fa22f2744025afb3b108d10954","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-19 19:36:39.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ad0939e046ce90f6a0fd81885c08ebdd853914b1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-21 03:29:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d166179a2329ff76009d07efda45c8e5bd1e3648","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-21 03:29:32.000000000","message":"Topic changed from bp/granular-api-policy to bp/policy-default-refresh","accounts_in_message":[],"_revision_number":8},{"id":"1f8823af2c50a8165e30460590ce429fcf767329","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-21 03:29:54.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"9f6c603c8fe11a49f0bec927bc81fad5cc9b59b5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-21 03:31:36.000000000","message":"Patch Set 9: Workflow-1\n\nworking on PoC to know how it will look like and migration plan of not breaking the existing defaults.","accounts_in_message":[],"_revision_number":9},{"id":"8ad10864965a3cf9e145a0e5b13536d64323b608","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 03:36:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/9/check/openstack-tox-docs/f724a54/html/ : SUCCESS in 5m 40s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/9/check/openstack-tox-pep8/4785efe/ : FAILURE in 4m 49s","accounts_in_message":[],"_revision_number":9},{"id":"12c17529e9b8d522c316d14b5f1d4aa504b2a27a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-03-21 03:40:27.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"96a3f70eb9d09d0521416b5108134da5ef48cba7","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-03-21 04:03:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f684efcac91491429410186543572ad816cde061","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-21 19:11:46.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4150822e0d675fd09aa2de2bff3d625b4d2d339a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-03-22 03:10:06.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"39f732bad782c536b81eabd9aa2f53e7162fac18","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-28 21:20:37.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4124e420171cea1a22e90c269a5def5e0856a1c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 21:31:23.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/10/check/openstack-tox-docs/76e15a2/ : FAILURE in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/10/check/openstack-tox-pep8/a788373/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":10},{"id":"c13768adcdfa305dc59de1d439cde641c31e8941","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-03-28 21:34:46.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"29ab862ae17f8443235bfd377db2f49f73d49b62","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-28 21:45:23.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"72052925473b93fb540705f031c45fba18ebaa63","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-03-29 02:13:03.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c3973933e340c9e625ac5b138fe40b74173d7986","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-29 19:09:45.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"6c972308d59914abbc9461f2ed3de85bec94d8da","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-29 19:13:59.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"97fc9712334fa10a8a26fe684ae5cb3dfd512222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 19:25:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/11/check/openstack-tox-docs/93a4e50/ : FAILURE in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/11/check/openstack-tox-pep8/d582c04/ : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":11},{"id":"1f94aec09892534ef01e7df2bb07646ae2758a0e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-29 21:44:48.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0c9cc4ef80e1ec9450e9293f9225d2a2b6c2ff4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 22:00:42.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/12/check/openstack-tox-docs/13e8ae4/ : FAILURE in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/12/check/openstack-tox-pep8/3e6cef0/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":12},{"id":"c301727e85f09e1446a24b26e0afc51b5128091e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-29 22:16:12.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"1bb0dad13baccf9de72aadab93c5757883790c35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 22:27:26.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/13/check/openstack-tox-docs/ed2b303/ : FAILURE in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/13/check/openstack-tox-pep8/e8cc534/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":13},{"id":"755c9feac867ba58130c88ca6dc0a1b06d7f0029","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-01 03:01:20.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"da9076e97d466149a3e5ae9bdba9614c92fd2002","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-01 03:05:16.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"63866152425ef5cd675a030ecf07e1f37450227b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 03:14:06.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/15/check/openstack-tox-docs/77bb633/html/ : SUCCESS in 8m 00s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/15/check/openstack-tox-pep8/54795fb/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":15},{"id":"ac13026064d2272716ac2dc2305a85837c13fea0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-04-01 14:01:09.000000000","message":"Patch Set 15:\n\n(9 comments)","accounts_in_message":[],"_revision_number":15},{"id":"c0f02f27986032de64ac3072afb358ca86a5cb4a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-02 18:59:57.000000000","message":"Patch Set 15:\n\n(10 comments)","accounts_in_message":[],"_revision_number":15},{"id":"cbe648c58024b7a460c2bcf28848dd3d6276a7c7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-02 19:18:30.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c25dc4901699f9247f1f85379819647116085d17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 19:27:51.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/16/check/openstack-tox-docs/6770220/html/ : SUCCESS in 8m 23s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/16/check/openstack-tox-pep8/04837bf/ : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":16},{"id":"d817dea2b914c800bbd242f1b746bc24cc919092","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-04-16 23:04:59.000000000","message":"Patch Set 16: Code-Review-1\n\n(6 comments)\n\nOverall, I\u0027m not able to understand how the problem description and use cases sections of this spec are related to the proposed changes described. I think they need to be fleshed out to describe what\u0027s wrong with the current state of policy today and how each piece of this spec: defaults, granular, scope types contributes to solving the problems.\n\nAlternatively, as I\u0027ve mentioned before, if this spec were broken into 3 separate specs, it would probably make the problem description and use cases sections easier to write and tie in to the proposed changes.","accounts_in_message":[],"_revision_number":16},{"id":"d422da371e3e9a61b69e3ec75e2e45ef79adfcf9","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-04-30 17:20:48.000000000","message":"Patch Set 16: Code-Review-1\n\n(3 comments)\n\nI want us to make these changes, I have attempt to described why. I also want to help with this effort.\n\nBut I agree with Mel\u0027s comments here, I think we can introduce the problems we are fixing here in a better way.","accounts_in_message":[],"_revision_number":16},{"id":"00380463630cc8e5d7ab83b81aa93526406587aa","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-04-30 17:22:18.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"102819e4957ad02bf02bd2e2491101009ffebc38","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-01 01:10:22.000000000","message":"Patch Set 16:\n\n(5 comments)","accounts_in_message":[],"_revision_number":16},{"id":"330175b8c80f7c30aefe6b2fe5cf0e5ea5b34d81","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-03 06:22:42.000000000","message":"Patch Set 16:\n\n(11 comments)","accounts_in_message":[],"_revision_number":16},{"id":"f618d1488eff2e8518628115348cd3619d8ddc1b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-03 06:35:43.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"5f477136d492965d9e13d6de9a5e293ba8a57cc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 06:42:44.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/17/check/openstack-tox-docs/7252468/html/ : SUCCESS in 6m 44s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/17/check/openstack-tox-pep8/905b9eb/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":17},{"id":"bb2de6b7efdda5b17aa6d613462440b3df0ad0b1","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-05-04 01:49:05.000000000","message":"Patch Set 17:\n\n(15 comments)\n\nNice spec, especially the policy separation between GET and POST/UPDATE/DELETE operations seems useful.","accounts_in_message":[],"_revision_number":17},{"id":"c792f4f8f74d95d626ef964654a4cc66a9f5759c","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-07 11:30:10.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"b54c9302a44e37767eed8868e265bbbe76dcf783","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-07 11:31:49.000000000","message":"Patch Set 18:\n\nUpdated after discussion at the PTG.","accounts_in_message":[],"_revision_number":18},{"id":"5777c11d4fe105909d2afcf5b3aacd30cb9c0be8","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-07 11:33:13.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"952e7ca392aadda30d0fd6b317243fb1b6247170","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 11:38:28.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/18/check/openstack-tox-docs/1b607a8/html/ : SUCCESS in 6m 58s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/18/check/openstack-tox-pep8/f1f243f/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":18},{"id":"592983d46a06fdf238169646ca84bdaaeeab7c75","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-10 23:34:42.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"ccd0c3189c7a264b399b2517eb8c930c99c13e8e","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-05-11 00:57:10.000000000","message":"Patch Set 18: Code-Review+1\n\n(8 comments)\n\nThanks for doing this, Mann san and Garbutt san.","accounts_in_message":[],"_revision_number":18},{"id":"69dac17117b6268eebb6052652b46ace3b21687c","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-13 16:17:32.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"3fd945478047732adf2023644e9ec1e6f807e968","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-13 16:20:25.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"f6226d2f49e7d182305c6a9e5af6950bd3f542c6","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-05-15 22:49:31.000000000","message":"Patch Set 18:\n\n(6 comments)\n\nInitial comments, will continue reading through.","accounts_in_message":[],"_revision_number":18},{"id":"0c0aa9c3757583e30f64eb2101a44c07d391ce7c","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-05-15 22:54:42.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"1ab4bd3211c929152b7db936c15579fc45380281","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-16 16:36:50.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"f57cc191939129e09432cd06b683fe4c8c20304a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-05-16 16:51:05.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"8d442e0b5d04f940b36b14dceecdc1a3861b6e35","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-22 15:43:54.000000000","message":"Patch Set 18:\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"c1dca541577c0e4bb383c15c784c81ef20863d24","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-27 05:49:00.000000000","message":"Patch Set 18:\n\n(7 comments)","accounts_in_message":[],"_revision_number":18},{"id":"07810a3f871b5475619b96a0d1111aca6a76d4e8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-27 06:45:29.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"53462e410265a2eaf7f05a588acdea7ae485f562","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 06:59:56.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/19/check/openstack-tox-docs/9176bdd/html/ : SUCCESS in 6m 59s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/19/check/openstack-tox-pep8/0c64097/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":19},{"id":"8cc8b1835167d3e1ceb9b69c240f274aeaaa4d9a","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-29 09:31:02.000000000","message":"Patch Set 19: Code-Review-1\n\n(5 comments)\n\nSo this feels close now... but I had some throughts on the details after trying out my own POC:\nhttps://review.opendev.org/#/q/topic:bp/policy-defaults-refresh+(status:open+OR+status:merged)\n\nI think we need to be more explcit about what the scope_types and rules will be and what the transition plan will be. I would have said this is implementation detail, but there are some gotcahs in there because all our APIs are admin_or_owner... so we need scope checks in both the rule_str and the scope_types for the majority of APIs.\n\nI also think we need to explicity call out correcting the target on all the policy calls. I might try work through this in code to get my head straight on the ideas, but I have tried to write out my current thinking.","accounts_in_message":[],"_revision_number":19},{"id":"1904dfcaecf35e9c40d24279315e9e1c4664812f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-30 14:27:38.000000000","message":"Patch Set 19:\n\n(6 comments)\n\nThanks John. Let me add the mapping of check_str etc.","accounts_in_message":[],"_revision_number":19},{"id":"ff1249826f7a81310bdfd62623f1a51b10e0e837","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-04 09:42:37.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"45ea04aacf4bbba7a1696dcac8dac044cb66ff99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 10:06:41.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/20/check/openstack-tox-docs/f9fb68b/html/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/20/check/openstack-tox-pep8/fcad683/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":20},{"id":"a2a784c4b3e905a27fc5c20a30ea0f581f25ea25","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-04 10:23:49.000000000","message":"Patch Set 20: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"67e65419f90377af2463f41850f81744121683d2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-04 11:24:21.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"8427acdb4b4da2adaa92dc4ded00fb60cd5178f2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-04 11:34:57.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"a39b627b8768fbf81c8afb9d07098b66ed976d62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 11:47:08.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/21/check/openstack-tox-docs/9e3b5a8/html/ : SUCCESS in 6m 35s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/21/check/openstack-tox-pep8/2425901/ : FAILURE in 4m 46s","accounts_in_message":[],"_revision_number":21},{"id":"e2da4d8b6642692ebc2c882c8fe3be5787633744","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-04 15:05:41.000000000","message":"Patch Set 21:\n\n(12 comments)","accounts_in_message":[],"_revision_number":21},{"id":"1a7a0b309f276898ada1ffb846417a873318d2f1","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-06 16:17:05.000000000","message":"Patch Set 21:\n\nAs a heads up, been working on this POC series to try out the ideas I am trying to describe here:\nhttps://review.opendev.org/#/c/663717\n\nIt seems like a little bit of testing goes a long way into debugging things. Feeling better about the coverage we can add in the functional tests. We start to see the 404 vs 403 error differences due to the way things currently work.","accounts_in_message":[],"_revision_number":21},{"id":"a92da233f615d254e090e9edea4e3413eee56b74","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-09 06:24:12.000000000","message":"Patch Set 21:\n\n(11 comments)","accounts_in_message":[],"_revision_number":21},{"id":"387a607a8dab3daec2d4177e630c3a7e51a8e02d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-09 07:05:46.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d2838c3a567d0ffd349f562939f59a8fef328f8d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-09 07:48:36.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"0cdf612a5b1ef45c59f2d9fd6d83d0ff50f521d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 07:58:45.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/22/check/openstack-tox-docs/7525a9d/html/ : SUCCESS in 7m 28s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/22/check/openstack-tox-pep8/0d70f4e/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":22},{"id":"7dbb55cdbcb55cc78487abf2978f198a08b52b59","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-27 16:35:54.000000000","message":"Patch Set 22: Code-Review+2\n\n(2 comments)\n\nI am happy with the way forward here. The details are quite... hairy, but I honestly think we agree enough about the approch here to move forward.\n\nI am read this too many times to be sure this gives enough context to other readers of the spec, but any other review will be checking that by default I guess.","accounts_in_message":[],"_revision_number":22},{"id":"252e7ad148d4bf3b963738b88e94bbcd35c9c5ca","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-27 16:38:40.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"01ae45e01cf5f98775d52385c5688fa7afa3b831","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-27 18:24:15.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"098d772d534f02db08cc1eecb2a1c00e29799ca1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-28 10:29:24.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"296427c2b82cd353a489247437273f72c35dfc01","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-07-02 08:47:38.000000000","message":"Patch Set 22: Code-Review+1\n\nWell detailed spec. Looks good to me but I don\u0027t have enough API expertise to be able to +2 it.","accounts_in_message":[],"_revision_number":22},{"id":"6bad5c0c608d1f8a63a9cab0da3a26c89d270154","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-07-02 21:57:23.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1\n\n(9 comments)\n\nThis aligns with the discussions we had at the PTG and nicely charts the course for modernizing our APIs to leverage keystone scope types and default roles.\n\nI think this spec is ready to go. It\u0027s full of detail, use cases, and walks through each change that will be carried out with the \u0027what\u0027, \u0027why\u0027, and \u0027how\u0027. A follow up will be provided by gmann to provide additional clarifications and I think we\u0027re ready to go forward with adding test coverage and implementation, and will work out any potential unforeseen details at that time.","accounts_in_message":[],"_revision_number":22},{"id":"ff0a1ca6596fdd6d6ff1e852541144aa02a09fdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 21:57:35.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"f01e8037563987ee4b97086a5f3f29cfc274df9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 22:22:59.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/547850/22/gate/openstack-tox-docs/b7e5ce1/html/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/50/547850/22/gate/openstack-tox-pep8/1e82ed9/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":22},{"id":"cceb8b01097af329200970ac97067be5dea2ba6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 22:22:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"749f1db4ebc3d07b3dd6164306277c1999720e8a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-04 14:07:46.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"4c2e45430952a47643e58e829faf5c3bd5ae5e5d","revisions":{"329762413bc8778f72aa2f410eb24be1ebd2ecfe":{"kind":"REWORK","_number":1,"created":"2018-02-25 17:48:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/1"}}},"commit":{"parents":[{"commit":"8e732868e79efb4e80b42a6976eb7179eb856219","subject":"Merge \"Re-propose use Neutron\u0027s new port binding API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e732868e79efb4e80b42a6976eb7179eb856219"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:48:30.000000000","tz":0},"subject":"Spec for Granular API policy","message":"Spec for Granular API policy\n\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/329762413bc8778f72aa2f410eb24be1ebd2ecfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/329762413bc8778f72aa2f410eb24be1ebd2ecfe"}]},"branch":"refs/heads/master"},"d9a6c310102990ee31dc50e4f718898fa4bc5007":{"kind":"REWORK","_number":2,"created":"2018-02-25 18:04:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/2"}}},"commit":{"parents":[{"commit":"8e732868e79efb4e80b42a6976eb7179eb856219","subject":"Merge \"Re-propose use Neutron\u0027s new port binding API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e732868e79efb4e80b42a6976eb7179eb856219"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"ghanshyam","email":"ghanshyammann@gmail.com","date":"2018-02-25 18:04:51.000000000","tz":180},"subject":"Spec for Granular API policy","message":"Spec for Granular API policy\n\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d9a6c310102990ee31dc50e4f718898fa4bc5007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d9a6c310102990ee31dc50e4f718898fa4bc5007"}]},"branch":"refs/heads/master"},"4ee62d33d6c9dd932ca630185a34878487809e38":{"kind":"REWORK","_number":3,"created":"2018-02-26 11:04:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/3"}}},"commit":{"parents":[{"commit":"8e732868e79efb4e80b42a6976eb7179eb856219","subject":"Merge \"Re-propose use Neutron\u0027s new port binding API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e732868e79efb4e80b42a6976eb7179eb856219"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"ghanshyam","email":"ghanshyammann@gmail.com","date":"2018-02-26 11:04:42.000000000","tz":180},"subject":"Spec for Granular API policy","message":"Spec for Granular API policy\n\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ee62d33d6c9dd932ca630185a34878487809e38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ee62d33d6c9dd932ca630185a34878487809e38"}]},"branch":"refs/heads/master"},"cb7137993073a369748a5d25ca5edf989ce2ec87":{"kind":"REWORK","_number":4,"created":"2018-03-02 13:30:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/4"}}},"commit":{"parents":[{"commit":"8e732868e79efb4e80b42a6976eb7179eb856219","subject":"Merge \"Re-propose use Neutron\u0027s new port binding API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e732868e79efb4e80b42a6976eb7179eb856219"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2018-03-02 13:30:33.000000000","tz":0},"subject":"Spec for Granular API policy","message":"Spec for Granular API policy\n\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cb7137993073a369748a5d25ca5edf989ce2ec87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cb7137993073a369748a5d25ca5edf989ce2ec87"}]},"branch":"refs/heads/master"},"ab5488ad4a31e081e06c75cf503268dc412ea5d4":{"kind":"REWORK","_number":5,"created":"2018-03-26 07:54:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/5"}}},"commit":{"parents":[{"commit":"8e732868e79efb4e80b42a6976eb7179eb856219","subject":"Merge \"Re-propose use Neutron\u0027s new port binding API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e732868e79efb4e80b42a6976eb7179eb856219"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2018-03-26 07:54:18.000000000","tz":0},"subject":"Spec for Granular API policy","message":"Spec for Granular API policy\n\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ab5488ad4a31e081e06c75cf503268dc412ea5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ab5488ad4a31e081e06c75cf503268dc412ea5d4"}]},"branch":"refs/heads/master"},"f7fd3225541092bc3c47eee6b1da7aadd793b5dd":{"kind":"REWORK","_number":6,"created":"2018-03-28 03:10:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/6"}}},"commit":{"parents":[{"commit":"8e732868e79efb4e80b42a6976eb7179eb856219","subject":"Merge \"Re-propose use Neutron\u0027s new port binding API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8e732868e79efb4e80b42a6976eb7179eb856219"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2018-03-28 03:09:22.000000000","tz":0},"subject":"Spec for Granular API policy","message":"Spec for Granular API policy\n\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7fd3225541092bc3c47eee6b1da7aadd793b5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7fd3225541092bc3c47eee6b1da7aadd793b5dd"}]},"branch":"refs/heads/master"},"baec6c562543a70964a25bc3ecb62ce74fb07a85":{"kind":"REWORK","_number":7,"created":"2018-09-20 10:47:31.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/50/547850/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/7"}}},"commit":{"parents":[{"commit":"f9ceb9c3c342960d747dcbe06f9dc2b9f148a611","subject":"Merge \"Add live-migration-force-after-timeout-stein spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9ceb9c3c342960d747dcbe06f9dc2b9f148a611"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2018-09-20 10:46:39.000000000","tz":60},"subject":"WIP: Spec for API policy updates","message":"WIP: Spec for API policy updates\n\nPartial implement blueprint granular-api-policy and\nblueprint policy-default-refresh which increases the scope\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/baec6c562543a70964a25bc3ecb62ce74fb07a85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/baec6c562543a70964a25bc3ecb62ce74fb07a85"}]},"branch":"refs/heads/master"},"0ffba698a93e94e3e9b7f282feb5f85ca05829bb":{"kind":"REWORK","_number":8,"created":"2019-03-21 03:29:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/8"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-03-21 03:28:03.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement policy-default-refresh\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0ffba698a93e94e3e9b7f282feb5f85ca05829bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0ffba698a93e94e3e9b7f282feb5f85ca05829bb"}]},"branch":"refs/heads/master"},"f82a7f8d1f284afd25c0d71ed693301f5c8612c9":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-03-21 03:29:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/9"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-03-21 03:29:51.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement blueprint policy-default-refresh\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f82a7f8d1f284afd25c0d71ed693301f5c8612c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f82a7f8d1f284afd25c0d71ed693301f5c8612c9"}]},"branch":"refs/heads/master"},"308cc5c2a4b9b76cf580f9d36f9af9778513ae5a":{"kind":"REWORK","_number":10,"created":"2019-03-28 21:20:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/10"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-03-28 21:20:22.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement blueprint policy-default-refresh\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/308cc5c2a4b9b76cf580f9d36f9af9778513ae5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/308cc5c2a4b9b76cf580f9d36f9af9778513ae5a"}]},"branch":"refs/heads/master"},"8d6621a63a17af57793dd47f0af07a7eb74e5e48":{"kind":"REWORK","_number":11,"created":"2019-03-29 19:13:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/11"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-03-29 19:13:56.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement blueprint policy-default-refresh\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d6621a63a17af57793dd47f0af07a7eb74e5e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8d6621a63a17af57793dd47f0af07a7eb74e5e48"}]},"branch":"refs/heads/master"},"33bd5779b32575eb782ca583d8dbc80cb6ab4982":{"kind":"REWORK","_number":12,"created":"2019-03-29 21:44:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/12"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-03-29 21:44:36.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement blueprint policy-default-refresh\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33bd5779b32575eb782ca583d8dbc80cb6ab4982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33bd5779b32575eb782ca583d8dbc80cb6ab4982"}]},"branch":"refs/heads/master"},"492499e54410d305b6849dd45683cf9b8139fd70":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-03-29 22:16:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/13"}}},"commit":{"parents":[{"commit":"32cf20135bbc9c3e3df0da21a12a4f79404c73a0","subject":"docs: Resolve issues with Sphinx 2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/32cf20135bbc9c3e3df0da21a12a4f79404c73a0"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-03-29 22:16:12.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement blueprint policy-default-refresh\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/492499e54410d305b6849dd45683cf9b8139fd70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/492499e54410d305b6849dd45683cf9b8139fd70"}]},"branch":"refs/heads/master"},"765ff792758ae861a1f030f3c5b16f7cc6f283dc":{"kind":"REWORK","_number":14,"created":"2019-04-01 03:01:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/14"}}},"commit":{"parents":[{"commit":"32cf20135bbc9c3e3df0da21a12a4f79404c73a0","subject":"docs: Resolve issues with Sphinx 2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/32cf20135bbc9c3e3df0da21a12a4f79404c73a0"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-04-01 03:01:04.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement blueprint policy-default-refresh\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/765ff792758ae861a1f030f3c5b16f7cc6f283dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/765ff792758ae861a1f030f3c5b16f7cc6f283dc"}]},"branch":"refs/heads/master"},"2b72ce3eaca0b5d7890af1f9b00badfb7ead3b8a":{"kind":"REWORK","_number":15,"created":"2019-04-01 03:05:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/15"}}},"commit":{"parents":[{"commit":"32cf20135bbc9c3e3df0da21a12a4f79404c73a0","subject":"docs: Resolve issues with Sphinx 2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/32cf20135bbc9c3e3df0da21a12a4f79404c73a0"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-01 03:05:14.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement blueprint policy-default-refresh\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b72ce3eaca0b5d7890af1f9b00badfb7ead3b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b72ce3eaca0b5d7890af1f9b00badfb7ead3b8a"}]},"branch":"refs/heads/master"},"cfae970808bc81c2d86be1f297dfc20d08a1a61c":{"kind":"REWORK","_number":16,"created":"2019-04-02 19:18:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/16"}}},"commit":{"parents":[{"commit":"32cf20135bbc9c3e3df0da21a12a4f79404c73a0","subject":"docs: Resolve issues with Sphinx 2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/32cf20135bbc9c3e3df0da21a12a4f79404c73a0"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-04-02 19:17:59.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement blueprint policy-default-refresh\nPartial implement blueprint granular-api-policy\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cfae970808bc81c2d86be1f297dfc20d08a1a61c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cfae970808bc81c2d86be1f297dfc20d08a1a61c"}]},"branch":"refs/heads/master"},"6d22cb3a5c812e5d624aa7b1215a3f10ce238ba3":{"kind":"REWORK","_number":17,"created":"2019-05-03 06:35:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/17"}}},"commit":{"parents":[{"commit":"14f84bf8632728ae63d0bc7d24e37bb2cf4c6272","subject":"Abandon flavor-classes backlog spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14f84bf8632728ae63d0bc7d24e37bb2cf4c6272"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-05-03 06:28:14.000000000","tz":0},"subject":"Spec for API policy updates","message":"Spec for API policy updates\n\nPartial implement blueprint policy-defaults-refresh\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6d22cb3a5c812e5d624aa7b1215a3f10ce238ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6d22cb3a5c812e5d624aa7b1215a3f10ce238ba3"}]},"branch":"refs/heads/master"},"7357ed2325df8439794fe8b6b7ff96b8d04cd0b3":{"kind":"REWORK","_number":18,"created":"2019-05-07 11:30:10.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/50/547850/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/18"}}},"commit":{"parents":[{"commit":"14f84bf8632728ae63d0bc7d24e37bb2cf4c6272","subject":"Abandon flavor-classes backlog spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14f84bf8632728ae63d0bc7d24e37bb2cf4c6272"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-05-07 11:29:04.000000000","tz":60},"subject":"Policy Default Refresh spec","message":"Policy Default Refresh spec\n\nSupport new roles of System Scoped Admin, Project Scoped Member,\nSystem Scoped Reader and Project Scoped Reader.\n\nPartial implement blueprint policy-defaults-refresh\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7357ed2325df8439794fe8b6b7ff96b8d04cd0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7357ed2325df8439794fe8b6b7ff96b8d04cd0b3"}]},"branch":"refs/heads/master"},"41498e81feff9fd53277e5b944447255034d4ba2":{"kind":"REWORK","_number":19,"created":"2019-05-27 06:45:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/19"}}},"commit":{"parents":[{"commit":"14f84bf8632728ae63d0bc7d24e37bb2cf4c6272","subject":"Abandon flavor-classes backlog spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14f84bf8632728ae63d0bc7d24e37bb2cf4c6272"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-05-27 06:45:17.000000000","tz":0},"subject":"Policy Default Refresh spec","message":"Policy Default Refresh spec\n\nSupport new roles of System Scoped Admin, Project Scoped Member,\nSystem Scoped Reader and Project Scoped Reader.\n\nPartial implement blueprint policy-defaults-refresh\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/41498e81feff9fd53277e5b944447255034d4ba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/41498e81feff9fd53277e5b944447255034d4ba2"}]},"branch":"refs/heads/master"},"d298e4ba480024963f92345d3481e14637a365e6":{"kind":"REWORK","_number":20,"created":"2019-06-04 09:42:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/20"}}},"commit":{"parents":[{"commit":"14f84bf8632728ae63d0bc7d24e37bb2cf4c6272","subject":"Abandon flavor-classes backlog spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14f84bf8632728ae63d0bc7d24e37bb2cf4c6272"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-06-04 09:42:35.000000000","tz":0},"subject":"Policy Default Refresh spec","message":"Policy Default Refresh spec\n\nSupport new roles of System Scoped Admin, Project Scoped Member,\nSystem Scoped Reader and Project Scoped Reader.\n\nPartial implement blueprint policy-defaults-refresh\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d298e4ba480024963f92345d3481e14637a365e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d298e4ba480024963f92345d3481e14637a365e6"}]},"branch":"refs/heads/master"},"e66766e1ba7ba71d44208cb103c0ef46abe2b335":{"kind":"REWORK","_number":21,"created":"2019-06-04 11:34:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/21"}}},"commit":{"parents":[{"commit":"14f84bf8632728ae63d0bc7d24e37bb2cf4c6272","subject":"Abandon flavor-classes backlog spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14f84bf8632728ae63d0bc7d24e37bb2cf4c6272"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-06-04 11:34:54.000000000","tz":0},"subject":"Policy Default Refresh spec","message":"Policy Default Refresh spec\n\nSupport new roles of System Scoped Admin, Project Scoped Member,\nSystem Scoped Reader and Project Scoped Reader.\n\nPartial implement blueprint policy-defaults-refresh\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e66766e1ba7ba71d44208cb103c0ef46abe2b335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e66766e1ba7ba71d44208cb103c0ef46abe2b335"}]},"branch":"refs/heads/master"},"4c2e45430952a47643e58e829faf5c3bd5ae5e5d":{"kind":"REWORK","_number":22,"created":"2019-06-09 07:48:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/547850/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/50/547850/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/50/547850/22"}}},"commit":{"parents":[{"commit":"37e4368601df54e4b01372b9919b65c807b3038c","subject":"Merge \"Amend count-quota-usage-from-placement to reflect implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/37e4368601df54e4b01372b9919b65c807b3038c"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-02-25 17:45:53.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-06-09 07:48:15.000000000","tz":0},"subject":"Policy Default Refresh spec","message":"Policy Default Refresh spec\n\nSupport new roles of System Scoped Admin, Project Scoped Member,\nSystem Scoped Reader and Project Scoped Reader.\n\nPartial implement blueprint policy-defaults-refresh\n\nAPIImpact\n\nChange-Id: Ie45afa4711a7a25e5f24ca0799eee6e1533b0481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4c2e45430952a47643e58e829faf5c3bd5ae5e5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4c2e45430952a47643e58e829faf5c3bd5ae5e5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
