)]}'
{"id":"openstack%2Fnova-specs~552924","triplet_id":"openstack%2Fnova-specs~master~I0c804743db77da5717c9c37e3c5ba57a9a3950ad","project":"openstack/nova-specs","branch":"master","topic":"bp/numa-topology-with-rps","hashtags":[],"change_id":"I0c804743db77da5717c9c37e3c5ba57a9a3950ad","subject":"Proposes NUMA topology with RPs","status":"MERGED","created":"2018-03-14 15:24:28.000000000","updated":"2020-02-17 10:38:45.000000000","submitted":"2020-02-15 10:25:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":629,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"552924-1581762312593-0cc33de8","meta_rev_id":"57a4fd0aed6f45146756398fe58c7f0f223053a2","_number":552924,"virtual_id_number":552924,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-15 10:25:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2020-02-15 10:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2020-02-14 17:21:35.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2020-02-15 10:13:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-27 02:22:14.000000000","updated_by":{"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},"reviewer":{"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},"state":"REVIEWER"},{"updated":"2018-03-27 12:51:28.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2018-03-29 05:29:35.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2018-04-25 14:24:38.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-09-12 21:16:43.000000000","updated_by":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"reviewer":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"state":"REVIEWER"},{"updated":"2018-10-02 15:08:27.000000000","updated_by":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"reviewer":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"state":"REVIEWER"},{"updated":"2019-04-16 15:11:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-04-21 16:46:30.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2019-06-24 15:16:09.000000000","updated_by":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"reviewer":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"state":"REVIEWER"},{"updated":"2019-07-11 12:17:48.000000000","updated_by":{"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},"reviewer":{"_account_id":30001,"name":"Ivens Zambrano","email":"ivens.zambrano@intel.com","username":"IvensZambrano"},"state":"REVIEWER"},{"updated":"2019-07-29 01:10:44.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-02-10 20:50:41.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2020-02-12 14:54:19.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-02-13 10:05:07.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-14 12:45:21.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-02-14 17:21:35.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-02-15 10:13:17.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-02-15 10:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8842b658eec94f870ade89fed783ddd65ea4f1dc","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-14 15:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"566a7a6c4d567693a7fdb776061b00ad9caf820d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 15:31:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/24/552924/1/check/build-openstack-sphinx-docs/4e75576/ : FAILURE in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/1/check/openstack-tox-pep8/4bdedf0/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"03dc1da45c3e77c8d00ac633ef36f790b62535b9","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-14 15:35:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a503244086c4d859b52b46c401e298ffb7f5aa6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-14 15:36:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab773a56221665950e66cd8a69e6caf440463c94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 15:49:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/24/552924/3/check/build-openstack-sphinx-docs/ec4c4d7/html/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/3/check/openstack-tox-pep8/b592bb5/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":3},{"id":"6a887ebdc8ef2edc99f5bc47108dd455e137b74d","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-14 16:44:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)\n\nDefinitely like the idea, but I think there are some issues that need to be addressed.","accounts_in_message":[],"_revision_number":3},{"id":"c05ff8932da7d6d9018ca28a3e4f1679ae81d299","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-14 19:21:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c7532b12f6216894eb3e397b96ec8b5af9b3baa6","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-14 21:45:58.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1b14c072fa3688ca049fb4ce3ee16ec72f9be116","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-03-15 10:13:29.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"184452f6c40b2794bfc20125fe7b7aabbb140b2e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-03-15 13:47:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5122bd5c205a09534dddb4870961acb1296a03f9","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-03-16 04:14:59.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d5cecfa575985cccb7ae15a476e4c04beffd79b4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-03-16 09:29:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fdb0767e4dfae20269b2dc596ca9af34d4a9bcee","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-16 17:20:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ab710323c6cd9361242769eaec83df59d5997692","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-19 22:08:41.000000000","message":"Patch Set 3: Code-Review-1\n\n(12 comments)\n\nSorry, maybe I\u0027m being dense, but I don\u0027t understand how this spec describes a way for me to request resources from the same NUMA node, when they\u0027re spread across multiple providers, without marking the node with a specific trait (NUMA_NODE_1) and requesting that trait specifically.\n\nAt the PTG, we discussed something like:\n\n- Mark a NUMA node with some trait (e.g. NUMA_NODE)\n- Add placement syntax to allow me to express that I want all the resources in my request group to come from the same subtree under a provider with a given trait (e.g. GET /allocation_candidates?resources4\u003dNUMA_CORE:2,MEMORY_MB:2048,VGPU:1\u0026subtree_traits4\u003dNUMA_NODE)\n- Mirror that syntax to flavor extra_specs","accounts_in_message":[],"_revision_number":3},{"id":"aaaf5f80629350d75b37edf209f4d9587c445b2f","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-20 19:53:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0fc45ce5eb5e55eb445a923a768ffa6509bf8399","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-03-23 17:00:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e588f442eb4d1616c7f960ceaba143cc816b5bfe","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-26 09:19:47.000000000","message":"Patch Set 3: Workflow-1\n\nThanks for all the feedback. Writing on another rev.","accounts_in_message":[],"_revision_number":3},{"id":"3d37312d8a6acf4216496e0a38599c8360597e86","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-26 21:25:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ffcb58f5e885b3b65e8d80aaa0d77d39a348aab4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-26 21:35:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/24/552924/4/check/build-openstack-sphinx-docs/513cf85/ : FAILURE in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/4/check/openstack-tox-pep8/4809d1e/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":4},{"id":"a7ba427b032163bfdf16eb185e3f2bbdfb19ae91","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-27 08:34:40.000000000","message":"Patch Set 4: Workflow-1\n\njust fixing the doc issues.","accounts_in_message":[],"_revision_number":4},{"id":"356bb73923b5f46d00df8d08ba79167d0bebb869","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-27 09:10:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"462f9253c1f1dfe17e5cb06f70b2ef36f04345fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-27 09:20:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/24/552924/5/check/build-openstack-sphinx-docs/a0d817f/html/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/5/check/openstack-tox-pep8/39de6f6/ : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":5},{"id":"754bf14dd7f054a2633f0658a15781d8118eb89e","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-03-27 12:51:28.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nLooks good thanks Sylvain","accounts_in_message":[],"_revision_number":5},{"id":"43f0dae01e45f53208e1f0b7885e33f65d25b587","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-27 13:16:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"230b350027f92be7c7efef133260c262b298d042","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-27 13:29:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/24/552924/6/check/build-openstack-sphinx-docs/684ab29/html/ : SUCCESS in 5m 19s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/6/check/openstack-tox-pep8/b88a02c/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":6},{"id":"4706c31e14ea699ab8eaf48faed3321cf48ab4e1","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-27 17:11:28.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nSome more discussion of hugepages.","accounts_in_message":[],"_revision_number":6},{"id":"6560946dd1ac64a75a8d8cdb2d6f0cbbba4c3c90","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-03-29 05:29:35.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8b80c67c1ad11c87bb45a470535c286df9ca9ad4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-03-29 15:33:20.000000000","message":"Patch Set 6:\n\n(12 comments)","accounts_in_message":[],"_revision_number":6},{"id":"73746da042011877ce1940482c1cfa5448ea4376","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-29 19:54:04.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"27b2d7b3a354b3b16c2b913739d7be504f4aeeb1","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-29 19:58:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d84bc35430517f4e0c69b73d44bb5bcac0603872","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-29 21:04:56.000000000","message":"Patch Set 6: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"452750add70a4a093c92a17975016e4bdb96f786","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-04-04 01:37:05.000000000","message":"Patch Set 6:\n\n(20 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fdbb530b1f35d535f5eb24dc4feefb69c1d5b062","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-04 13:33:11.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"9e646d246b4b5b47f2eacfa3a2039b982ccfadc8","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-04-04 16:55:01.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fd78700705e602ea40f88e0247a7d09295be8697","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-04-04 17:16:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"93bfbc787ef432454a428520093785aefb79b025","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-04-10 09:54:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cd64d2eeb9e405f00ee2bf9a5f0552eea31d4f32","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-04-10 14:12:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"99e34df492a2bc5714718b50c5abbb6967ff004c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 14:26:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/24/552924/7/check/build-openstack-sphinx-docs/bb74016/html/ : SUCCESS in 6m 30s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/7/check/openstack-tox-pep8/512a4ee/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":7},{"id":"57f0d1d0c644271d80eb3583ffc9dc0292dc120e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-04-10 14:36:24.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"94c5880954d454f7adf56795858a8dfc60765eb9","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-10 15:15:59.000000000","message":"Patch Set 7: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9935b7ced5b3ce273ab14da6672aff153032c669","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-04-10 16:13:30.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4be5dab69ff69056474bcec79eec8aef45f39517","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-11 04:00:58.000000000","message":"Patch Set 7: Code-Review-1\n\n(17 comments)\n\nAs I see it, there are four use cases you *might* be wanting to cover, and I\u0027m not completely clear which ones this spec is addressing.\n\n- Landing all of an instance\u0027s VCPU and/or MEMORY_MB on the same NUMA node.  This we can do with the modeling you\u0027ve suggested by requesting the resources in the same numbered request group.\n- Landing VCPU/MEMORY_MB *and* some device (VGPU, VF, etc.) on the same NUMA node.  This we can NOT do without further invention when the resources are in separate providers.  We would need some concept of \"same subtree\".\n- Explicit sharding across more than one NUMA node.  E.g. I want *exactly* X VCPUs on one NUMA node and Y VCPUs on a *different* one.  This we can\u0027t do via placement without adding the concept of \"force resources1 and resources2 to land on *separate* providers\".  We could do it post-placement (e.g. in the NUMATopologyFilter or elsewhere in the scheduler) by filtering out the allocation candidates that aren\u0027t appropriately sharded.\n- Maximizing the chance of my instance landing *somewhere* when I don\u0027t care about NUMA affinity, and I don\u0027t mind whether it lands on a host that\u0027s set up for NUMA or not.  This we can do under the modeling you\u0027ve suggested (but non-NUMA hosts still modeled with all their VCPU in the root CN RP) by translating VCPU\u003dN into resources1\u003dVCPU:1...resourcesN\u003dVCPU:1.","accounts_in_message":[],"_revision_number":7},{"id":"d46d1913f7e9e70dc6a411041f29e8057103d208","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-04-11 15:15:09.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e77cfec188cf4e59fc667e2b18a645ede43a119d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-04-11 15:15:21.000000000","message":"Patch Set 7: Workflow-1 -Code-Review\n\nshit, wrong -W","accounts_in_message":[],"_revision_number":7},{"id":"000611bc5e085a52207d6fbac45b2bc50560b9e7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-04-16 13:19:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9c0c16dda7e10712209920196e530f63cf832dbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-16 13:26:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/24/552924/8/check/build-openstack-sphinx-docs/5bd7a91/ : FAILURE in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/8/check/openstack-tox-pep8/24075b1/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":8},{"id":"097d26867a1256b52707df1247ab2c591aeefef6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-04-16 13:31:06.000000000","message":"Patch Set 7:\n\n(16 comments)","accounts_in_message":[],"_revision_number":7},{"id":"74f9b9d9890940ca1ab5abb271d5abba087f8071","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-04-18 13:46:11.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b7d0e74e4348c8683c19f8b0bec477f9cc525796","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-18 14:01:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/24/552924/9/check/build-openstack-sphinx-docs/20c2cd9/html/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/9/check/openstack-tox-pep8/edfe8cd/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":9},{"id":"1f662c4003dc9c736949b3a27ec70f80a0c7584d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-18 14:03:25.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)\n\nInline, plus still some unaddressed comments from PS7.","accounts_in_message":[],"_revision_number":9},{"id":"4d4a4351693457ecf166af6fd3458e81a33bd590","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-04-18 17:08:54.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)\n\nThere\u0027s some good stuff here, but I\u0027m not fully comprehending what is actually being proposed other than: it would be nice to use nested providers for numa, and there should be a config item that controls which resources classes are to be considered as candidate numa inventory.\n\nPretty much \"yes\" to those, with the caveat that the stuff around line 311 about numa discovery being _not_ a virt driver thing sounds very wrong.","accounts_in_message":[],"_revision_number":9},{"id":"bbbb3063e351af31bf7bf53d4aa757b9c27b3025","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-04-25 14:24:38.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7ba6c0b3e61fde00b49cb337a0e06eb07bb4ac50","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-05-18 09:08:13.000000000","message":"Patch Set 9: Workflow-1\n\nWorking on a new revision","accounts_in_message":[],"_revision_number":9},{"id":"ed0dd400db85b474b33b0bc2ef1fb3f7cd6c290a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-05-18 09:10:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"de91861bdd0bd722270aba44439cec9963614b3f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-09-08 21:47:06.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"2a76d2ca50b897ece0edd456ee5c542971fe2d02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-08 21:53:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/24/552924/9/check/openstack-tox-docs/ba26cb6/html/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/9/check/openstack-tox-pep8/e86b4cb/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":9},{"id":"3fb2e7a37ef401f4dd3664e1b7008f155f487b62","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-09-11 20:09:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7fd7de03aa9dfec928acdfe94d470e54f74f4f43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-11 20:15:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/24/552924/10/check/openstack-tox-docs/819445e/html/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/10/check/openstack-tox-pep8/30113c8/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":10},{"id":"fc95d3adfb598e7e5019655f469093290611a2b1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-09-11 20:37:03.000000000","message":"Patch Set 10:\n\n(5 comments)\n\nsome nits inline but over all this looks reasonable to me.","accounts_in_message":[],"_revision_number":10},{"id":"9f40fda6a7e5f8e87d1555037d62a81192c34eb8","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-11 22:58:10.000000000","message":"Patch Set 10: Code-Review+1\n\nI\u0027m +1 overall...if you have to respin you could fix Sean\u0027s terminology nits.","accounts_in_message":[],"_revision_number":10},{"id":"4af83c6a1784608158b89ad595732e9b591fe81c","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-09-12 19:59:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c3b9e6f939d284daa8d222eae2b2a22adcd5f59d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-09-12 20:14:03.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b9c06dba87800858f8daeb30cff2e67f5e57e1b0","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-09-12 20:38:23.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e68229abd357f667327d1215a3cb8544cdafaf15","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-09-12 21:30:39.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"15e10edf3cff7498e90bb2dc7743ec622635ba96","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-12 22:58:56.000000000","message":"Patch Set 10: -Code-Review\n\n(1 comment)\n\nremoving +1 based on Sylvain\u0027s comments.  Needs minor tweaks.","accounts_in_message":[],"_revision_number":10},{"id":"818b172a26998448150d02297040f787f0696376","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-09-14 15:59:27.000000000","message":"Patch Set 10: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"60057294cefa21d94836570b5ef6232bda690584","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-09-14 16:23:25.000000000","message":"Patch Set 10: Workflow-1\n\nOh shit, wrong button","accounts_in_message":[],"_revision_number":10},{"id":"19b572a55a63b35c2f200300f0e321d73eaf489e","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-17 16:55:12.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"43d30827ffe8584d3a435023b5e84345d4e7bbfc","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2018-09-17 20:43:22.000000000","message":"Patch Set 10:\n\n(5 comments)\n\nI know you\u0027ll respin this, so no pointless -1 from me, but some thoughts and questions inline.","accounts_in_message":[],"_revision_number":10},{"id":"1d7222c0ed660d8c7569c12429c2c939b518f01b","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-17 22:53:58.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"66a66f0bbed67db369257311894d8d9de3456778","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-18 15:29:55.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"76594c6ac7823fc79bece4f005761844ef96afa2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-09-19 10:18:12.000000000","message":"Patch Set 10: Code-Review-1\n\n(27 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d7f13a8e92454890a60836b5b41e7989c82b68e1","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-10-01 21:37:23.000000000","message":"Patch Set 10: Code-Review-1\n\n(12 comments)\n\nAgree with much of stephenfin\u0027s review comments, plus left a number of my own. In particular, I added some color commentary on what I think the provider naming convention should entail.\n\n-jay","accounts_in_message":[],"_revision_number":10},{"id":"eca9c899e1784e3d8f13b68a40a9d2620a91a18c","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-10-02 15:08:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"78eac928b8553451a97f5cc9e2e037612511a70d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-10-23 01:24:50.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a25a54770ab715cb28f1255f0d281039e10e1c88","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-10-23 11:03:38.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7130a2f959c4eba48026920ebb4f409275cb02b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-23 11:43:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/24/552924/11/check/openstack-tox-docs/2118fad/html/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/11/check/openstack-tox-pep8/ccfda34/ : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":11},{"id":"93d0de2eebea7c236e67f61dd478296e007c7c8e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-10-23 19:33:30.000000000","message":"Patch Set 11: Code-Review-1\n\n(12 comments)\n\nDownvote is mainly to clarify whether we\u0027re doing the [numa]resource_classes thing here or not.","accounts_in_message":[],"_revision_number":11},{"id":"f2ea7ad263b7b3f105942b2d4ba08f452841de7a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-10-24 14:20:44.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"74d32e9911236eb38c162ee35a7c97a9061876d5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-10-24 14:42:22.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)\n\nThanks for the updates.\n\nThis seems like a solid and workable incremental step toward supporting NUMA via placement.","accounts_in_message":[],"_revision_number":12},{"id":"f411d171c272e83c92dc73270db2bbc6c29501fa","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-10-24 15:15:47.000000000","message":"Patch Set 12:\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"89414e69e34daa155d7a97f9474da017e70733b0","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-10-24 15:16:07.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"2c3fdffcc10e87f5bcf6ab4222265751e431ee2f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-10-24 15:40:11.000000000","message":"Patch Set 13: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"4ba2b4cf180435139edeef1ccfc4fabbd5fdde6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 16:48:56.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/24/552924/13/check/openstack-tox-docs/67a07a4/html/ : SUCCESS in 5m 55s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/13/check/openstack-tox-pep8/e338a45/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":13},{"id":"751a1f89c409db38ee036c4ae3ae412522b74e7d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-26 13:41:35.000000000","message":"Patch Set 13: Code-Review+1\n\n(4 comments)\n\nThis has been here long enough. Let\u0027s just get some code up","accounts_in_message":[],"_revision_number":13},{"id":"f59704a0475932ceab70dd54b7280c8fe190d9ea","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-10-30 12:59:21.000000000","message":"Patch Set 13: Code-Review-1\n\n(8 comments)\n\nI have some questions inline. I\u0027m -1 mostly due to the affinity question \n\ncheers,\ngibi","accounts_in_message":[],"_revision_number":13},{"id":"e4983f5b3a3cc7b333fbf9fd6a8b4c2575f04123","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-11-16 00:03:26.000000000","message":"Patch Set 13: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":13},{"id":"350ef4d274405a3aa1b4fe7e1ab56bdfe03233a1","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-12-21 19:09:11.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"fdafeacf3661e01e60b6c5e4bf1088371f39c3e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 19:15:51.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/24/552924/14/check/openstack-tox-docs/f86ff05/html/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/14/check/openstack-tox-pep8/aaa75ee/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":14},{"id":"dd43746a2e93a6c3a2cea4018217984ca848f1e9","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-02-19 07:59:32.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"1a07981c6cbed0687435939832b0281aff103743","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-15 13:27:18.000000000","message":"Patch Set 14: Code-Review-1\n\n(16 comments)\n\nover all i like the dirctionof this spec but\ni have added a number of edge cases to the spec taht are not currently covered. there is a lot of work i would like to do in this area on the nova side that i have higlight that shoudl not be blockers for this spec but i want to ensure this spec does not prevent that work in the future.\n\nwe coudl declare many of them out of scope for a first version of numa in placement provided we do not break novas ablity to supprot these via the numa resouce tracker and numa topology filter.","accounts_in_message":[],"_revision_number":14},{"id":"7eac80960e415301c7eb4e2562de57ca8d4d883e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-16 14:52:14.000000000","message":"Patch Set 14: Code-Review-1\n\n(6 comments)\n\nMy biggest concern is the proposed SRIOV PF modelling as it probably requires nova - neutron interworking due to already having an RP that represents the SRIOV PF (added for the bandwidth resource).","accounts_in_message":[],"_revision_number":14},{"id":"9c3853a42c5eac6623df144beeddb2a99584c405","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-16 14:58:05.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"171f0bf04e07101f3c7ec1701e5e92cedcca4c1f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-04-16 15:11:07.000000000","message":"Patch Set 14: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":14},{"id":"cf86c02c6c34f9783478b6a2b8566d55063498e8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-16 15:32:21.000000000","message":"Patch Set 14:\n\nrecheck and rerender","accounts_in_message":[],"_revision_number":14},{"id":"cdae6e15900b6e9a7bb30b20ebee98ecc066d300","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 15:46:25.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/24/552924/14/check/openstack-tox-docs/e4b2333/html/ : SUCCESS in 8m 01s\n- openstack-tox-pep8 http://logs.openstack.org/24/552924/14/check/openstack-tox-pep8/bcb4370/ : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":14},{"id":"bba6431532ddb1eac964ebd244a9a8e4184ec09a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-16 16:45:55.000000000","message":"Patch Set 14: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":14},{"id":"72cd42685477e18c83d0ebf8da4b1afba0de612b","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-04-16 16:57:46.000000000","message":"Patch Set 14:\n\n(1 comment)\n\nJust leaving a comment on \"when to reshape\".","accounts_in_message":[],"_revision_number":14},{"id":"438a42ed2448433ad229fc3f8783663081e0cab4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-16 17:07:46.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5cc68ea759e04dc0bac8f02286fefdb488c57ec0","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-21 16:46:30.000000000","message":"Patch Set 14: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":14},{"id":"bd96c04d31e56cb712b85fd962f5f76d82e4110b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-29 22:15:06.000000000","message":"Abandoned\n\nThis change is being abandoned because it proposes a spec against an old release. If the spec should still be considered, please restore the change and move the spec to the current release (please note that spec freeze is in effect for Train - see [1] for the exception process) or the backlog/ directory [2].\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/008060.html\n[2] http://specs.openstack.org/openstack/nova-specs/readme.html#backlog-specifications","accounts_in_message":[],"_revision_number":14},{"id":"9ccfe4e6278b851f39f53a8446eb7c188f542b60","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-15 11:15:37.000000000","message":"Restored\n\nreopening the spec for Ussuri","accounts_in_message":[],"_revision_number":14},{"id":"ed21b7293b03ea822fd58c7bfdaa64ba3139e9c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 11:28:18.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088f573e2169490d975ec01ed382705d : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbc338a41cc544539e88d0f52f7e3f68 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":14},{"id":"8dfadd7421c1453aae589a21382ebc3a235f485e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-15 12:48:57.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"f2056dc3d7b1d7bb27f4099cdd13c78b18d076f7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-15 15:39:14.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"825ffc9eb5476504db6a03ec8af5f5cfd9c7bf9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:52:56.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11ad9ad6de8c478fac406c1be4a6125b : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f0f1ac088e445be914c745b24f79460 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":15},{"id":"e458369e99dbd78815235b190140c47299509de1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-15 15:54:55.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)\n\ni was going to review but i dont have time today so ill come back to this tomorrow. the approch that dan is advocating we take with mixed pinned and floaign cpu in the same instance using placment resouce:* makes me very concerned with the use of granduarl groups and numa in placmenent so i think this need very carful taught.\n\ni am internally debating if we shoudl remove granular resource request form the flavour entirely or find another way to express them as the current strait passthough would be probelematic when we model numa in placment.\n\nthat is not to say nova shoudl not generage groups internally and pass them in the placment query but i am not sure we should admins to specify groups in the flaovr.","accounts_in_message":[],"_revision_number":15},{"id":"34ddc8109dee5eb0e2e2ee42fc8b45db96206371","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-27 17:03:44.000000000","message":"Patch Set 15: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"41a959a66fcbdd0b030539e3ec6cd05efa3d3132","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-28 13:46:50.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"f2f9750c222282fb4e8b1d31949f8c36928c9c46","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-29 10:27:17.000000000","message":"Patch Set 15: Code-Review-1\n\n(7 comments)\n\nI need to consider the group_policy issue and which kind of API call we should rather make.","accounts_in_message":[],"_revision_number":15},{"id":"aa21e4b986d75da1ed4b4b23b2b4207e570ebc0e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-01-29 10:27:35.000000000","message":"Patch Set 15: -Code-Review Workflow-1\n\nwhoops s/-1/-W","accounts_in_message":[],"_revision_number":15},{"id":"653276d0d9e9ede01dd0b2c4436fe47203f92800","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-29 11:01:02.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"3e10c36f82f9b0883e9d86b8da58d7b75409102e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-29 11:09:40.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f767e72a953743bcd7f0f65634110aab03baf3a7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-29 11:32:20.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"602ac5baefc534e71f4b548dcd73657555ab04a9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-29 21:23:55.000000000","message":"Patch Set 15: Code-Review-1\n\n(8 comments)\n\nAfter discussions in IRC today, I feel like we have a solid path forward, which simplifies a number of the rough spots in this spec.","accounts_in_message":[],"_revision_number":15},{"id":"3bcb6dae65937ef97502fd617ec7a9338578faff","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-29 21:30:32.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2d7319e4acdcf70950cbf3eef423ac6d7bec5ae8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-30 13:18:20.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a6759df4545cd182688633f2f29a839013f74ca4","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-01-31 03:33:29.000000000","message":"Patch Set 15:\n\n(2 comments)\n\nI feel like we haven\u0027t fully thought through the upgrade implications of this, or maybe my puny brain can\u0027t wrap itself around them.","accounts_in_message":[],"_revision_number":15},{"id":"e74ff41beb707b61923bbf1dab0e52ae4a409495","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-01-31 03:34:42.000000000","message":"Patch Set 15:\n\n\u003e (2 comments)\n \u003e \n \u003e I feel like we haven\u0027t fully thought through the upgrade\n \u003e implications of this, or maybe my puny brain can\u0027t wrap itself\n \u003e around them.\n\nPS: That was me coming back to this spec after a long time - I grok most of it, and the memory model seems to make sense, but I\u0027m not caught up enough yet for a full review.","accounts_in_message":[],"_revision_number":15},{"id":"72248e74fec053967426d0cc9067aa1d1e75c233","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-31 11:35:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"618fd3e72644ab83b61bf88a7fdf81fa75d5e6e7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-06 15:46:15.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"1b58909c793155481c339e1e00ce6e1e12a829a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 16:02:26.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7f0b4a1add045f6b666b8a39a57136c : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b42f7f3827d743728285fd0b68bd53b0 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":16},{"id":"8bfc45cfb55cf0ea75d28a50250cadf2bb577495","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-10 17:01:42.000000000","message":"Patch Set 16: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":16},{"id":"9928e61b374c22b2375886371f75b5776de358d2","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-10 18:52:43.000000000","message":"Patch Set 16: Code-Review-1\n\n(16 comments)\n\nThis is getting real close, thanks Sylvain.\n\nThere are a few details that definitely need to be explained, since they\u0027re non-obvious and the result of extensive discussions that we\u0027ll forget otherwise.\n\nAlso suggested an alternative for the conf toggle.","accounts_in_message":[],"_revision_number":16},{"id":"31faa37c3b6c0c441854cf92b02098c0f8852140","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-10 20:19:04.000000000","message":"Patch Set 16:\n\n(5 comments)\n\njust quickly skimmed over this ill review it again tomorrow","accounts_in_message":[],"_revision_number":16},{"id":"45d682709166f989bf72d0685cdb0f8407bca444","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-02-10 20:50:41.000000000","message":"Patch Set 16: Code-Review-1\n\n(15 comments)\n\nI guess I\u0027ll pile on to the -1s - though it\u0027s mostly nits from me.","accounts_in_message":[],"_revision_number":16},{"id":"304b118005077e32db65100a55966fbeb90685c8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-11 09:52:01.000000000","message":"Patch Set 16:\n\n(7 comments)","accounts_in_message":[],"_revision_number":16},{"id":"5ab16c254438cd2c4c6607b106f91ebd8900cba3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-11 13:20:30.000000000","message":"Patch Set 16: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"61faf6af7f6b10a2cc1c76633b97b7532fa2ea17","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-11 14:14:22.000000000","message":"Patch Set 16:\n\n(20 comments)","accounts_in_message":[],"_revision_number":16},{"id":"90e50b59574872e8464bb83dcf1ad0db3afae3ae","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-11 14:39:02.000000000","message":"Patch Set 16:\n\n(6 comments)","accounts_in_message":[],"_revision_number":16},{"id":"804cb27a45d426d0d9a88a89cd6c91974e9364a3","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-11 15:22:11.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"cf99e147fc9799b4474e25963be47c760961b344","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-11 15:24:29.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"0839246f57f974ebd65987d5960fd08a336fc15e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:33:44.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467d7154eadd4f3abc920ecebcc407ba : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb5cfe82c0014840b22949bdb82a5ad8 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":17},{"id":"0d8d9b0a14fc48e85c28eb634664abb695bb0f40","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-11 16:06:07.000000000","message":"Patch Set 17: Code-Review+2\n\n(4 comments)\n\nLGTM! All may issues was addressed. Thanks you Sylvain!","accounts_in_message":[],"_revision_number":17},{"id":"01779e262cc2ea625c7a9e074c793613836ced58","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-11 16:35:12.000000000","message":"Patch Set 17:\n\n(10 comments)","accounts_in_message":[],"_revision_number":17},{"id":"c31e744e3efd900bfdd7eca7d5c186c86aca1ac4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-11 16:41:29.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7348b2faae818ee5433fa59d6230cdfd59f1bcce","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-11 16:55:32.000000000","message":"Patch Set 17: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"067f87f0d7c599d3d776084f5eed0bedb397ba02","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-11 17:34:55.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"afeb44094a1ec1847c174fd2769737d462578bfe","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-11 17:42:34.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"9177bfd45ddd72cfde50badeff2266fa08bd73aa","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-11 20:22:03.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)\n\nI\u0027m quite opposed to the idea of having zero motivation to switch over to NUMA-modeled hosts in U.\n\nAfter discussing the upgrade picture in IRC, some of us came up with an alternative suggestion, summarized inline.","accounts_in_message":[],"_revision_number":17},{"id":"f827dbf3303a961dd0d95f014259031e4628e8db","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-11 23:13:41.000000000","message":"Patch Set 17: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"19529ea9a1b3f5638ce0eecfdc0383e906a7707b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-11 23:21:13.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"06e6ce261b22aee391d085a91c5406fb5ccd2d67","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-12 02:27:36.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"04b0895fb7458cda3baa31a6be52c15c52f92ae7","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-12 02:35:20.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"3635c70cb62d8d4988a7302d5fbdbf48b09778f9","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-12 02:39:29.000000000","message":"Patch Set 17:\n\nI guess I just can\u0027t form a upgrade path for the new modeling(all instance are NUMA aware) in my mind.","accounts_in_message":[],"_revision_number":17},{"id":"1b214eb79d725d9fe9af12ab274cf4ea0860b94a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-12 03:09:02.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"26547ca3377761ed9e852e1e74ae1edcba2ad389","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-12 03:14:22.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"13db8e9f9be922a9f12f9da20b711a6671baccc4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-12 09:09:51.000000000","message":"Patch Set 16:\n\n(36 comments)","accounts_in_message":[],"_revision_number":16},{"id":"02c4341d8be515cc15790070e4824cadc6592878","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-12 11:43:57.000000000","message":"Patch Set 17: Code-Review-1\n\n(9 comments)\n\nPushing a new rev as I speak.","accounts_in_message":[],"_revision_number":17},{"id":"370769cec085028f12610b56a6ab519ab62320ae","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-12 11:44:03.000000000","message":"Patch Set 17: -Code-Review Workflow-1","accounts_in_message":[],"_revision_number":17},{"id":"09f9323e35f22048954b2d1c44c74be0e856796a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-12 11:48:01.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"7dd85899d0fa4e1c72e0dfed18c7204462c267ff","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-12 11:52:58.000000000","message":"Patch Set 18: Workflow-1\n\n(1 comment)\n\nputting -W as I need to update the controversial upgrade impact section.","accounts_in_message":[],"_revision_number":18},{"id":"ac0b9b280f50a037ab165f05cc1eefb81540a231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:03:31.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8fc10a1ffaf4e05a15d73d96db8d884 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e18ca5fb3bc84bcbac6181114a942ea4 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":18},{"id":"b77e57b67896aa8f05ac8e23c8f90db761491bf1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-12 13:59:25.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a9121cbae5fde47facf243822c5e3a89aed8743f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-12 14:28:13.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"2ed5bdf1ffc9a91d1c96594d91f07678ae18a5a6","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-12 14:54:19.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"060e310947c1249716b48c9b254ea2ecf4ff88a7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-12 15:13:11.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"00bdbccc938665a20f480a937a2ca1a16ee992f8","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-12 18:03:08.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"d9711f3cd8f89fd696fabf185c523f731677293f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:24:32.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9bc71151a3947bdb2f26cd94a787287 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f55e12e3ff44ec82c84f8d617b10b4 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":19},{"id":"aa897430bf2e1d1a1b9287192c2e270735b8d18a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-12 20:34:54.000000000","message":"Patch Set 19: Code-Review-1\n\n(4 comments)\n\nthis is very close in my view.\n\nbasicaly make the second config option an int that is defaulted to 0\nand then mention the progressive spliting earlier or leave it where it is and just put a link to it earlier.\n\n\ni noticed some english grammer isue too but i am not good at reviewing for those and i think they can always be fixed later so i did not highlight them.","accounts_in_message":[],"_revision_number":19},{"id":"ead047ea7b827e3a9e464cb1916ebe5a9626322d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-13 10:05:07.000000000","message":"Patch Set 19: Code-Review-1\n\n(14 comments)\n\nSo I\u0027m not going to block this, but my gut feeling on this is that we need to do \u0027can_split\u0027 for this to proceed. Looking at the notes cdent recorded on that decision, it seems the reason we didn\u0027t proceed with that work was because there wasn\u0027t significant feedback from operators when asked about NUMA in placement [1]. If we\u0027ve now decided that we really care about NUMA in placement then why aren\u0027t we reviving the \u0027can_split\u0027 work as part of that? Without doing this, we end up having to develop hacky multiple query workarounds like the one suggested here. I think if we spent time adding that functionality to placement this cycle, we\u0027d be able to close this out easily. In fact, we could still proceed with the memory modelling rework this cycle, in the knowledge that the NUMATopologyFilter would still be needed to add NUMA semantics to things.\n\nLmk what you think\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006726.html","accounts_in_message":[],"_revision_number":19},{"id":"7057b593d0baee4bb9ae6a9846e62b4ce9e6e44e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-13 10:06:14.000000000","message":"Patch Set 19:\n\nMore notes on can_split at [1][2][3]\n\n[1] https://anticdent.org/placement-update-19-21.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006726.html\n[3] https://review.opendev.org/#/c/658510/","accounts_in_message":[],"_revision_number":19},{"id":"801506e5183147e193f3ba8c751b1faf133d3284","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-13 10:09:27.000000000","message":"Patch Set 19:\n\nThe alternative to can_split is, IMO, to clearly delineate between NUMA and non-NUMA hosts. Only NUMA guests (explicit or implicit) will land on the former, and only non-NUMA guests will land on the latter. In short, if you want to use huge pages or CPU pinning, you need to accept that you might have to configure explicit NUMA topologies for things like larger guests.","accounts_in_message":[],"_revision_number":19},{"id":"7ff7435c9c7a5b772e589467b97c275377e90754","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-13 12:34:16.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"9bad72972635d7b234327e68b87058f68021ff50","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-13 12:40:24.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"6d381443944d2cdacc1c70badc1bd46fca170be7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-13 13:31:09.000000000","message":"Patch Set 19: Workflow-1\n\nworking on a rev","accounts_in_message":[],"_revision_number":19},{"id":"7e2c54abb3a2b1ca43a4ad5a317cbe3a3a1e97d8","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-13 13:56:05.000000000","message":"Patch Set 19:\n\n(16 comments)","accounts_in_message":[],"_revision_number":19},{"id":"a97590b38afc07f361404dcac88989495b7fa970","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-13 14:06:47.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"7f812e65372632b96f1d91ebdf6ff2e322480d2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:23:05.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff849adca434a358702ac47c14ffde5 : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a47d4fbb40b04a568f8e6d46db759ea4 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":20},{"id":"8124556605db991ec500d17c59ff2f9d2a749044","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-13 16:19:54.000000000","message":"Patch Set 20: Code-Review-1\n\n(8 comments)\n\nI didn\u0027t mark them all inline, but there were a number of places \"defer to implementation\" was suggested. For something like \"the name of the workaround option\" that\u0027s fine, but some of the issues absolutely need to be agreed upon here in the spec.\n\nMy -1 is for the conf opt defaults, which I feel pretty strongly are backwards as currently written.\n\n@Stephen:\n\n\u003e If we\u0027ve now decided that we really care about NUMA in placement then why aren\u0027t we reviving the \u0027can_split\u0027 work as part of that? Without doing this, we end up having to develop hacky multiple query workarounds like the one suggested here.\n\nFor me, it\u0027s a matter of moving the ball. The proposed design (hacky multiple query workarounds included) gets us 80% of the way there *relatively* easily. If we block on can_split, we\u0027ll stall this effort for at least another release. Not because putting features into Placement is hard -- case(s) in point: nested magic 1 -- but because can_split is hard.\n\n\u003e In fact, we could still proceed with the memory modelling rework this cycle, in the knowledge that the NUMATopologyFilter would still be needed to add NUMA semantics to things.\n\nRight, that\u0027s where we were at PS16, until we started considering the upgrade implications more closely.","accounts_in_message":[],"_revision_number":20},{"id":"642b95acc3b7a8b673361a5a1a92add5ef4bbcfd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-13 16:38:08.000000000","message":"Patch Set 20: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"740372a5623b6a6068a7053be6e0d7eb047e3c25","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-13 18:28:01.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"6d12f0f2f6eb64bd953726fe34492d8e05262b30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 18:42:31.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0ec8bca5154ffa9b6e1a24bf98f12f : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25baeaa9c4d245b4ac11b5274038091b : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":21},{"id":"6691f3fe646732b6bfb4c07e5437fccf355f6604","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-13 20:06:22.000000000","message":"Patch Set 21: Code-Review-1\n\n(7 comments)\n\nThis gels with the re-re-re-reversal we decided upon in IRC today. However, we lost some crucial details that need to be added back into the document; and there\u0027s also a missing design element that I\u0027m not sure we noticed/discussed before.","accounts_in_message":[],"_revision_number":21},{"id":"3ec63c24c3a1680f0dc0dd7c486c397504f9686e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-13 22:04:51.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"9644a48cc4261004e9db8919a07fe7eba8418026","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-13 22:17:11.000000000","message":"Patch Set 21:\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"34dd0ddea9d2d1e2c37caa809e8cdcc95bd24d7c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-13 22:22:14.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"2c2e47d466c43a245f2fac256dc33f9f61d009c9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-14 00:27:58.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4f2cc11397c2e1f12db8d84530b2dfa1f282bd6e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-14 00:33:07.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"1115edec1ea66f0d1d4bc59567a8993490a12d9a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-14 10:23:07.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"ac4d5bad5a5ae7922a8b02b4a50dea014b75f69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 10:38:19.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf4f28208ac402fa51f3d50097b9400 : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647321e74b0b49c6859837914cec4652 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":22},{"id":"bb760a010e47806414d5ac494a2f26b296394953","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-14 12:45:21.000000000","message":"Patch Set 22: Code-Review+1\n\ni am off today so i wont be able to interact with this again before spec freeze. i have enough tibal knoladge of the discussions that i understand how the fallback query exctra will work and with forbiden aggreates i think we can adress the \"numa\u003dfalse hosts shoudl not return allcoation candiates for numa instance\" case so i would be good with merging this and then adressing the rest as a fix up patch.","accounts_in_message":[],"_revision_number":22},{"id":"55a7ce34cd85b5ad3c24cf89c7f8d8319c34cf3b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-14 13:38:21.000000000","message":"Patch Set 22: Code-Review+1\n\nworks for me but honestly I lost track of all the reasoning of all the complexity we are building into the solution. Still I trust you that this is necessary complexity.","accounts_in_message":[],"_revision_number":22},{"id":"0593c7cd00f31893b47fe7d86a05079fcf0af22c","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-14 13:48:16.000000000","message":"Patch Set 21:\n\n(6 comments)","accounts_in_message":[],"_revision_number":21},{"id":"d62f977ce35b06f2cd32da626592a90e3dcd6770","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-14 16:01:36.000000000","message":"Patch Set 22: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"9e7c851cfe637e658c4631531f4c32679eef42d3","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-14 16:47:59.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"8e17f461856ff6e45e7212314926904e5a84482d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-14 17:04:47.000000000","message":"Patch Set 23: Code-Review-1\n\n(3 comments)\n\nVery very close.","accounts_in_message":[],"_revision_number":23},{"id":"c6f04d99cc8c20e5251a6d52e821da8a4c9f0781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:04:57.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e345fe6abfe54a7b82b5778655daca1d : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d00c1eac23c411eb7a55f3e12bfc7df : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":23},{"id":"5be118cb29c50e973cdb31e14277f078a27cc239","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-14 17:18:22.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"5d7dcb1e6f007351722f18902704bd2980f875cc","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-14 17:21:35.000000000","message":"Patch Set 24: Code-Review+2\n\nThanks for sticking with this Sylvain. Looks good to me now.","accounts_in_message":[],"_revision_number":24},{"id":"27f2c9ed38e144bd386e5966f646a30752d0a6c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:37:02.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fc69c5773c40c9a00aed861f92da77 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8e6eb7e6e4451f9f733e80b66393d3 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":24},{"id":"cf572781ccc9bc31b74d068c12544701c1fefd6f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-15 10:13:17.000000000","message":"Patch Set 24: Code-Review+2 Workflow+1\n\n(4 comments)\n\nNothing deal breaker here for me. My comments can be fixed in a follow up as they does not really change the agreed direction of the spec.","accounts_in_message":[],"_revision_number":24},{"id":"b3bdce2dfcf6ddf3a73476457ce3e5fff22d984f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 10:13:28.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"15fb6f4c6bf35db161fd9cb6cb634bdf9e523a9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 10:25:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":24},{"id":"657a2304d73262ee983b24d34813d1478ffac806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 10:25:12.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a7520f2cd1042c2b044d97c10ccae7e : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132bef89d97e4cf8a568d493112bb546 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":24},{"id":"e7a916613a6d12641e6bff7bec9fc64f3bcd4c79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 10:26:45.000000000","message":"Patch Set 24:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/a5b7d536be274280b2faf60128d54cde : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":24},{"id":"51b5541b67e2d2718421088fcbd64991d00b36b2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-02-17 10:38:45.000000000","message":"Patch Set 24:\n\n(6 comments)","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"7d320517b494a6134f361df51b1dc04b5e4748a9","revisions":{"f2d3256ee7da387041f6f688e8dc9ebbc7704553":{"kind":"REWORK","_number":1,"created":"2018-03-14 15:24:28.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/1"}}},"commit":{"parents":[{"commit":"557896dca78bf6ed862fee2cefba444d4ce7fed2","subject":"Merge \"Express forbidden traits in placement API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/557896dca78bf6ed862fee2cefba444d4ce7fed2"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nThat\u0027s the spec for blueprint numa-topology-with-rps\n\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f2d3256ee7da387041f6f688e8dc9ebbc7704553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f2d3256ee7da387041f6f688e8dc9ebbc7704553"}]},"branch":"refs/heads/master"},"5f953606c71d970a48ae9b8b31725fc856dc3a9e":{"kind":"REWORK","_number":2,"created":"2018-03-14 15:35:12.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/2"}}},"commit":{"parents":[{"commit":"557896dca78bf6ed862fee2cefba444d4ce7fed2","subject":"Merge \"Express forbidden traits in placement API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/557896dca78bf6ed862fee2cefba444d4ce7fed2"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:34:59.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nThat\u0027s the spec for blueprint numa-topology-with-rps\n\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f953606c71d970a48ae9b8b31725fc856dc3a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f953606c71d970a48ae9b8b31725fc856dc3a9e"}]},"branch":"refs/heads/master"},"416731f5d080ec8e57a389c05ec003762a8dfe76":{"kind":"REWORK","_number":3,"created":"2018-03-14 15:36:24.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/3"}}},"commit":{"parents":[{"commit":"2ea1f01db975c50bf958d9c16e69e9f32d3dd746","subject":"Merge \"Allow for merging traits and aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2ea1f01db975c50bf958d9c16e69e9f32d3dd746"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:36:06.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nThat\u0027s the spec for blueprint numa-topology-with-rps\n\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/416731f5d080ec8e57a389c05ec003762a8dfe76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/416731f5d080ec8e57a389c05ec003762a8dfe76"}]},"branch":"refs/heads/master"},"0aacb54d03d7582663000151c73d8fbee7a40867":{"kind":"REWORK","_number":4,"created":"2018-03-26 21:25:55.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/4"}}},"commit":{"parents":[{"commit":"2ea1f01db975c50bf958d9c16e69e9f32d3dd746","subject":"Merge \"Allow for merging traits and aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2ea1f01db975c50bf958d9c16e69e9f32d3dd746"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-26 21:25:38.000000000","tz":120},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0aacb54d03d7582663000151c73d8fbee7a40867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0aacb54d03d7582663000151c73d8fbee7a40867"}]},"branch":"refs/heads/master"},"83cbc3a14ac624e963a761d393762c40222e3bd8":{"kind":"REWORK","_number":5,"created":"2018-03-27 09:10:30.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/5"}}},"commit":{"parents":[{"commit":"2ea1f01db975c50bf958d9c16e69e9f32d3dd746","subject":"Merge \"Allow for merging traits and aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2ea1f01db975c50bf958d9c16e69e9f32d3dd746"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-27 09:10:13.000000000","tz":120},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/83cbc3a14ac624e963a761d393762c40222e3bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/83cbc3a14ac624e963a761d393762c40222e3bd8"}]},"branch":"refs/heads/master"},"7d2a8902eed926f448ad2b9b07d8984dcf8c3d5d":{"kind":"REWORK","_number":6,"created":"2018-03-27 13:16:12.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/6"}}},"commit":{"parents":[{"commit":"2ea1f01db975c50bf958d9c16e69e9f32d3dd746","subject":"Merge \"Allow for merging traits and aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2ea1f01db975c50bf958d9c16e69e9f32d3dd746"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-27 13:16:02.000000000","tz":120},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d2a8902eed926f448ad2b9b07d8984dcf8c3d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d2a8902eed926f448ad2b9b07d8984dcf8c3d5d"}]},"branch":"refs/heads/master"},"3ec980e9d2a14e3f41eb3bd150f94896418894fe":{"kind":"REWORK","_number":7,"created":"2018-04-10 14:12:49.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/7"}}},"commit":{"parents":[{"commit":"2ea1f01db975c50bf958d9c16e69e9f32d3dd746","subject":"Merge \"Allow for merging traits and aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2ea1f01db975c50bf958d9c16e69e9f32d3dd746"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-04-10 14:12:35.000000000","tz":120},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3ec980e9d2a14e3f41eb3bd150f94896418894fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3ec980e9d2a14e3f41eb3bd150f94896418894fe"}]},"branch":"refs/heads/master"},"e3f11fda1d5652cd4781611d7b0466114cb7f6e8":{"kind":"REWORK","_number":8,"created":"2018-04-16 13:19:01.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/8"}}},"commit":{"parents":[{"commit":"2ea1f01db975c50bf958d9c16e69e9f32d3dd746","subject":"Merge \"Allow for merging traits and aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2ea1f01db975c50bf958d9c16e69e9f32d3dd746"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-04-16 13:18:47.000000000","tz":120},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3f11fda1d5652cd4781611d7b0466114cb7f6e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3f11fda1d5652cd4781611d7b0466114cb7f6e8"}]},"branch":"refs/heads/master"},"3f3c3cceae107ba70de79aa5e641eabcf6e7d93c":{"kind":"REWORK","_number":9,"created":"2018-04-18 13:46:11.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/9"}}},"commit":{"parents":[{"commit":"2ea1f01db975c50bf958d9c16e69e9f32d3dd746","subject":"Merge \"Allow for merging traits and aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2ea1f01db975c50bf958d9c16e69e9f32d3dd746"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-04-18 13:45:57.000000000","tz":120},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3f3c3cceae107ba70de79aa5e641eabcf6e7d93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3f3c3cceae107ba70de79aa5e641eabcf6e7d93c"}]},"branch":"refs/heads/master"},"f9b9c51845e713af88a269f0c78e7dac3feba301":{"kind":"REWORK","_number":10,"created":"2018-09-11 20:09:29.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/10"}}},"commit":{"parents":[{"commit":"f9ceb9c3c342960d747dcbe06f9dc2b9f148a611","subject":"Merge \"Add live-migration-force-after-timeout-stein spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9ceb9c3c342960d747dcbe06f9dc2b9f148a611"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-09-11 20:09:21.000000000","tz":-360},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9b9c51845e713af88a269f0c78e7dac3feba301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9b9c51845e713af88a269f0c78e7dac3feba301"}]},"branch":"refs/heads/master"},"5ac130dae30c195a01dc6a4a2d26250fbf58cd90":{"kind":"REWORK","_number":11,"created":"2018-10-23 11:03:38.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/11"}}},"commit":{"parents":[{"commit":"f9ceb9c3c342960d747dcbe06f9dc2b9f148a611","subject":"Merge \"Add live-migration-force-after-timeout-stein spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9ceb9c3c342960d747dcbe06f9dc2b9f148a611"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-10-23 11:03:24.000000000","tz":120},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5ac130dae30c195a01dc6a4a2d26250fbf58cd90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5ac130dae30c195a01dc6a4a2d26250fbf58cd90"}]},"branch":"refs/heads/master"},"662a4278c862190b94a37b769631c6fdf12f3790":{"kind":"REWORK","_number":12,"created":"2018-10-24 14:20:44.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/12"}}},"commit":{"parents":[{"commit":"f9ceb9c3c342960d747dcbe06f9dc2b9f148a611","subject":"Merge \"Add live-migration-force-after-timeout-stein spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9ceb9c3c342960d747dcbe06f9dc2b9f148a611"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-10-24 14:20:33.000000000","tz":120},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/662a4278c862190b94a37b769631c6fdf12f3790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/662a4278c862190b94a37b769631c6fdf12f3790"}]},"branch":"refs/heads/master"},"c4ae8fe22f51dde94b2147203941b83d3f4bdf04":{"kind":"REWORK","_number":13,"created":"2018-10-24 15:16:07.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/13"}}},"commit":{"parents":[{"commit":"f9ceb9c3c342960d747dcbe06f9dc2b9f148a611","subject":"Merge \"Add live-migration-force-after-timeout-stein spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9ceb9c3c342960d747dcbe06f9dc2b9f148a611"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-10-24 15:15:58.000000000","tz":120},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4ae8fe22f51dde94b2147203941b83d3f4bdf04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4ae8fe22f51dde94b2147203941b83d3f4bdf04"}]},"branch":"refs/heads/master"},"6fa84c5b056e1936056b7a448a07754ba124e5c8":{"kind":"REWORK","_number":14,"created":"2018-12-21 19:09:11.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/14"}}},"commit":{"parents":[{"commit":"f9ceb9c3c342960d747dcbe06f9dc2b9f148a611","subject":"Merge \"Add live-migration-force-after-timeout-stein spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f9ceb9c3c342960d747dcbe06f9dc2b9f148a611"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-12-21 19:08:58.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6fa84c5b056e1936056b7a448a07754ba124e5c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6fa84c5b056e1936056b7a448a07754ba124e5c8"}]},"branch":"refs/heads/master"},"e82d864d931155b67daf8b1495ba64a1548b0586":{"kind":"REWORK","_number":15,"created":"2020-01-15 15:39:14.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/15"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-01-15 15:39:03.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e82d864d931155b67daf8b1495ba64a1548b0586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e82d864d931155b67daf8b1495ba64a1548b0586"}]},"branch":"refs/heads/master"},"388723de0407042e5b5cfade001131d0c1df1c76":{"kind":"REWORK","_number":16,"created":"2020-02-06 15:46:15.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/16"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-02-06 15:45:55.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/388723de0407042e5b5cfade001131d0c1df1c76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/388723de0407042e5b5cfade001131d0c1df1c76"}]},"branch":"refs/heads/master"},"07e6b8feda440a989dab8d2480100a92725c0a47":{"kind":"REWORK","_number":17,"created":"2020-02-11 15:22:11.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/17"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-02-11 15:22:00.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/07e6b8feda440a989dab8d2480100a92725c0a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/07e6b8feda440a989dab8d2480100a92725c0a47"}]},"branch":"refs/heads/master"},"adbd5203826b9c7c596d192f3f9e4381188670fe":{"kind":"REWORK","_number":18,"created":"2020-02-12 11:48:01.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/18"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-02-12 11:47:50.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/adbd5203826b9c7c596d192f3f9e4381188670fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/adbd5203826b9c7c596d192f3f9e4381188670fe"}]},"branch":"refs/heads/master"},"565191f33536b6c6c2882fd37ad25323a6b28c78":{"kind":"REWORK","_number":19,"created":"2020-02-12 18:03:08.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/19"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-02-12 18:02:57.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/565191f33536b6c6c2882fd37ad25323a6b28c78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/565191f33536b6c6c2882fd37ad25323a6b28c78"}]},"branch":"refs/heads/master"},"4dba90a39488eb6094c862403b01defdab85c6d4":{"kind":"REWORK","_number":20,"created":"2020-02-13 14:06:47.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/20"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-02-13 14:06:37.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4dba90a39488eb6094c862403b01defdab85c6d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4dba90a39488eb6094c862403b01defdab85c6d4"}]},"branch":"refs/heads/master"},"69ed0a4508017184daf56c7b00f70ca2ee072d36":{"kind":"REWORK","_number":21,"created":"2020-02-13 18:28:01.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/21"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-02-13 18:27:49.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/69ed0a4508017184daf56c7b00f70ca2ee072d36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/69ed0a4508017184daf56c7b00f70ca2ee072d36"}]},"branch":"refs/heads/master"},"d2d9cf4de7e948c96f7522fb85dd7722e1526f2d":{"kind":"REWORK","_number":22,"created":"2020-02-14 10:23:07.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/22"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-02-14 10:22:56.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2d9cf4de7e948c96f7522fb85dd7722e1526f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2d9cf4de7e948c96f7522fb85dd7722e1526f2d"}]},"branch":"refs/heads/master"},"1a5fb1f0f442cbb8fb1a2914e0d55f25ed05b7a8":{"kind":"REWORK","_number":23,"created":"2020-02-14 16:47:59.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/23"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-02-14 16:47:37.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a5fb1f0f442cbb8fb1a2914e0d55f25ed05b7a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a5fb1f0f442cbb8fb1a2914e0d55f25ed05b7a8"}]},"branch":"refs/heads/master"},"7d320517b494a6134f361df51b1dc04b5e4748a9":{"kind":"REWORK","_number":24,"created":"2020-02-14 17:18:22.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/24/552924/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/24/552924/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/24/552924/24"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2018-03-14 15:21:10.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2020-02-14 17:12:54.000000000","tz":60},"subject":"Proposes NUMA topology with RPs","message":"Proposes NUMA topology with RPs\n\nSpec for explaining the relationship between a root RP and\nand one or more Non-Uniform Memory Access (NUMA) nodes (aka. cells)\n\nSpec-for: blueprint numa-topology-with-rps\nChange-Id: I0c804743db77da5717c9c37e3c5ba57a9a3950ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d320517b494a6134f361df51b1dc04b5e4748a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d320517b494a6134f361df51b1dc04b5e4748a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
