)]}'
{"id":"openstack%2Fnova-specs~555081","triplet_id":"openstack%2Fnova-specs~master~I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c","project":"openstack/nova-specs","branch":"master","topic":"bp/cpu-resources","hashtags":[],"change_id":"I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c","subject":"Standardize CPU resource tracking","status":"MERGED","created":"2018-03-21 22:13:41.000000000","updated":"2019-06-17 09:05:18.000000000","submitted":"2019-05-15 15:24:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":880,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"555081-1557933891708-cf8f5779","meta_rev_id":"76515efe53c9f6e71ad69f9d1975d6a7c76003c4","_number":555081,"virtual_id_number":555081,"owner":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-02 16:11:45.000000000","_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":2,"date":"2019-05-15 15:24:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2019-04-29 21:06:30.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"date":"2019-05-15 14:32:05.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":28007,"name":"haokunsheng","email":"haokunsheng@126.com"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2019-05-06 11:13:24.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"date":"2019-04-26 02:33:05.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"date":"2019-04-25 23:58:39.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":2,"date":"2019-04-25 15:30:03.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":1,"date":"2019-04-25 11:32:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},{"value":0,"date":"2019-05-02 16:37:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":28007,"name":"haokunsheng","email":"haokunsheng@126.com"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2019-05-02 16:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-06-17 09:05:18.000000000","post_submit":true,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":28007,"name":"haokunsheng","email":"haokunsheng@126.com"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":1,"date":"2019-05-15 15:14:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":28007,"name":"haokunsheng","email":"haokunsheng@126.com"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"_account_id":28007,"name":"haokunsheng","email":"haokunsheng@126.com"},{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-23 15:24:42.000000000","updated_by":{"_account_id":28007,"name":"haokunsheng","email":"haokunsheng@126.com"},"reviewer":{"_account_id":28007,"name":"haokunsheng","email":"haokunsheng@126.com"},"state":"REVIEWER"},{"updated":"2018-05-31 01:23:01.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2018-09-27 14:00:29.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-09-28 13:44:19.000000000","updated_by":{"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},"reviewer":{"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},"state":"REVIEWER"},{"updated":"2018-10-26 17:46:24.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2018-11-27 18:48:04.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2018-12-13 08:58:49.000000000","updated_by":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"reviewer":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"state":"REVIEWER"},{"updated":"2018-12-19 12:15:49.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2019-03-23 20:57:38.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2019-04-04 15:24:52.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-04-08 15:38:01.000000000","updated_by":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"reviewer":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"state":"REVIEWER"},{"updated":"2019-04-09 17:37:42.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-04-23 08:26:15.000000000","updated_by":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"reviewer":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"state":"REVIEWER"},{"updated":"2019-04-25 11:32:08.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-04-25 23:58:39.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-04-26 02:33:05.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-29 21:06:30.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-05-02 16:11:45.000000000","updated_by":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"reviewer":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"state":"REVIEWER"},{"updated":"2019-05-02 16:37:05.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2019-05-06 11:13:24.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-05-15 14:32:05.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-15 15:14:27.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-05-15 15:24:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-17 09:05:18.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"0104a25686b157e688838c463edb3ff5b1888d6b","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-21 22:13:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"495385a48287187ced64eb954cd7e79b466297fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-21 22:40:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/1/check/build-openstack-sphinx-docs/0eebe7e/html/ : SUCCESS in 5m 17s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/1/check/openstack-tox-pep8/4461d03/ : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"88037052cc25633db25fd332290dd910ef120b24","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-21 23:31:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"127efe031d59f6b1d4fd4f88aac112b22dc4cead","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-21 23:43:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/2/check/build-openstack-sphinx-docs/3064bb6/html/ : SUCCESS in 6m 29s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/2/check/openstack-tox-pep8/0901007/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a6052852374384e3aa93bc6d87418834bd3d8fd7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-03-22 00:14:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(10 comments)\n\nthis adds some new fucntionality but we are also losing the ablity for end users to configure pinning via uploading there own image with metadata.","accounts_in_message":[],"_revision_number":2},{"id":"808497dc0ca29471b54a478fc85c91bb1e12d28d","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-03-22 02:55:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)\n\nThank you so much for doing this spec!\nI have couple of comments and questions.","accounts_in_message":[],"_revision_number":2},{"id":"ef2c5d84e8c652aded373e9e267c55dd5757b610","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-03-22 08:46:01.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4c3eff83b22105a4fadff8352650063567119344","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-22 14:38:19.000000000","message":"Patch Set 2:\n\n(15 comments)","accounts_in_message":[],"_revision_number":2},{"id":"14d646a0edb0b66feeca23391dd0a4e387781524","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-22 15:27:20.000000000","message":"Patch Set 2:\n\n(12 comments)\n\nA few summary points:\n\n1) I find the VCPU and PCPU terminology slightly confusing since they\u0027re all virtual cpus just with different affinity.  But that\u0027s not a showstopper, I can live with it.\n\n2) I think we need a way to specify *which* guest cpus are \"shared\" vs \"dedicated\".\n\n3) Do we have a requirement to specify whether emulator threads run on \"shared\" or \"dedicated\" host cpus?  The only scenario I can think of where this might make sense is if someone had an instance that they really wanted to be fast with the absolute minimum of impact from other guests, regardless of the financial cost.\n\n4) I think that the sum of VCPU+PCPU should add up to the number of cpus visible within the guest.  If we need to account for the emulator threads we should do it implicitly.\n\n5) Do we expect the emulator thread overhead to scale with the number of guests, or with the number of guest cpus?  If the latter, then when using \"shared\" host cpus to run the emulator threads the overhead is essentially already accounted for in the cpu allocation ratio and we don\u0027t need to do anything with placement.  If the former, then it would make sense to allocate a VCPU in placement to account for the host cpu resources consumed by the emulator threads.\n\n6) We need to figure out what we\u0027re going to do with instances that currently can float over the whole compute node.","accounts_in_message":[],"_revision_number":2},{"id":"3f5c9660c688b7f47c19438f49b4ee77c85e8572","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-03-22 15:43:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54b7ee072ed618275b8a07b58de31a8ba61c3616","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-22 15:44:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dfc02f9791bc69732ed06608d99bbecd6725645c","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-22 16:32:17.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1f1c21be68bf38c394ac13ca55577ea064978c07","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-22 19:17:04.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nI have looked at this.\n\nPretty sure \"pinsets\" as you\u0027ve described them aren\u0027t at all how the Power platform expresses things.  Normally I would red-flag for that reason.  But I *think* we don\u0027t care in this case, because we don\u0027t let you do pinning anyway (as I understand it, the platform does affinity for you, and does a better job than if you tried to do it yourself).  We *do* have the concept of shared vs dedicated, but we do that dynamically - you don\u0027t have to preallocate procs to either category.  We\u0027ll have to figure out how to model all of that in placement, of course, but I think not having to deal with NUMA will allow us to do something way simpler than what\u0027s described herein.\n\nThat said, other \"nonstandard\" (read: not libvirt) driver owners ought to look at this from that perspective and scream accordingly.\n\nAlso, I could be completely wrong about the above, so I reserve the right to change my mind later.\n\nBut for now, deliberate +0.","accounts_in_message":[],"_revision_number":2},{"id":"b25717a585ce58d6deb1eeba396cfee51fbbfe74","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-03-23 08:16:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3766fa51433aeb047e20c1dee943329931ce757","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-03-23 10:13:24.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0e5243a7e276a9bb51a2a19e43f5840621661373","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-23 14:14:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ebacf62e2208e977e7a882b76185537574d0668b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-23 14:21:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/3/check/build-openstack-sphinx-docs/aabd1dd/html/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/3/check/openstack-tox-pep8/b8edd54/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2d8a52b637f533b82531f1d17b118d8f2b712f30","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-03-23 14:35:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)\n\nSounds perfect that is addressing all use cases I have in my head. Thakns Jay !","accounts_in_message":[],"_revision_number":3},{"id":"1e8c9c47a9e0b2aab17eb221e9de07f9dda647d3","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-23 14:40:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"032b4d9bb7b1f386d796c0fc258397b86570947c","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-23 14:47:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)\n\nI only reviewed the NUMA specific bits since another spec I own will care about the details","accounts_in_message":[],"_revision_number":3},{"id":"a83880e7ae5274df872fe540ee3f466b3f676bff","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-03-23 15:07:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)\n\nthe -1 is really -0.5 so yes i am much happier with this version\nsome nits inline","accounts_in_message":[],"_revision_number":3},{"id":"045d4db0370a848cbfa439fa5688f4e38f3b90f1","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-23 15:10:49.000000000","message":"Patch Set 3:\n\n(8 comments)\n\nBased on the discussion we had on IRC, just a few notes about the fact that even if a host system is NUMA-based, we leave the operator choosing which kind of workloads it wants to run on top of it (and leave details for my own spec)","accounts_in_message":[],"_revision_number":3},{"id":"97b3bc656a29b056ad08ee7ef26d9ac0ab74a82e","author":{"_account_id":28007,"name":"haokunsheng","email":"haokunsheng@126.com"},"date":"2018-03-23 15:24:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3599acc24d7557c18f6cd3d6b3a94aeb2458c7fd","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-23 17:23:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(13 comments)\n\nA few minor fixes still, and some questions.","accounts_in_message":[],"_revision_number":3},{"id":"c4bcb6f4a09ff86827250e39e24440e4231b87d9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-03-24 21:10:38.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b07309a7f2844ea6b3e422d7f6b81308cae429df","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-25 13:23:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"35ee574703a9ea4c12d375a05436b4d0129677fb","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-25 13:23:26.000000000","message":"Patch Set 3:\n\n(22 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f1b7e6f09eba21ee8534a0647bafb5c8506c198d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-25 13:29:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/4/check/build-openstack-sphinx-docs/99a5d64/ : FAILURE in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/4/check/openstack-tox-pep8/6122ad5/ : FAILURE in 3m 57s","accounts_in_message":[],"_revision_number":4},{"id":"aaaabf5a5ed1f97844b619dfbeca8950cd22957e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-25 18:30:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(9 comments)\n\nI like how the pinset conf options translate into resource providers, as a means of guiding how things should be modeled.  But beyond that I don\u0027t think we\u0027ve actually solved any of the affinity/pinning problems we started with.","accounts_in_message":[],"_revision_number":4},{"id":"d855d652ec4ab095c74d76b9f7e2d63379629b1b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-25 18:32:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5af246d5a48b4bb62800e787fe3b1435ce53e8c3","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-03-26 04:11:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0275ad004d68124e4d000a82031a521cbf33483c","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-03-26 10:30:17.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4ce6662a8fb216c24f01d90b7f8d663c2a2fb405","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-26 16:47:28.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8f0a1449cd4b506e3b6b2a0a66ca88a8c692272c","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-26 16:59:48.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2f14fcd9e9508f003f457d87a66175695d6a1e34","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-26 20:54:12.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"36b724175f96d54978651d835348877fc8887607","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-26 22:44:03.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"350578369d55eb911e73fa12e16d8455dd0ecde2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-03-27 09:08:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(10 comments)\n\nSince most of the NUMA concerns are shared with https://review.openstack.org/#/c/552924 I\u0027d be in favor of not discussing about those concerns in that spec, just providing some high-level comments but not detailed here, and do the conversation on the other spec. \nThat\u0027s confusing and hard to follow since we have the same group of people voicing on two different threads.\nThat said, that spec is the perfect place for discussing about cpu pinning, guest hyperthreads and other things I clearly excluded from my own spec.","accounts_in_message":[],"_revision_number":4},{"id":"65d344425b8a4bae5d4b71ad57a62d10a1c6f86c","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2018-03-27 13:43:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f1eba6fd886d02e092aa05c5fc9389f946b0f16a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-03-27 16:17:07.000000000","message":"Patch Set 4:\n\nStill reviewing this. It\u0027ll probably take me into tomorrow... :)","accounts_in_message":[],"_revision_number":4},{"id":"53ebe4fe7658fa885dff17cf5599be98f66efe81","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-27 16:37:45.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"345a616b99a1418f230f273932ea47dfa44a41e5","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-29 18:55:16.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d9b610e38b8175db3872035e2d3ef50e356a5ec9","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-29 18:55:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"14834246904a6a1645de5016b18cef54b72ddbe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-29 19:04:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/5/check/build-openstack-sphinx-docs/567f359/ : FAILURE in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/5/check/openstack-tox-pep8/46ad9ed/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":5},{"id":"f8f65b149429b6a304a48e8b0c1d9ad4d2d3fe75","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-29 19:14:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"38fde6234de080162f59e7bad11bb11f40962765","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-29 19:42:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/6/check/build-openstack-sphinx-docs/3ab67cb/html/ : SUCCESS in 6m 15s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/6/check/openstack-tox-pep8/cc2968f/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":6},{"id":"66bda73d88a900589e849df26bdf32965e1ab8c5","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-03-29 20:11:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b310ad251a82248a73057a0dcd4eeab678510c9d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-29 22:38:05.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nLots of discussion from PS4 hasn\u0027t been resolved yet.","accounts_in_message":[],"_revision_number":6},{"id":"d8b1b4229fc96aa1f283c618dbdf40671bf07f44","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-30 02:23:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52c831c73f17b8eac80456c4728daaa08e49ef7d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-03-30 13:10:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e67fdb65868841d6e6415e3bf09a8684d573ac0","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-04-02 14:50:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8c2001be0295f7ba726e5d8816be03d3d0e39b89","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-02 15:08:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82427fae9ed24938778c2659975d1ea8a04ece93","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-04-03 02:20:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"44e39772608b9c00d227c6c3eba9822cf31ba92f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-03 13:39:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4969cac6394e1638065ea67f77e4632842f994b8","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-04-03 19:44:08.000000000","message":"Patch Set 6:\n\n(7 comments)\n\nNice spec. I started typing in questions, read ahead, found them answered. Learned a lot about crazy cpu stuff too.\n\nThe comment at line 111 is probably the only one that is really relevant. The rest of it is pretty much notes to myself.\n\nFrom a big picture standpoint the model described doesn\u0027t prevent, say (say), a virt driver that wouldn\u0027t be able to make sense of a cpuset in nova.conf to represent the many cpu sets it might be aware of by other means, so that\u0027s good.\n\nIf that\u0027s not the case, we probably need to talk about that.","accounts_in_message":[],"_revision_number":6},{"id":"181cbe01c6f86f35f5b3e07a005a0bef26d62d30","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-04-03 22:32:57.000000000","message":"Patch Set 6: Code-Review+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7f1ad4c0a06c10fa71e6981ebe0e2f8e8ec3ef36","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-05 17:28:22.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"77853a222df7b00f8f6e89aa6cb25db2024e8bd2","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2018-04-06 15:53:19.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e633c4b183c23d17a9918376c0eb867ddd5138c9","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-04-09 09:23:06.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nPretty neat and clean specs!!! Thank you.\n\nI have some comments on guest resource request.\nVoted -1 to clear things upfront before implementation.","accounts_in_message":[],"_revision_number":6},{"id":"5553e5737614f4510e728dc7b1f404ee2b43d8fe","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-09 21:10:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4150dc0d562ea8114ba4126b6b32db44023ee472","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-04-10 01:44:16.000000000","message":"Patch Set 6: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"37039a8111f44a46fe44df734a1c01d9ab276564","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-11 20:07:41.000000000","message":"Patch Set 6:\n\n(18 comments)","accounts_in_message":[],"_revision_number":6},{"id":"219e42c7fa59df72fc0d6fd54fdda1f7e4384995","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-12 16:01:36.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2299bfd035d6fd9ee2e4e80b4792a1db57b981e7","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-04-12 16:34:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a4b5c4c647eb7d960b843f7b2ec6bd91b0f83e18","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-12 17:13:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"02b2be868e1afa79a2518fec5ea7d40fc0e4b134","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-12 17:17:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a0c66fb86ec31894c60ba0bcb93f8d3949cbca5d","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-04-12 17:18:20.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b3fe3066ee79fc22c86e7fd91c4874d54fd0b194","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 19:26:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/7/check/build-openstack-sphinx-docs/d1cf093/html/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/7/check/openstack-tox-pep8/7099e65/ : FAILURE in 3m 50s","accounts_in_message":[],"_revision_number":7},{"id":"9b47c039e7a97544b2ccce87aa559987adaf9454","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-12 20:05:40.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)\n\nIssues inline, plus unaddressed comments from PS6, plus pep (probably for long lines).","accounts_in_message":[],"_revision_number":7},{"id":"2bc02c599237ccc6810144a124a89a66e8bf5fbf","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-12 20:46:22.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e71d23c0e413b0b856278f45a7d8a8ccc1b9367f","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-12 20:47:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"541f8989a0473385f6650c9ba8865129eaddea64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 21:23:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/8/check/build-openstack-sphinx-docs/c777548/html/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/8/check/openstack-tox-pep8/efe65ec/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":8},{"id":"3ccd495b7a770740b40993f4084f4c7136b9801c","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-04-16 11:07:14.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3b1e40d94fc841abf54f74bceb39feb499bd31d8","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-04-16 11:08:15.000000000","message":"Patch Set 8: Code-Review+1\n\nNothing leaps out as dire-ly problematic. Eric\u0027s suggestions for housekeeping improvements as well as more detail in the alternatives section make sense.\n\nI know we want to get this stuff (complex cpu setups) right and not bound ourselves into a bad precedent, but on the other hand, this is complex enough that being able to experiment with some beginning of it is going to take us a long way. So if we can figure out the common denominators and get them out in the world that would be dandy.","accounts_in_message":[],"_revision_number":8},{"id":"b95307febd7059684940b39d8a24a98fdb0a52e8","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-04-16 11:08:24.000000000","message":"Patch Set 8:\n\nNothing leaps out as dire-ly problematic. Eric\u0027s suggestions for housekeeping improvements as well as more detail in the alternatives section make sense.\n\nI know we want to get this stuff (complex cpu setups) right and not bound ourselves into a bad precedent, but on the other hand, this is complex enough that being able to experiment with some beginning of it is going to take us a long way. So if we can figure out the common denominators and get them out in the world that would be dandy.","accounts_in_message":[],"_revision_number":8},{"id":"aafaa957d836779b277dbf4b55f820e1642f6a0b","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-04-16 11:43:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c4b83b4a7a59ec40a0adaf3b17a078884b414277","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-16 12:30:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ff3cfb8d48673ffff44fa5baaf2465b2764c91d7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-16 22:40:59.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nThis spec refers to the original change set proposing the rocky granular resource requests spec [1]; but it relies on semantics [2] that are not in that change set or in the published spec.  The dependency on [2] needs to be made crystal clear, including how that semantic is going to be used.  For example, it is implied, but not stated, that the \"don\u0027t care about NUMA topology\" case will use the config option / query param allowing the un-numbered group\u0027s resources to be \"spread\".\n\nI\u0027m still not on board with [2], and have proposed an alternative [3] which still allows you to satisfy the use cases described in this spec.\n\n[1] https://review.openstack.org/#/c/540179/\n[2] https://review.openstack.org/#/c/560974/\n[3] https://review.openstack.org/#/c/561717/","accounts_in_message":[],"_revision_number":8},{"id":"84d1d8fe062c781fcfe0f05c1d2e679823b330c6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-16 22:41:05.000000000","message":"Patch Set 8: Code-Review-1","accounts_in_message":[],"_revision_number":8},{"id":"6e96c1988ddd9c501c10c0234f879b6c0f87bf0b","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-04-17 13:44:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5e613cf1ca9d09a64b56e87c3bd1ab65d0fba458","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-04-17 13:52:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8c67b18995675ee8c3f45d29eccd4168ee599f1e","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-04-18 07:00:30.000000000","message":"Patch Set 8:\n\n@Eric, For the \"don\u0027t care the NUMA topology\" case, I don\u0027t see why we depend on this [2] https://review.openstack.org/#/c/560974/\", since we always try to pack the \"NUMA topo careless\" instance into single physical NUMA node.","accounts_in_message":[],"_revision_number":8},{"id":"14b40c5d33f59617e768772adb14793b5580c670","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-18 12:44:01.000000000","message":"Patch Set 8:\n\n@Alex\n\n\u003e we always try to pack the \"NUMA topo careless\" instance into single physical NUMA node.\n\nThat\u0027s the reason.  We still want to be able to land our instance even if there is no single NUMA node (anywhere in the cloud) with room for all the requested procs.  At that point, we need some way to say X procs from one NUMA node and Y procs from another (on the same host).  And if that doesn\u0027t work, split them further.  Etc.  The referenced spec mod provides a query param that will try such splits iteratively if it doesn\u0027t find a match the first time around.\n\nHowever, we may be going with [1], which would mean you would put each proc in its own granular request group to satisfy this \"any fit\" scenario.\n\n[1] https://review.openstack.org/#/c/561717/","accounts_in_message":[],"_revision_number":8},{"id":"69152f1f2d8ab3ff806daf225c9967f1a46aa45d","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-04-18 13:45:36.000000000","message":"Patch Set 8:\n\n@Eric, so that isn\u0027t a thing we do today, so that is why I think it shouldn\u0027t be a hard dependence.","accounts_in_message":[],"_revision_number":8},{"id":"b7081f3a490ac645b2d606154ac6e99041c5f69c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-18 14:09:43.000000000","message":"Patch Set 8:\n\n\u003e that isn\u0027t a thing we do today,\n\nOkay; I\u0027ve just been getting the impression that it was.\n\n\u003e so that is why I think it shouldn\u0027t be a hard dependence.\n\nIf that\u0027s so, the spec should say that we\u0027re not trying to support that use case.  If we support the use case, we have the dependency; if not, we don\u0027t.","accounts_in_message":[],"_revision_number":8},{"id":"9be6d1b636547a2bc4e4551c17bc6324c498a3af","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-27 18:50:17.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nWhat\u0027s the status on this, reviewers?","accounts_in_message":[],"_revision_number":8},{"id":"bf9cfc873e092acfb0c59a4e5322a5ac5f0480ab","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-27 19:57:32.000000000","message":"Patch Set 8:\n\n(9 comments)\n\n@Jay, earlier comments still haven\u0027t been addressed.  I\u0027m obviously not going to hold up on line width if pep doesn\u0027t care (though why doesn\u0027t it??) - it\u0027s the other stuff.\n\nNamely, it\u0027s still not clear whether/how we\u0027re dealing with the \"don\u0027t care about NUMA topology\" cases.  The text implies that specifying VCPU:4 in the unnumbered group could get you VCPUs from multiple NUMA nodes.  That\u0027s either untrue, or there\u0027s going to be some change in the translation from flavor to placement qparams beyond what\u0027s being done today.  I marked the relevant spots inline.","accounts_in_message":[],"_revision_number":8},{"id":"b7de1a5162b871b71500faa440380f80204be3b3","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-27 20:09:46.000000000","message":"Patch Set 8:\n\nack, will fix up tomorrow and resubmit.","accounts_in_message":[],"_revision_number":8},{"id":"4c936f748b4579693e02e7ed9b6fbd5eec0c2f48","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-30 17:00:28.000000000","message":"Patch Set 8:\n\n(10 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3fc32eb694da5c22a73d4dfb10deb34afd872d2a","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-04-30 17:00:57.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f99655e02f253f5331a526119bdb0bed6f837879","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-30 17:11:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/9/check/build-openstack-sphinx-docs/2c3411d/html/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/9/check/openstack-tox-pep8/e9c62d7/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":9},{"id":"2ee4ad29bdd1d59b0c5138a715532545e50484d7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-04-30 17:20:45.000000000","message":"Patch Set 9: Code-Review+1\n\nThanks Jay, this is clear now.","accounts_in_message":[],"_revision_number":9},{"id":"a34ed09114268efa6f80ba2c2060e7192beb7e92","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-05-14 10:59:42.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)\n\nLGTM. Just one question to ask.","accounts_in_message":[],"_revision_number":9},{"id":"f85bdd574b06bb9699decbb057ed162ee9d4c8a4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-05-14 13:49:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6f598a49cb2fdfbca10fbad868a19cb20e60e566","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-05-14 14:59:16.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"818f84be67cdb42bc3e41016ed5678674f51b6f4","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-05-14 19:25:30.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e25f4a4e56401a801c5551cf71122353433a963d","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-05-15 06:45:36.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ca87986b3a21e96e5d577d9cdee44805bc5ed83c","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-05-15 07:35:04.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)\n\nI think we are good you can\u0027t really address every implementation details in that spec. We have something good to make a good start. Thanks for that Jay.","accounts_in_message":[],"_revision_number":9},{"id":"365c168e8372ce3cb2f597dc32441cadd4875886","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-05-29 10:46:41.000000000","message":"Patch Set 9: Code-Review-1\n\n(22 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bbed215470bc35d03bcd18bc5f1f1c350273810e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-05-29 10:46:50.000000000","message":"Patch Set 9:\n\nAll the comments :)","accounts_in_message":[],"_revision_number":9},{"id":"b3ff8c099a84e6a64ed8e0bc2258b6c5a5e7f732","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-05-29 14:03:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"910fa1c7632864c55f7ad0d839a68aa2f435a43d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-05-29 15:50:57.000000000","message":"Patch Set 9:\n\n(10 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3e5c82a127d109dc5101704d2b44805d40d02601","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-05-29 16:37:29.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"275243163c254487ebee29135eb3414378a0b0fc","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-05-29 20:33:54.000000000","message":"Patch Set 9:\n\n(22 comments)","accounts_in_message":[],"_revision_number":9},{"id":"eddb3081fabf348ec83a953b90f023b6c30fd292","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-05-29 21:00:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2e3141fa6916d6aacd888d8005b0dad604a38143","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-29 21:10:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/10/check/build-openstack-sphinx-docs/36c2881/html/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/10/check/openstack-tox-pep8/1316710/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":10},{"id":"833e984ee34369461cb7fbfb894538608a7d4e87","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-05-30 08:16:55.000000000","message":"Patch Set 10:\n\n(2 comments)\n\nHi Jay,\n\nCould you please answer below queries?","accounts_in_message":[],"_revision_number":10},{"id":"5f21caa9d306ff9ea4f90f1f6e44d39019832ada","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-05-30 09:08:14.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e752c89c41cf52efa56fbd1e2198f11af0913400","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-05-30 10:18:08.000000000","message":"Patch Set 10: Code-Review-1\n\n(7 comments)\n\n-1 for the open questions on upgrade path and how the two new pinset options will replace the older ones. Couple of nits inline too","accounts_in_message":[],"_revision_number":10},{"id":"0638c825c5acb732090166309081f494003f3a93","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-05-30 16:26:48.000000000","message":"Patch Set 10:\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6098028ee6cf05e96775fd49b54191745b3c02b0","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-05-30 16:27:21.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f14cd499983c3b3ab3f2e71d7f87b879fee68f34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-30 16:36:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/11/check/build-openstack-sphinx-docs/acf7651/html/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/11/check/openstack-tox-pep8/3a8c8f7/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":11},{"id":"789e9fe61f6a007007bf8b4bd41cf3e938859453","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-05-30 21:57:50.000000000","message":"Patch Set 11: Code-Review-1\n\n(35 comments)","accounts_in_message":[],"_revision_number":11},{"id":"cf19c8d4f914f7dd8e334f6c6225131386be3ecb","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-05-31 07:24:36.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d7d6d93bef40eb8e09cb4ce89083e9a7164ce013","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-05-31 10:13:31.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"67c716dbb6e94272d9244e61439b24ead5a9b017","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-05-31 10:44:00.000000000","message":"Patch Set 11: Code-Review-1\n\n(16 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9576b23c18b00c74bb331218b4b0125b4591eea7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-06-01 17:44:30.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d42210a3b1cbaf916338a2633ed8706d510a3751","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-06-01 20:15:08.000000000","message":"Patch Set 11: Code-Review-1\n\n(29 comments)\n\nby the way i may have missed a conversation where all numa related topics were declared out of scope of this spec but from my persecivie i think that would be a mistake.","accounts_in_message":[],"_revision_number":11},{"id":"9579ad3859f2a4fb2dd1da1583cfbe2ee43ab971","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-06-04 12:40:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2dfc69223392c1b0290deb7379139b8d6f74663e","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-04 18:55:16.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"843524c2eb2308189d8c5755d8a9055d8ec28ce0","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-04 18:55:28.000000000","message":"Patch Set 11:\n\n(48 comments)","accounts_in_message":[],"_revision_number":11},{"id":"d57aa7bf4f9cdeeb05047624f069c9f02fc29bed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-04 19:29:48.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/12/check/build-openstack-sphinx-docs/d215a2f/ : FAILURE in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/12/check/openstack-tox-pep8/c60c7c6/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":12},{"id":"8932e883c37635c6c587f0ab710b2e296c574416","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-04 20:18:39.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b20aaba78ae9e25a2388071d35451e120e0085bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-04 20:48:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/13/check/build-openstack-sphinx-docs/27b7793/html/ : SUCCESS in 5m 32s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/13/check/openstack-tox-pep8/152fa95/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":13},{"id":"dae08859b0b00b5f5b9adf242f2263f4c7e183db","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-06-05 10:06:46.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7c81e55de1f8e363fd8942d866d49e81352ad817","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-06-05 10:20:55.000000000","message":"Patch Set 13: Code-Review+1\n\n(7 comments)\n\nI\u0027ve a few nits in there but I think the bones of this are good to go now. Will be interested in seeing what Sean comes back with :)","accounts_in_message":[],"_revision_number":13},{"id":"20e6f58e977265cb9b89a4cb31357b6b32d860ef","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-06-05 10:57:47.000000000","message":"Patch Set 11:\n\n(3 comments)\n\njust some responese to jay on patchset 11\nill review the latest draft later today.","accounts_in_message":[],"_revision_number":11},{"id":"b099ff3f4664a85d147908409ac303fce5970115","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-05 12:56:32.000000000","message":"Patch Set 13:\n\n(7 comments)","accounts_in_message":[],"_revision_number":13},{"id":"6473eabdc84ae6f8932946e92516c7ed74239517","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-05 13:00:03.000000000","message":"Patch Set 13:\n\n(6 comments)\n\nSome questions inline, nothing justifies a -1.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":13},{"id":"18f8b75915028e243417898d303d7275f975acfe","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-05 13:00:23.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e840a9bbb11381e063bebe62160c5b4f6c7b49cf","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-05 13:00:38.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ffc524c16c46dee6ed1a238dbd2a81db7ff21555","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 13:08:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/14/check/build-openstack-sphinx-docs/b933e11/html/ : SUCCESS in 6m 07s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/14/check/openstack-tox-pep8/abb89d9/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":14},{"id":"1aa8b4bc0fb27a64ff33c8b903b4be1adbd6a2de","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-05 16:35:47.000000000","message":"Patch Set 13:\n\n(5 comments)","accounts_in_message":[],"_revision_number":13},{"id":"e563ca99b7542327fa45cac2c1abe864d13489e9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-06 13:21:44.000000000","message":"Patch Set 13:\n\n(5 comments)","accounts_in_message":[],"_revision_number":13},{"id":"33217178a145a582fdce946cc0c8d6c92521af36","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-06 15:40:24.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"56a5d6405c1abf1bb0982215e14edab003a92716","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-06 15:46:52.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"61cb950d650bfeb56fb28f6cd07a087cd2e700b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-06 15:57:11.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/15/check/build-openstack-sphinx-docs/9b90f94/html/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/15/check/openstack-tox-pep8/717e637/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":15},{"id":"f027dd9b8891f003562673ce96e80dd29896b981","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-06-11 09:39:45.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nHi Jay,\n\nCould you please answer below query?","accounts_in_message":[],"_revision_number":15},{"id":"51b6e6f9c531db3d54bf16c3b10e5c03d5e1b9db","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-06-15 21:04:56.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"db06ac77c7102002cba163372a245b391d699888","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-06-25 20:40:11.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3e490cee170f480b1cddc2bd5a4d56984b587892","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-06-26 02:11:11.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3eb4d9d4442c9a0c385ad06b07dd6910a7466bad","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-06-26 10:41:10.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"fa0cd4d8f096cd4de122e812549f1e5eb230f5fc","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-06-26 15:24:30.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"592e9502f10f240e4e8f5f205cb98ab057064d7d","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-06-27 08:13:32.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4ac89ed1c833f8c0a3d4bae544abca78befbf577","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-06-27 13:52:44.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1028dbbf999e4fd446a2bc854fb8a2ac4f67639d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-08-27 12:19:25.000000000","message":"Patch Set 15:\n\nGuess this needs to be re-proposed for Stein now","accounts_in_message":[],"_revision_number":15},{"id":"af8acf1597e992398c0743d99cad6c7c9eca70ba","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-28 13:53:43.000000000","message":"Patch Set 15:\n\n\u003e Guess this needs to be re-proposed for Stein now\n\nYeah, I\u0027m super looking forward to debating this stuff to death for another cycle.\n\n-jay","accounts_in_message":[],"_revision_number":15},{"id":"b359cc060dba39de63390e464c1cd65b651b83a2","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-04 19:08:19.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"4f897090901185516be73bd856469320c2b53947","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-04 19:32:47.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/81/555081/16/check/build-openstack-sphinx-docs/9459c56/html/ : SUCCESS in 6m 07s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/16/check/openstack-tox-pep8/86124e2/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":16},{"id":"54d0cd43cc5fe2f48e31933e306a40470895b8a4","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-10 18:40:33.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"3a04c1e65fed73274d97d50ca190a04272c2f94f","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-10 21:29:57.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"0174d39bfdcf6419e29cc99a8125846da6f4cabf","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-10 21:30:18.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"9b90ac754ecc8561b88eabd0fe59e7de5b15a65d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-10 22:43:19.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"2d5ea36b735f83940287b11c751dd6a5bd961976","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-09-13 16:51:46.000000000","message":"Patch Set 16: Code-Review-1\n\n(7 comments)\n\nFirst notes","accounts_in_message":[],"_revision_number":16},{"id":"73e550cb8d2f2a4e74ea6eb00fa274a4d883b903","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-13 21:32:48.000000000","message":"Patch Set 16: -Code-Review\n\n(4 comments)\n\nI guess there are a couple small things that could be clarified.","accounts_in_message":[],"_revision_number":16},{"id":"0179b5847c1cfc30739521219bd5c62617be719a","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-14 15:01:05.000000000","message":"Patch Set 16:\n\n(5 comments)\n\n@bauzas: I\u0027m not entirely sure what your -1 is due to.","accounts_in_message":[],"_revision_number":16},{"id":"2f6299aca8702c2d6d5302f267a7767b124de82e","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-14 15:48:28.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"1e74425620a9d7e7ec9b743af027b58e7708500b","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-14 15:57:15.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"7ee9dd35494ffdd641e01308b6255ee9b035e348","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-09-14 16:04:33.000000000","message":"Patch Set 18: Code-Review+2\n\n(2 comments)\n\nJay covered my concerns about upgrade path. I feel we\u0027re in a good shape and the few details I still have in mind are implementation details, not requiring to hold the spec. Thanks Jay, very good work.\n\nNote again, how NUMA models things is a separate spec, comments on this should be reserved there.","accounts_in_message":[],"_revision_number":18},{"id":"00358d7c161c1beaf9f25ca661c05b6f8b405e6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-14 16:08:05.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/18/check/openstack-tox-docs/27dca57/html/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/18/check/openstack-tox-pep8/1a5dbed/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":18},{"id":"ada68c0595d9d142a71a5447ee63c9854776c26e","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-17 16:59:06.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)\n\nLooks good, thanks Jay.","accounts_in_message":[],"_revision_number":18},{"id":"eff965c598f9c6dee1c4580b8991b01bd2cdda67","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-09-18 17:00:41.000000000","message":"Patch Set 18:\n\n(13 comments)\n\nNothing -1 worthy but a couple of ideas and questions in place","accounts_in_message":[],"_revision_number":18},{"id":"f051d868e7af3f7a46d903c5e98dbc94fdb7ed74","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-18 17:49:24.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"07cbc1260f5b155ca7001889c5730fdf98f16b7b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-09-19 08:50:59.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"667ca70e7221220b12a0f15f9a8b8024ada8839b","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-09-19 14:48:20.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"8b07fadee6afcd5971143ff7e74d02580df7d3ce","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-19 15:39:19.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"55c67104f3b5d4e254437803bebcc16de3109f70","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-09-19 16:17:03.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"2a9bbc620702204a139dc11fcf383196437bcc73","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-19 16:44:32.000000000","message":"Patch Set 18:\n\n(5 comments)","accounts_in_message":[],"_revision_number":18},{"id":"be4cdf07a4ae2020a4557a3ee76b909d8ea6cc14","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-19 17:01:41.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"758914ddd1b5f0c69075ef7d08072d11984e8dc6","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-19 17:55:07.000000000","message":"Patch Set 18:\n\n(5 comments)","accounts_in_message":[],"_revision_number":18},{"id":"33a93b52bc590182091e8e808118dafe19e54927","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-19 22:56:18.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/19/check/openstack-tox-docs/778d3a1/html/ : SUCCESS in 7m 41s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/19/check/openstack-tox-pep8/6a25f16/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":19},{"id":"5a2cf9f5b09d21468677637168145179834e04a5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-09-20 10:41:31.000000000","message":"Patch Set 19: Code-Review+1\n\nA\u0027ight, good enough. Let\u0027s start on code","accounts_in_message":[],"_revision_number":19},{"id":"bd94282d789a37d20fd8c7466f5216465c171009","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2018-09-21 16:45:43.000000000","message":"Patch Set 19: Code-Review+1\n\nAgreed - this is clear enough to move forward.","accounts_in_message":[],"_revision_number":19},{"id":"742a6f0ff81708968869729959ecc0b9c629eb91","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-21 18:46:17.000000000","message":"Patch Set 19: Code-Review+1\n\nyeppers, time for code","accounts_in_message":[],"_revision_number":19},{"id":"6fa926dc301b2125251064cd966436ecbabe22ea","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-09-27 14:00:29.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"d75852329df899993547a172604eadc97a36f761","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-27 18:25:13.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"810d35b8579861e0fc420a811b075d29fbc61be0","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2018-09-27 18:45:39.000000000","message":"Patch Set 19:\n\n(7 comments)\n\nI\u0027m obviously coming to this very late (though I did review the precursor spec [1]), so I\u0027m not sure I have the balls to -1 at this point, but I have some concerns (and some nits that can be fixed later).\n\n[1] https://review.openstack.org/#/c/543805/","accounts_in_message":[],"_revision_number":19},{"id":"d3645eb3a2edb3b707e927127b568f46f5538432","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-27 18:58:16.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"b8d5fa052fbf1669a8630ac59bb91918a6c6c2a6","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-09-27 19:03:49.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"66109a707df40a9015008588b53327ad2fef988e","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2018-09-27 19:15:42.000000000","message":"Patch Set 19:\n\n(2 comments)\n\nThanks for taking the time to clarify :)","accounts_in_message":[],"_revision_number":19},{"id":"8d1c71d156b2eee918657bc0a29752e1bf07f5ee","author":{"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},"date":"2018-09-27 19:43:20.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"47284d4c794e745da886bb054d8ea39a5c351591","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-27 20:08:14.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"018bcf4d1ed4afcf55ed77c228e6475f841798f3","author":{"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},"date":"2018-09-27 20:26:31.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"ce9c56efe4e663ec22c56a378d223b2291b474b8","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-27 21:04:11.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1d94e4da002a58409761aeadebc880f0e845f3a2","author":{"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},"date":"2018-09-27 23:10:51.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"2c98704cefe39a9c67a0b637530d921d02bfc8c7","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-09-27 23:48:29.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f387162ddd15715bb87d22d811a81ace5640c257","author":{"_account_id":16881,"name":"Munish","email":"mm6021@att.com","username":"munish.mehan@att.com"},"date":"2018-09-28 13:44:19.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"8d39210a82aced9aca2a9a91e536071397c08e7a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-03 10:55:58.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"1b5fdf3c20d6b627a28de263cbab71d09db146be","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-10-08 08:46:33.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)\n\none question","accounts_in_message":[],"_revision_number":19},{"id":"38f881e997f7372d58d060573d9ee56a7fb74ba0","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-10-08 13:52:22.000000000","message":"Patch Set 19:\n\n(5 comments)","accounts_in_message":[],"_revision_number":19},{"id":"dfbeed167dd85338f67ef43b91f08fa6b2e03d07","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-10-08 13:52:30.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"0329dd36a6cc64824c20d8a541b394450a329648","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 14:14:07.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/20/check/openstack-tox-docs/a281db0/html/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/20/check/openstack-tox-pep8/6bf6f4f/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":20},{"id":"a8c151037d1c434223598fd6a955ef1024adee28","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-10-09 04:49:58.000000000","message":"Patch Set 20: Code-Review-1\n\n(4 comments)\n\nSorry, Jay, I want to +2 to this spec, but more questions coming up. One is also for guest without NUMA, I didn\u0027t see we are talking about where to track the pinned PCPU, I guess that is part of data model change we should doc in the spec.\n\nAnother question is for the upgrade issue.","accounts_in_message":[],"_revision_number":20},{"id":"20dc50a98b1ec5abc66f2b1627f9af1ec463845d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-10-23 13:17:48.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"a3317c665a0618dc78a03201726e3d616e61f2fc","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-10-23 13:51:22.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"681c61024e38c6764440da7ae0bbc751f0394fbb","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-10-23 13:55:42.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"ed9f4039b3095b15baef7f99fe9044e32ae8b685","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-10-23 15:22:08.000000000","message":"Patch Set 20: -Code-Review\n\n(1 comment)\n\nstill not sure the thing resources:PCPU doesn\u0027t mean you get vcpu really pin to the a pcpu, that confuses for the end user. so leave my point to 0 first.","accounts_in_message":[],"_revision_number":20},{"id":"3c201eadb842fbced3c107fe7b501101bdc8641b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-10-23 17:05:12.000000000","message":"Patch Set 20: Code-Review-1\n\n(21 comments)\n\nThis is nice and clear, thanks Jay.\n\nMain holdups:\n- Clarify default behaviors (when either/both new conf options are absent)\n- Clarify what happens with reserved values (implicit vs. config)","accounts_in_message":[],"_revision_number":20},{"id":"dff148422bf807c80f4afd8b160db191de344981","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2018-10-25 20:01:11.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)\n\nI haven\u0027t re-read the whole thing, but wanted to address one of Eric\u0027s points.","accounts_in_message":[],"_revision_number":20},{"id":"48ef3f0102d2bc2e591c56a5ca6aa624c493ab66","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-10-25 20:29:58.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"02c432aa118c7d8d158722cc2c25692b01fe9561","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-10-26 13:08:25.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"59849479916941279ff3e232921904bcc036fb2e","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-10-26 13:19:39.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"50c62542ef4cd103621bcf3c53f81371370065cd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-26 13:29:17.000000000","message":"Patch Set 20:\n\n(6 comments)","accounts_in_message":[],"_revision_number":20},{"id":"4ece588b2e3d50f9494d499ee7cc3841bce6b468","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-10-26 13:53:46.000000000","message":"Patch Set 20:\n\n(13 comments)","accounts_in_message":[],"_revision_number":20},{"id":"ff9a0b062dc1a0181f54ec6701aa526c3a3ebf4e","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-10-26 14:09:37.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"6debc5981a8855c7572dec3d0dcca08e716081a2","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-10-26 15:12:19.000000000","message":"Patch Set 21: Code-Review+1\n\nchanges look good.","accounts_in_message":[],"_revision_number":21},{"id":"97c8a002cb02a0582332895b63f44731545eaf48","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-26 15:41:36.000000000","message":"Patch Set 21: Code-Review+1\n\nThis has been around long enough and I think the biggest changes have been documented","accounts_in_message":[],"_revision_number":21},{"id":"585d4aa59483d1eb2989743739ceeda6cfd6de65","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2018-10-26 17:46:24.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"09ab2678f8500dd1470c57fe845f387c5870e5e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-26 18:14:06.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/21/check/openstack-tox-docs/dcc70c4/html/ : SUCCESS in 5m 41s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/21/check/openstack-tox-pep8/59a35d7/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":21},{"id":"31d87e1be72fd5c032c8f6e6844ebb2c4c5048af","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-11-19 05:10:13.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"145148c8c6d9581b6ab4fc56368b859171f2dc9f","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2018-11-27 18:48:04.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"617ce0c2b50fa778ec273248ce8ac41d3afb565a","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2018-12-04 00:22:14.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4be7d405e020371016feb6c28731669d28b53332","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2018-12-13 08:58:49.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"13d645b506a9b66bef93bfd7f33351d5ded13f9d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-12-19 02:09:58.000000000","message":"Patch Set 21: Code-Review-1\n\n(12 comments)\n\nsorry jay.\nthis is largely fine but there are several items \nthat i had reaised in previous version and at the denver ptg that i dont think have been addressed. most of these change would be trival to code but are miniung full to get right form an upgrading and data modeling point of view.","accounts_in_message":[],"_revision_number":21},{"id":"d83a59869109805b4c8bee68864fa5382cc85cf2","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-12-19 12:15:49.000000000","message":"Patch Set 21: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"8fbf2171f95c08821bd7c0a3860fa51e084f315b","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2018-12-20 17:03:08.000000000","message":"Patch Set 21: Code-Review-1\n\n(6 comments)\n\nThat\u0027s a soft -1 but I feel we clarify the upgrade path, in particular given some config options are being reused/changed.\nAFAICU, the Stein release will allow old flavors to keep same behaviour, that\u0027s correct? I haven\u0027t seen in the spec any kind of compatible mechanism description that would explain how we plan such thing.","accounts_in_message":[],"_revision_number":21},{"id":"b03f413097e4ba419f00c742282bd5bd634a4f5b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-01-10 16:09:15.000000000","message":"Patch Set 21: Code-Review-1\n\n(7 comments)\n\nGood catch, Sean. Guess that needs to be clarified before we go any further","accounts_in_message":[],"_revision_number":21},{"id":"d1081bc2c160d3c3f139672cf5ef89065bf80d44","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-01-11 13:05:24.000000000","message":"Patch Set 21:\n\n(4 comments)\n\nso damn tired of this spec. this is why nothing ever seems to improve or get refactored with the NUMA and CPU topology stuff...","accounts_in_message":[],"_revision_number":21},{"id":"a27f7089f1666975d23fd3639b8bfcdafc47b069","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-01-11 14:15:08.000000000","message":"Patch Set 21: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d41b7bdcbb7219ac38c49c8ede23a568188befec","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-02-17 05:06:53.000000000","message":"Patch Set 21: -Code-Review\n\nremove my +2","accounts_in_message":[],"_revision_number":21},{"id":"0a76256e231184c1a1164feb4743227316c4538d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-07 22:43:52.000000000","message":"Patch Set 21: Code-Review+1\n\n(8 comments)\n\nI think this is good enough to go.\n\nDebate about reserved value is not worth holding up for IMO.\n\nUpgrade concerns are the usual catch-22 between backward compatibility and progress. IMO the paths described here are not overly aggressive.","accounts_in_message":[],"_revision_number":21},{"id":"2dda97805526283c5d4f051e01c7f1d3d58754dd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-02 14:27:27.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"1397a4f32f61eb7a818d1617b1a8a3cc49691846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 14:40:58.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/22/check/openstack-tox-docs/a065569/html/ : SUCCESS in 7m 14s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/22/check/openstack-tox-pep8/99f0115/ : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":22},{"id":"f0517fe8fc36fe2178e8699375474fab487f6a5b","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-04-04 15:24:52.000000000","message":"Patch Set 22:\n\n(5 comments)\n\nThe underlying mechanisms here (the conf settings, the resource classes, their meanings) all seem sane.\n\nI\u0027ve got some questions within, but nothing blocking. There\u0027s a bit of attention that will need to be paid to placement with regard to nested rps, noted within, but for the most part this can \"just work\" from that side of things.","accounts_in_message":[],"_revision_number":22},{"id":"2635b90d0672e90e5ae12bcc8f87d7ac68808616","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-04-09 17:08:34.000000000","message":"Patch Set 22: Code-Review-1\n\n(11 comments)\n\nMy major concerns go on the upgrade procedure, besides the nitty s/Stein/Train needed update.","accounts_in_message":[],"_revision_number":22},{"id":"131b809744d152dab57a46049fc8c176947e8888","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-04-09 17:37:10.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"5187411bcaadedbafb55507308968d5979dcc6a1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-04-09 17:37:42.000000000","message":"Patch Set 22:\n\n(1 comment)\n\nThis needs to be updated to reflect the Stein and Train release timing being Train and U right?","accounts_in_message":[],"_revision_number":22},{"id":"a30634203d784fd28d16254e33cc1fc3ec79cb3e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-04-10 08:33:23.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"dcb698ab2f1eedbb9866f864d6498ec61af5342e","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-04-12 00:44:58.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"5edf6ba44ea0ddf4e9e9c8a421d786a02f26dbf5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-15 16:07:27.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"80b82ab686ab4e043200b2f6298dc06193ebbadd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-15 16:07:31.000000000","message":"Patch Set 22:\n\n(5 comments)","accounts_in_message":[],"_revision_number":22},{"id":"2ae66ee76d8c3ed68023e868e9477d86c399feaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 16:26:01.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/23/check/openstack-tox-docs/65d7b8a/html/ : SUCCESS in 8m 55s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/23/check/openstack-tox-pep8/965af53/ : FAILURE in 6m 35s","accounts_in_message":[],"_revision_number":23},{"id":"b7b103fc1750681aad75247a25c5af74489c8362","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-15 18:12:45.000000000","message":"Patch Set 23:\n\n(17 comments)","accounts_in_message":[],"_revision_number":23},{"id":"4d054517b86ad512c7b54f4494056415af1ad0f5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-16 14:15:34.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"e97d0372b91ec8436fea0e05dc2906aee4c86b08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 14:29:30.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/24/check/openstack-tox-docs/a4867dd/html/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/24/check/openstack-tox-pep8/7cfe729/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":24},{"id":"757033c8ec4291e2890493343d4117aec428d0b4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-04-16 14:31:37.000000000","message":"Patch Set 24:\n\n(3 comments)","accounts_in_message":[],"_revision_number":24},{"id":"ba6742529b28fb0c4d9d6d4e0fa87646b3d1eba9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-16 14:48:44.000000000","message":"Patch Set 24:\n\n(3 comments)","accounts_in_message":[],"_revision_number":24},{"id":"5aff72f9fb9db284d4501e7fd621f32dbaa479f3","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-04-16 17:05:13.000000000","message":"Patch Set 24: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"1b1cabb6e16e8986adf41e3150b1b2c31c1dc2e7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-16 17:09:04.000000000","message":"Patch Set 24: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"15b2f9388f2bd903e5e47af29d5af7b3a1508010","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-17 19:16:59.000000000","message":"Patch Set 24: Code-Review-1\n\n(7 comments)\n\nI\u0027ll have to think about the upgrade situation with a fresh brain, but I\u0027ve left some thoughts on the rest.","accounts_in_message":[],"_revision_number":24},{"id":"54e19d4a9e6656245edba210e55bdd7a54395827","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-24 13:08:30.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"bf7f2f940673473cc7ba23284d2fb3f4468136ef","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-24 14:44:11.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"196419370a21b6e561a43b2106760efee93d787c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-24 14:57:05.000000000","message":"Patch Set 24:\n\n(19 comments)","accounts_in_message":[],"_revision_number":24},{"id":"1dea115f47353277e78fe3f9b4e5f51350bffb10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:58:32.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/25/check/openstack-tox-docs/b7480cd/html/ : SUCCESS in 8m 07s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/25/check/openstack-tox-pep8/4ce62fa/ : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":25},{"id":"a7b85a50e6ed393e055181e9d3a34b448f24e2ce","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-24 15:30:19.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"28ff7b16662b0b5642872111e2225a444bcdddbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:46:42.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/26/check/openstack-tox-docs/d9bfdc7/html/ : SUCCESS in 7m 08s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/26/check/openstack-tox-pep8/708b8b1/ : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":26},{"id":"3be89f626e0d6c7eddfaf32169f0300d878b0850","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-24 22:26:42.000000000","message":"Patch Set 26: Code-Review-1\n\n(7 comments)\n\nMy initial reaction is that we\u0027ve taken a step backward in both functionality and UX from PS23-ish.\n\nThe UX is simpler, yes, but IMO it is very confusing that I need to ask for vcpus when I want PCPU resources.\n\nAnd I think not being able to ask for VCPU and PCPU in the same request is going to make it impossible for the CPU priorities spec [1] to use Jay\u0027s suggested approach.\n\n[1] https://review.opendev.org/#/c/649882/","accounts_in_message":[],"_revision_number":26},{"id":"5bb7bc1e9e6ce156ff0d87cd896a9085ad54e849","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-25 00:08:19.000000000","message":"Patch Set 26:\n\n(6 comments)","accounts_in_message":[],"_revision_number":26},{"id":"1051596e1bbb6d0cc2da1ccd5a931b7e562259a9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-25 10:26:05.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"c78a10c919d5ab805eff2ded30d097e54b6bfd65","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-25 10:29:49.000000000","message":"Patch Set 26:\n\n(7 comments)\n\n\u003e My initial reaction is that we\u0027ve taken a step backward in both\n \u003e functionality and UX from PS23-ish.\n\nThis is intentional. This spec is waaay too complicated, as shown but how long it\u0027s dragged on, and I actually want to get started with something during this cycle. As such, I think it wise to forget about NUMA and mixed VCPU/PCPU instances for now and focus on the bare minimum of getting tracking of PCPUs into placement. We can build upon this after the fact but we need _something_.\n\n \u003e The UX is simpler, yes, but IMO it is very confusing that I need to\n \u003e ask for vcpus when I want PCPU resources.\n\nIt is, but ultimately that\u0027s a terminology thing. Unless we want to rename VCPU (the resource) to something like CPU_SHARED (is this a thing?), or go with something like VCPU_DEDICATED instead of PCPU, we\u0027re stuck with this. I think we can deal with it, tbh.\n\n \u003e And I think not being able to ask for VCPU and PCPU in the same\n \u003e request is going to make it impossible for the CPU priorities spec\n \u003e [1] to use Jay\u0027s suggested approach.\n \u003e \n \u003e [1] https://review.opendev.org/#/c/649882/\n\nAs noted by Sean, this shouldn\u0027t be an issue. I\u0027m happy to propose a dependent spec that re-adds that functionality but I really want to keep it out of this one, which is already bloated enough","accounts_in_message":[],"_revision_number":26},{"id":"4606f75b7ad152983b1f10394d093baca68da628","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 10:38:16.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/27/check/openstack-tox-docs/e306154/html/ : SUCCESS in 6m 53s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/27/check/openstack-tox-pep8/486f6e5/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":27},{"id":"46c2b415c3bdda9eb3a2c3fdb00ac14d0933970b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-25 10:46:38.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"90725fe8b0754a79d72d1290efd306f7b5513484","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 10:56:33.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/28/check/openstack-tox-docs/6314f69/html/ : SUCCESS in 6m 59s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/28/check/openstack-tox-pep8/56b8690/ : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":28},{"id":"9ee9a8c30374478d28ff24779e5d5e8bd4941b16","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-25 11:32:08.000000000","message":"Patch Set 28: Code-Review+1\n\n(1 comment)\n\ni am more or less ok with this as it is.\n\nthere are a few bits that i might personally change but i can live with this versions so i think we should move forward with it and then consider other exetntion like mixing floating and pinned core in the same instance or cpu prioity pool separately after this is all implemented.","accounts_in_message":[],"_revision_number":28},{"id":"a75c91246bd24faf532a7d81f217bb7fe14139a7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-25 13:26:45.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"6aba4a91946a47a21adc8152ccace05d42855825","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-25 14:03:56.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"9ad5fddd07c6d6165e4f3e1eb13047832b044821","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-25 15:30:03.000000000","message":"Patch Set 28: Code-Review+2\n\n(8 comments)\n\ngood to go.\n\nI don\u0027t agree with not deprecating the hw:cpu_policy and hw:cpu_thread_policy specs, but it is what it is. :)\n\n-jay","accounts_in_message":[],"_revision_number":28},{"id":"8a2faa9f75951bc300c3c3dc4a1a2a3a66f9d22d","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-25 17:03:37.000000000","message":"Patch Set 26:\n\n(1 comment)\n\nJust reply the comment which @ me. I\u0027m also interested in reviewing the detail of how you isolate the shared and dedicated vcpu. but I didn\u0027t get chance today, and it is very late. I\u0027m also in the travel to denver tomorrow.","accounts_in_message":[],"_revision_number":26},{"id":"295f0d18310b3e01fae4a28a04a33fc9e787e261","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 23:05:41.000000000","message":"Patch Set 28:\n\n(3 comments)\n\nOkay, I\u0027m going to officially abstain from voting on this spec, so don\u0027t wait for me if you want to approve it.\n\nAlex tells me the prioritized-CPU thing doesn\u0027t need to have a VM with a mix of dedicated and share CPUs, so not blocking on that.\n\nAs for the rest, if y\u0027all are happy...","accounts_in_message":[],"_revision_number":28},{"id":"72df0a16cbbd2e0d6cd5c36b2da966477adf3a5e","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-25 23:58:39.000000000","message":"Patch Set 28:\n\n(1 comment)\n\nyea, no blocking from me also, sorry I need time get to sync with internal. Just one question about the NUMACell obj, I feel we should tuning the obj.  But if you think it is the implement detail, then it is ok. Also no point yet, I didn\u0027t review the full spec, since I need to catch flight morning. Free to go if anyone want to approve it.","accounts_in_message":[],"_revision_number":28},{"id":"1fd6c4f2997ce448a0218ff2943ef724080b006a","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-26 00:59:19.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"2d72e9942f3c52bd4b8c84f94caffe2ee03a5505","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-26 02:33:05.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"b9c433735644569a4a39d676aecd7a1231816769","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-26 09:45:45.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"ab19e7b0b059c44b1b395bd43837ef3d3f2f335a","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-26 18:40:14.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"37333bba5d8e83140630b64e73a04c8c2bcd87c0","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-04-27 10:21:10.000000000","message":"Patch Set 28:\n\n(3 comments)\n\nThank you for updating.\nQuestions within before the voting.","accounts_in_message":[],"_revision_number":28},{"id":"d8362f5f328e0812becffa56c0b79720c5012f27","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-27 22:01:19.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"68bb77bb47a68ca2df2ea82a73a51ea102946427","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-04-29 21:06:30.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"238390350df490080344956158956cd0617ff78d","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-02 02:34:38.000000000","message":"Patch Set 28: Code-Review-1\n\n(3 comments)\n\nI think I finally grok the overarching issue with this spec. Currently we\u0027re in model A where resource inventory is defined by two things:\n\n1. the flavor(s) of the instance(s) running on a host\n2. certain config options (ex: vcpu_pin_set)\n\nAnd we\u0027d like to move to model B where resource inventory is defined *only* by config options, and resource *usage* is defined by the instances and their flavors.\n\nCurrently, what this spec would call the PCPU inventory of a host is dependant on the number of instances it has with a dedicated CPU policy. Not helping is the fact that pinning is used for different things: a pinned CPU is a VCPU resource unless its instance has the dedicated CPU policy, in which case it\u0027s a PCPU resource.\n\nI think we need to address this fundamental paradigm shift (buzzword bingo!) before we think about the details of config options and flavor extra specs.\n\nI don\u0027t have an answer, but I think I\u0027ve finally had the epiphany of why the upgrade situation with this spec always sat wrong with me.","accounts_in_message":[],"_revision_number":28},{"id":"7b8a43994bff8cc1a8a139a8068fe6ad7cbb6271","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-02 16:11:00.000000000","message":"Patch Set 28: -Code-Review","accounts_in_message":[],"_revision_number":28},{"id":"19b2cffa52744770c3e7aae4a8e81bec2e9e63a6","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2019-05-02 16:11:45.000000000","message":"Patch Set 28:\n\nCurrently the inventory doesn\u0027t depend on the flavors (in the sense that we don\u0027t modify the inventory depending on what instances are on a node), but I think the key is that we overload the \"VCPU\" count to mean both virtual and physical CPUs and rely on host aggregates to distinguish hosts intended for shared instances vs hosts intended for dedicated instances.","accounts_in_message":[],"_revision_number":28},{"id":"10577729e092bd0de9521a89fff29bc471949e7a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-05-02 16:25:47.000000000","message":"Patch Set 28: Code-Review+2\n\n(8 comments)\n\nAs discussed at the PTG, we feel this spec is in a good enough review state with basic consensus on the direction while some details still are subject to debates.\nI just feel we can move on and work on the implementation that could potentially show some design concerns that we haven\u0027t identified yet. That\u0027s cool, we also said at the PTG that it\u0027s part of the implementation pace, and we shouldn\u0027t just hold spec approval by being afraid of potential issues we could get in the future.\n\n+2\u0027ing the spec as of now accordingly.","accounts_in_message":[],"_revision_number":28},{"id":"9e46d6bcec4396d78e7aef7075eb203dbd934dc9","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-02 16:37:05.000000000","message":"Patch Set 28:\n\nSo I think my confusion was coming from not understanding that aggregates are for *dedicated* CPUs vs floating CPUs. That will indeed allow the \"double-reporting and rely on aggregates to have allocations for only 1 kind of thing\" upgrade strategy work. In any case, let\u0027s merge this and thing about it in smaller chunks during code review.","accounts_in_message":[],"_revision_number":28},{"id":"520fe70a2fd7dd1fb5da956ccd33a17af100f92a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-06 11:13:24.000000000","message":"Patch Set 28:\n\n(6 comments)","accounts_in_message":[],"_revision_number":28},{"id":"179196452df607ed0feaba79475a7adbfc9afbf8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-05-15 09:45:16.000000000","message":"Patch Set 28:\n\n(2 comments)\n\nit was pointed out on irc that we still have not adress the reserved value.\ni have not remved my +1 but i think what is proposed in the spec is incorrect  and we shoudl adress it in a follow up patch.","accounts_in_message":[],"_revision_number":28},{"id":"766145d48a6eb39cc7ae62b787a659815af41b8d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-15 14:32:05.000000000","message":"Patch Set 28:\n\n(10 comments)","accounts_in_message":[],"_revision_number":28},{"id":"c2990eda14b02bcc24bc61eebd77c4e113def22d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-05-15 15:14:27.000000000","message":"Patch Set 28: Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"dfc953e8221d64efa3d4e07fc479d6840c6c9d40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 15:14:40.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"e4d6248a38cd05f3ec34efd270e66556dd00c044","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-05-15 15:14:55.000000000","message":"Patch Set 28:\n\nWe agreed at the PTG that even if the spec needs some follow-up, it\u0027s still good to get it.","accounts_in_message":[],"_revision_number":28},{"id":"17c6540f08e922a020d277eddae6e5a937bfcb4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 15:24:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":28},{"id":"3449016acfaa4f6c7fb6f1fff5600a2d4ef9dc46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 15:24:51.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/555081/28/gate/openstack-tox-docs/45def98/html/ : SUCCESS in 7m 48s\n- openstack-tox-pep8 http://logs.openstack.org/81/555081/28/gate/openstack-tox-pep8/494d605/ : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"0fcfb762567ebdc1d5db35f2db75659d4e8d6a04","revisions":{"6c0e40d32b15a5b53d4d6c218f5087f09f839690":{"kind":"REWORK","_number":1,"created":"2018-03-21 22:13:41.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/1"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6c0e40d32b15a5b53d4d6c218f5087f09f839690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6c0e40d32b15a5b53d4d6c218f5087f09f839690"}]},"branch":"refs/heads/master"},"7e2ddeb4184aa20a27a27b8c3acba5eaffdc0f3b":{"kind":"REWORK","_number":2,"created":"2018-03-21 23:31:17.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/2"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 23:31:09.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7e2ddeb4184aa20a27a27b8c3acba5eaffdc0f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7e2ddeb4184aa20a27a27b8c3acba5eaffdc0f3b"}]},"branch":"refs/heads/master"},"a16fd915f15b62b69f7d176ca4c9a34cea9d884e":{"kind":"REWORK","_number":3,"created":"2018-03-23 14:14:26.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/3"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-23 14:14:18.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a16fd915f15b62b69f7d176ca4c9a34cea9d884e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a16fd915f15b62b69f7d176ca4c9a34cea9d884e"}]},"branch":"refs/heads/master"},"b703c1748ef662fd9978a97921af52627ef8387e":{"kind":"REWORK","_number":4,"created":"2018-03-25 13:23:10.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/4"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-25 13:23:03.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b703c1748ef662fd9978a97921af52627ef8387e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b703c1748ef662fd9978a97921af52627ef8387e"}]},"branch":"refs/heads/master"},"9c2a5299b29e373bde31d4850c06a54040afb98c":{"kind":"REWORK","_number":5,"created":"2018-03-29 18:55:26.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/5"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-29 18:55:11.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9c2a5299b29e373bde31d4850c06a54040afb98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9c2a5299b29e373bde31d4850c06a54040afb98c"}]},"branch":"refs/heads/master"},"5219626163d0c81818cd2284614911938386e85b":{"kind":"REWORK","_number":6,"created":"2018-03-29 19:14:52.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/6"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-29 19:14:44.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5219626163d0c81818cd2284614911938386e85b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5219626163d0c81818cd2284614911938386e85b"}]},"branch":"refs/heads/master"},"e865aed089197ba348a60ec801976c8a85da1449":{"kind":"REWORK","_number":7,"created":"2018-04-12 17:13:22.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/7"}}},"commit":{"parents":[{"commit":"45e038f4d19bbaf4c4c6a55c2217f11b26e40b2f","subject":"Merge \"Allow abort live migrations in queued status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45e038f4d19bbaf4c4c6a55c2217f11b26e40b2f"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-04-12 17:13:10.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e865aed089197ba348a60ec801976c8a85da1449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e865aed089197ba348a60ec801976c8a85da1449"}]},"branch":"refs/heads/master"},"3d1c944f805ea68bbf331a357458bd81279ebdd4":{"kind":"REWORK","_number":8,"created":"2018-04-12 20:47:33.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/8"}}},"commit":{"parents":[{"commit":"45e038f4d19bbaf4c4c6a55c2217f11b26e40b2f","subject":"Merge \"Allow abort live migrations in queued status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45e038f4d19bbaf4c4c6a55c2217f11b26e40b2f"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-04-12 20:47:25.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3d1c944f805ea68bbf331a357458bd81279ebdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3d1c944f805ea68bbf331a357458bd81279ebdd4"}]},"branch":"refs/heads/master"},"23f3dd7422212c53186427d2b8450da4b2394c17":{"kind":"REWORK","_number":9,"created":"2018-04-30 17:00:57.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/9"}}},"commit":{"parents":[{"commit":"45e038f4d19bbaf4c4c6a55c2217f11b26e40b2f","subject":"Merge \"Allow abort live migrations in queued status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/45e038f4d19bbaf4c4c6a55c2217f11b26e40b2f"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-04-30 17:00:51.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/23f3dd7422212c53186427d2b8450da4b2394c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/23f3dd7422212c53186427d2b8450da4b2394c17"}]},"branch":"refs/heads/master"},"1712e28bbe895839599bb1af08fa8ee8a3f772ea":{"kind":"REWORK","_number":10,"created":"2018-05-29 21:00:44.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/10"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-05-29 21:00:34.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1712e28bbe895839599bb1af08fa8ee8a3f772ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1712e28bbe895839599bb1af08fa8ee8a3f772ea"}]},"branch":"refs/heads/master"},"33d37fc22fc4cb8d6b3017c5d67cd64f80183c07":{"kind":"REWORK","_number":11,"created":"2018-05-30 16:27:21.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/11"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-05-30 16:27:06.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33d37fc22fc4cb8d6b3017c5d67cd64f80183c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33d37fc22fc4cb8d6b3017c5d67cd64f80183c07"}]},"branch":"refs/heads/master"},"4396177347d547a61bed8adcca84a5aa6d14441d":{"kind":"REWORK","_number":12,"created":"2018-06-04 18:55:16.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/12"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-06-04 18:55:09.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4396177347d547a61bed8adcca84a5aa6d14441d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4396177347d547a61bed8adcca84a5aa6d14441d"}]},"branch":"refs/heads/master"},"79c314f4da074aae8de55d53df44c751d5d260f6":{"kind":"REWORK","_number":13,"created":"2018-06-04 20:18:39.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/13"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-06-04 20:18:29.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/79c314f4da074aae8de55d53df44c751d5d260f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/79c314f4da074aae8de55d53df44c751d5d260f6"}]},"branch":"refs/heads/master"},"4f540c043fc4b73da618542ebf7376ef8f493095":{"kind":"REWORK","_number":14,"created":"2018-06-05 13:00:38.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/14"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-06-05 13:00:32.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f540c043fc4b73da618542ebf7376ef8f493095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f540c043fc4b73da618542ebf7376ef8f493095"}]},"branch":"refs/heads/master"},"95f1b954bd90431e0dcd8d570353c73fc9b9527d":{"kind":"REWORK","_number":15,"created":"2018-06-06 15:46:52.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/15"}}},"commit":{"parents":[{"commit":"3e21df3365910280d73651e790d023ae482c3cbd","subject":"Merge \"Update placement aggregates spec to clarify generation handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3e21df3365910280d73651e790d023ae482c3cbd"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-06-06 15:40:35.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/95f1b954bd90431e0dcd8d570353c73fc9b9527d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/95f1b954bd90431e0dcd8d570353c73fc9b9527d"}]},"branch":"refs/heads/master"},"0e4a607af5828b331a6623ab89adba4fde09f77a":{"kind":"REWORK","_number":16,"created":"2018-09-04 19:08:19.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/16"}}},"commit":{"parents":[{"commit":"732075880e4f211bdc1b49ab2550bd6de1d63a57","subject":"Merge \"Move rocky implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/732075880e4f211bdc1b49ab2550bd6de1d63a57"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-09-04 19:07:42.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0e4a607af5828b331a6623ab89adba4fde09f77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0e4a607af5828b331a6623ab89adba4fde09f77a"}]},"branch":"refs/heads/master"},"e58a91ba1e7fd95c091e95972a8d0e3c1f43f758":{"kind":"REWORK","_number":17,"created":"2018-09-14 15:48:28.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/17"}}},"commit":{"parents":[{"commit":"732075880e4f211bdc1b49ab2550bd6de1d63a57","subject":"Merge \"Move rocky implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/732075880e4f211bdc1b49ab2550bd6de1d63a57"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-09-14 15:48:22.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e58a91ba1e7fd95c091e95972a8d0e3c1f43f758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e58a91ba1e7fd95c091e95972a8d0e3c1f43f758"}]},"branch":"refs/heads/master"},"587bb5b10af62404f3003a503caf1dbee063a6eb":{"kind":"REWORK","_number":18,"created":"2018-09-14 15:57:15.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/18"}}},"commit":{"parents":[{"commit":"732075880e4f211bdc1b49ab2550bd6de1d63a57","subject":"Merge \"Move rocky implemented specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/732075880e4f211bdc1b49ab2550bd6de1d63a57"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-09-14 15:57:09.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/587bb5b10af62404f3003a503caf1dbee063a6eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/587bb5b10af62404f3003a503caf1dbee063a6eb"}]},"branch":"refs/heads/master"},"5442f7273709f987871a1693a233237b702ae603":{"kind":"REWORK","_number":19,"created":"2018-09-19 17:01:41.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/19"}}},"commit":{"parents":[{"commit":"35f62a53aa1bfcd4e0714dc9992d58703b3f0205","subject":"Merge \"import zuul job settings from project-config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/35f62a53aa1bfcd4e0714dc9992d58703b3f0205"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-09-19 17:01:33.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5442f7273709f987871a1693a233237b702ae603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5442f7273709f987871a1693a233237b702ae603"}]},"branch":"refs/heads/master"},"2ba9eaa914c7dc211b2b2bc7ae123ce7b7d530c9":{"kind":"REWORK","_number":20,"created":"2018-10-08 13:52:30.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/20"}}},"commit":{"parents":[{"commit":"35f62a53aa1bfcd4e0714dc9992d58703b3f0205","subject":"Merge \"import zuul job settings from project-config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/35f62a53aa1bfcd4e0714dc9992d58703b3f0205"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-10-08 13:52:04.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2ba9eaa914c7dc211b2b2bc7ae123ce7b7d530c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2ba9eaa914c7dc211b2b2bc7ae123ce7b7d530c9"}]},"branch":"refs/heads/master"},"5953b568c20e642729eb62e838bd9e9a957e3c9e":{"kind":"REWORK","_number":21,"created":"2018-10-26 14:09:37.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/81/555081/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/21"}}},"commit":{"parents":[{"commit":"35f62a53aa1bfcd4e0714dc9992d58703b3f0205","subject":"Merge \"import zuul job settings from project-config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/35f62a53aa1bfcd4e0714dc9992d58703b3f0205"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-10-26 14:09:29.000000000","tz":-240},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5953b568c20e642729eb62e838bd9e9a957e3c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5953b568c20e642729eb62e838bd9e9a957e3c9e"}]},"branch":"refs/heads/master"},"1a31ec568380d799e4c173dab859c374c8c4ac81":{"kind":"REWORK","_number":22,"created":"2019-04-02 14:27:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/555081/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/22"}}},"commit":{"parents":[{"commit":"32cf20135bbc9c3e3df0da21a12a4f79404c73a0","subject":"docs: Resolve issues with Sphinx 2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/32cf20135bbc9c3e3df0da21a12a4f79404c73a0"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-04-02 14:27:20.000000000","tz":60},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a31ec568380d799e4c173dab859c374c8c4ac81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a31ec568380d799e4c173dab859c374c8c4ac81"}]},"branch":"refs/heads/master"},"c4e33910cb8aa9466d01e60e965480f1b8e81276":{"kind":"REWORK","_number":23,"created":"2019-04-15 16:07:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/555081/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/23"}}},"commit":{"parents":[{"commit":"e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1","subject":"Merge \"Spec: Use in_tree getting allocation candidates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-04-15 16:07:17.000000000","tz":60},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4e33910cb8aa9466d01e60e965480f1b8e81276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4e33910cb8aa9466d01e60e965480f1b8e81276"}]},"branch":"refs/heads/master"},"dee3acaa82a14a48378469fcb083825af70fdb1a":{"kind":"REWORK","_number":24,"created":"2019-04-16 14:15:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/555081/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/24"}}},"commit":{"parents":[{"commit":"e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1","subject":"Merge \"Spec: Use in_tree getting allocation candidates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-04-16 14:15:26.000000000","tz":60},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dee3acaa82a14a48378469fcb083825af70fdb1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dee3acaa82a14a48378469fcb083825af70fdb1a"}]},"branch":"refs/heads/master"},"775a23bc960e3e5f4a6ff2eba7021a9bd268dc13":{"kind":"REWORK","_number":25,"created":"2019-04-24 14:44:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/555081/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/25"}}},"commit":{"parents":[{"commit":"e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1","subject":"Merge \"Spec: Use in_tree getting allocation candidates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-04-24 14:43:48.000000000","tz":60},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/775a23bc960e3e5f4a6ff2eba7021a9bd268dc13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/775a23bc960e3e5f4a6ff2eba7021a9bd268dc13"}]},"branch":"refs/heads/master"},"83f403a2abc594705d233190a1f380017bdba927":{"kind":"REWORK","_number":26,"created":"2019-04-24 15:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/555081/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/26"}}},"commit":{"parents":[{"commit":"e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1","subject":"Merge \"Spec: Use in_tree getting allocation candidates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-04-24 15:30:10.000000000","tz":60},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/83f403a2abc594705d233190a1f380017bdba927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/83f403a2abc594705d233190a1f380017bdba927"}]},"branch":"refs/heads/master"},"fe68b9cd8098b002bc4c9bc43f0f3e4279701254":{"kind":"REWORK","_number":27,"created":"2019-04-25 10:26:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/555081/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/27"}}},"commit":{"parents":[{"commit":"e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1","subject":"Merge \"Spec: Use in_tree getting allocation candidates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-04-25 10:25:11.000000000","tz":60},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fe68b9cd8098b002bc4c9bc43f0f3e4279701254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fe68b9cd8098b002bc4c9bc43f0f3e4279701254"}]},"branch":"refs/heads/master"},"0fcfb762567ebdc1d5db35f2db75659d4e8d6a04":{"kind":"REWORK","_number":28,"created":"2019-04-25 10:46:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/555081/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/555081/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/555081/28"}}},"commit":{"parents":[{"commit":"e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1","subject":"Merge \"Spec: Use in_tree getting allocation candidates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e0ea0e7a8d0596a641e7be1f14e3af23a722ddd1"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-03-21 22:12:50.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-04-25 10:46:31.000000000","tz":60},"subject":"Standardize CPU resource tracking","message":"Standardize CPU resource tracking\n\nWe would like to both simplify the configuration of a compute node with\nregards to CPU resource inventory as well as make the quantitative\ntracking of dedicated CPU resources consistent with the tracking of\nshared CPU resources via the placement API.\n\nChange-Id: I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\nblueprint: cpu-resources\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0fcfb762567ebdc1d5db35f2db75659d4e8d6a04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0fcfb762567ebdc1d5db35f2db75659d4e8d6a04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
