)]}'
{"id":"openstack%2Fnova-specs~572805","triplet_id":"openstack%2Fnova-specs~master~Icfcdf203ccce16f417ae043d71381f344e0feaaa","project":"openstack/nova-specs","branch":"master","topic":"bp/nova-image-download-via-rbd","hashtags":[],"change_id":"Icfcdf203ccce16f417ae043d71381f344e0feaaa","subject":"Add spec for downloading images via RBD","status":"MERGED","created":"2018-06-06 15:16:08.000000000","updated":"2020-05-18 16:08:40.000000000","submitted":"2020-05-18 16:06:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":98,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"572805-1589818005838-02fb7b31","meta_rev_id":"b83113207736cb65910d318eef3dc7514cb09cff","_number":572805,"virtual_id_number":572805,"owner":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 16:06:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},{"value":0,"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":2,"date":"2020-05-18 15:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},{"value":0,"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2020-05-18 15:30:25.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":1,"date":"2020-05-18 15:45:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},{"value":0,"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},{"value":0,"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-06 15:16:25.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2018-06-06 16:04:56.000000000","updated_by":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"reviewer":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"state":"REVIEWER"},{"updated":"2018-06-08 12:47:35.000000000","updated_by":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"reviewer":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"state":"REVIEWER"},{"updated":"2018-06-11 14:59:11.000000000","updated_by":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"reviewer":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"state":"REVIEWER"},{"updated":"2018-06-11 15:24:47.000000000","updated_by":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"reviewer":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"state":"REVIEWER"},{"updated":"2018-10-23 19:38:46.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2018-10-24 15:28:04.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-12-13 14:30:51.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-12-13 14:42:53.000000000","updated_by":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"reviewer":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"state":"REVIEWER"},{"updated":"2019-12-13 23:37:39.000000000","updated_by":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"reviewer":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"state":"REVIEWER"},{"updated":"2020-01-30 18:34:34.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-05-12 14:49:27.000000000","updated_by":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"reviewer":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"state":"REVIEWER"},{"updated":"2020-05-13 07:18:27.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-05-18 15:03:27.000000000","updated_by":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"reviewer":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"state":"REVIEWER"},{"updated":"2020-05-18 15:30:25.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-05-18 15:45:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-05-18 16:06:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9243326ebecf9b43605739503ad0ef26262f446","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2018-06-06 15:16:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"891f912a81223c51cd82eedcf3e9b7b0a4970eb2","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2018-06-06 15:16:25.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\n    Your patch will now be tested automatically by OpenStack testing frameworks\n    and once the automatic tests pass, it will be reviewed by other friendly\n    developers. They will give you feedback and may require you to refine it.\n\n    People seldom get their patch approved on the first try, so don\u0027t be\n    concerned if requested to make corrections. Feel free to modify your patch\n    and resubmit a new change-set.\n\n    Patches usually take 3 to 7 days to be reviewed so be patient and be\n    available on IRC to ask and answer questions about your work. Also it\n    takes generally at least a couple of weeks for cores to get around to\n    reviewing code. The more you participate in the community the more\n    rewarding it is for you. You may also notice that the more you get to know\n    people and get to be known, the faster your patches will be reviewed and\n    eventually approved. Get to know others and become known by doing code\n    reviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\n    Thanks again for supporting OpenStack, we look forward to working with you.\n\n    IRC: https://wiki.openstack.org/wiki/IRC\n    Workflow: http://docs.openstack.org/infra/manual/developers.html\n    Commit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"e3f61ea6216698c05160b0451c87478d862eef4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-06 15:26:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/572805/1/check/build-openstack-sphinx-docs/d0ddd7c/ : FAILURE in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/05/572805/1/check/openstack-tox-pep8/0cfbfbf/ : FAILURE in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"dbfa97b93169997813aca55fda32fd10cb75d72b","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2018-06-06 16:02:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82e84bedc4a70910d6c6a1853ca3bb584961ad9f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-06 16:04:19.000000000","message":"Patch Set 2: Code-Review-1\n\nYou need to move the spec to the rocky directory, queens is released.","accounts_in_message":[],"_revision_number":2},{"id":"80d393da889bd48955e9a4defe090f90c8c84922","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2018-06-06 16:05:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d380dec4046c3544c884006d801f70a76c78e88c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-06 16:21:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/572805/3/check/build-openstack-sphinx-docs/d47f57e/ : FAILURE in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/05/572805/3/check/openstack-tox-pep8/0490f5b/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"96c58a549be9c10be0775de91338b8ccf0c481c8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-06 16:22:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4e9a9456b3db000c07071a7a3bddcc38c9b10741","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2018-06-06 17:45:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1b6b1e5e93b1cea9b0413e336c4b201225b6dca2","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2018-06-06 17:45:19.000000000","message":"Patch Set 3:\n\n(11 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d9d37e5b36d31b7a53cda948a778729dc27e5f3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-06 17:53:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/572805/4/check/build-openstack-sphinx-docs/68efa26/ : FAILURE in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/05/572805/4/check/openstack-tox-pep8/b23cade/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":4},{"id":"f65c826726852351a9a2452ba84f1b66b672add2","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2018-06-07 15:45:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5cd29fec966d2cec2d71807f04aa3ec305656897","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 15:56:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/572805/5/check/build-openstack-sphinx-docs/31fdf00/ : FAILURE in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/05/572805/5/check/openstack-tox-pep8/bd6d7e8/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":5},{"id":"21517ccea2c0171d8ffc5fa471a8a6c1955141ff","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-07 17:47:25.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)\n\nLooks good, just a few small things inline.","accounts_in_message":[],"_revision_number":5},{"id":"b07eb02e7dd4872729c73802b2c443dc6c8f6169","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-07 17:52:26.000000000","message":"Patch Set 5: Workflow-1\n\n(4 comments)\n\nI\u0027m addressing the small nits.","accounts_in_message":[],"_revision_number":5},{"id":"8af8a0f51b2698d12fab57ece7d6882964d8b25a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-07 17:57:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2bfdc738ab9ad7a3ef6c4ab60551cfe8842431a6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-07 17:58:11.000000000","message":"Patch Set 6: Code-Review+2\n\nI\u0027m OK with this at a high level, the devil is going to be in the implementation details, but I\u0027m happy to get this in now so we can undeprecated the image handler extension point since someone came forward with a use case for it.","accounts_in_message":[],"_revision_number":6},{"id":"4287c8c598f3b3e78a4a6f03e752ccdb617baf2b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-07 18:19:10.000000000","message":"Patch Set 6:\n\nWould like mdbooth to look at this given the image cache implications.","accounts_in_message":[],"_revision_number":6},{"id":"aa00f5bd1158ffca745780e14fc1587cadee62c7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-07 18:45:40.000000000","message":"Patch Set 6: -Code-Review\n\nSounds like we might give a week exception to this so we can get some others to look at this to determine if it\u0027s the direction we\u0027d want to go, and if the extension point mechanism is still needed for this, which it might not be, i.e. if you\u0027re using a non-rbd image backend, but nova.conf is configured for ceph stuff in the libvirt driver, we could just check for images from ceph if they aren\u0027t in the cache.","accounts_in_message":[],"_revision_number":6},{"id":"3761a1c24cceda33f4353dde2567330ccfbeb967","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-07 18:48:19.000000000","message":"Patch Set 6:\n\nConversation from IRC today:\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2018-06-07.log.html#t2018-06-07T17:58:50","accounts_in_message":[],"_revision_number":6},{"id":"04726236e4e1b094b7f2e9f7d8331c829680455f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-06-07 18:55:59.000000000","message":"Patch Set 6:\n\n(2 comments)\n\nI would like to get a read on this from someone who understands the image cache code (i.e. mdbooth) before we slam it in. Definitely agree that from a high level this makes sense and we should do it.","accounts_in_message":[],"_revision_number":6},{"id":"38b33e0266186bf035a7ab272188d185b82c7ac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 18:59:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/572805/6/check/build-openstack-sphinx-docs/0a348a3/html/ : SUCCESS in 6m 49s\n- openstack-tox-pep8 http://logs.openstack.org/05/572805/6/check/openstack-tox-pep8/b3c60e8/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":6},{"id":"93aae01db123d1d04e7ba0e5b55139d41f7128ce","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2018-06-07 19:19:29.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4441534e07b1f9eac863b6d89cfdd781b908db71","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-07 21:02:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"31c5f5b948609cc5f71b1dc5aeb847f48937b09f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-07 21:06:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"824b769572dce298f4abedc6ec6701d527b4a844","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2018-06-08 12:47:35.000000000","message":"Patch Set 6:\n\n(4 comments)\n\nI\u0027m generally in favour of this if we can decide where the code should live.\n\nI wonder if we shouldn\u0027t be poking glance\u0027s cinder backend for this instead, though. It\u0027s marked experimental, but we *could* implement host-attach for ceph if it\u0027s not already there. It would be more flexible than a ceph-only optimisation.","accounts_in_message":[],"_revision_number":6},{"id":"81882439377f26b9d1e4f3cc4e9cb2e09a2a9fde","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2018-06-11 15:24:47.000000000","message":"Patch Set 6:\n\n(8 comments)\n\nI am very interested in this. We have a custom \"nova.image.download\" driver that is a mere 50 LOC and it makes a huge difference in performance in certain workloads, where a new image is built and immediately used on many compute hosts that haven\u0027t had the chance to cache it yet.\n\nI can also propose our implementation using the now-deprecated nova.image.download module as PoC. It\u0027s using Python Ceph bindings, not `rbd export` though.","accounts_in_message":[],"_revision_number":6},{"id":"858841cf2827ebb47d61605d2421fc5eef301c4b","author":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"date":"2018-06-11 15:39:46.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"82f7ac98950e0d5d3f57fc0f3c753ec5946798bb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-10-23 18:49:30.000000000","message":"Patch Set 6:\n\nI\u0027m not sure what we\u0027re waiting on, if anything, with this spec. Maybe waiting for mnaser to come back around it, but at least two people have expressed an interest in using something generic for faster rbd image download. The spec would have to be re-targeted to stein at this point, but beyond that I\u0027m not sure what else we need - maybe more operator input from ceph deployments saying this would be useful to them? Maybe that could be floating around the operators list again, or an ops meetup, or the public cloud SIG during the Stein Forum? I know vexxhost and city network is using ceph.","accounts_in_message":[],"_revision_number":6},{"id":"d30b08b8e406eaad8989513b5d71a3a833f76f58","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-10-23 19:38:46.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nI think the use case here makes sense and I\u0027m +1 on adding ceph direct download functionality. But I think we should just add it to the rbd imagebackend instead of preserving the extension point.\n\nSo -1 for only the extension point part of this spec. I\u0027d rather update this spec to describe adding direct download into the rbd imagebackend.","accounts_in_message":[],"_revision_number":6},{"id":"e6a9907cc81b302cbf1f931ff1dcbd949880a697","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-10-24 15:28:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c4612e62b9b8a4fb70e2d84bb577bafab04f336c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-29 22:15:11.000000000","message":"Abandoned\n\nThis change is being abandoned because it proposes a spec against an old release. If the spec should still be considered, please restore the change and move the spec to the current release (please note that spec freeze is in effect for Train - see [1] for the exception process) or the backlog/ directory [2].\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/008060.html\n[2] http://specs.openstack.org/openstack/nova-specs/readme.html#backlog-specifications","accounts_in_message":[],"_revision_number":6},{"id":"a1234d2e9f7c8a3574ac3ebaf097edd255009b39","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-12-13 13:22:40.000000000","message":"Restored","accounts_in_message":[],"_revision_number":6},{"id":"d032f56a566499b192c6746779ddb5b6661c5db2","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2019-12-13 13:27:46.000000000","message":"Patch Set 6: Code-Review-1\n\nI\u0027ll update this to the latest release.","accounts_in_message":[],"_revision_number":6},{"id":"51e0bbe0e77175550066d3a0f50bc46a63783fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 13:37:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365e56dffb65429da38bf15a1522eb84 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12350e8cd2f84fb6b830f7c5fc3f794b : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":6},{"id":"d9301bef7b4df93478a5a59dab16df84375848f5","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2019-12-13 13:38:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6320724399cef3eafdcff0a0be7c31288daa5625","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2019-12-13 13:39:14.000000000","message":"Patch Set 7:\n\nPS7 moves the spec to ussuri","accounts_in_message":[],"_revision_number":7},{"id":"43e7588eb4783deae05ea7a8e6742165dcb8335d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 13:57:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a60d67c3ddfd43e39758556a0e1abff5 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab2692db5a4439a955bf8a374ae2e0e : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":7},{"id":"13d8d84014c008e247d1a26d928cdf2fbc4f2064","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-12-13 14:30:51.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b14d211b334e96974ef587f9d62e7b5692e5841f","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2019-12-13 14:42:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4b825ca6e91a1076602f22311607548becf28154","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-30 18:34:34.000000000","message":"Patch Set 7:\n\nIs this still being pursued? Note that spec freeze is in two weeks (Feb 13).","accounts_in_message":[],"_revision_number":7},{"id":"e44d46eefe15aa19441865c79782617a57cf4ff7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-01-31 17:28:58.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nI think we need to decide whether or not we\u0027re going to un-deprecate an external plugin interface just so we can add this in as an in-tree plugin, or whether or not the functionality gets integrated more directly.\n\nThe plugin plumbing (which is currently deprecated) definitely sits in front of the regular code and just bails on the regular paths if the download succeeds via the plugin. Since ceph is a pretty common use-case, and a low-latency local-disk hypervisor configuration would also be common, it seems weird to have this was-external-but-now-in-tree plugin type behavior for what should be something we can just do natively.","accounts_in_message":[],"_revision_number":7},{"id":"990558a47516d6cc2625de91135735825c081f63","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-10 13:31:41.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"99cafa529ef396097aca01a3958aa40c5543205b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-13 16:54:54.000000000","message":"Abandoned\n\n10:49:52 AM - efried: tobiash: What\u0027s the word on https://review.opendev.org/#/c/572805/ ? Spec freeze is today.\n10:51:19 AM - tobiash: efried: sorry, I was busy with other other things in the meantime, I fear it has to be postponed to the next release :(\n10:51:34 AM - efried: tobiash: Okay, thanks, I\u0027ll do that.\n10:51:41 AM - tobiash: thanks a lot\n\nTo repropose for a subsequent $release please\n- Restore (or ask a nova-core to do so)\n- Move the spec from specs/ussuri/ to specs/$release/\n- Update the blueprint\u0027s Series Goal","accounts_in_message":[],"_revision_number":7},{"id":"696287a8ddb501af8f5a7f320e37e94c2e7c8c5b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-04-28 15:52:31.000000000","message":"Restored\n\nPer gmann","accounts_in_message":[],"_revision_number":7},{"id":"4bf6a34105771038b16cc31e67188c3963cd56e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:16:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cafb3c9a30b4ee88f32bbc6fd4bc786 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad389011a61948b49e30129da52ad94e : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":7},{"id":"6468e6d5917cb79a423dc682ec40e16e21ea272c","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-04-29 07:58:56.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"d55a69ce0b76982487583a21fe5be3aee4edd9c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 08:14:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3aa67dd3c404818ba15d702d3605cf2 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a24b6f56f54ba69739a5c9f225e166 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":8},{"id":"ada977bc327348f20b5af4584e8548cbfbd5a2a8","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-04-29 08:53:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"62c9296e3247454b0c612a748a9624ea6a6911c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 09:09:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69bbd1547c54e848f24fbf945225e5a : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75720c9b20ec4f58823da7bcfd3f1489 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":9},{"id":"267c4c6653c23be8c1e30031c0fae084474bfaf5","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-04 11:41:00.000000000","message":"Patch Set 9:\n\nHi! So what else is needed to get this spec approved?","accounts_in_message":[],"_revision_number":9},{"id":"6f21b5e509299763c46ff225c7472d5bd8a398cf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-04 15:06:19.000000000","message":"Patch Set 9: Code-Review-1\n\nI\u0027m still -1 on undeprecating an external hook to add something internally. Either convince me that doing so makes sense or re-jigger this to be an integrated feature please.","accounts_in_message":[],"_revision_number":9},{"id":"c8e196005c2c50cb275d930fe6f9555383fb55bf","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-07 14:32:53.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"91a0af0decf17ce891123a16cc1f73a3299f79b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:48:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6d559ea53c24cff82e346a60c26530c : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825b75fa50c843b78be1a95b621ea355 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":10},{"id":"abb22f146eccb925c2cf6c730a2e431ac9bedb3a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-11 14:06:08.000000000","message":"Patch Set 10: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"01545b56dc9101e973207b0e651eeeda1a88d0de","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-11 14:28:58.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5376d37eb5bcdaffc65f5fd51dff3f7726e8fa29","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-12 09:55:47.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"84080edf48086da2c1d5a1d3a2e268020dfab086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 10:10:07.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ade92ede454fd5ac45e5d6ee882da4 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ed757f91e445ac9834896d029b859d : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":11},{"id":"d81c8d344b62ca3408fbfacd136aca7c8ed4c6d9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-12 13:46:22.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)\n\nI\u0027d still like to see the words around the test job strengthened, but the intent is there so we can work it out with the code I think.","accounts_in_message":[],"_revision_number":11},{"id":"036c915346394b18409e54f7a6653a4ad873287c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-12 13:47:35.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)\n\nSorry, I just realized one of my requested additional bits of information didn\u0027t get added.","accounts_in_message":[],"_revision_number":11},{"id":"78bf50cd9b936dd646c7bd40064566ec698ae254","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-12 14:03:43.000000000","message":"Patch Set 11:\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"51b66f0c23b032c6c459f372c8d5e31d12543675","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-12 14:18:16.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"26c6a9463220ad5475dbda7690dd9a9666973a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 14:31:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317d09101fb34d18875babd766af6721 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edeb41cff6324010bde92950eaf2fe75 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":12},{"id":"3ea7483efeaa9fcaf1b224aab5668afc514c4ffe","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-12 14:31:50.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"aaad01cbf6278240e27112ccba0751bdf2d8fa0f","author":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"date":"2020-05-12 14:49:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1ba7d78ea49404f38dc1fd9b5b299470cf8fa3d8","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-12 15:12:53.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"60ed833ffec6c9c9dcfd1f52f13d3252ed94307f","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-12 15:32:49.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5c55063b8be0217638c873c7fa86c66eec73833d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 15:46:54.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8fe58070cca423783f6cad199ec2858 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d994323c8712446e9955c7b925d60579 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":13},{"id":"ab0d5e195a04b7d3fdadc4d705b5717e22b40ef8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-12 15:48:05.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"ac9c0ae9274eee5248e86132f36119c9d57fe809","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-13 01:05:59.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"da29e3e7a6e6a92efa535a310fcb561b5bb79ce5","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-13 06:57:58.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2559bfeb0a3ca71bef26c63610988943e341c168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 07:14:47.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42eae22dbf7a4b598c1636ae7cfbac3a : SUCCESS in 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31c015ef71ef4344a0f3a094af55d515 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":14},{"id":"f9d40a492fd7e4573ff4f1be145876ba30f723d9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-13 07:18:27.000000000","message":"Patch Set 14: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":14},{"id":"79a08216cc63a9af070a8dcd6e0b8660a0066639","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-14 14:08:32.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"91cc9888f70b8dd0aea19b9b73510fca194baec5","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-05-18 13:10:32.000000000","message":"Patch Set 14: Code-Review-1\n\n(7 comments)\n\nLGTM overall, -1 as I\u0027d like the unrelated ``[libvirt]/images_rbd_pool`` configurable dropped if you could respin.","accounts_in_message":[],"_revision_number":14},{"id":"d96b2e46b817277c4124e9a98188f4107214f6e6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-18 14:08:23.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c1b61727e77a640bafdab4d8f1e419ccd96bd429","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-18 14:27:37.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"629ed80ae1af7d3eab1aade4edd6370563cfaf13","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-05-18 14:44:17.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"bd3f3c112be110ab8898bd829902ee8abc343557","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-18 14:58:06.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"09b84ec4ed56bf128cb50b8ddced23153459717d","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2020-05-18 15:03:27.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"cd345a2eed746af3740ecfebb1695973265bc934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 15:08:10.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59c1c6cda80d4ba2bf5016cbef68ee98 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ec25c66ceb4fc0b4c45ba4db4a5ba8 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":15},{"id":"108f4c1c858fc64c8ceb29587a745d0920fdad62","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-05-18 15:30:25.000000000","message":"Patch Set 15: Code-Review+2\n\nLGTM now thanks","accounts_in_message":[],"_revision_number":15},{"id":"9c6b2bf5a2dc728dda38de387bf3be12deb3e52b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-05-18 15:45:59.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"666db01b98dea57425057a1564f5d3d193f2db99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 15:46:11.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"492b94435a88d35290cfe4fe0e0dc8a2174b4a09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 16:06:45.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a5837d32c1c4f38a83c11b83cedc9c1 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0a5c70395174f9bb570d9b57845ca44 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":15},{"id":"499d4b01b4f140bd4bec2eb835dc70bf6c3aaba2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 16:06:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"5500b968937f9abd22ae7bf1efea174be61c0caf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 16:08:40.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/1074145669df49dbbab9cfcd7b76bfdd : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"07254dba6bfc96d929f930c69481bca097602e7f","revisions":{"b1c76bf32dcf0e439dd0057810edc738d279480f":{"kind":"REWORK","_number":1,"created":"2018-06-06 15:16:08.000000000","uploader":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"ref":"refs/changes/05/572805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/1"}}},"commit":{"parents":[{"commit":"63d703df2c455382573027f62994019caac9b9c2","subject":"Merge \"Add \u0027numa-aware-vswitches\u0027 spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/63d703df2c455382573027f62994019caac9b9c2"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:15:26.000000000","tz":-300},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b1c76bf32dcf0e439dd0057810edc738d279480f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b1c76bf32dcf0e439dd0057810edc738d279480f"}]},"branch":"refs/heads/master"},"079875bee642780e596e8b8a9d004ad59cea1a9e":{"kind":"REWORK","_number":2,"created":"2018-06-06 16:02:44.000000000","uploader":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"ref":"refs/changes/05/572805/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/2"}}},"commit":{"parents":[{"commit":"63d703df2c455382573027f62994019caac9b9c2","subject":"Merge \"Add \u0027numa-aware-vswitches\u0027 spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/63d703df2c455382573027f62994019caac9b9c2"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 16:03:06.000000000","tz":-300},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/079875bee642780e596e8b8a9d004ad59cea1a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/079875bee642780e596e8b8a9d004ad59cea1a9e"}]},"branch":"refs/heads/master"},"ddf34f4a97989683f2d331578f718e34ac60b105":{"kind":"REWORK","_number":3,"created":"2018-06-06 16:05:41.000000000","uploader":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"ref":"refs/changes/05/572805/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/3"}}},"commit":{"parents":[{"commit":"63d703df2c455382573027f62994019caac9b9c2","subject":"Merge \"Add \u0027numa-aware-vswitches\u0027 spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/63d703df2c455382573027f62994019caac9b9c2"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 16:06:06.000000000","tz":-300},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ddf34f4a97989683f2d331578f718e34ac60b105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ddf34f4a97989683f2d331578f718e34ac60b105"}]},"branch":"refs/heads/master"},"7401f514a68b005ee4b0496f9d5630a2fec0c2a9":{"kind":"REWORK","_number":4,"created":"2018-06-06 17:45:16.000000000","uploader":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"ref":"refs/changes/05/572805/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/4"}}},"commit":{"parents":[{"commit":"63d703df2c455382573027f62994019caac9b9c2","subject":"Merge \"Add \u0027numa-aware-vswitches\u0027 spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/63d703df2c455382573027f62994019caac9b9c2"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 17:42:24.000000000","tz":-300},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7401f514a68b005ee4b0496f9d5630a2fec0c2a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7401f514a68b005ee4b0496f9d5630a2fec0c2a9"}]},"branch":"refs/heads/master"},"d2e6902b1e6bf1836f1b1472de1662f25d7f8d43":{"kind":"REWORK","_number":5,"created":"2018-06-07 15:45:28.000000000","uploader":{"_account_id":28397,"name":"Curt Moore","email":"curt.moore@gmail.com","username":"jcmoore"},"ref":"refs/changes/05/572805/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/5"}}},"commit":{"parents":[{"commit":"63d703df2c455382573027f62994019caac9b9c2","subject":"Merge \"Add \u0027numa-aware-vswitches\u0027 spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/63d703df2c455382573027f62994019caac9b9c2"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-07 15:45:52.000000000","tz":-300},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2e6902b1e6bf1836f1b1472de1662f25d7f8d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2e6902b1e6bf1836f1b1472de1662f25d7f8d43"}]},"branch":"refs/heads/master"},"3b1de56c1faefd279745b45e5d6c47df61b14531":{"kind":"REWORK","_number":6,"created":"2018-06-07 17:57:32.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/05/572805/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/6"}}},"commit":{"parents":[{"commit":"63d703df2c455382573027f62994019caac9b9c2","subject":"Merge \"Add \u0027numa-aware-vswitches\u0027 spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/63d703df2c455382573027f62994019caac9b9c2"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-06-07 17:57:31.000000000","tz":-240},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3b1de56c1faefd279745b45e5d6c47df61b14531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3b1de56c1faefd279745b45e5d6c47df61b14531"}]},"branch":"refs/heads/master"},"a8786283964f4250f085a41a1c71a2f36f0a13f4":{"kind":"REWORK","_number":7,"created":"2019-12-13 13:38:40.000000000","uploader":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"ref":"refs/changes/05/572805/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/7"}}},"commit":{"parents":[{"commit":"b4c98982014b197b891200a3cb087b342522cbed","subject":"Merge \"Virtual instance rescue with stable disk devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b4c98982014b197b891200a3cb087b342522cbed"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","date":"2019-12-13 13:38:29.000000000","tz":60},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a8786283964f4250f085a41a1c71a2f36f0a13f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a8786283964f4250f085a41a1c71a2f36f0a13f4"}]},"branch":"refs/heads/master"},"c59383e4951d0df1671b0e039f8713cf16614f98":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-04-29 07:58:56.000000000","uploader":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"ref":"refs/changes/05/572805/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/8"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Jiří Suchomel","email":"jiri.suchomel@suse.com","date":"2020-04-29 07:58:16.000000000","tz":120},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c59383e4951d0df1671b0e039f8713cf16614f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c59383e4951d0df1671b0e039f8713cf16614f98"}]},"branch":"refs/heads/master"},"8ff3bbd23f787995e11776db55589be5bca03404":{"kind":"REWORK","_number":9,"created":"2020-04-29 08:53:16.000000000","uploader":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"ref":"refs/changes/05/572805/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/9"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Jiří Suchomel","email":"jiri.suchomel@suse.com","date":"2020-04-29 08:53:03.000000000","tz":120},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8ff3bbd23f787995e11776db55589be5bca03404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8ff3bbd23f787995e11776db55589be5bca03404"}]},"branch":"refs/heads/master"},"7977fb270e463cc791b428c8a13517f2cfc50759":{"kind":"REWORK","_number":10,"created":"2020-05-07 14:32:53.000000000","uploader":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"ref":"refs/changes/05/572805/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/10"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Jiří Suchomel","email":"jiri.suchomel@suse.com","date":"2020-05-07 14:32:30.000000000","tz":120},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7977fb270e463cc791b428c8a13517f2cfc50759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7977fb270e463cc791b428c8a13517f2cfc50759"}]},"branch":"refs/heads/master"},"c4e509edc1a30f64eb5a72b30feb66d8c8ccead4":{"kind":"REWORK","_number":11,"created":"2020-05-12 09:55:47.000000000","uploader":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"ref":"refs/changes/05/572805/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/11"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Jiří Suchomel","email":"jiri.suchomel@suse.com","date":"2020-05-12 09:55:34.000000000","tz":120},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4e509edc1a30f64eb5a72b30feb66d8c8ccead4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4e509edc1a30f64eb5a72b30feb66d8c8ccead4"}]},"branch":"refs/heads/master"},"c3a3aacedb58cd16721121a89050c2a4c4097aac":{"kind":"REWORK","_number":12,"created":"2020-05-12 14:18:16.000000000","uploader":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"ref":"refs/changes/05/572805/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/12"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Jiří Suchomel","email":"jiri.suchomel@suse.com","date":"2020-05-12 14:18:01.000000000","tz":120},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c3a3aacedb58cd16721121a89050c2a4c4097aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c3a3aacedb58cd16721121a89050c2a4c4097aac"}]},"branch":"refs/heads/master"},"17829ac290c3ad0f4abbf62adf519dbcf2a1ccc9":{"kind":"REWORK","_number":13,"created":"2020-05-12 15:32:49.000000000","uploader":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"ref":"refs/changes/05/572805/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/13"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Jiří Suchomel","email":"jiri.suchomel@suse.com","date":"2020-05-12 15:32:30.000000000","tz":120},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/17829ac290c3ad0f4abbf62adf519dbcf2a1ccc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/17829ac290c3ad0f4abbf62adf519dbcf2a1ccc9"}]},"branch":"refs/heads/master"},"7d8ddd5177f34bcca03d67c1b2f7c63218b090e1":{"kind":"REWORK","_number":14,"created":"2020-05-13 06:57:58.000000000","uploader":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"ref":"refs/changes/05/572805/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/14"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Jiří Suchomel","email":"jiri.suchomel@suse.com","date":"2020-05-13 06:57:42.000000000","tz":120},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d8ddd5177f34bcca03d67c1b2f7c63218b090e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7d8ddd5177f34bcca03d67c1b2f7c63218b090e1"}]},"branch":"refs/heads/master"},"07254dba6bfc96d929f930c69481bca097602e7f":{"kind":"REWORK","_number":15,"created":"2020-05-18 14:58:06.000000000","uploader":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"ref":"refs/changes/05/572805/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/05/572805/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/05/572805/15"}}},"commit":{"parents":[{"commit":"1d8709a091de8b33b754dcd9bf0f3e1af6899e9c","subject":"Merge \"Update SPEC: Using PUT instead of PATCH API to update delete_on_termination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1d8709a091de8b33b754dcd9bf0f3e1af6899e9c"}]}],"author":{"name":"Curt Moore","email":"curt.moore@garmin.com","date":"2018-06-06 15:12:10.000000000","tz":-300},"committer":{"name":"Jiří Suchomel","email":"jiri.suchomel@suse.com","date":"2020-05-18 14:57:54.000000000","tz":120},"subject":"Add spec for downloading images via RBD","message":"Add spec for downloading images via RBD\n\nThis spec adds details on the changes needed to allow nova the ability\nto download Glance images into the libvirt image cache via a direct RBD\ndownload instead of over the Glance HTTP API.\n\nChange-Id: Icfcdf203ccce16f417ae043d71381f344e0feaaa\nBlueprint: nova-image-download-via-rbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/07254dba6bfc96d929f930c69481bca097602e7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/07254dba6bfc96d929f930c69481bca097602e7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
