)]}'
{"id":"openstack%2Fnova-specs~580336","triplet_id":"openstack%2Fnova-specs~master~If06bd29a8fcad322aed337bb7fdb54aae304c8dc","project":"openstack/nova-specs","branch":"master","topic":"bp/destroy-instance-with-datavolume","hashtags":[],"change_id":"If06bd29a8fcad322aed337bb7fdb54aae304c8dc","subject":"Support re-configure deleted_on_termination in server","status":"MERGED","created":"2018-07-05 10:07:54.000000000","updated":"2020-02-23 01:23:46.000000000","submitted":"2020-02-22 02:11:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":120,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"580336-1582337468305-5f1f21a0","meta_rev_id":"d070fa94f6ba21bbc0345ebc5f86e141f35ce5e0","_number":580336,"virtual_id_number":580336,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22378,"name":"Kevin Qi","email":"qiky@inspur.com","username":"qiky"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-22 02:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2020-02-21 02:58:40.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22378,"name":"Kevin Qi","email":"qiky@inspur.com","username":"qiky"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":2,"date":"2020-02-21 15:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-2,"date":"2020-02-23 01:23:46.000000000","post_submit":true,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-02-21 23:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22378,"name":"Kevin Qi","email":"qiky@inspur.com","username":"qiky"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2020-02-22 02:00:02.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22378,"name":"Kevin Qi","email":"qiky@inspur.com","username":"qiky"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"value":0,"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22378,"name":"Kevin Qi","email":"qiky@inspur.com","username":"qiky"},{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-14 20:22:37.000000000","updated_by":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"reviewer":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"state":"REVIEWER"},{"updated":"2018-10-29 10:14:44.000000000","updated_by":{"_account_id":22378,"name":"Kevin Qi","email":"qiky@inspur.com","username":"qiky"},"reviewer":{"_account_id":22378,"name":"Kevin Qi","email":"qiky@inspur.com","username":"qiky"},"state":"REVIEWER"},{"updated":"2018-11-06 08:21:29.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2019-07-25 05:59:48.000000000","updated_by":{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},"reviewer":{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},"state":"REVIEWER"},{"updated":"2019-07-25 06:00:03.000000000","updated_by":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"reviewer":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"state":"REVIEWER"},{"updated":"2019-07-25 06:00:20.000000000","updated_by":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"reviewer":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"state":"REVIEWER"},{"updated":"2019-07-25 06:01:15.000000000","updated_by":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"reviewer":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"state":"REVIEWER"},{"updated":"2019-07-25 06:06:27.000000000","updated_by":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"reviewer":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"state":"REVIEWER"},{"updated":"2019-07-25 06:14:37.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"REVIEWER"},{"updated":"2019-07-25 14:02:20.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-10-29 20:37:32.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-10-30 01:13:59.000000000","updated_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"reviewer":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"state":"REVIEWER"},{"updated":"2019-12-01 02:55:56.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-12-16 02:20:51.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-12-16 11:51:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-09 14:22:52.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-02-14 17:40:21.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-02-20 14:01:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-20 16:01:21.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-02-21 15:35:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-02-21 23:09:12.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-02-22 02:00:02.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-02-22 02:11:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e942d3cbf3fb73ddde744a1a1b1e0c99efe57122","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-07-05 10:07:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e842268c7e541a55ae2aa84f842b597e78a446dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-05 10:15:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/36/580336/1/check/build-openstack-sphinx-docs/c6f64b4/html/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/1/check/openstack-tox-pep8/57c4d56/ : FAILURE in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2eec9a5c08c23bd741bf70a79a65cbf78dd1bb06","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-07-05 12:07:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ebc968042bda40d4f9140c5b31cfd116f01a9ead","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-05 12:17:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/36/580336/2/check/build-openstack-sphinx-docs/4134519/html/ : SUCCESS in 7m 08s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/2/check/openstack-tox-pep8/4cecac0/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"93631578d789b23ccb07c56500c6c2fa0712d0ba","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-08-16 07:03:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"bb3d064b67b7c2e21971c74be3c62a8918cdfbc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-16 07:11:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/36/580336/3/check/build-openstack-sphinx-docs/93fc558/html/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/3/check/openstack-tox-pep8/45643db/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"15ab3db4f3be803520ee4da78ee282db1f18789d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-08-16 18:17:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nHi, we\u0027re just wrapping up the Rocky release, so this needs to be updated to be proposed to the specs/stein directory (you might have to rebase, the stein directory was added yesterday).","accounts_in_message":[],"_revision_number":3},{"id":"88130acdb46edc1b1ce6b442b73705c27e69af4d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-08-17 01:40:43.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nHi melanie, thank you for your review, I\u0027ll change it.","accounts_in_message":[],"_revision_number":3},{"id":"88f0a0889793d28d12ad1c22b6ca3c2b1def2a1e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-08-17 01:50:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4c70632b67dd82a1c751e72d2f847aff242906ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-17 02:00:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/36/580336/4/check/build-openstack-sphinx-docs/10c4145/html/ : SUCCESS in 5m 23s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/4/check/openstack-tox-pep8/bdcf938/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":4},{"id":"b3e5bd4c77bd3065499c4c678b13abb6ef617c2b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-10-23 21:35:41.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)\n\nI definitely don\u0027t like the idea of doing this within the PUT /servers/{server_id} API. It would much more naturally fit in the os-volume_attachments API, but the PUT method on that resource is already being used for the swap volume operation. We could entertain the idea of changing swap volume to a server action, since that\u0027s really what it is, but it would be a much larger change. Alternatively, if we allowed simply specifying delete_on_termination in the volume attach API, if you needed to change that value, you could just detach and re-attach with the new value - I realize that\u0027s a bit heavy handed, but it\u0027s an alternative.","accounts_in_message":[],"_revision_number":4},{"id":"ae9f4e1f3bd54e22459e85a407cc4b606c2c0ddd","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-10-23 23:09:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\n\u003e (5 comments)\n \u003e \n \u003e I definitely don\u0027t like the idea of doing this within the PUT\n \u003e /servers/{server_id} API. It would much more naturally fit in the\n \u003e os-volume_attachments API, but the PUT method on that resource is\n \u003e already being used for the swap volume operation. We could\n \u003e entertain the idea of changing swap volume to a server action,\n \u003e since that\u0027s really what it is, but it would be a much larger\n \u003e change. Alternatively, if we allowed simply specifying\n \u003e delete_on_termination in the volume attach API, if you needed to\n \u003e change that value, you could just detach and re-attach with the new\n \u003e value - I realize that\u0027s a bit heavy handed, but it\u0027s an\n \u003e alternative.\n\nAgreed that I don\u0027t think adding this to PUT /servers/\u003cserver\u003e is appropriate.\n\nAnd I don\u0027t see realistically changing the swap volume API to a server action, especially not solely for the ability to modify delete_on_termination. If we moved swap volume to a server action and re-purposed volume PUT for updating delete_on_termination, would we use it to allow other updates to block device mappings? I\u0027m thinking about use cases where operators have changed storage backend credentials and need to update block device mapping connection_info (though I\u0027m not sure they would find volume PUT friendly enough for that). If there is other usefulness in updating block device mappings, I could maybe see it being worth changing swap volume to a server action and re-purposing volume PUT.\n\nI think the alternative mentioned here makes sense. Adding the ability to specify delete_on_termination would bring parity with the server create API. But agreed it would be a bit of a pain, having to detach/attach to change it.","accounts_in_message":[],"_revision_number":4},{"id":"3fb1d206cc362d04ac77992429417e322245a2a8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-10-24 08:45:17.000000000","message":"Patch Set 4:\n\n(6 comments)\n\nThanks mriedem and melwitt.\nI agree with Matt\u0027s suggestion, and I have not found a better API method than the PUT method in the nova API.\nOr add an API method just to set the value of the delete_on_termination field of the data volume?\nIf I do this, I prefer to use the PUT method instead of adding a new method. I think this is easy to accept for API changes.\n\nIf the volume attach API supports delete_on_termination, there is also a need to add a new microversion, I think it is not appropriate to put in the current specs, I submitted another specs application, please see[1].\n\n[1] https://review.openstack.org/#/c/612949/","accounts_in_message":[],"_revision_number":4},{"id":"d342be729cec6c2c88cfe415bfb5b5f023c450df","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-10-24 08:54:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"faf5b5ff46c0ccfa977f74a664e189a0acf557ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 10:11:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/5/check/openstack-tox-docs/a26439c/html/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/5/check/openstack-tox-pep8/3c4b817/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":5},{"id":"00be10c670c9cd27d01b8b225aa8908f58004143","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-10-24 12:32:00.000000000","message":"Patch Set 5:\n\n\u003e If the volume attach API supports delete_on_termination, there is also a need to add a new microversion\n\nYou will need a new microversion either way. If the request body changes anywhere in the API, it requires a new microversion because we need to validate the schema (and for interoperability).","accounts_in_message":[],"_revision_number":5},{"id":"152150da37c83b4b8c4c409bf162649fa4b4b76b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-10-25 00:49:15.000000000","message":"Patch Set 5:\n\n\u003e \u003e If the volume attach API supports delete_on_termination, there is\n \u003e also a need to add a new microversion\n \u003e \n \u003e You will need a new microversion either way. If the request body\n \u003e changes anywhere in the API, it requires a new microversion because\n \u003e we need to validate the schema (and for interoperability).\n\nI know this, I mean that the update_on_termination of the configuration data volume needs to use the update server API, the other specs are used for the volume attach API, and the delete_on_termination can be configured when the volume is attached for the instance.:P","accounts_in_message":[],"_revision_number":5},{"id":"44cf41e93df8a44175fc5b6a32d7e7cb1e5e3c58","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-10-25 00:53:56.000000000","message":"Patch Set 5:\n\nHi Matt. I am sorry for the error sentence, use the following words instead of the above reply.\n\nI mean that the \u0027delete_on_termination\u0027 of the configuration data volume needs to use the update server API, the other specs is used for the volume attach API, and the \u0027delete_on_termination\u0027 can be configured when the volume is attached for the instance.\nSo I add the another specs.","accounts_in_message":[],"_revision_number":5},{"id":"1ad4afc582751804063a97a93a934f5a9735a4d4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-11-01 15:30:50.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"127f68be87c1d09097877571776daca2f343f2c0","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-02 09:56:35.000000000","message":"Patch Set 5:\n\n\u003e (1 comment)\n\nThanks Matt, I have some questions on [1], I hope you can reply :P\n\n[1] http://paste.openstack.org/show/733923/","accounts_in_message":[],"_revision_number":5},{"id":"273ac4816d492ac394236a0a62d4babc63ef222c","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-11-06 08:21:29.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7306d13979705c81f3b4e30fe46104f9fe9db506","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-06 11:28:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ecdda43547fb6f3fc903a5348ff2fce94461947a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-06 11:49:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b675a34b8ce8108f1c13086b662914bfe3193a37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-06 11:59:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/6/check/openstack-tox-docs/db1df1b/html/ : SUCCESS in 5m 50s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/6/check/openstack-tox-pep8/926e475/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":6},{"id":"27aa815f6c7620ab1488f25b1721b1d2d28d862d","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-11-28 07:54:15.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c976aefbf623c6cb46f8565c2b0d7892ffdb75be","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-11-28 08:07:16.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8954a834bdf6d976ed9558f47946e1ed0f696687","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-11-28 08:09:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"acff8b0efa8ac590325b18174d0d59e940da6516","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-28 08:40:38.000000000","message":"Patch Set 6:\n\n(4 comments)\n\nThanks Alex.","accounts_in_message":[],"_revision_number":6},{"id":"b7d1b50fae9aeeb6e54b06618fc7f904309463e9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-28 11:33:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"73e7de481ca78e4e868f67a4d190717ced1b9292","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 12:56:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/7/check/openstack-tox-docs/8b84b6e/html/ : SUCCESS in 7m 21s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/7/check/openstack-tox-pep8/79e7867/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":7},{"id":"2fb32e0934295c679a075ad683f33eaec1c03cd5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-11-29 00:11:16.000000000","message":"Patch Set 7:\n\nThis came up at the summit in Berlin, you can read the notes here:\n\nhttps://etherpad.openstack.org/p/BER-bfv-improvements\n\nTo summarize, there are kind of mixed feelings on this, and there are two parts to the change:\n\n1. Specifying delete_on_termination when attaching a volume - there was generally more agreement on this being OK since it would have parity with the server create block_device_mapping_v2 flag, but I would not say there was a lot of enthusiasm for adding this in general.\n\n2. Being able to update the delete_on_termination flag on existing attachments got a less than favorable reception, mostly because of the complexity and implications it could have to things like Heat stacks. Also, there are several alternatives to the use case for this (and the use case of the user having to manually delete volumes later isn\u0027t that strong to begin with).\n\nSo just FYI but I don\u0027t think you\u0027ll probably get much support for the latter proposal. Maybe I shouldn\u0027t have asked you to combine the specs. :)","accounts_in_message":[],"_revision_number":7},{"id":"f688eef1423d2c071f07c2742cd91489605732b8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-29 01:40:17.000000000","message":"Patch Set 7:\n\nHi Matt, I read the information on the etherpad, which is consistent with what you summarized above. As mentioned before the merger, personally, I insisted on separating the two features to discuss, so I need to restore *that* patch[1], and roll back the current patch to PS5 (copy the contents of PS5 to PS8)? Just wasting some time.\n\nIn addition, after the merger of these two patches, it is a challenge for the reviewer, which is more confusing.","accounts_in_message":[],"_revision_number":7},{"id":"62d123bfb0f04475b4503b691e9a16b0ed0f091a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-29 02:18:32.000000000","message":"Patch Set 7:\n\n\u003e Hi Matt, I read the information on the etherpad, which is\n \u003e consistent with what you summarized above. As mentioned before the\n \u003e merger, personally, I insisted on separating the two features to\n \u003e discuss, so I need to restore *that* patch[1], and roll back the\n \u003e current patch to PS5 (copy the contents of PS5 to PS8)? Just\n \u003e wasting some time.\n \u003e \n \u003e In addition, after the merger of these two patches, it is a\n \u003e challenge for the reviewer, which is more confusing.\n\n[1] https://review.openstack.org/#/c/612949","accounts_in_message":[],"_revision_number":7},{"id":"3db8989682eb8c1b02118fc37e69c5e9cd86193b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-11-29 14:29:20.000000000","message":"Patch Set 7:\n\nYou can separate the specs if you want, or ask about it in the mailing list to get wider input. As I said, I\u0027m not sure that *either* proposed change has much support, but you can try. Sorry for wasting your time on this with the back and forth.","accounts_in_message":[],"_revision_number":7},{"id":"c9c2dd46bee3cdc7d9b5459ea73d4acda171a66f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-30 02:33:54.000000000","message":"Patch Set 7:\n\nHi Matt, this is nothing :). I will take apart the two specs first, then investigate the *heat* detection mechanism. If necessary, I will initiate the mailing list.","accounts_in_message":[],"_revision_number":7},{"id":"693b2f2491e17898838f9aa019f520cf785bdee6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-30 03:15:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"eb5a5bfd9ff7781f4cc8de0dece278ebc7c37d86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 03:33:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/8/check/openstack-tox-docs/d989349/html/ : SUCCESS in 6m 40s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/8/check/openstack-tox-pep8/d2e305c/ : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":8},{"id":"f68cf20cc753cda93b1df9dac16e49198d184363","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-11-30 03:39:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"491037d66e8fdb30e5e7fc7ef9213dbe15d716af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 04:08:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/9/check/openstack-tox-docs/2af43a1/html/ : SUCCESS in 6m 11s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/9/check/openstack-tox-pep8/da68e08/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":9},{"id":"4da26aac79aa39e7fcf8878979239542da0ef20d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-03-05 00:58:02.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"33dd790ce8ff34386056cbd23f80920d217f41da","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-03-05 00:59:32.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0d0c7c705ba71ad9f601c7a8856283359701224d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 01:06:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/11/check/openstack-tox-docs/364183d/html/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/11/check/openstack-tox-pep8/3d5d911/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":11},{"id":"d18dbfa7ef376b05358616cabaa3bbb126c379d6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-03-06 00:39:14.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7cbc2bd6a721d3835041c3fb96940343b00557a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 00:50:58.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/12/check/openstack-tox-docs/2940acd/html/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/12/check/openstack-tox-pep8/d2544d9/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":12},{"id":"fd70970781bfe259c4ff4015e60175cd6d6139c7","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-04-09 23:40:23.000000000","message":"Patch Set 12: Code-Review-1\n\n(4 comments)\n\nOverall, I\u0027m hesitant about the idea of adding a different mode to the swap volume API that will not swap a volume but instead update block device mapping. The volumeAttachment request parameter isn\u0027t option, and it sounds like this proposal would need to make it optional _and_ add a block_device_mapping_v2 parameter at the same time. It sounds messy and confusing to me.","accounts_in_message":[],"_revision_number":12},{"id":"fdfa1f6e7c736eb74c0d6da0d7c0469b4ac6d945","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-10 10:54:55.000000000","message":"Patch Set 12:\n\n(4 comments)\n\nHi melwitt, thanks for your review.\nYes. this will use the volume attachment API (PUT), which looks at the current API that interacts with the volume. But this requires adding new parameters to its request body. Considering that an instance will have multiple volumes, block_device_mapping_v2 is used, which is an array that can record information about multiple volumes.\n\nI hope to explain your confusion of this spec description :)","accounts_in_message":[],"_revision_number":12},{"id":"5804ac77c4faf697da412921d656b5e7387c546b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-05-20 10:57:43.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"e82a0e0399d2f1d90b0d6d6735ba99df16c0910b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:12:27.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/13/check/openstack-tox-docs/9d0ac94/html/ : SUCCESS in 11m 18s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/13/check/openstack-tox-pep8/b42cc6b/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":13},{"id":"ae734f13f85ac1c5e150ddbbb1a38e2c620f13d3","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-05-23 12:30:18.000000000","message":"Patch Set 13: Code-Review-1\n\nAs discussed at PTG I\u0027d like to deprecate the swap volume logic from this API, potentially making the process of simply updating existing attachments much easier. Until then I really don\u0027t think we should be overloading this API anymore so I\u0027m -1. I\u0027ll get my proposal written up in a spec shortly.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/005895.html","accounts_in_message":[],"_revision_number":13},{"id":"6a9117c370e784d82f010e143b6ef9e79fd56b91","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-05-24 08:18:22.000000000","message":"Patch Set 13:\n\n\u003e As discussed at PTG I\u0027d like to deprecate the swap volume logic\n \u003e from this API, potentially making the process of simply updating\n \u003e existing attachments much easier. Until then I really don\u0027t think\n \u003e we should be overloading this API anymore so I\u0027m -1. I\u0027ll get my\n \u003e proposal written up in a spec shortly.\n \u003e \n \u003e [1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/005895.html\n\nThanks for your review, Lee. Looking forward to your proposal.","accounts_in_message":[],"_revision_number":13},{"id":"665a00b2b3987640318966417afefc72596be41b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-12 07:05:53.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"666f7e5225a3ab476f34d5290f76a34369c2fe96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:51:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/14/check/openstack-tox-docs/d791448/html/ : SUCCESS in 7m 36s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/14/check/openstack-tox-pep8/813d1d5/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":14},{"id":"76b998803c3081e5be36deaec8bcd2395a9928bd","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-25 00:07:45.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"1a77d13b0074906842d1b817a8eee6d9a886f63f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 00:25:17.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/15/check/openstack-tox-docs/d3c2023/html/ : SUCCESS in 6m 11s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/15/check/openstack-tox-pep8/03a3d87/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":15},{"id":"cdd4572e80b8395bea5ef5e6e975d126a255c5a4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-04 10:41:26.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"9df670d8d3951d8989df83691cccb1f16ca98f14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:00:35.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/16/check/openstack-tox-docs/a6e27a7/html/ : SUCCESS in 5m 42s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/16/check/openstack-tox-pep8/9d678ec/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":16},{"id":"02da07424fc28e7b403bfb9ce4d286fa5435616d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-11 03:05:24.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"2dbce9ab6f9c15c6bc2efbe0c543ace8488b5959","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 03:12:04.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/17/check/openstack-tox-docs/5d34ca2/html/ : SUCCESS in 5m 36s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/17/check/openstack-tox-pep8/dfa6e3d/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":17},{"id":"1b08a3fa90b191e0642ec8855940418e46106c5c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-24 06:50:57.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"a4750e289e7ecb0d69510986a7eccc3c26639534","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 07:02:06.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/18/check/openstack-tox-docs/a5d2e8a/html/ : SUCCESS in 5m 32s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/18/check/openstack-tox-pep8/772c895/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":18},{"id":"935d5672d307fd365e15d62a5893abce39f27fc0","author":{"_account_id":30200,"name":"qinhaizhong","email":"qinhaizhong01@inspur.com","username":"qinhaizhong01"},"date":"2019-07-25 05:59:48.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"a301bef7f51cc9f01ba1c04a01b8ab0f2b8eea7d","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-07-25 06:00:03.000000000","message":"Patch Set 18: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":18},{"id":"1a5d7a79b455a57637983e0f2d56a84f9aa90650","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-07-25 06:00:20.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"6d9a87c2165c89b6e36915ca79ae1e4a3a59472b","author":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"date":"2019-07-25 06:01:15.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"65f5d236a52f5139f9e68b1e35dade01b3202a08","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-07-25 06:06:27.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"4cd9d44b1e832c300dd3ed500d9597fbb56e7923","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-07-25 06:14:37.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"d75df93586924b31a54d74098fc8c6cf1d5297ef","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2019-07-25 09:10:49.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"7b30890713afc3ac4252dcb7bce8fdaf61c1b17f","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-07-25 14:02:20.000000000","message":"Patch Set 18: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":18},{"id":"0d8aff6982c6ca49ce443d652674e6c743da9de4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-25 21:03:36.000000000","message":"Patch Set 18: Code-Review-1\n\n\u003e \u003e As discussed at PTG I\u0027d like to deprecate the swap volume logic\n \u003e \u003e from this API, potentially making the process of simply updating\n \u003e \u003e existing attachments much easier. Until then I really don\u0027t think\n \u003e \u003e we should be overloading this API anymore so I\u0027m -1. I\u0027ll get my\n \u003e \u003e proposal written up in a spec shortly.\n \u003e \u003e\n \u003e \u003e [1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/005895.html\n \u003e \n \u003e Thanks for your review, Lee. Looking forward to your proposal.\n\nDid you re-proposed it somewhere else or planning to update this spec after first part is merged in separate spec ? - https://review.opendev.org/#/c/612949","accounts_in_message":[],"_revision_number":18},{"id":"6ae456b2f31f6dd17506fb1815a43b3ee7444fad","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-28 09:48:05.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"8f11d8e2e030cf925ea2ae2fdaff0d4a784172f4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-28 09:50:02.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"107c62ac591f13af5043bd8b13bf8c3585a579c5","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-28 09:53:23.000000000","message":"Patch Set 20:\n\n\u003e Did you re-proposed it somewhere else or planning to update this\n \u003e spec after first part is merged in separate spec ? -\n \u003e https://review.opendev.org/#/c/612949\n\nThanks gmann, based on spec [1], this has been updated.\n[1] https://review.opendev.org/#/c/612949","accounts_in_message":[],"_revision_number":20},{"id":"c65f42036eee95d73859924f7a7e82cdec281d82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 10:13:48.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/580336/20/check/openstack-tox-docs/2d71465/html/ : SUCCESS in 6m 32s\n- openstack-tox-pep8 http://logs.openstack.org/36/580336/20/check/openstack-tox-pep8/0e2cd47/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":20},{"id":"db16e48a03e3c6bf1ff1cd2acac995201ded787c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-29 22:26:29.000000000","message":"Abandoned\n\nThis change proposes a spec for the Train release, which is currently under spec freeze. If you still wish your spec to be considered for Train, you may restore it and follow the spec freeze exception process per [1]. Alternatively, you may propose your spec to the backlog per [2], or wait until the U release opens and propose it there.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/008060.html\n[2] http://specs.openstack.org/openstack/nova-specs/readme.html#backlog-specifications","accounts_in_message":[],"_revision_number":20},{"id":"e2e07a5e2f5b3181835506c0c2cfe59442ff6680","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-08-12 06:43:13.000000000","message":"Restored\n\nmove it to backlog","accounts_in_message":[],"_revision_number":20},{"id":"83c19e02c385f5819b33830d1a4062d1bec00fe8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 06:55:18.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/36/580336/20/check/openstack-tox-docs/2b87c70/html/ : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://logs.opendev.org/36/580336/20/check/openstack-tox-pep8/90c3c27/ : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":20},{"id":"93211376cd9e9b7b9e222a243c406e7d25e3dd60","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-08-12 07:40:41.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"5bcdd6a6c6ed7dc34b513b5c279ae3c142528868","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 07:53:14.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/36/580336/21/check/openstack-tox-docs/64979ec/html/ : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://logs.opendev.org/36/580336/21/check/openstack-tox-pep8/c35bffd/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":21},{"id":"58f9737e34cf23921193f935f8ba5ea3bc667fc0","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-02 07:11:08.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"969c3aa13b000972f1019b1d8c7cef4201db159a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 07:29:39.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/044d8a9dbf5f4926b318d6f87f42238b : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780aa0880fd14ee59f8e9618570c4b1f : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":22},{"id":"203d84fb06b6ee0f58cbc0aa5e67906585fd6eaf","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-28 11:15:59.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"cf14f61b19592b37cc9ab59e75ccef82fcb9c7d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 11:26:07.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e2ee5178e9453da9bb866228a4af78 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b8d2060847742e88763e86d42284eba : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":23},{"id":"b02f1a3388fad4fe98a4d46df5bbdf6666f23831","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-29 18:54:41.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"6d009585a70fcb4839c5cd08bac980334f8bcc3c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-29 19:00:00.000000000","message":"Patch Set 23: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"3bed7d240020766dc9a0220563ad747d5ac6c9de","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-29 19:04:41.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"27373f96c736c492b4bfff3e9d15009c36a0e598","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-10-29 19:46:09.000000000","message":"Patch Set 23: Code-Review-1\n\n(1 comment)\n\nI am quite against this due to what we said here about trying to reduce the scope of Nova (its an old doc, but I think the concerns are more accute now than when we wrote the doc):\nhttps://docs.openstack.org/nova/latest/contributor/project-scope.html#no-more-orchestration\n\nNow I would change my mind if there was a use case that isn\u0027t fixed by using Ansible, Terraform or Heat to mange resource, like most users really should be. If this is something people are really hitting using Horizon, it might be worth looking in Horizon at improving terminate instance, and add in details about helping users tidy up orpahed ports and volumes that they may have got confused about.\n\nmriedem\u0027s point on snapshotting the volume to help (logically) detach a volume from a running instance where you happen to have chosen the wrong delete volume on termination flag in the first place.","accounts_in_message":[],"_revision_number":23},{"id":"c9177ab9479f4647a6ab7c9d1880e044071fd8e2","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-10-29 20:34:53.000000000","message":"Patch Set 23: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"f1a4449da2505ecf1654062fada4a5f9d716845e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-10-29 20:37:32.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"d6772b70084673fce5b82f71ba371b9454a89bdd","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2019-10-30 01:13:59.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"a4833afbee99f224627e5bf89252156563ef697c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-30 12:48:54.000000000","message":"Patch Set 23:\n\n(4 comments)\n\nThanks Matt, John.","accounts_in_message":[],"_revision_number":23},{"id":"b767cdb0d9e5ae726a3b9c4be156a102c2f089d3","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-30 13:10:13.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"5668ec0f6895133e9d98ad71468186556c890cec","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-30 13:14:30.000000000","message":"Patch Set 24:\n\n\u003e Uploaded patch set 24.\n\nIn PS23 https://review.opendev.org/#/c/580336/23/specs/ussuri/approved/destroy-instance-with-datavolume.rst@72 need to be discussed, so no updates that.","accounts_in_message":[],"_revision_number":24},{"id":"f73ef959d28247da1410b8bd9d6cf7b73101d008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 13:25:55.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8850c14c802e4275b3df1fc607b0c6e9 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b294b0c70d854c43903e68cedc86e698 : FAILURE in 3m 47s","accounts_in_message":[],"_revision_number":24},{"id":"1170a2abfe63e562ef25394f4fca3d6673c52d3d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-30 14:13:17.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"84dd1bfc55c4df1a144df0b6c1e13f4fb76c00c7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-30 14:15:06.000000000","message":"Patch Set 24: Code-Review-1\n\nSee my comments in PS23, I think you misunderstood what I was saying about volume_id in the path and volumeId in the body.\n\npep8 is failing because you removed the \"Feature Liaison\" section.","accounts_in_message":[],"_revision_number":24},{"id":"62fb768bd5a4be47fbaecd2b116c10e0559b4c8f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-31 02:04:52.000000000","message":"Patch Set 23:\n\n(2 comments)\n\nThanks Matt.","accounts_in_message":[],"_revision_number":23},{"id":"07eaf75bd84beba737496296ffd795053f8f91a4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-31 03:05:23.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"24757bb0779e608e7ea4d8ed974ea55c8975830c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 03:16:56.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8812525b382244efad278b05e5d467b4 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e8e4233d4f45d9a0ae0854649ab3d6 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":25},{"id":"f49078de959beff3c6ead3bd64f48dae0afb0def","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-04 21:52:52.000000000","message":"Patch Set 25:\n\nMy feeling on this is we don\u0027t really need this and there are suitable alternatives available but I realize doing this is relatively easy since the plumbing is already there for the most part. So I\u0027m a no-vote for now since I\u0027m not really in favor of doing it but I\u0027m not strongly opposed either. If there were more than one group of users/vendors asking for this then I\u0027d be more persuaded to do this.","accounts_in_message":[],"_revision_number":25},{"id":"5bd1d96d7fe0017d537cb616f8b0a05a4936fbd3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-11-06 08:22:16.000000000","message":"Patch Set 25: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"542991071070b821356e71da5e0c6d1bfe443e9a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-11-29 00:03:41.000000000","message":"Patch Set 25:\n\n(1 comment)\n\nThanks Alex.\nThe \"volumeId\" will be changed to optional, and it was described.","accounts_in_message":[],"_revision_number":25},{"id":"55652ee271daf7095356e8565d1e9072b7ec1dc5","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-12-02 13:01:24.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"07aae472695d9940df003e8da63ee0347f114095","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-12-02 13:03:43.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"f3fe13eece638c55e3ef6cb2ac894e2f7e955a5a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-03 02:50:08.000000000","message":"Patch Set 25:\n\n(1 comment)\n\nThanks Alex.","accounts_in_message":[],"_revision_number":25},{"id":"dbc3a2c8bd6df2773ded15f9244f39f220fc7c51","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-12-03 06:19:40.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"9aa9ac52ebf842fdfd2a06faabbaede2e927d46d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-04 05:46:29.000000000","message":"Patch Set 25:\n\n(1 comment)\n\nThanks Alex, I will update your suggestions in the next patch.","accounts_in_message":[],"_revision_number":25},{"id":"100333221d662076caf99e2ab5da2d1f1f684a8c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-04 05:47:25.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"f25de5e75ba0d6ce03c06d81fbfa64843530e8f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 05:55:12.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c7d975ae76444d9abe70dd87f88dec : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17442de6e757464cb192760414ef3698 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":26},{"id":"3532cd338e6ef5ab8e3f6019e792bd9f73eb3f96","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-12-15 13:26:22.000000000","message":"Patch Set 26: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":26},{"id":"6243b4453eedd6d6f4f57e26e4f5cc605245d9ec","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-12-15 22:34:49.000000000","message":"Patch Set 26: Code-Review-1\n\nWait..., there is an issue I just found, the volume attachment API is admin only API, and as I remember it doesn\u0027t suppose to be used by non-admin user https://openstack.nimeyo.com/75343/openstack-cinder-volume-updating-attached-volume-function.\n\nIf it is admin user only API, then it doesn\u0027t match your usecase. And we shouldn\u0027t expose swap volume to non-admin user","accounts_in_message":[],"_revision_number":26},{"id":"bd27739cbd1dab3bc4f428ff858d3f18d6b06973","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-16 02:20:51.000000000","message":"Patch Set 26:\n\nSwap volume is and should remain admin only. Updating fields in a volume attachment without swapping it could be separated from swap volume via enforcement of the request body and a new policy rule. The question all along has been is the complexity worth it for this feature when we already have alternatives.","accounts_in_message":[],"_revision_number":26},{"id":"b66bdc2572d54a176f41a7fb3a51bca916cca981","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-16 03:19:33.000000000","message":"Patch Set 26:\n\n\u003e The question all along has been is the complexity worth it for this\n \u003e feature when we already have alternatives.\n\nIMO, it\u0027s complicated but practical.\n\nWhen there are a lot of servers in the envirement (e.g. in Government Cloud env more than 1000+), of course there are also need some volumes, and these volume are specially named to retain important data, if we want to cleanup the environment, and free mem, cpu and other host resources, I think this change is better (it just a small usecase).\n\nEmm ... In the customer environment, there are their own management methods, such as making an application, requesting approval, and retaining designated resources/data.","accounts_in_message":[],"_revision_number":26},{"id":"6c7e70cdd928abf015bc38a9d55538f77e02d3c4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-12-31 01:16:26.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"50def68a8eaf85802f9d0b9a5945dd570e98c7be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 01:29:35.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7418f6573324d33a6f51d55d1c7c458 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c575f77e7fb449b1b368096af9214736 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":27},{"id":"73b6064ddb155a714bb0e697e4185802c997b5d2","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-04 06:29:00.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"db1e66a36eba9f9a40d647fdf771a2ecb9027e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 06:43:55.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ead7abfccf42c4abda449ba5b5a2ae : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/314a7cae900e4928aab65275d7c563a3 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":28},{"id":"c0b6d4882ad49568fb3601543eb633e753b14f98","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-01-09 14:22:52.000000000","message":"Patch Set 28:\n\nI\u0027ll repeat my earlier concern that we are overloading the update volume API here and that it would be much cleaner if we could separate the admin only swap_volume flow from this before we proceed with this user driven usecase.\n\nI appreciate I never followed up with this after the Denver PTG so I\u0027m not going to -1 this but I just wanted to air my concern again.","accounts_in_message":[],"_revision_number":28},{"id":"46eb193d7f6333629bdcdcd8e47bf22c40ffece9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-13 21:15:18.000000000","message":"Patch Set 28:\n\nWe\u0027re at spec freeze. I\u0027m going to defer this if it hasn\u0027t been approved by EOB Friday Feb 14.","accounts_in_message":[],"_revision_number":28},{"id":"2f9e330edcda960b3c326abf5a328bea37c993cf","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-14 15:48:12.000000000","message":"Patch Set 29: Patch Set 28 was rebased","accounts_in_message":[],"_revision_number":29},{"id":"8ac6fb63589f33bb45992af6d0c5a83b55f82da0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-14 15:56:06.000000000","message":"Patch Set 29: Code-Review-1\n\n(3 comments)\n\nTo API related comment inline.","accounts_in_message":[],"_revision_number":29},{"id":"2431e593b15ac7b2c36a58025203fbf6d25e9e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 16:07:04.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb904032f07a4a7ea8d8fa85ce1b52e6 : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449e0da4f42049aea4fa1bc3f3e07914 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":29},{"id":"6d16d94546ac81e4698d151a410f805f93d14d01","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-14 16:18:35.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"5989e21027dd9c3441661a9f95f6f605cf574903","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-14 16:32:19.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"ec71965bfda318a038006a6d441657ec4c2e4aaf","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-14 16:33:13.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"5329d597d9cd4f393fa420ad4597bb9a22565bb3","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-14 16:51:54.000000000","message":"Patch Set 30: Published edit on patch set 29.","accounts_in_message":[],"_revision_number":30},{"id":"b8942bd77bc7abef41bfc1bdf436246826233658","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-14 16:57:56.000000000","message":"Patch Set 30:\n\n\u003e I am curious why it was not done during supporting the same in POST /servers/{server_id}/os-volume_attachments/{volume_id}\n\n \u003e https://blueprints.launchpad.net/openstack/nova/+spec/support-delete-on-termination-in-server-attach-volume\n\nIt was split from https://review.opendev.org/#/c/612949/ (of you paste specification), you can review that patch.","accounts_in_message":[],"_revision_number":30},{"id":"c8c0599854d8f2a66dd9c345a53989615c871a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:05:26.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5deaaceb09e646d0981805c2d0594e34 : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59f8cd274d2d440d807a2610907b36c7 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":30},{"id":"3f87036a33e307c204c06fc5a9e50a0dc6adfe15","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-14 17:40:21.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"26636a165fc40166824a2bf4a6b9003fc6fd9e7a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-15 06:11:22.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"8973671ea161a348dc0470e4af54881bf1a88385","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-15 10:33:09.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"bb66088dae436c3a1fac43263ae205fde3b670c9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-15 10:34:25.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"5ad0293ee21ae9e0d1d8f3b91ca4271a7c774fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 10:46:14.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed8976e5a6794406b8aa83a6c5975039 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a853e6558fa047baa846b18d66ea6c30 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":32},{"id":"c9141b7f7eb717ea4565b239d1e1309213470225","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-17 17:20:35.000000000","message":"Patch Set 32: Code-Review-2\n\nWe\u0027re past spec freeze, so I\u0027m going to put a procedural hold on this. It will need a freeze exception and spec approvals granted within the week in order to be considered for Ussuri. Please send a note to openstack-discuss with [nova][sfe] in the subject line.","accounts_in_message":[],"_revision_number":32},{"id":"57d694a9592a0f6a046d3fc584afd4d7ad2eeb2c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-20 02:02:26.000000000","message":"Patch Set 32: Code-Review-1\n\n(14 comments)\n\nsoft -1 to address the wording but i think the general direction is fine. that said i am not an expert on cinder nova interaction so i defer to lee and other but this seams sane to me.\n\ni have one question regarding why we cant just do a put to the volume attachment api but i believe the answer is we already stole that for swap volume so now we are trying to use this for both.","accounts_in_message":[],"_revision_number":32},{"id":"0a640ad457a7aa3f0aa69cab0b0653a757db5c3c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-20 13:06:54.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"980a5f44dc032532e7ba894693a725354d525d32","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-20 13:07:45.000000000","message":"Patch Set 32: Code-Review+1\n\nI\u0027m OK with the conditional swap proposed in the current patch set","accounts_in_message":[],"_revision_number":32},{"id":"444bca429a31084bb0fdd1a4b51b0ac057f10a95","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-20 13:11:22.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"edf836611f6cf63371c34f34f25f624d2f4d359a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-20 13:13:04.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"95e22f664c5fabedf669b0c4f1c1bf846716c1a6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-20 14:01:05.000000000","message":"Patch Set 32: Code-Review-1\n\n(2 comments)\n\nPATCH option looks good to me and keep the swap API as it is.","accounts_in_message":[],"_revision_number":32},{"id":"3bd7ac1d94fe4b2dc8ce3bfee41f9ab93669a0df","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-20 14:24:12.000000000","message":"Patch Set 32: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"7f0891fe020d28e12cd5978ac894882000fd4b25","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-20 14:58:29.000000000","message":"Patch Set 32: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"e147aae0a704a946fbc1dc5a2879ee17779b2392","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-20 15:04:07.000000000","message":"Patch Set 32:\n\nAgreed in the nova meeting to approve the spec freeze exception if this has two +2s by EOB Friday 20200221 [1]. Alex and gibi are on the hook for those approvals, assuming the spec can be brought to an approvable state. If that happens, I\u0027ll remove CR-2 and add +W during my daytime. Otherwise we\u0027ll defer this to Victoria.\n\n[1] http://eavesdrop.openstack.org/meetings/nova/2020/nova.2020-02-20-14.00.log.html#l-62","accounts_in_message":[],"_revision_number":32},{"id":"208c6cefd6a015f51653efad52180537d6c6aac7","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-20 15:34:42.000000000","message":"Patch Set 32:\n\n(15 comments)\n\nThanks gibi, alex_xu, gmann, sean-mooney, efried comments on.","accounts_in_message":[],"_revision_number":32},{"id":"769b3ba542d3f2770736dd8475c85116795e60de","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-20 15:46:22.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"fc1d4c9c6e5d321901262515d35ad2fc9ce2c00a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 16:00:17.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab14ab383c445aea8c6d9d1a972838e : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/569e75a6250d444eaa4880f85891a626 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":33},{"id":"cb776b68af825a3c3c99cd08ddbd85cc3d174b00","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-02-20 16:01:21.000000000","message":"Patch Set 33:\n\n(2 comments)","accounts_in_message":[],"_revision_number":33},{"id":"f4238739514738d1ceec5cb6fbd87839f2853371","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-20 16:30:36.000000000","message":"Patch Set 33: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"c90b6d202a9d0ff4c9a897b8147d00cee1943544","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-20 16:34:43.000000000","message":"Patch Set 33: -Code-Review\n\nDropping -1 as alex was against the PATCH during the meeting http://eavesdrop.openstack.org/meetings/nova/2020/nova.2020-02-20-14.00.log.html#l-82\n\nI think we are still at an impasse.  :/","accounts_in_message":[],"_revision_number":33},{"id":"1757af7e1b746cb119b23cc3e3b0ce545b64118f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-21 02:18:32.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"dde2a530b191aedd4e92be0c3ed58362d01bed45","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-21 02:34:39.000000000","message":"Patch Set 34: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":34},{"id":"64d9d6368444aec742a3ab20ff78616b1862f9f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 02:34:51.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aef62d40747498db8c8335790a48983 : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c8d747b1aeb4c9fb328cda1eaef0ab9 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":34},{"id":"478543a49003242391fe01966b84fe7dca5ec4ec","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-21 02:57:11.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"7189c7d9110d7e26e508d9b5c3b4d4cad3a030a9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-21 02:58:40.000000000","message":"Patch Set 35:\n\n(4 comments)\n\nThanks Alex, all comments are addressed.","accounts_in_message":[],"_revision_number":35},{"id":"594fbbdfb967524593008cd0126c5f2b10a6d852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 03:10:34.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd09adbbaf74b69a7aa0633a47cfac2 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/881345da0ff045be8d841b5462b54bf9 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":35},{"id":"3691824570d13bd27971229dbf0d9d4d7c66863e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-21 15:35:16.000000000","message":"Patch Set 35: Code-Review+2\n\nworks for me.","accounts_in_message":[],"_revision_number":35},{"id":"8461c12c7b6249aa245534c26800e9ba02961d4d","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-21 23:09:12.000000000","message":"Patch Set 35: Code-Review+2\n\n(1 comment)\n\nSorry for the late, I thought I +2 yesterday afternnon. But it may interrupt by the network. \n\nSo I\u0027m ok with PATCH, that won\u0027t mess the existing PUT. But maybe in the futue we should move the swap API to an action API, make the PUT API as normal API for updating properites. But yes, I also feel this is simple feature, it doesn\u0027t worth such change, this will force our client update their code also. So maybe just make it works and keep it simple for now.","accounts_in_message":[],"_revision_number":35},{"id":"5c1db9a8fec03d9bd343d3fd4ff736143bed36f0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-22 02:00:02.000000000","message":"Patch Set 35: -Code-Review Workflow+1\n\nMerging as agreed with gibi and Alex +2s.","accounts_in_message":[],"_revision_number":35},{"id":"c7f4e8f53c7f690f0b45e5b71d7ba435814bb53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 02:00:13.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"a1a8f1c0037de37a2e6d30971240fe8bccb53bf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 02:11:07.000000000","message":"Patch Set 35: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2bc8706724d44499856e67de6a9ca87 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a65a7850d2c4d089ed3ca359a548e61 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":35},{"id":"b80ac322056b526339daec89462d6c78164a5fc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 02:11:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":35},{"id":"b17f115eed6538db56a1d29a8ccd67fc341dabf9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 02:12:51.000000000","message":"Patch Set 35:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/1a970f4adc534bb7bf4e28c333bd9dd0 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":35},{"id":"d1b770403235a25cbb41d8b8c7d0f68a6cf94a0a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-23 01:23:46.000000000","message":"Patch Set 35:\n\n(2 comments)\n\nsorry for the late response. I am +1 on this approach.","accounts_in_message":[],"_revision_number":35}],"current_revision_number":35,"current_revision":"3dbc3d940427b4d12672e34c5e2ae6f9d9f0422d","revisions":{"cb1ac04868cf9ff18f190710673ab87304a09268":{"kind":"REWORK","_number":1,"created":"2018-07-05 10:07:54.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/1"}}},"commit":{"parents":[{"commit":"33ca21c026a3ed4bb3605b0ec80183576eac37b3","subject":"Merge \"Spec: Handling Reshaped Provider Trees\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33ca21c026a3ed4bb3605b0ec80183576eac37b3"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:56.000000000","tz":-240},"subject":"Support deleting data volume when destroy instance","message":"Support deleting data volume when destroy instance\n\nThis spec proposed to add support deleting data volumes while destroying\ninstance.\n\nAPIImpact\nDocImpact\n\nbp: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cb1ac04868cf9ff18f190710673ab87304a09268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cb1ac04868cf9ff18f190710673ab87304a09268"}]},"branch":"refs/heads/master"},"f4d4d61360392c627d89d95b54e47644113416f6":{"kind":"REWORK","_number":2,"created":"2018-07-05 12:07:06.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/2"}}},"commit":{"parents":[{"commit":"33ca21c026a3ed4bb3605b0ec80183576eac37b3","subject":"Merge \"Spec: Handling Reshaped Provider Trees\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/33ca21c026a3ed4bb3605b0ec80183576eac37b3"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 12:06:47.000000000","tz":-240},"subject":"Support deleting data volume when destroy instance","message":"Support deleting data volume when destroy instance\n\nThis spec proposed to add support deleting data volumes while destroying\ninstance.\n\nAPIImpact\nDocImpact\n\nbp: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f4d4d61360392c627d89d95b54e47644113416f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f4d4d61360392c627d89d95b54e47644113416f6"}]},"branch":"refs/heads/master"},"ed42198b577d784d496abae95d608bc4aba804d0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-08-16 07:03:57.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/3"}}},"commit":{"parents":[{"commit":"5409bb5de3560f8a7b99562ac7be43e4aacc26bd","subject":"Create specs directory for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5409bb5de3560f8a7b99562ac7be43e4aacc26bd"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2018-08-16 07:03:57.000000000","tz":0},"subject":"Support deleting data volume when destroy instance","message":"Support deleting data volume when destroy instance\n\nThis spec proposed to add support deleting data volumes while destroying\ninstance.\n\nAPIImpact\nDocImpact\n\nbp: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed42198b577d784d496abae95d608bc4aba804d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ed42198b577d784d496abae95d608bc4aba804d0"}]},"branch":"refs/heads/master"},"aea0e7bdf1c6605cf97d38cba6186406994bdeea":{"kind":"REWORK","_number":4,"created":"2018-08-17 01:50:04.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/4"}}},"commit":{"parents":[{"commit":"5409bb5de3560f8a7b99562ac7be43e4aacc26bd","subject":"Create specs directory for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5409bb5de3560f8a7b99562ac7be43e4aacc26bd"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-08-17 01:49:50.000000000","tz":-240},"subject":"Support deleting data volume when destroy instance","message":"Support deleting data volume when destroy instance\n\nThis spec proposed to add support deleting data volumes while destroying\ninstance.\n\nAPIImpact\nDocImpact\n\nbp: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aea0e7bdf1c6605cf97d38cba6186406994bdeea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aea0e7bdf1c6605cf97d38cba6186406994bdeea"}]},"branch":"refs/heads/master"},"8a5f2fe7e76578994656101a4897094c5fa1b572":{"kind":"REWORK","_number":5,"created":"2018-10-24 08:54:59.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/5"}}},"commit":{"parents":[{"commit":"786543cca732657ea806e4c0a484dad2a25bbff9","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/786543cca732657ea806e4c0a484dad2a25bbff9"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-10-24 08:51:43.000000000","tz":-240},"subject":"Support deleting data volume when destroy instance","message":"Support deleting data volume when destroy instance\n\nThis spec proposed to add support deleting data volumes while destroying\ninstance.\n\nAPIImpact\n\nbp: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8a5f2fe7e76578994656101a4897094c5fa1b572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8a5f2fe7e76578994656101a4897094c5fa1b572"}]},"branch":"refs/heads/master"},"4ae8e68e14d5e934fd603dbd3d223276a0059c7b":{"kind":"REWORK","_number":6,"created":"2018-11-06 11:49:52.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/6"}}},"commit":{"parents":[{"commit":"786543cca732657ea806e4c0a484dad2a25bbff9","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/786543cca732657ea806e4c0a484dad2a25bbff9"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-11-06 11:49:42.000000000","tz":-300},"subject":"Support deleting data volume when destroy instance","message":"Support deleting data volume when destroy instance\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot/data volumes, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAnother change is to add the ``delete_on_termination`` request parameter\nin the attach a volume to an instance API.\n\nAPIImpact\n\nbp: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ae8e68e14d5e934fd603dbd3d223276a0059c7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4ae8e68e14d5e934fd603dbd3d223276a0059c7b"}]},"branch":"refs/heads/master"},"7ec18067b0c68a312abedc925b220065b6c21260":{"kind":"REWORK","_number":7,"created":"2018-11-28 11:33:03.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/7"}}},"commit":{"parents":[{"commit":"786543cca732657ea806e4c0a484dad2a25bbff9","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/786543cca732657ea806e4c0a484dad2a25bbff9"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-11-28 11:32:56.000000000","tz":-300},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot/data volumes, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAnother change is to add the ``delete_on_termination`` request parameter\nin the attach a volume to an instance API.\n\nAPIImpact\n\nbp: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ec18067b0c68a312abedc925b220065b6c21260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ec18067b0c68a312abedc925b220065b6c21260"}]},"branch":"refs/heads/master"},"eafb4a5b5de8ca25d5a3abf56d82480767742b44":{"kind":"REWORK","_number":8,"created":"2018-11-30 03:15:43.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/8"}}},"commit":{"parents":[{"commit":"786543cca732657ea806e4c0a484dad2a25bbff9","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/786543cca732657ea806e4c0a484dad2a25bbff9"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-11-30 03:14:24.000000000","tz":-300},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eafb4a5b5de8ca25d5a3abf56d82480767742b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eafb4a5b5de8ca25d5a3abf56d82480767742b44"}]},"branch":"refs/heads/master"},"3d23d2758497af1c5dfb6a84dab6a41f4470da25":{"kind":"REWORK","_number":9,"created":"2018-11-30 03:39:50.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/9"}}},"commit":{"parents":[{"commit":"786543cca732657ea806e4c0a484dad2a25bbff9","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/786543cca732657ea806e4c0a484dad2a25bbff9"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-11-30 03:37:53.000000000","tz":-300},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3d23d2758497af1c5dfb6a84dab6a41f4470da25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3d23d2758497af1c5dfb6a84dab6a41f4470da25"}]},"branch":"refs/heads/master"},"1557733a86378d7322798284b7ca33fc45560e3f":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-03-05 00:58:02.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/10"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-03-05 00:58:02.000000000","tz":0},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1557733a86378d7322798284b7ca33fc45560e3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1557733a86378d7322798284b7ca33fc45560e3f"}]},"branch":"refs/heads/master"},"10e2d26f7517915268713f7bdd69b7c8884e1c08":{"kind":"REWORK","_number":11,"created":"2019-03-05 00:59:32.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/11"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-03-05 00:59:41.000000000","tz":480},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/10e2d26f7517915268713f7bdd69b7c8884e1c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/10e2d26f7517915268713f7bdd69b7c8884e1c08"}]},"branch":"refs/heads/master"},"351cf15f630fd53a92df7614a64533c95f2bbf0c":{"kind":"REWORK","_number":12,"created":"2019-03-06 00:39:14.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/12"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-03-06 00:39:13.000000000","tz":480},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/351cf15f630fd53a92df7614a64533c95f2bbf0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/351cf15f630fd53a92df7614a64533c95f2bbf0c"}]},"branch":"refs/heads/master"},"5a919476c638bd55cb247d88a4ceab9edf8590cf":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-05-20 10:57:43.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/13"}}},"commit":{"parents":[{"commit":"2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9","subject":"Merge \"Follow up for I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-05-20 10:57:43.000000000","tz":0},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5a919476c638bd55cb247d88a4ceab9edf8590cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5a919476c638bd55cb247d88a4ceab9edf8590cf"}]},"branch":"refs/heads/master"},"91730d907a4247d076069e4f5952d7e4f7fcb8fa":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-06-12 07:05:53.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/14"}}},"commit":{"parents":[{"commit":"93904799878bef37eb26db463767da0cb657bbc6","subject":"Clarify the error response in proposed change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/93904799878bef37eb26db463767da0cb657bbc6"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-06-12 07:05:53.000000000","tz":0},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/91730d907a4247d076069e4f5952d7e4f7fcb8fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/91730d907a4247d076069e4f5952d7e4f7fcb8fa"}]},"branch":"refs/heads/master"},"336d10f0146533601d8385ff13f076e5d46ae662":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-06-25 00:07:45.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/15"}}},"commit":{"parents":[{"commit":"84cf12416bfb8009bc73cbfb4f13c4586f86599a","subject":"Merge \"Change new image property to hw_mem_encryption\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/84cf12416bfb8009bc73cbfb4f13c4586f86599a"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-06-25 00:07:45.000000000","tz":0},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/336d10f0146533601d8385ff13f076e5d46ae662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/336d10f0146533601d8385ff13f076e5d46ae662"}]},"branch":"refs/heads/master"},"4721914b792ea0d8dc5a82ac3c207d50f5a5d17d":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-07-04 10:41:26.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/16"}}},"commit":{"parents":[{"commit":"d74e8586a73d669f614cc9a1184e8056a12a7e7d","subject":"Merge \"Add \"Secure Boot support for KVM \u0026 QEMU guests\" spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d74e8586a73d669f614cc9a1184e8056a12a7e7d"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-07-04 10:41:26.000000000","tz":0},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4721914b792ea0d8dc5a82ac3c207d50f5a5d17d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4721914b792ea0d8dc5a82ac3c207d50f5a5d17d"}]},"branch":"refs/heads/master"},"676cd0ad3e8babe18cbc3c7f7d2d263e1e03705b":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-07-11 03:05:24.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/17"}}},"commit":{"parents":[{"commit":"42ba77612edefafa5c366a759f003e70ee8090c5","subject":"Merge \"Fix the cross_az_attach value in AZ-unshelved spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/42ba77612edefafa5c366a759f003e70ee8090c5"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-07-11 03:05:24.000000000","tz":0},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/676cd0ad3e8babe18cbc3c7f7d2d263e1e03705b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/676cd0ad3e8babe18cbc3c7f7d2d263e1e03705b"}]},"branch":"refs/heads/master"},"d2ef6ba2a913ec3ab4517d27fac2ba4955ece7de":{"kind":"REWORK","_number":18,"created":"2019-07-24 06:50:57.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/18"}}},"commit":{"parents":[{"commit":"1a8e0c431f370a4d6179a27e241bce40bbc21f68","subject":"Merge \"Boot a VM with an unaddressed port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a8e0c431f370a4d6179a27e241bce40bbc21f68"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-07-24 06:49:44.000000000","tz":480},"subject":"Support for changing deleted_on_termination after boot","message":"Support for changing deleted_on_termination after boot\n\nThis spec proposed to support configuring ``delete_on_termination`` for\nroot or data volume, if ``delete_on_termination`` is **True**, you can\ndelete the attached volume when you delete the instance.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2ef6ba2a913ec3ab4517d27fac2ba4955ece7de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2ef6ba2a913ec3ab4517d27fac2ba4955ece7de"}]},"branch":"refs/heads/master"},"2362cf227c23278549dbf3c99b73e0b88fd8ad4d":{"kind":"REWORK","_number":19,"created":"2019-07-28 09:48:05.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/19"}}},"commit":{"parents":[{"commit":"1a8e0c431f370a4d6179a27e241bce40bbc21f68","subject":"Merge \"Boot a VM with an unaddressed port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a8e0c431f370a4d6179a27e241bce40bbc21f68"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-07-28 09:47:49.000000000","tz":0},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2362cf227c23278549dbf3c99b73e0b88fd8ad4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2362cf227c23278549dbf3c99b73e0b88fd8ad4d"}]},"branch":"refs/heads/master"},"720a7145db15933068c029a3bce1f3ee569c752f":{"kind":"REWORK","_number":20,"created":"2019-07-28 09:50:02.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/20"}}},"commit":{"parents":[{"commit":"1a8e0c431f370a4d6179a27e241bce40bbc21f68","subject":"Merge \"Boot a VM with an unaddressed port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a8e0c431f370a4d6179a27e241bce40bbc21f68"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-07-28 09:48:59.000000000","tz":0},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/720a7145db15933068c029a3bce1f3ee569c752f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/720a7145db15933068c029a3bce1f3ee569c752f"}]},"branch":"refs/heads/master"},"864dcfab623be101a2e2bcaec47628ffce1b0fd9":{"kind":"REWORK","_number":21,"created":"2019-08-12 07:40:41.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/21"}}},"commit":{"parents":[{"commit":"1a8e0c431f370a4d6179a27e241bce40bbc21f68","subject":"Merge \"Boot a VM with an unaddressed port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a8e0c431f370a4d6179a27e241bce40bbc21f68"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-08-12 07:12:59.000000000","tz":480},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/864dcfab623be101a2e2bcaec47628ffce1b0fd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/864dcfab623be101a2e2bcaec47628ffce1b0fd9"}]},"branch":"refs/heads/master"},"30861c947ebc827920c2809bb8e7df138952ce86":{"kind":"REWORK","_number":22,"created":"2019-09-02 07:11:08.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/22"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-09-02 07:06:57.000000000","tz":480},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/30861c947ebc827920c2809bb8e7df138952ce86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/30861c947ebc827920c2809bb8e7df138952ce86"}]},"branch":"refs/heads/master"},"5404766f61c89211e7eb5f306764bbcbf05cee65":{"kind":"REWORK","_number":23,"created":"2019-10-28 11:15:59.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/23"}}},"commit":{"parents":[{"commit":"8459c3b3821243489b932563dd0dc7a1342b55f4","subject":"Create specs directory for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8459c3b3821243489b932563dd0dc7a1342b55f4"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-10-28 11:15:52.000000000","tz":0},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5404766f61c89211e7eb5f306764bbcbf05cee65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5404766f61c89211e7eb5f306764bbcbf05cee65"}]},"branch":"refs/heads/master"},"aab2442c5161a1c977ff19b3710d27787b89aa70":{"kind":"REWORK","_number":24,"created":"2019-10-30 13:10:13.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/24"}}},"commit":{"parents":[{"commit":"7147d2d8925e4edd15ec52aebeb33769a98d8164","subject":"Remove the todo in the migrations spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7147d2d8925e4edd15ec52aebeb33769a98d8164"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-10-30 13:07:12.000000000","tz":480},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aab2442c5161a1c977ff19b3710d27787b89aa70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/aab2442c5161a1c977ff19b3710d27787b89aa70"}]},"branch":"refs/heads/master"},"00acf055d60103241aec6bbccb34b49ddeb5a482":{"kind":"REWORK","_number":25,"created":"2019-10-31 03:05:23.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/25"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-10-31 03:03:19.000000000","tz":480},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/00acf055d60103241aec6bbccb34b49ddeb5a482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/00acf055d60103241aec6bbccb34b49ddeb5a482"}]},"branch":"refs/heads/master"},"dc6c3b9205ed934739cf0bf67120a33627f6d407":{"kind":"REWORK","_number":26,"created":"2019-12-04 05:47:25.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/26"}}},"commit":{"parents":[{"commit":"477fb6859f6b9880abd91f0abce36ecf42a350e5","subject":"Merge \"Re-propose policy-defaults-refresh spec for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/477fb6859f6b9880abd91f0abce36ecf42a350e5"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-12-04 05:47:21.000000000","tz":480},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dc6c3b9205ed934739cf0bf67120a33627f6d407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/dc6c3b9205ed934739cf0bf67120a33627f6d407"}]},"branch":"refs/heads/master"},"3471d4d5b56d004febc7fa9811d1722519040269":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2019-12-31 01:16:26.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/27"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2019-12-31 01:16:26.000000000","tz":0},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3471d4d5b56d004febc7fa9811d1722519040269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3471d4d5b56d004febc7fa9811d1722519040269"}]},"branch":"refs/heads/master"},"5f8f254b0d0bfe5c94ab6eec1c44e05912749012":{"kind":"REWORK","_number":28,"created":"2020-01-04 06:29:00.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/28"}}},"commit":{"parents":[{"commit":"36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38","subject":"[ussuri][goal] Drop python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/36b1b7fbf52360c5e7e335c78a3a8e0da8d2cf38"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-01-04 06:28:49.000000000","tz":0},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f8f254b0d0bfe5c94ab6eec1c44e05912749012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f8f254b0d0bfe5c94ab6eec1c44e05912749012"}]},"branch":"refs/heads/master"},"d4dd5b6cb99b29feacdb2dfd1f0725bc8a44da39":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2020-02-14 15:48:12.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/29"}}},"commit":{"parents":[{"commit":"9bb4b8316c4832a5532ff488d21a10fb79f19263","subject":"Merge \"Add nova-audit spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bb4b8316c4832a5532ff488d21a10fb79f19263"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-02-14 15:48:12.000000000","tz":0},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d4dd5b6cb99b29feacdb2dfd1f0725bc8a44da39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d4dd5b6cb99b29feacdb2dfd1f0725bc8a44da39"}]},"branch":"refs/heads/master"},"5c71ff9a6a87024ab9d4d79878f709c358e33049":{"kind":"REWORK","_number":30,"created":"2020-02-14 16:51:54.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/30"}}},"commit":{"parents":[{"commit":"9bb4b8316c4832a5532ff488d21a10fb79f19263","subject":"Merge \"Add nova-audit spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bb4b8316c4832a5532ff488d21a10fb79f19263"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-02-14 16:51:50.000000000","tz":0},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5c71ff9a6a87024ab9d4d79878f709c358e33049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5c71ff9a6a87024ab9d4d79878f709c358e33049"}]},"branch":"refs/heads/master"},"20c82a792a69c352d1646c0b4be2137e3e8d82af":{"kind":"REWORK","_number":31,"created":"2020-02-15 10:33:09.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/31"}}},"commit":{"parents":[{"commit":"9bb4b8316c4832a5532ff488d21a10fb79f19263","subject":"Merge \"Add nova-audit spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bb4b8316c4832a5532ff488d21a10fb79f19263"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-02-15 10:32:55.000000000","tz":0},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/20c82a792a69c352d1646c0b4be2137e3e8d82af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/20c82a792a69c352d1646c0b4be2137e3e8d82af"}]},"branch":"refs/heads/master"},"4209441be7ff28ee5ddef8f30eb5d0776a3f451d":{"kind":"REWORK","_number":32,"created":"2020-02-15 10:34:25.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/32"}}},"commit":{"parents":[{"commit":"9bb4b8316c4832a5532ff488d21a10fb79f19263","subject":"Merge \"Add nova-audit spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bb4b8316c4832a5532ff488d21a10fb79f19263"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-02-15 10:34:18.000000000","tz":0},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposed to changing the ``deleted_on_termination`` attribute\nafter boot an instance, that support deleting the target volumes while\nthe instance is destroyed.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4209441be7ff28ee5ddef8f30eb5d0776a3f451d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4209441be7ff28ee5ddef8f30eb5d0776a3f451d"}]},"branch":"refs/heads/master"},"af4ba6138a26371bb61e7305335f44d83bd5f6ec":{"kind":"REWORK","_number":33,"created":"2020-02-20 15:46:22.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/33"}}},"commit":{"parents":[{"commit":"9bb4b8316c4832a5532ff488d21a10fb79f19263","subject":"Merge \"Add nova-audit spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bb4b8316c4832a5532ff488d21a10fb79f19263"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-02-20 15:45:32.000000000","tz":480},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposes to allow changing the ``deleted_on_termination``\nattribute of a volume after an instance is booted.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/af4ba6138a26371bb61e7305335f44d83bd5f6ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/af4ba6138a26371bb61e7305335f44d83bd5f6ec"}]},"branch":"refs/heads/master"},"4f47aa9424f1663edb11ad545194996a30ee87b3":{"kind":"REWORK","_number":34,"created":"2020-02-21 02:18:32.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/34"}}},"commit":{"parents":[{"commit":"9bb4b8316c4832a5532ff488d21a10fb79f19263","subject":"Merge \"Add nova-audit spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bb4b8316c4832a5532ff488d21a10fb79f19263"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-02-21 02:18:13.000000000","tz":480},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposes to allow changing the ``deleted_on_termination``\nattribute of a volume after an instance is booted.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f47aa9424f1663edb11ad545194996a30ee87b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4f47aa9424f1663edb11ad545194996a30ee87b3"}]},"branch":"refs/heads/master"},"3dbc3d940427b4d12672e34c5e2ae6f9d9f0422d":{"kind":"REWORK","_number":35,"created":"2020-02-21 02:57:11.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/36/580336/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/36/580336/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/36/580336/35"}}},"commit":{"parents":[{"commit":"9bb4b8316c4832a5532ff488d21a10fb79f19263","subject":"Merge \"Add nova-audit spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bb4b8316c4832a5532ff488d21a10fb79f19263"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-07-05 10:04:45.000000000","tz":-240},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-02-21 02:56:38.000000000","tz":480},"subject":"Support re-configure deleted_on_termination in server","message":"Support re-configure deleted_on_termination in server\n\nThis blueprint proposes to allow changing the ``deleted_on_termination``\nattribute of a volume after an instance is booted.\n\nAPIImpact\n\nBlueprint: destroy-instance-with-datavolume\nChange-Id: If06bd29a8fcad322aed337bb7fdb54aae304c8dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3dbc3d940427b4d12672e34c5e2ae6f9d9f0422d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3dbc3d940427b4d12672e34c5e2ae6f9d9f0422d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
