)]}'
{"id":"openstack%2Fnova-specs~592755","triplet_id":"openstack%2Fnova-specs~master~Icf75ce9d27a6ca5cdfcc147698d79669bbfbdba8","project":"openstack/nova-specs","branch":"master","topic":"move_implemented_specs-delete-approved-template","hashtags":[],"change_id":"Icf75ce9d27a6ca5cdfcc147698d79669bbfbdba8","subject":"Delete approved template in move_implemented_specs","status":"MERGED","created":"2018-08-16 22:16:34.000000000","updated":"2019-04-23 16:42:12.000000000","submitted":"2019-04-23 16:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"592755-1556037732106-a663f06a","meta_rev_id":"ef1d949a7de6bcf188cb32e7de18e731006b6d34","_number":592755,"virtual_id_number":592755,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-04-23 16:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2019-04-23 16:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-04-23 16:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2019-04-23 16:22:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-11 18:58:06.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-04-23 16:09:55.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-04-23 16:22:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-23 16:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b8c0a07196e4863561377ba06efc0ab015b38d0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-16 22:16:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6aba23c456ca7c7100051d9eff15a4f582454e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-16 22:28:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/55/592755/1/check/build-openstack-sphinx-docs/50ba4f1/html/ : SUCCESS in 5m 45s\n- openstack-tox-pep8 http://logs.openstack.org/55/592755/1/check/openstack-tox-pep8/3f860c8/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"50d7eefd883388ebf1e73f17985ef122d5c21362","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-16 22:56:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36d78728922effb8429b6a7d4004835c113a78d9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-23 16:05:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c1683d722a99d8f7835cf78eb38341f363ebbab8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-23 16:09:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8b8504b51ffd90f0508a02355721be3c3d59ae8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-23 16:22:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fc8bd92976efed07c1949f1105e6888eec7f8e06","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-23 16:23:07.000000000","message":"Patch Set 2:\n\nJust got to make sure there\u0027s something in the folder first, which will clearly be the case when clearing stuff out like this","accounts_in_message":[],"_revision_number":2},{"id":"863a5e4d1d595da22208eef6e1d0ae11f49f2317","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:23:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/592755/2/check/openstack-tox-docs/9597c7b/html/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/55/592755/2/check/openstack-tox-pep8/0c01c31/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"1adff27938ef3211713c1f3dce898f33eb52247b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:23:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b58e87cae47ab32307f418a7748c0a2f95ac91e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:42:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/592755/2/gate/openstack-tox-docs/8e9bd25/html/ : SUCCESS in 9m 43s\n- openstack-tox-pep8 http://logs.openstack.org/55/592755/2/gate/openstack-tox-pep8/0775545/ : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":2},{"id":"eeb643ecc6c8b7ad8fdd8e27962ed28cd8439c9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:42:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2206aeb5b0c267fa73f863e128b83ba0925f796","revisions":{"3fc6135a2f3903a25f6b940933acfde108283d34":{"kind":"REWORK","_number":1,"created":"2018-08-16 22:16:34.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/55/592755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/592755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/592755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/592755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/592755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/592755/1"}}},"commit":{"parents":[{"commit":"5409bb5de3560f8a7b99562ac7be43e4aacc26bd","subject":"Create specs directory for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5409bb5de3560f8a7b99562ac7be43e4aacc26bd"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2018-08-16 22:14:19.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2018-08-16 22:16:32.000000000","tz":-300},"subject":"Delete approved template in move_implemented_specs","message":"Delete approved template in move_implemented_specs\n\nThe move_implemented_specs tool was cleverly removing the\nimplemented/\u003crelease\u003e-template.rst link so it didn\u0027t clutter up the\nTOC. This patch whacks the approved/\u003crelease\u003e-template.rst link as well.\n\nChange-Id: Icf75ce9d27a6ca5cdfcc147698d79669bbfbdba8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3fc6135a2f3903a25f6b940933acfde108283d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3fc6135a2f3903a25f6b940933acfde108283d34"}]},"branch":"refs/heads/master"},"a2206aeb5b0c267fa73f863e128b83ba0925f796":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-23 16:05:14.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/55/592755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/592755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/592755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/592755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/592755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/592755/2"}}},"commit":{"parents":[{"commit":"d20bff938555c36207db240a02986af3e5751ca3","subject":"Merge \"Tools \u0026 docs for backlog \u0026 abandoned spec process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d20bff938555c36207db240a02986af3e5751ca3"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2018-08-16 22:14:19.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-23 16:05:14.000000000","tz":0},"subject":"Delete approved template in move_implemented_specs","message":"Delete approved template in move_implemented_specs\n\nThe move_implemented_specs tool was cleverly removing the\nimplemented/\u003crelease\u003e-template.rst link so it didn\u0027t clutter up the\nTOC. This patch whacks the approved/\u003crelease\u003e-template.rst link as well.\n\nChange-Id: Icf75ce9d27a6ca5cdfcc147698d79669bbfbdba8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2206aeb5b0c267fa73f863e128b83ba0925f796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2206aeb5b0c267fa73f863e128b83ba0925f796"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
