)]}'
{"id":"openstack%2Fnova-specs~603955","triplet_id":"openstack%2Fnova-specs~master~Ibecee0d46e296813e90f855379a134b84d43a55f","project":"openstack/nova-specs","branch":"master","topic":"bp/nova-cyborg-interaction","hashtags":[],"change_id":"Ibecee0d46e296813e90f855379a134b84d43a55f","subject":"Nova Cyborg interaction specification.","status":"MERGED","created":"2018-09-20 02:33:47.000000000","updated":"2019-06-20 16:03:09.000000000","submitted":"2019-06-20 16:03:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":676,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"603955-1561046589153-eb0806c3","meta_rev_id":"c7587e90baa7b9d517dc2d4098da2b53abde7a07","_number":603955,"virtual_id_number":603955,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2019-06-20 16:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 15:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2019-06-19 01:48:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-06-20 15:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":1,"date":"2019-06-20 15:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-26 21:13:22.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2018-10-10 08:50:05.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2018-10-10 17:56:34.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2018-11-14 14:21:29.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2018-11-14 22:16:49.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2019-04-09 09:36:04.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2019-04-09 15:56:13.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-04-15 12:52:18.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2019-04-20 03:57:50.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2019-04-22 03:23:28.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2019-04-30 16:56:59.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-05-07 04:11:49.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"},{"updated":"2019-05-13 15:22:09.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2019-06-04 10:43:12.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-13 13:18:16.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-06-19 01:48:18.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-06-20 15:45:13.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-06-20 15:45:20.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-06-20 16:03:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fac9ee5ad79429c71d46128cae8a2b6430fd3753","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-09-20 02:33:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"282dd3c363b36845772654a82a42892ce5846c23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-20 06:35:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/1/check/openstack-tox-docs/deea48f/ : RETRY_LIMIT in 5m 41s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/1/check/openstack-tox-pep8/aa0f87c/ : FAILURE in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5146079f94401dbf76d0ea5f6cf01f97b78e8383","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-09-20 07:27:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(54 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b40e9f3d0a255f28fde945e611ab7b4172e5fe06","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-09-21 22:07:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(19 comments)\n\nThis is largely sane and gels with what we agreed upon in Denver. Only a couple of blocking issues for my part inline. Other bits and pieces open for discussion.","accounts_in_message":[],"_revision_number":1},{"id":"89b671437e1988a49853aef7916d357bad20541a","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2018-09-26 21:13:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f77b20a084f6cb9b70f46a8993fe54c546c540dd","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-09-28 13:21:33.000000000","message":"Patch Set 1: Workflow-1\n\n(49 comments)\n\nThanks for the comments. I will post a updated version in the next few days.","accounts_in_message":[],"_revision_number":1},{"id":"92cdc8dd309b6998b9a097cacf5239dcb892e24c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-02 04:44:23.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nThanks for the comments, Ed.","accounts_in_message":[],"_revision_number":1},{"id":"0151e19806f554c959e63c0cd3fa33a1d1470c75","author":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"date":"2018-10-03 01:30:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"04bdd386244655663a0b67a3fbc2ef47a15f5415","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-04 00:45:39.000000000","message":"Patch Set 1: -Workflow\n\n(11 comments)\n\nMore responses.","accounts_in_message":[],"_revision_number":1},{"id":"e9df9d5359ee3c589810f04a28d02a871494369b","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-04 00:46:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"67e70d80415b43e45708e03a8d92e8c31320734e","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-04 01:04:27.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nThanks, Li Liu.","accounts_in_message":[],"_revision_number":1},{"id":"557f055e3cb9024ad81348011487a11c06fd94d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-04 02:47:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/2/check/openstack-tox-docs/aab68f8/html/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/2/check/openstack-tox-pep8/bf8d55c/ : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f9211c355947275d41525f20f64e7708112dc6ec","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-04 03:09:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"527434f5e555b6574573e070c4a738992520c669","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-04 04:07:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/3/check/openstack-tox-docs/8598913/html/ : SUCCESS in 6m 01s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/3/check/openstack-tox-pep8/3d6e708/ : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"183de70c42ce57a829e94de555b8db7c1bb2ab05","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2018-10-09 08:40:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":3},{"id":"58913e7b38ce4edf31e42837d928655a01d31845","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-10-09 18:03:17.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2b94e871b71ba631c162f3b854343d74e6b032f4","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-10 03:09:37.000000000","message":"Patch Set 3:\n\n(13 comments)\n\nThanks for the comments, Dan and Xinran.","accounts_in_message":[],"_revision_number":3},{"id":"4c5e301cad9c7012803b63da84ab52ebc0d1d989","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-10 04:27:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b5226b79cee8aeb1ddc712006f1fbc1736861d88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 04:43:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/4/check/openstack-tox-docs/b3980ab/html/ : SUCCESS in 6m 51s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/4/check/openstack-tox-pep8/c3e7106/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":4},{"id":"5c3efb6920b608e03dce1ad78e6fbd53ffa92904","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2018-10-10 06:22:24.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"be6378786caf8039ce420c4f2a3dff11f0dbeaeb","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-10-10 08:50:05.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)\n\nI didn\u0027t finish all the reivews. But I feel VAR is an oslo obj used in os-acc, confuse on why we code VAR into the cyborg API.","accounts_in_message":[],"_revision_number":4},{"id":"d9cfd0c9e4ee26444810758619069025b78838cc","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-10 16:46:44.000000000","message":"Patch Set 4:\n\n(6 comments)\n\nThanks, Alex.","accounts_in_message":[],"_revision_number":4},{"id":"3389cd8fba4d80e95d2ff9e69205b77d3ee86460","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-10 16:48:15.000000000","message":"Patch Set 4:\n\n\u003e (6 comments)\n \u003e \n \u003e I didn\u0027t finish all the reivews. But I feel VAR is an oslo obj used\n \u003e in os-acc, confuse on why we code VAR into the cyborg API.\n\n  The os-acc does not have a db. VARs are stored in Cyborg db, and are therefore accessed via Cyborg APIs.","accounts_in_message":[],"_revision_number":4},{"id":"861ca5f368cd580a2e9d3b571957dfe3362f21ff","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-10 17:03:06.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThanks Xinran.","accounts_in_message":[],"_revision_number":3},{"id":"5236025e3e564e04ae3dc8fc7db029b7041d9830","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-10-10 17:16:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3317133679961f1221416457820a440ca0cf99bb","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-10-10 17:56:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"177f3d44d0b2562ab42b565cbaa70cc962b233b0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-10-10 18:11:47.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f8f441f74aca7dded426617cc18a221b676cd26b","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-11 05:08:03.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThanks, Melanie and Dan.","accounts_in_message":[],"_revision_number":3},{"id":"30818beae61a095049543d8b358168f270245576","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-10-11 14:33:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c8045fdd39d6f0416874b54879cb575e14d3435c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-16 15:26:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f541acea3da199934a4f196cf34977d0a0c6f3a5","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-16 15:27:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"02b9f089a9f549175c4c14ad986cc018b2f4173a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-16 17:39:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/5/check/openstack-tox-docs/8330a4f/html/ : SUCCESS in 6m 07s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/5/check/openstack-tox-pep8/26a43e4/ : FAILURE in 4m 46s","accounts_in_message":[],"_revision_number":5},{"id":"9448eda8fda4d5118e2ab4c8c50339946e9b175b","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-16 20:32:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"76e5760107e767a83ee0df434d4bc1c80c4dfedf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-16 22:47:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/6/check/openstack-tox-docs/e03de8e/html/ : SUCCESS in 6m 32s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/6/check/openstack-tox-pep8/22986ad/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":6},{"id":"65952bde099eb30998b7f09530e1d6877c9a5aea","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-18 22:47:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"edfb20e3bae88221d3d397638d100374e3951c14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-18 23:13:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/7/check/openstack-tox-docs/48790c1/html/ : SUCCESS in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/7/check/openstack-tox-pep8/f6e33af/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":7},{"id":"0c662ec0d5e4d7ebe7939a8db9b28a230e3ff89a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-10-19 13:03:59.000000000","message":"Patch Set 7: Code-Review-1\n\n(32 comments)\n\nWell detailed spec. I have some comments / questions inline.\nCheers,\ngibi","accounts_in_message":[],"_revision_number":7},{"id":"b11f050aa0ab63b7daf3be3ca08a3d7542501e50","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-10-19 13:08:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a8f50bb030c0cdaacd56934ce452d82d05031f08","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-10-23 01:01:24.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)\n\ni need to fully review this again in the morning but i dont think this spec should mandate or descibe indirect accelerators as fundamentally\nthey are not manged via nova since they are not consuemd by the instance.","accounts_in_message":[],"_revision_number":7},{"id":"e26a8791973be759264ce48778d1662c49c29e70","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-23 18:26:26.000000000","message":"Patch Set 6:\n\n(26 comments)\n\nThanks, Gibi.","accounts_in_message":[],"_revision_number":6},{"id":"d40a7188b17af6f66f15b2f6069e5e72aafdd71a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-23 18:27:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5badb95dd49e4a2ba9db13ad0669eb487a197d7b","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-23 18:35:16.000000000","message":"Patch Set 7:\n\n(2 comments)\n\nThanks, Sean. Looking forward to your full review.","accounts_in_message":[],"_revision_number":7},{"id":"5ebeeef4a612da8b7d522f6915ae51f4d68a5ad9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-23 19:56:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/8/check/openstack-tox-docs/507b52a/html/ : SUCCESS in 7m 27s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/8/check/openstack-tox-pep8/22abb97/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":8},{"id":"430f17ae8b41d2c7b3b132ebc8427065223bb8e5","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-10-27 00:45:58.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ff57040f7138be570604b30811115c84117bfd8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-27 03:21:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/9/check/openstack-tox-docs/193d11b/html/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/9/check/openstack-tox-pep8/4e7b3b6/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":9},{"id":"2e437f3e41813834e7628dc5f2ab94e603fbfc85","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-10-29 09:28:39.000000000","message":"Patch Set 9:\n\n(7 comments)\n\nI think this is getting in shape now. My only concerns is about the REST APIs. I\u0027d like to see some discussion with Chris or Ed from the API WG.","accounts_in_message":[],"_revision_number":9},{"id":"3fcfb8aa64ce0f028f0d8f6a187b8ad9e7a46312","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-10-29 16:10:34.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)\n\nSome comments within about the API. The -1 is for the 401, which should be corrected.\n\nThe rest of it is commentary on the shape of the API, which may or may not be correct so doesn\u0027t warrant a -1, but may need some more discussion. In general it seems like the resources and representations thereof are too closely tied to the specific needs of the API, rather than a general way of providing an API to the resources involved.","accounts_in_message":[],"_revision_number":9},{"id":"421f1edbab66b4a00e609050e5f7f079ec8000db","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2018-10-31 09:51:29.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2cb5b1f1b8b14e35bc26b6ca55a3949b753d2296","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-10-31 12:08:39.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"257ffbf5dff7f8eb849f4e5aff331b2d930f11eb","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-11-05 13:51:53.000000000","message":"Patch Set 9:\n\n(5 comments)\n\nThanks, Chris. \n\nRe. \"resources and representations thereof are too closely tied to the specific needs of the API\", the general intention is to minimize the number of Nova -\u003e Cyborg API calls for two reasons:\n* More APIs \u003d\u003d more touchpoints between Nova/Cyborg\n* Fewer calls will (hopefully) help better scaling. Some of these calls are made in the Nova scheduler context and we don\u0027t want to affect instance creation/deletion throughput.\n\nIf you have ideas for improvement, please LMK.","accounts_in_message":[],"_revision_number":9},{"id":"bb4119eab7c20098b222460da116f74de82a70c6","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-11-05 13:52:30.000000000","message":"Patch Set 9:\n\nThanks, Chris. \n\nRe. \"resources and representations thereof are too closely tied to the specific needs of the API\", the general intention is to minimize the number of Nova -\u003e Cyborg API calls for two reasons:\n* More APIs \u003d\u003d more touchpoints between Nova/Cyborg\n* Fewer calls will (hopefully) help better scaling. Some of these calls are made in the Nova scheduler context and we don\u0027t want to affect instance creation/deletion throughput.\n\nIf you have ideas for improvement, please LMK.","accounts_in_message":[],"_revision_number":9},{"id":"fe73cef10f8c392bb119b663212c262786b78fee","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-11-05 14:15:37.000000000","message":"Patch Set 9:\n\n(9 comments)\n\nThanks, Xinran and Sean. Will update the spec.","accounts_in_message":[],"_revision_number":9},{"id":"d6fc181c0b1f380a9a7df893a32710c42d3d49e7","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-11-05 20:27:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"81934f744fba58b81a968b241cdcdb30c5283426","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-11-06 23:14:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"929b5a090f607b75f5df07259b3f119a6a5edd49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-06 23:25:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/10/check/openstack-tox-docs/801f5e2/html/ : SUCCESS in 6m 05s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/10/check/openstack-tox-pep8/2595546/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":10},{"id":"5b202ead3bc27f69bcac8bb7a03c968fe763e4ad","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-11-14 09:13:39.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nMore comments forthcoming on the next patch too.","accounts_in_message":[],"_revision_number":9},{"id":"136d0d01f2405412689f52e855cd7469b7b2ae4e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-11-14 09:37:47.000000000","message":"Patch Set 10:\n\n(2 comments)\n\ncouple more comments on the state of the HTTP API","accounts_in_message":[],"_revision_number":10},{"id":"48f8cde9f7e30141d7420a32ad64bbef038ea7a6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-11-14 14:21:29.000000000","message":"Patch Set 10: Code-Review-1\n\n(23 comments)\n\nI\u0027m trying to provide some feedback to this while at summit, so apologies for not catching up on all the conversation between now and the last time I looked at this. IMHO, this is really way too verbose and detailed. I know that\u0027s my usual thing, but... yeah.\n\nI\u0027m -1ing at least because I found a typo, but I hope this can be trimmed quite a bit. If you want another spec reviewer to come along and be able to digest and approve this in a reasonable amount of time, I think it needs to be cut down a lot.","accounts_in_message":[],"_revision_number":10},{"id":"2e3e6dda8bc8723b90faef30c4705cc553d53413","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-11-14 15:45:12.000000000","message":"Patch Set 10: Code-Review-1\n\n(21 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0f0e57e3d8dd97c7a62cbdf44db46522b78fa772","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-11-14 16:02:06.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"008306342bfba08a10e7d313f4e52b219b04f686","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-11-14 22:16:49.000000000","message":"Patch Set 10: Code-Review-1\n\n(21 comments)\n\nSundar, I know you\u0027re not going to like my review here, but I have a whole lot of problems with this proposal.\n\nIn summary, I don\u0027t believe the proposed way of representing the concepts in Cyborg is a sound proposal. I\u0027m uncomfortable with the fact that the device profile contains multiple resource requests groups. I\u0027m uncomfortable with the concept of an \"ARQ\" when the thing it describes is more of just an attachment context. And I\u0027m uncomfortable with the idea that os-acc is not focused on the low-level device plug/unplug functionality and is instead just a REST client for the Cyborg API (which I also don\u0027t think is well-thought-through at this time).\n\nThis all goes back to what I asked for in Denver from the Cyborg team: less focus on a REST API (which is mostly unnecessary at this point, IMHO) and more focus on getting an *actual* working low-level device attachment library going. I thought os-acc was going to be that thing, but I see no evidence yet that there is anything in os-acc or in the works for os-acc that will functionally allow a VM to attach a device to an FPGA resource context, a specialized SR-IOV VF, or any other type of accelerated device.\n\nUntil I see some evidence of a device attachment library that actually attaches and configures an accelerator device to a VM, I\u0027m not supportive of any of the proposed changes to nova contained in this spec.\n\n-jay","accounts_in_message":[],"_revision_number":10},{"id":"a74eca14b8f5b8f0075f08a944623e169e1fc3e6","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2018-11-19 09:15:18.000000000","message":"Patch Set 10:\n\n(43 comments)\n\nThanks, Jay and Sean. (Will respond to other comments and update the spec shortly.)","accounts_in_message":[],"_revision_number":10},{"id":"ff649544558254a6b8210eaafed45724d2ce35da","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2018-11-19 17:24:34.000000000","message":"Patch Set 10:\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"5e84d73bd0efd6ffaa6f710705bbdf6302b05882","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-11-19 17:29:51.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"733e821b9ab6753bd1ac8764e04ac672186435f6","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-08 04:15:43.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8c8de486e92a3e6b8795936ef8418da8a7334b35","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-08 04:16:23.000000000","message":"Patch Set 10:\n\n(11 comments)\n\nThanks.","accounts_in_message":[],"_revision_number":10},{"id":"544d0340b320869b54d4a0a0ade953c0b093a4da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 04:34:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/11/check/openstack-tox-docs/022b7fd/html/ : SUCCESS in 7m 16s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/11/check/openstack-tox-pep8/b432725/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":11},{"id":"44ffbc6e207afefe275d6a4b1840cfdfaa8520f8","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-04-09 09:36:04.000000000","message":"Patch Set 11:\n\n(3 comments)\n\nHi Sundar, thx for the efforts for cyborg! Please see my comments in line.","accounts_in_message":[],"_revision_number":11},{"id":"2312af9645de97af78cb5aea4f19f7e73f1bed2d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-09 10:05:40.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"14ed37c363910f8c853c735422964ab53af9236f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-04-09 15:56:13.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)\n\nI think the testing section needs more.","accounts_in_message":[],"_revision_number":11},{"id":"37dc02302f6e3231c4c0fccbac43131c6566f3b6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-10 12:52:31.000000000","message":"Patch Set 11: Code-Review-1\n\n(8 comments)\n\nCouple of things inline.\n\ncheers,\ngibi","accounts_in_message":[],"_revision_number":11},{"id":"f4eacf27526173d003cbd54d1ee42a379a824dc2","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-11 02:06:31.000000000","message":"Patch Set 11:\n\n(9 comments)\n\nThanks.","accounts_in_message":[],"_revision_number":11},{"id":"3158e47108f645552a05687081d8c32f305dabf8","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-04-15 12:52:18.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5ca209c23055c44ed1ee0e03d6d245fa538c9368","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-04-20 03:57:50.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ae98da0a7e34e64254d851e6ebae979bfe423470","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-04-22 03:23:28.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c0af5924923e18fc30791b01644f81fd37d8fae8","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-25 15:20:38.000000000","message":"Patch Set 11:\n\n(5 comments)\n\nThanks for the reviews.","accounts_in_message":[],"_revision_number":11},{"id":"3359aa6a779d65321e71d2e511200ac203aa288c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-04-25 15:26:16.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nThanks for the comments.","accounts_in_message":[],"_revision_number":11},{"id":"52c1b5d5c22507390983ee2f0bcbe004907d7111","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-03 02:06:44.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)\n\nMy questions were answered. Looking forward for the demo tomorrow.","accounts_in_message":[],"_revision_number":11},{"id":"b8e3d98049d13a41ac3752f988ba79885d4a7479","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-06 22:42:44.000000000","message":"Topic changed from specs/nova-cyborg-interaction.rst to bp/nova-cyborg-interaction","accounts_in_message":[],"_revision_number":11},{"id":"3c3d46032adbb7137e9465bb541efce90f4058d6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-06 22:43:43.000000000","message":"Patch Set 11:\n\nQuick note: this spec should propose names for any standard resource classes and traits it wants to use.","accounts_in_message":[],"_revision_number":11},{"id":"2b6ea4de27888d9a0dba9d0cbd0d969cbc8f5d0e","author":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"date":"2019-05-07 04:11:49.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"548588ed791183aea6fc3cef63b33d81a056d2bd","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-07 20:21:19.000000000","message":"Patch Set 11: Code-Review-1\n\n(9 comments)\n\n-1 for the instance lifecycle operations thing, because I don\u0027t want to end up in a situation like we had for NUMA live migration, where it was half-working for years. Nova should either outright refuse to do a thing, or if it accepts it, handle it fully.","accounts_in_message":[],"_revision_number":11},{"id":"ec97c6e4626d5a5350a8c0d81bf4ae5a1debe56a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-10 22:49:06.000000000","message":"Patch Set 11:\n\n(10 comments)\n\nThanks for the insightful comments.","accounts_in_message":[],"_revision_number":11},{"id":"0a8fb4349eab19910a1dd196245999572c6ecb6c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-10 22:54:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"4f8d858e42d0155fe13ab2b4f4d24ed10a7a8f3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 23:02:31.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/12/check/openstack-tox-docs/d170c1b/html/ : SUCCESS in 7m 51s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/12/check/openstack-tox-pep8/0c25c73/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":12},{"id":"f0c43c3bac1d439d8ce92d9e469b9e35641d3ddd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-13 15:07:42.000000000","message":"Patch Set 12: Code-Review-1\n\n(8 comments)\n\nThis is very close. We agreed on the broad strokes in Denver. There were a few tweaks suggested, at least some of which need to be accounted for here.","accounts_in_message":[],"_revision_number":12},{"id":"cd488ab603662cde9e06607d14602f4d33462b88","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-13 15:16:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"6e8a72fe62a5de0f5e4449a885c30c822acf1012","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-13 15:22:09.000000000","message":"Patch Set 12:\n\n(6 comments)\n\nI\u0027m generally OK with this, no +1 yet because Eric still has concerns and we may want to figure out the polling vs. external events thing in the spec before we start on the implementation.","accounts_in_message":[],"_revision_number":12},{"id":"9558eae415d38c0bdef036865956743d145f61c8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-13 15:31:17.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8a4f02f7e340d1bc8819a43c47d28703380a306e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-13 16:11:30.000000000","message":"Patch Set 12: Code-Review-1\n\n(8 comments)\n\nTwo things worry me:\n* move bind to nova-compute\n* use external events API rather than polling\n\nI think we should either use this approach, or explain why that wouldn\u0027t work in the alternatives section. It is all about using similar patterns to how we want Cinder and Neutron interactions to work. I think it will be easier to maintain and operate if they all do the same kind of thing.","accounts_in_message":[],"_revision_number":12},{"id":"5f8723a60b0e9c217fcc01f6aba88cac1d049503","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-14 05:45:16.000000000","message":"Patch Set 12:\n\n(5 comments)\n\nResponding to some of the major comments. Will get to the rest later.","accounts_in_message":[],"_revision_number":12},{"id":"f05c8daa7fa3ee4466b21046cad75d57864dd487","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-14 08:32:40.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b8b5203b0d5aa5dce79ecd8dea222114c5b66920","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-14 13:13:20.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"478c85b2349876519586998cb55efa0a44674ec6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-14 18:36:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f8710f852b99a12a51677735e2718f9a335cc2cb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-05-15 07:15:27.000000000","message":"Patch Set 12: Code-Review-1\n\n(11 comments)\n\ni think this is quite close at this point but some nits inline mainly to update this to what we agreed at the ptg.","accounts_in_message":[],"_revision_number":12},{"id":"fbdc1a6b9ae37a5fa003644adba42980d31397bf","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-22 08:04:43.000000000","message":"Patch Set 12:\n\n(13 comments)\n\nTrying to get some concrete time measurements. Started with a 1-node devstack setup. Working on a 2-node setup now, starting multiple VMs together.\n\nMeanwhile, my responses are below.","accounts_in_message":[],"_revision_number":12},{"id":"a5685bfd21ade85419d9a2277e64bfb473f6db1f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-05-22 11:27:03.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"5570d4351d97226bfd7ba9f6574f4400f33469fb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-05-22 11:29:14.000000000","message":"Patch Set 12:\n\nsundar unless you are going to deploy on a 300 node cluseter and do some real scale testing i dont think you are going to be able to produce meaninging results using your two node setup. at least not results that are indicitive of a production deployment so i would not spend two much time on that effort vs the rest of what needs to get done to complete this spec.","accounts_in_message":[],"_revision_number":12},{"id":"f779a49d951c41da447921d48b04676a2e1fd906","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-24 04:58:19.000000000","message":"Patch Set 12:\n\n(10 comments)\n\nThanks for all the comments.","accounts_in_message":[],"_revision_number":12},{"id":"cc555c0d332e89942c02377327d803b6ef3f035b","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-24 04:58:25.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6f781d366c4408c766ad5da78eb62686f5d226c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 05:10:22.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/13/check/openstack-tox-docs/4e69519/html/ : SUCCESS in 7m 03s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/13/check/openstack-tox-pep8/31649bf/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":13},{"id":"57c770b11b0d71475db5962e17557195c2af4ade","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-30 03:15:21.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"2a9d79b653d7b4dd9be34f497cefcfbcae098db7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-03 22:20:33.000000000","message":"Patch Set 13: Code-Review-1\n\n(22 comments)\n\nWell -1 for the sequence diagram not rendering, that\u0027s an easy one.\n\nLots of other questions inline and some corrections on what I think you mean, based on how the nova port requested resources stuff works. There are obviously lots of implementation details that will need to be sorted out in code review but I\u0027ve left some notes on high level things I think we\u0027ll want to sort out early, like which nova service creates ARQs (conductor?) and which one binds them (compute).","accounts_in_message":[],"_revision_number":13},{"id":"d2bbe9923a86493cbd458cf785fd66b239b889d1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-04 05:31:26.000000000","message":"Patch Set 13:\n\n(19 comments)\n\nThanks, Matt. Responded to most of your comments. Not sure why the diagram will not render; it passed \u0027tox -edocs\u0027. Will update the spec after your response and further discussion.","accounts_in_message":[],"_revision_number":13},{"id":"77cf106d1ad1f1cea74d639bf07b1a383535482a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-04 10:43:12.000000000","message":"Patch Set 13:\n\n(4 comments)\n\nI didn\u0027t go through this thoroughly, just looked at what is needed to fix the seqdiag, see inline.","accounts_in_message":[],"_revision_number":13},{"id":"ce098f3e8b5a9c48b72c62204e8da122235cc6ce","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-04 16:20:30.000000000","message":"Patch Set 13:\n\n(8 comments)\n\nI think my main concerns at this point are:\n\n1. Nova only creating ARQs during server create rather than the opposite of having changes to the API only for passing a pre-created ARQ id and having nova attach that, like an SR-IOV port. I\u0027m not sure how much that has been debated already in the past, but assuming quite a bit. I\u0027m not blocking on it but it\u0027s a concern.\n\n2. To a lesser extent, and is an implementation detail, creating the ARQs in the conductor rather than the compute because as noted inline I think we want to avoid some of the problems we\u0027ve had with managing external resources in the compute service (ports and volumes). But again I\u0027m not blocking on that since it\u0027s an implementation detail.","accounts_in_message":[],"_revision_number":13},{"id":"d96abd66f60628d554a6f2b66b0a8c21d38b7e03","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-04 16:36:38.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8d4f81f33ce10657447fb5555f344600d423984b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-04 16:39:54.000000000","message":"Patch Set 13:\n\n@Sundar: before making changes based on my latest comments you might want to read this short(ish) IRC discussion I had with Sean and Dan today:\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-06-04.log.html#t2019-06-04T16:22:01\n\nThat might help clear up any confusion or worry about needing to re-write the API interaction based on my comments. tl;dr it sounds like the spec matches what has been previously discussed and agreed for the first version, I\u0027m just catching up on all that. I think my concerns could be noted in the Alternatives section but we don\u0027t need to change course here.","accounts_in_message":[],"_revision_number":13},{"id":"97bdf1bcd916a885ed25913a4b18d223780f9bf8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-05 13:44:41.000000000","message":"Patch Set 13:\n\n(10 comments)\n\nhonestly i think this is pretty close to merging.\nwe coudl fix up some of the nits inline in a follow up patch.\n\nthere are other changes that should be made before this merges so\nill hold back on +1 untill matts cleanups are addressed.","accounts_in_message":[],"_revision_number":13},{"id":"a37ee81c8c7e3d5438ca1a6bca18510f00094f1f","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-05 22:13:16.000000000","message":"Patch Set 13:\n\n(11 comments)\n\nThanks, everybody. Based on the comments, I hope we are close to merging the spec.","accounts_in_message":[],"_revision_number":13},{"id":"d20293a7c6ee63c2de9eda14dbb276232de03413","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-05 22:13:49.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d91a7041d020e7e134f7478887d3873d0d921d47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 22:23:26.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/14/check/openstack-tox-docs/f99aff5/html/ : SUCCESS in 6m 59s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/14/check/openstack-tox-pep8/f5f6365/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":14},{"id":"f3f88dde566ded6cc05b99e51787f81e09d94dd7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-05 22:40:32.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"044c409b4adba452d6d07147f4693d31f4e6dfff","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-06 20:26:30.000000000","message":"Patch Set 14: Code-Review+2\n\n(10 comments)\n\nI have some nits inline which can be addressed in a follow up, but I don\u0027t want to hold this up anymore since I think the agreed-to direction is stated in here. The major thing to decide at this point it seems is where the ARQ creation should happen (conductor or compute) which we can sort out in the ML or code review.","accounts_in_message":[],"_revision_number":14},{"id":"2f3dcda1a369a8bfe3f89ec4daf8bf1d0866585c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-06 23:11:43.000000000","message":"Patch Set 14: Code-Review+1\n\nok the rest of the open points can be handeled as part of code review so i also think its good to merge this now since we seam to have reached agreement on the majory points and leave the rest for the mailing list and code reveiw.","accounts_in_message":[],"_revision_number":14},{"id":"e44182c496179d950088d5d959ba75f677b03d68","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-13 10:15:28.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)\n\nLooks good to me. I have one question before I +2 it.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":14},{"id":"b2522417a59be92a77c540072422f8149078d7b4","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-13 13:18:16.000000000","message":"Patch Set 14: Code-Review+2\n\n(4 comments)\n\nFor me I think we have enough agreement to move forward, and follow up with the loose ends we have found here. Will leave this open for now, incase we can resolve the questions quite quickly.","accounts_in_message":[],"_revision_number":14},{"id":"acca31f104c99e9ca6e3e2793160b45788cb7299","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-18 22:57:17.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e4cc633e783bfacbb7dc19c04ebc4ceea3ee875d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-18 22:58:32.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"2929483f649df18a7e9546ec1821dfbf9dfb4364","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 23:11:34.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/15/check/openstack-tox-docs/800b776/html/ : SUCCESS in 7m 45s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/15/check/openstack-tox-pep8/2a257ff/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":15},{"id":"7a1eeaf158546139d7affe14149ed29b3ade4911","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-19 01:44:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5ba22d14694e71c6448082be10fee4d67b0b2db3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-19 01:48:18.000000000","message":"Patch Set 15: Code-Review+1\n\nim still ok with this version. im glad you remove the ..._is_fatal config option as i think making it always fatal will simplify things in the long run.\n\nif i could remove the config we have for neutron i would be we have too much legacy tech debt with out of tree dirvers.","accounts_in_message":[],"_revision_number":15},{"id":"09b91fba57d3e6795d4600c66f3b89e322839f48","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-19 18:12:54.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nI\u0027m not so sure that arq_binding_is_fatal should have been removed. I left comments in PS14. It would have been nice to address the formatting issues I also mentioned in PS14 when pushing up PS15.","accounts_in_message":[],"_revision_number":15},{"id":"dac8062d75543f4f900bcfdf0e7230fe24813bdb","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-20 01:31:41.000000000","message":"Patch Set 14:\n\n(1 comment)\n\n\u003e It would have been nice to address the formatting issues.\n\nI took them as nits. I will fix them in the next version. But I\u0027d like a conclusion on the arq_*_is_fatal tunable before my next push.","accounts_in_message":[],"_revision_number":14},{"id":"43f271c08bdaa965eb9989923974c123e0558391","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-20 15:07:37.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"aa14c26fcff6c358a5be922eee3d896176b2ed04","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-20 15:25:04.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5664a838c637c3583c0b93a78b6965c173be6dbd","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-20 15:30:43.000000000","message":"Patch Set 14:\n\n(1 comment)\n\nThanks. mriedem and gibi.","accounts_in_message":[],"_revision_number":14},{"id":"6f3fd9ad0e2b557143dbdb3babec2cd1273c14d1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-20 15:32:26.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b4fc7755918214f0536368aae13ad47fb478b76b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-20 15:35:55.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f7ec82c476e740e0d1e2ee1bb7a2181f459b5a57","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-20 15:44:12.000000000","message":"Patch Set 15: Code-Review+2\n\nI think we\u0027ve clarified the confusion over the \"binding timeout is fatal\" config option - we don\u0027t have the information needed to configure the guest domain xml if we timeout waiting for binding to complete so we should just fail and don\u0027t need an option for that. The formatting nits can be cleaned up in a follow up change.","accounts_in_message":[],"_revision_number":15},{"id":"b59b8eb901c70eb1723389ab48eec339de355e30","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-20 15:45:13.000000000","message":"Patch Set 15: Code-Review+2\n\nLGTM!","accounts_in_message":[],"_revision_number":15},{"id":"1fe42fda92921a0d3913c17f07e42ba7460e95f9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-20 15:45:20.000000000","message":"Patch Set 15: Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"40809e290ed322a3beacb88c78f4475ed4584230","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 15:45:34.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"38030fd6118e919492d81c35976cf286cad0f8bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 16:03:08.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/603955/15/gate/openstack-tox-docs/3f27a41/html/ : SUCCESS in 12m 31s\n- openstack-tox-pep8 http://logs.openstack.org/55/603955/15/gate/openstack-tox-pep8/430cad9/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":15},{"id":"c7587e90baa7b9d517dc2d4098da2b53abde7a07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 16:03:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"d0bae70b911d503384a321a162ced2100eb63218","revisions":{"c0737b343f7ff132d77873479780f0e9f9400d7e":{"kind":"REWORK","_number":1,"created":"2018-09-20 02:33:47.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/1"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:32:19.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c0737b343f7ff132d77873479780f0e9f9400d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c0737b343f7ff132d77873479780f0e9f9400d7e"}]},"branch":"refs/heads/master"},"c9548388fc885a70c4275a975c4d27054171a520":{"kind":"REWORK","_number":2,"created":"2018-10-04 00:46:19.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/2"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-10-04 00:42:24.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c9548388fc885a70c4275a975c4d27054171a520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c9548388fc885a70c4275a975c4d27054171a520"}]},"branch":"refs/heads/master"},"7822f852b257b533c4082febb21ab19aac34c73b":{"kind":"REWORK","_number":3,"created":"2018-10-04 03:09:52.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/3"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-10-04 03:07:57.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7822f852b257b533c4082febb21ab19aac34c73b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7822f852b257b533c4082febb21ab19aac34c73b"}]},"branch":"refs/heads/master"},"b6080be405b2c396b6626389e6d9b715859ca59e":{"kind":"REWORK","_number":4,"created":"2018-10-10 04:27:35.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/4"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-10-10 04:25:28.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b6080be405b2c396b6626389e6d9b715859ca59e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b6080be405b2c396b6626389e6d9b715859ca59e"}]},"branch":"refs/heads/master"},"e55bb924f97299c9b8fe92bae23c177a2824dd47":{"kind":"REWORK","_number":5,"created":"2018-10-16 15:27:24.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/5"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-10-16 15:23:45.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e55bb924f97299c9b8fe92bae23c177a2824dd47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e55bb924f97299c9b8fe92bae23c177a2824dd47"}]},"branch":"refs/heads/master"},"f23cdb2c9061b86f694dd4323649938a49b0e507":{"kind":"REWORK","_number":6,"created":"2018-10-16 20:32:24.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/6"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-10-16 20:30:19.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f23cdb2c9061b86f694dd4323649938a49b0e507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f23cdb2c9061b86f694dd4323649938a49b0e507"}]},"branch":"refs/heads/master"},"341ee872e993dd84903a2f5eeb97099cd7490eb7":{"kind":"REWORK","_number":7,"created":"2018-10-18 22:47:37.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/7"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-10-18 22:45:19.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/341ee872e993dd84903a2f5eeb97099cd7490eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/341ee872e993dd84903a2f5eeb97099cd7490eb7"}]},"branch":"refs/heads/master"},"9bb3f95e3023e39f6eeaeeb366eafe950bfc3640":{"kind":"REWORK","_number":8,"created":"2018-10-23 18:27:46.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/8"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-10-23 18:24:54.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bb3f95e3023e39f6eeaeeb366eafe950bfc3640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9bb3f95e3023e39f6eeaeeb366eafe950bfc3640"}]},"branch":"refs/heads/master"},"e5cb89565834b3ec1f34c4a55f4df7ecd6ee4e0b":{"kind":"REWORK","_number":9,"created":"2018-10-27 00:45:58.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/9"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-10-27 00:42:57.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e5cb89565834b3ec1f34c4a55f4df7ecd6ee4e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e5cb89565834b3ec1f34c4a55f4df7ecd6ee4e0b"}]},"branch":"refs/heads/master"},"6e07ef67b1b4fef1e9bc1a7f8df758f37817fc4c":{"kind":"REWORK","_number":10,"created":"2018-11-06 23:14:54.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/10"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-11-06 23:12:10.000000000","tz":-480},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6e07ef67b1b4fef1e9bc1a7f8df758f37817fc4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6e07ef67b1b4fef1e9bc1a7f8df758f37817fc4c"}]},"branch":"refs/heads/master"},"cb9e58da70cb2471f3674ea078bad8784e911a43":{"kind":"REWORK","_number":11,"created":"2019-04-08 04:15:43.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/11"}}},"commit":{"parents":[{"commit":"06b5446e53c728bf7266696e6bedda2b92b498c0","subject":"Merge \"Add a script for counting blueprints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/06b5446e53c728bf7266696e6bedda2b92b498c0"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-04-08 05:13:46.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cb9e58da70cb2471f3674ea078bad8784e911a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cb9e58da70cb2471f3674ea078bad8784e911a43"}]},"branch":"refs/heads/master"},"d25929cd7c6dddfa6462017928e3ba73fce6d45f":{"kind":"REWORK","_number":12,"created":"2019-05-10 22:54:21.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/12"}}},"commit":{"parents":[{"commit":"06b5446e53c728bf7266696e6bedda2b92b498c0","subject":"Merge \"Add a script for counting blueprints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/06b5446e53c728bf7266696e6bedda2b92b498c0"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-11 00:53:51.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d25929cd7c6dddfa6462017928e3ba73fce6d45f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d25929cd7c6dddfa6462017928e3ba73fce6d45f"}]},"branch":"refs/heads/master"},"e994fa31a23a228878110d900b5d6dabfbd7da8d":{"kind":"REWORK","_number":13,"created":"2019-05-24 04:58:25.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/13"}}},"commit":{"parents":[{"commit":"06b5446e53c728bf7266696e6bedda2b92b498c0","subject":"Merge \"Add a script for counting blueprints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/06b5446e53c728bf7266696e6bedda2b92b498c0"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-24 06:56:25.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e994fa31a23a228878110d900b5d6dabfbd7da8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e994fa31a23a228878110d900b5d6dabfbd7da8d"}]},"branch":"refs/heads/master"},"ef53e908fd9f7fe0670f527589b6cf50dcfc2199":{"kind":"REWORK","_number":14,"created":"2019-06-05 22:13:49.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/14"}}},"commit":{"parents":[{"commit":"06b5446e53c728bf7266696e6bedda2b92b498c0","subject":"Merge \"Add a script for counting blueprints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/06b5446e53c728bf7266696e6bedda2b92b498c0"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-06-06 00:10:54.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ef53e908fd9f7fe0670f527589b6cf50dcfc2199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ef53e908fd9f7fe0670f527589b6cf50dcfc2199"}]},"branch":"refs/heads/master"},"d0bae70b911d503384a321a162ced2100eb63218":{"kind":"REWORK","_number":15,"created":"2019-06-18 22:58:32.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/55/603955/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/55/603955/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/55/603955/15"}}},"commit":{"parents":[{"commit":"06b5446e53c728bf7266696e6bedda2b92b498c0","subject":"Merge \"Add a script for counting blueprints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/06b5446e53c728bf7266696e6bedda2b92b498c0"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2018-09-20 02:29:21.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-06-19 00:57:13.000000000","tz":-420},"subject":"Nova Cyborg interaction specification.","message":"Nova Cyborg interaction specification.\n\nDescribes the Nova - Cyborg interaction needed to create and manage\ninstances with accelerators, and the changes needed in Nova to\naccomplish that.\n\nChange-Id: Ibecee0d46e296813e90f855379a134b84d43a55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0bae70b911d503384a321a162ced2100eb63218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d0bae70b911d503384a321a162ced2100eb63218"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
