)]}'
{"id":"openstack%2Fnova-specs~603969","triplet_id":"openstack%2Fnova-specs~master~I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d","project":"openstack/nova-specs","branch":"master","hashtags":[],"change_id":"I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d","subject":"Spec for API inconsistency cleanup","status":"MERGED","created":"2018-09-20 04:16:51.000000000","updated":"2019-05-30 17:20:06.000000000","submitted":"2019-05-30 17:20:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":140,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"603969-1559236806955-60c58a6a","meta_rev_id":"1dc0fed2883797814a629cddca1dca09aca269da","_number":603969,"virtual_id_number":603969,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":2,"date":"2019-05-30 17:20:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-05-30 16:45:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-05-30 16:45:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-21 07:31:07.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2018-09-24 08:34:08.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2018-09-25 00:48:44.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2018-10-12 00:15:08.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-04-08 21:59:11.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-04-09 15:44:19.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-04-09 20:04:23.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2019-04-18 16:52:41.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-19 00:56:09.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2019-04-26 14:01:18.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-05-07 09:53:31.000000000","updated_by":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"reviewer":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"state":"REVIEWER"},{"updated":"2019-05-28 07:40:43.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-05-28 08:40:48.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-05-30 16:45:12.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-05-30 17:20:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce277b3e68f81302354b148acdc248e42792c3cb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-09-20 04:16:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6c26441144555969476da4d42ac74c621410a5e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-09-20 04:28:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2bd412f130cd3b64c6f2efcf9856e698635587d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-20 06:05:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/2/check/openstack-tox-docs/1bbbca6/html/ : SUCCESS in 5m 40s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/2/check/openstack-tox-pep8/871c010/ : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a50c5f5473cdba6d2abe2bf17be7a9fc7597fd1e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-09-20 10:58:24.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f403a7bafe39b56cd7f92d621e81764f45e7c1fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-20 13:01:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/3/check/openstack-tox-docs/226449d/html/ : SUCCESS in 6m 08s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/3/check/openstack-tox-pep8/22bafc8/ : FAILURE in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"436ef1e370d3bbb57e0ec3131db2adcc8be6ddf2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-09-20 16:13:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5611c40e00993a681a9be4e7875a3082e86d1e89","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-09-21 07:31:07.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nSounds like a huge job, I would like to help out if needed.","accounts_in_message":[],"_revision_number":3},{"id":"d28107f1ee399699098ca765f11eef1297c1ae80","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-09-24 08:34:08.000000000","message":"Patch Set 3:\n\n(6 comments)\n\nCouple of grammar/spelling nits but nothing major. Idea seems good. Not +1ing because there\u0027s still some work to do, naturally","accounts_in_message":[],"_revision_number":3},{"id":"42bb07f7c8c9d8edbbcaa93b69c108c9ec490ee6","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-09-25 00:48:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"888145ca88191f1c47740cd93ff71cb1c44f5a10","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-09-26 05:23:43.000000000","message":"Patch Set 3: Workflow-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1e0dffcf5603f390c19f9fc60fd90f8e5a89952b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-10-12 00:15:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"306f997990edee92af5f8cbb955b1c03acab8d16","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-11-28 08:48:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8d3b4644e20dbe3935e3662eac66773b9ed5aad8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-11-28 09:22:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a7568341b600ac97bddf3294244c1bb9b6b4dc49","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-11-28 09:23:39.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"92afac7d191e22a5a2412bebe747029ee296640f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 10:10:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/5/check/openstack-tox-docs/4c28a8b/html/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/5/check/openstack-tox-pep8/5f69859/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":5},{"id":"0fe95de5629bf949a5a44b7b3c7f66832101afc6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-18 01:35:46.000000000","message":"Patch Set 5:\n\nlet\u0027s discuss this in Train PTG if we can do these cleanup in single microversion bump or not.","accounts_in_message":[],"_revision_number":5},{"id":"e4d1a90c0fab965f64e5448993a6d192dfc3d69a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-04 02:37:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dbb08eeaa076adccf9ac7855100855337a7c5a21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 02:49:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/6/check/openstack-tox-docs/8453c8a/ : FAILURE in 10m 00s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/6/check/openstack-tox-pep8/23e334d/ : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":6},{"id":"2f0faaa5df4e5722c0065134241ab73c2b141a7f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-08 21:48:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"eeb4a2bc4654bf72406eece24fc1d0818cad4216","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-08 21:59:03.000000000","message":"Patch Set 7:\n\n(10 comments)\n\nSome questions inline, but I worry this might be trying to do too much in a single microversion. I would definitely love to see the OS-EXT* prefix stuff go away though.","accounts_in_message":[],"_revision_number":7},{"id":"035bdd2e16666847de37157d3eb9a7c32010e3c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 22:43:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/7/check/openstack-tox-docs/3929276/html/ : SUCCESS in 8m 28s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/7/check/openstack-tox-pep8/a4f48b0/ : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":7},{"id":"ffbf17bd06c9f364369c51d82ca6f72da378d0b3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 01:01:35.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"afce2669b01b15760b519eb74f3fe2e0da796d88","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 01:23:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"56f9171dc1e4eb36876d25d2fbf24a8c2ce86b88","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 04:21:50.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"cf9a7d32fde7dd231ad7d07417e5d06a1cc49cc9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 04:44:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"13132325b4fb4f1e24b4b3eb5432b561e2023f89","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 04:46:07.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"4a9795d92c869554a12745052ab79c2cdfd7cbac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 04:54:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/9/check/openstack-tox-docs/726e874/html/ : SUCCESS in 7m 10s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/9/check/openstack-tox-pep8/efb8cc3/ : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":9},{"id":"9043ad5866100958841d0eff85b92947ad558977","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-09 15:43:38.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"dc4cfd86bbf984ee886bfef663ea055153c566c4","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-04-09 15:44:19.000000000","message":"Patch Set 9:\n\n(4 comments)\n\nThis is clearly something that needs to happen and almost impossible to make a complete spec about. I think what\u0027s there does a good job of painting a clear overview of what\u0027s going so, yeah, make it so.\n\nthere\u0027s a line length problem","accounts_in_message":[],"_revision_number":9},{"id":"f3534f8427a7eb16b02df4ad86b57c2509e920d5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-09 15:48:29.000000000","message":"Patch Set 9: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e47ea8aca4bb6d87ce4d61ff7b891e627a40e2d0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 16:14:30.000000000","message":"Patch Set 9:\n\n(15 comments)","accounts_in_message":[],"_revision_number":9},{"id":"99f4cc0830e342c2005944dbc4c24abef197ac0c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 16:49:01.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bd41c895b67c81054d28338fa2fc5e92c0f5660d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 17:10:08.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f3eaa88f11b71dbc3927e230e7cd1aab26b2aed0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-09 17:20:05.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b818b259d24a324d1998c2841760806691899dd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 17:24:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/10/check/openstack-tox-docs/4b9b65a/html/ : SUCCESS in 7m 45s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/10/check/openstack-tox-pep8/6e4c8c1/ : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":10},{"id":"c50243c2f4f4284fa5d2b31e6190f8d682f507fe","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2019-04-09 20:04:23.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"c6d0d33ba18127058d6061f976bdb25ccffab53d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 20:05:15.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7e129cb6d48fcc74d3bb2cf01863cc73d9dfb2b8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-09 20:38:16.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c09b544e4bb84489ced281c1e9fd987bfd03a3e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 20:48:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/11/check/openstack-tox-docs/ad0d00d/html/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/11/check/openstack-tox-pep8/4531b8e/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":11},{"id":"292503a12df77517c78c3f63f3df95fe4498abd9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-16 22:20:25.000000000","message":"Patch Set 11:\n\nI\u0027ll echo what Chris said about it being impossible for this to be complete. That being the case, we should just draw a line, do a bunch of stuff in a microversion, and accept that we\u0027ll have to do more later in another microversion. Otherwise we\u0027ll spin our wheels forever trying to be exhaustive and never get anything done.\n\nI think the spec as it stands does a good job delineating three classes of things to be done. So, cool.","accounts_in_message":[],"_revision_number":11},{"id":"9059407301591d37f9ac6714c0c5a0780bffbc1f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-17 00:02:21.000000000","message":"Patch Set 11:\n\n(9 comments)","accounts_in_message":[],"_revision_number":11},{"id":"4e9ed21560849e8373baaa1f07b04c218ff16d03","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-17 02:53:45.000000000","message":"Patch Set 11:\n\n(8 comments)","accounts_in_message":[],"_revision_number":11},{"id":"1bd386b1121c488228938a3c8345853400f4547b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-17 02:54:15.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2c23715a1ac71eb018174971c32a3df8bd17e30c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-17 02:55:09.000000000","message":"Patch Set 11:\n\n\u003e I\u0027ll echo what Chris said about it being impossible for this to be\n \u003e complete. That being the case, we should just draw a line, do a\n \u003e bunch of stuff in a microversion, and accept that we\u0027ll have to do\n \u003e more later in another microversion. Otherwise we\u0027ll spin our wheels\n \u003e forever trying to be exhaustive and never get anything done.\n \u003e \n \u003e I think the spec as it stands does a good job delineating three\n \u003e classes of things to be done. So, cool.\n\ntrue. I was targetting more initially which has always the risk of regression also.","accounts_in_message":[],"_revision_number":11},{"id":"947afaa68f01442f913e89d65c64739a71a42cbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 03:05:16.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/12/check/openstack-tox-docs/0a6272b/html/ : SUCCESS in 8m 48s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/12/check/openstack-tox-pep8/6bf1494/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":12},{"id":"a99989515ae811fb28fc939e1208eee3541e91f1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-17 18:56:22.000000000","message":"Patch Set 12: Code-Review-1\n\n(13 comments)\n\nAlmost there. Mostly docs/formatting/wording nits inline, but I think there are a couple of issues worth fixing for clarity and also we need to determine what, if anything, should be done about scheduler hints.","accounts_in_message":[],"_revision_number":12},{"id":"d2d85b343638f3fad6e2d4e91a47753f19ecbf0a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-17 19:20:44.000000000","message":"Patch Set 12:\n\n(12 comments)","accounts_in_message":[],"_revision_number":12},{"id":"2066b3e0468dd4112c6de68c6d684d00843b9c25","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-17 19:20:49.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"b11ee8da4f49349df07e1c09694b725f28ffad94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 19:49:31.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/13/check/openstack-tox-docs/f6e426e/ : FAILURE in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/13/check/openstack-tox-pep8/936a663/ : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":13},{"id":"2ae69c255a33d0b774a6cfde0852b8f1fa01c486","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-18 14:18:38.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b28a43542b0f11e80852c49d69602551818fc76a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 14:32:22.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/14/check/openstack-tox-docs/05f673c/html/ : SUCCESS in 8m 11s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/14/check/openstack-tox-pep8/f068379/ : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":14},{"id":"8d054e19103746c0e59057db495d9ee035fd704f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-18 16:28:06.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"361626c61d57311eb06a79a56b4df7289d3ca471","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-18 16:38:13.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"dd28d2f874bf2809cac2ef0530062ba2065944c5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-18 16:42:09.000000000","message":"Patch Set 15: Code-Review+1\n\nThere might be other prefixes and API wrinkles that I\u0027m missing with this, I haven\u0027t done an exhaustive audit, but don\u0027t want to hold this up on me doing that audit either - it can be done during code review, and hopefully most of these prefixed parameters are centrally located in the view builder code anyway.\n\nI\u0027m just a +1 for now since this is a pretty big change and I want to make sure other cores (and devs/users/ops) are at least aware of this and either in general agreement or opt out of caring before we move forward - especially since we\u0027re just under 2 weeks from the Train PTG.","accounts_in_message":[],"_revision_number":15},{"id":"a6fdea9aec80336101e7aeed6230068b66a0bebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 16:49:08.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/15/check/openstack-tox-docs/9669b8e/html/ : SUCCESS in 7m 05s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/15/check/openstack-tox-pep8/01ff38e/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":15},{"id":"e0be997845ed8550ed29156ea53bb09de488ace9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-18 16:52:41.000000000","message":"Patch Set 15: Code-Review+1\n\nRegistering \"aware of this and ... in general agreement\"","accounts_in_message":[],"_revision_number":15},{"id":"3ca89165739cf2e715127073444d358fb4b3b0ae","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2019-04-19 00:56:09.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"bb9ea97981c8425032a279d523ebd4b548b6d1be","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-26 14:01:18.000000000","message":"Patch Set 15: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":15},{"id":"e028b1fc722129d5728ed7d63888694bc4ce60aa","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-05-03 03:43:40.000000000","message":"Patch Set 15: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":15},{"id":"16d0d12980129c4792fbcb6b5dab2ae882747be8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-06 19:13:58.000000000","message":"Patch Set 15: Code-Review-1\n\n(3 comments)\n\nThis needs to be updated based on the Train PTG discussion.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-May/005824.html","accounts_in_message":[],"_revision_number":15},{"id":"e82898b75e7b01061932a21a8b34e86a8c6720cb","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-05-07 00:53:07.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"93027b3642690d5b148808d30bd7d689d36d65ec","author":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"date":"2019-05-07 09:53:31.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"b66a33571bb545632168115a4d03d64edc90ba40","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-08 16:23:08.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"984eb6240828df8715bb7b7d1f99eb8154911c87","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-05-10 06:54:00.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"8e11b50303f54a00d5cd4688b5e7c01d506304aa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-27 06:47:51.000000000","message":"Patch Set 15: Workflow-1\n\nupdating with the Train PTG agreement.\n-  http://lists.openstack.org/pipermail/openstack-discuss/2019-May/005824.html","accounts_in_message":[],"_revision_number":15},{"id":"90c4186fbb38cd1ac98f2ce088bc9b40831871e8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-27 09:13:54.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"54015428a174b270f46c2018e740925c842de7fd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-27 09:18:55.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"8346b8136c837d4c8a375df1f20d65bc8faf69a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 09:29:17.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/16/check/openstack-tox-docs/eb228ee/html/ : SUCCESS in 7m 33s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/16/check/openstack-tox-pep8/7cc7322/ : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":16},{"id":"3baa81dba23d5d963eccf6e15a594ba4e9266c2e","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-05-28 07:40:43.000000000","message":"Patch Set 16: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":16},{"id":"80e9be0d2c0a8f5e17bca874eb1943a7fdf873b6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-05-28 08:40:48.000000000","message":"Patch Set 16: Code-Review+1\n\nThanks, LGTM.","accounts_in_message":[],"_revision_number":16},{"id":"48e54fa2d20d4ed12106e8bf3b97cf4d5a51f49b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-05-28 08:45:56.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4ec79f542f16993e4a5c2d0e08bd262ba0b99449","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-30 16:26:21.000000000","message":"Patch Set 16:\n\n(7 comments)","accounts_in_message":[],"_revision_number":16},{"id":"d7f0845772a52166099ef3e8b07aaf2c513c2192","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-30 16:27:14.000000000","message":"Patch Set 16:\n\nI don\u0027t want to -1 this but I\u0027d like to see my comments addressed so this gets cleaned up. I\u0027ll fast approve it.","accounts_in_message":[],"_revision_number":16},{"id":"74c9f2b75a691b6d476dfb9012b1a8a52df4549f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-30 16:40:11.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"a4abc09aafb6ce0963a6f5420fc96be2d1dc82fa","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-30 16:45:12.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\nLGTM. Alex was +2 on PS16 and PS17 is just cleaning up some stuff, so I\u0027m going to fast approve.","accounts_in_message":[],"_revision_number":17},{"id":"a8a7c378297fc202f3d96afc9b264a14699442fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 17:07:33.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/17/check/openstack-tox-docs/a7376b5/html/ : SUCCESS in 9m 05s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/17/check/openstack-tox-pep8/08b1cfd/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":17},{"id":"8be9d32a32a836ffd797e5fc2d37582e47f12333","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 17:07:47.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"d32080f6d962cb0e98322a67f614fe23ddaa8246","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 17:20:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"1dc0fed2883797814a629cddca1dca09aca269da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 17:20:06.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/603969/17/gate/openstack-tox-docs/25dc58f/html/ : SUCCESS in 7m 29s\n- openstack-tox-pep8 http://logs.openstack.org/69/603969/17/gate/openstack-tox-pep8/ac07272/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"f81524d40e7bc71bef86d5f1f860a7d1c6a9df7d","revisions":{"2795ac0953afe6535b7073ea61cdb49d6a102cef":{"kind":"REWORK","_number":1,"created":"2018-09-20 04:16:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/1"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:16:49.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2795ac0953afe6535b7073ea61cdb49d6a102cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2795ac0953afe6535b7073ea61cdb49d6a102cef"}]},"branch":"refs/heads/master"},"a9e90e7933cce723a585a5847e07409333a647e8":{"kind":"REWORK","_number":2,"created":"2018-09-20 04:28:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/2"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:28:17.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a9e90e7933cce723a585a5847e07409333a647e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a9e90e7933cce723a585a5847e07409333a647e8"}]},"branch":"refs/heads/master"},"0330af5504f1131c16baf5cd8a708e48ae1cec2e":{"kind":"REWORK","_number":3,"created":"2018-09-20 10:58:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/3"}}},"commit":{"parents":[{"commit":"cce50c82878a035c685eb8c3f4b70351567c82c4","subject":"Merge \"fix spelling mistake\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cce50c82878a035c685eb8c3f4b70351567c82c4"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2018-09-20 10:58:19.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0330af5504f1131c16baf5cd8a708e48ae1cec2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0330af5504f1131c16baf5cd8a708e48ae1cec2e"}]},"branch":"refs/heads/master"},"f650377e69d2db59226eeca386bd17bd7958cc4f":{"kind":"REWORK","_number":4,"created":"2018-11-28 09:22:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/4"}}},"commit":{"parents":[{"commit":"6da301304c6467b8bd4fafadde16f119b1291d32","subject":"Merge \"Use conductor groups to partition nova-compute services for Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6da301304c6467b8bd4fafadde16f119b1291d32"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-11-28 09:22:02.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f650377e69d2db59226eeca386bd17bd7958cc4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f650377e69d2db59226eeca386bd17bd7958cc4f"}]},"branch":"refs/heads/master"},"e3d229923395b4d076ec0b6e56ce469083acbe88":{"kind":"REWORK","_number":5,"created":"2018-11-28 09:23:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/5"}}},"commit":{"parents":[{"commit":"6da301304c6467b8bd4fafadde16f119b1291d32","subject":"Merge \"Use conductor groups to partition nova-compute services for Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6da301304c6467b8bd4fafadde16f119b1291d32"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2018-11-28 09:23:34.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3d229923395b4d076ec0b6e56ce469083acbe88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3d229923395b4d076ec0b6e56ce469083acbe88"}]},"branch":"refs/heads/master"},"1a45b7032c33fbe7a24084261f9e51fdbd1198de":{"kind":"REWORK","_number":6,"created":"2019-04-04 02:37:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/6"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-04-04 02:37:11.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a45b7032c33fbe7a24084261f9e51fdbd1198de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/1a45b7032c33fbe7a24084261f9e51fdbd1198de"}]},"branch":"refs/heads/master"},"ae6a7f091c357f5f9eb3ece00e0202edb72ccd01":{"kind":"REWORK","_number":7,"created":"2019-04-08 21:48:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/7"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-04-08 21:47:50.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae6a7f091c357f5f9eb3ece00e0202edb72ccd01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ae6a7f091c357f5f9eb3ece00e0202edb72ccd01"}]},"branch":"refs/heads/master"},"38206779b8ee8c48eb358ba3577084f651b59d21":{"kind":"REWORK","_number":8,"created":"2019-04-09 04:44:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/8"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-04-09 04:44:46.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/38206779b8ee8c48eb358ba3577084f651b59d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/38206779b8ee8c48eb358ba3577084f651b59d21"}]},"branch":"refs/heads/master"},"9dd8ad5a805a1f6a26f7ef608fe89f56ab755bc4":{"kind":"REWORK","_number":9,"created":"2019-04-09 04:46:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/9"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-09 04:46:04.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9dd8ad5a805a1f6a26f7ef608fe89f56ab755bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9dd8ad5a805a1f6a26f7ef608fe89f56ab755bc4"}]},"branch":"refs/heads/master"},"77808d683c10561fb3d81d4c70cfa15114492db8":{"kind":"REWORK","_number":10,"created":"2019-04-09 17:10:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/10"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-04-09 17:09:55.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/77808d683c10561fb3d81d4c70cfa15114492db8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/77808d683c10561fb3d81d4c70cfa15114492db8"}]},"branch":"refs/heads/master"},"5f9b0353c05e9a306daa6442ad34fc076b30f945":{"kind":"REWORK","_number":11,"created":"2019-04-09 20:38:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/11"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-04-09 20:31:22.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f9b0353c05e9a306daa6442ad34fc076b30f945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f9b0353c05e9a306daa6442ad34fc076b30f945"}]},"branch":"refs/heads/master"},"c7732cc6a3dc77cd924cf8f4c17a1b0dae36d66d":{"kind":"REWORK","_number":12,"created":"2019-04-17 02:54:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/12"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-17 02:54:12.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c7732cc6a3dc77cd924cf8f4c17a1b0dae36d66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c7732cc6a3dc77cd924cf8f4c17a1b0dae36d66d"}]},"branch":"refs/heads/master"},"d432b43bae417f3d0e1d86318e39a50bc4f7b5d5":{"kind":"REWORK","_number":13,"created":"2019-04-17 19:20:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/13"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-17 19:20:33.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d432b43bae417f3d0e1d86318e39a50bc4f7b5d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d432b43bae417f3d0e1d86318e39a50bc4f7b5d5"}]},"branch":"refs/heads/master"},"f7f300463c959233f9d0ad0334529a17be4d61f8":{"kind":"REWORK","_number":14,"created":"2019-04-18 14:18:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/14"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-18 14:18:26.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7f300463c959233f9d0ad0334529a17be4d61f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f7f300463c959233f9d0ad0334529a17be4d61f8"}]},"branch":"refs/heads/master"},"f1a0d7e65b493589b9c102d17cfe7ed23bc13d30":{"kind":"REWORK","_number":15,"created":"2019-04-18 16:38:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/15"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-18 16:38:11.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1a0d7e65b493589b9c102d17cfe7ed23bc13d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f1a0d7e65b493589b9c102d17cfe7ed23bc13d30"}]},"branch":"refs/heads/master"},"529ec9c298d0f7738f272cdfad08e9b7153ed187":{"kind":"REWORK","_number":16,"created":"2019-05-27 09:18:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/16"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-05-27 09:18:41.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/529ec9c298d0f7738f272cdfad08e9b7153ed187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/529ec9c298d0f7738f272cdfad08e9b7153ed187"}]},"branch":"refs/heads/master"},"f81524d40e7bc71bef86d5f1f860a7d1c6a9df7d":{"kind":"REWORK","_number":17,"created":"2019-05-30 16:40:11.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/603969/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/69/603969/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/69/603969/17"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2018-09-20 04:15:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-05-30 16:40:07.000000000","tz":0},"subject":"Spec for API inconsistency cleanup","message":"Spec for API inconsistency cleanup\n\nAPIImpact\n\nChange-Id: I7b71f5b3ec00b5f59fca6b1b0e239c1dd9dfa75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f81524d40e7bc71bef86d5f1f860a7d1c6a9df7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f81524d40e7bc71bef86d5f1f860a7d1c6a9df7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
