)]}'
{"id":"openstack%2Fnova-specs~624689","triplet_id":"openstack%2Fnova-specs~master~I0d4855c100a43548019cd0330ec325ba8b8c686d","project":"openstack/nova-specs","branch":"master","topic":"bp/support-specifying-az-when-restore-shelved-server","hashtags":[],"change_id":"I0d4855c100a43548019cd0330ec325ba8b8c686d","subject":"Specifying az when restore shelved server","status":"MERGED","created":"2018-12-12 12:56:22.000000000","updated":"2019-07-05 13:52:19.000000000","submitted":"2019-06-10 06:01:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":95,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"624689-1560146482457-b5d80b2d","meta_rev_id":"90adf22bd09f6319e2d4e60ed2e94d2a350a9197","_number":624689,"virtual_id_number":624689,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-06-07 00:30:38.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2019-06-10 06:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2019-06-06 23:55:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2019-06-10 05:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2019-06-10 05:52:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 08:32:51.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2019-06-06 21:27:41.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-06-06 23:55:05.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-06-10 05:52:52.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-06-10 06:01:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ff70a905aef3173798733163ffe5a1cb0d9617f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-12-12 12:56:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c077c656ae96f3182d09da7ab9c7856439223ebc","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2018-12-12 12:58:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"24d320529985066e3b7d59c2d3deb1d71e6fa226","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-12 13:05:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/2/check/openstack-tox-docs/23b2664/html/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/2/check/openstack-tox-pep8/7b75d72/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"2b691b51d6a301e68dca1c697ccf114446dd4272","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-01-03 23:54:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nSoft -1 for getting more details, but I don\u0027t see a major reason not to allow this since unshelve is really just like server create except it\u0027s an existing server with existing ports/volumes.\n\nIt might be worth socializing this in the mailing list to see if other users are looking for this functionality, or if operators have had requests for this from their users.","accounts_in_message":[],"_revision_number":2},{"id":"25311010f6a35748dfaa9512fa65227651bc5224","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-01-04 08:55:00.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nThanks Matt, add the necessary details and update this.","accounts_in_message":[],"_revision_number":2},{"id":"3d8d415f08736a9719ef64941d0fb29d622cbced","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-01-04 09:45:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"14768d449b07376ae58a0fb38e99b250487686b3","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-01-04 09:47:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0b8220dac20825e47a176c83788f2fd347109dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-04 09:52:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/3/check/openstack-tox-docs/cba740b/html/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/3/check/openstack-tox-pep8/c177cc3/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"ef0a0c920b65c79b1c2bd299688dc2d31c370089","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-01-04 22:11:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88d602753aa003deeaf6d7bb4cc32ffea5b31715","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-01-04 22:23:01.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dfa1bd11eb01414b8187e111f05165c0c05a73cb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-01-04 22:25:38.000000000","message":"Patch Set 3:\n\nTo be clear, my -1 is mostly about cleaning up the spec, not about the proposal itself.","accounts_in_message":[],"_revision_number":3},{"id":"6e5ead675ed80a698a09b49d2d3657045c21ac69","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-01-07 03:06:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"395306c31e77cc677f670e5892407f3ad458afc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-07 03:15:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/4/check/openstack-tox-docs/d849ab7/html/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/4/check/openstack-tox-pep8/d8e0564/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":4},{"id":"4d51a6f887875e3f3af7bf57360745e831f346d5","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-01-07 03:30:39.000000000","message":"Patch Set 4:\n\n(10 comments)\n\n\u003e To be clear, my -1 is mostly about cleaning up the spec, not about\n \u003e the proposal itself.\n\nThanks Matt.\nI considered and discussed it in our group and thought there are two options here (for reference only):\n1. Do not clean up RequestSpec.availability_zone, determine whether AZ has changed during unshelve, and then select the unshelve processing method - using the user specifying AZ to restore, or using [DEFAULT]/default_schedule_zone (when not specifying AZ);\n2. Set RequestSpec.availability_zone is None, but this will clear the RequestSpec.availability_zone set when first created, and can\u0027t track changes in server availability_zone (not what I think).","accounts_in_message":[],"_revision_number":4},{"id":"8236ebd5ace6d0d2067db80ed694133e2d891a5c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-02 02:13:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3080b9a0ae8b6eaf3058e09f4a3c8384b0a91c67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 02:23:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/5/check/openstack-tox-docs/b262b10/ : FAILURE in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/5/check/openstack-tox-pep8/9cf96d0/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":5},{"id":"ef23ddf8ef13ba082c634d01343cab9a2997de6f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-02 05:59:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"27b94315c716ac1033be65fd9537df491968d40b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 06:14:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/6/check/openstack-tox-docs/94cc8ff/html/ : SUCCESS in 8m 35s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/6/check/openstack-tox-pep8/df2959f/ : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":6},{"id":"dc99d72e76adccaf76522c1f3c9a9760292b061d","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2019-04-18 08:32:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b7c9ba1b6f63d3d127f07b4e4ea4fd43f4ff65a4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-18 09:43:51.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThanks Kevin.","accounts_in_message":[],"_revision_number":6},{"id":"3b08c16622448291f5fc8d914777b94c079e251f","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-19 06:39:55.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5ff0531ea925a6f48ca897b020ac3fc33a4ce3b1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-19 08:23:01.000000000","message":"Patch Set 6:\n\n(4 comments)\n\n\u003e (3 comments)\n\nThanks Alex.","accounts_in_message":[],"_revision_number":6},{"id":"cb3c7706e5a5fa9cf215558aa35884439c54f73a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-19 09:14:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52240386c9b416f981414477d4c4107e4f88be9d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-19 09:38:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8c7f8eca3f199e8d98d6700e9e0521576db687cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 10:47:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/7/check/openstack-tox-docs/b8fceb1/html/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/7/check/openstack-tox-pep8/33e1c25/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":7},{"id":"cc66eaa52dbfdfc28506d5f89b430bb5c0ccfb61","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-23 19:59:47.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)\n\nI would support this if it weren\u0027t for the ZONE:HOST:NODE thing where the HOST:NODE forcefully bypass the scheduler filters and put the server on the given host/node, which we\u0027re trying to do away with:\n\nhttps://specs.openstack.org/openstack/nova-specs/specs/train/approved/add-host-and-hypervisor-hostname-flag-to-create-server.html\n\nhttps://specs.openstack.org/openstack/nova-specs/specs/stein/implemented/remove-force-flag-from-live-migrate-and-evacuate.html\n\nAnd we also pushed back on adding the ability to force a specified target host during cold migrate.","accounts_in_message":[],"_revision_number":7},{"id":"c4ecadc83c395014e21faabad821948230d9ce4e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-23 20:03:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"067f357632c9ca14fa271f925440ed74a0f9db61","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-23 20:06:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ce219b6d5035614c5230eaaf5d80f5e40fa650f7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-23 20:12:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cba50357ef00fedb4720524851fc3a5de192131e","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-24 04:49:55.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e796fb8fdb1f13322e6e65e85bd1592ca3d4c19e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-29 14:34:26.000000000","message":"Patch Set 7:\n\n(7 comments)\n\nThanks Matt and Alex, I will update these suggestions, and adopt your suggestion, using \u0027host_name\u0027 and \u0027hypervisor_hostname\u0027 instead of force host and/or node, I also prefer this way.","accounts_in_message":[],"_revision_number":7},{"id":"1254fcc7acf7331fa53ac50a3d2069a0e7e7ac47","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-30 00:56:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6f4c206e6f0c949aa0399367e3790fe88214d68e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-30 01:01:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e4b48a076bfcf2579e89031cf83adff07c72b1c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 01:08:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/8/check/openstack-tox-docs/c104d82/html/ : SUCCESS in 7m 04s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/8/check/openstack-tox-pep8/48b9fd6/ : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":8},{"id":"97c4c99536e914355379007c831371f871997e63","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-30 01:58:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fa2e7b6c29a123a089e6cbabd8afde8c2cec8bed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 02:20:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/9/check/openstack-tox-docs/69dd3c1/html/ : SUCCESS in 8m 48s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/9/check/openstack-tox-pep8/2a9ae58/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":9},{"id":"f4481ba4b3d87990b25f0aafa9a24754891565ba","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-04 20:31:16.000000000","message":"Patch Set 9: Code-Review-1\n\n(13 comments)\n\nThis is looking mostly OK but some things need to be corrected within and I think it\u0027s a bit confusing over whether or not the forced host/node behavior is going to be supported with this or not. I think from earlier discussions we said we would not support the zone:host:node format to force the host/node and bypass the scheduler, but the spec doesn\u0027t read that way to me. If we want to clarify and get wider input we should ask in the mailing list.","accounts_in_message":[],"_revision_number":9},{"id":"f37e96390eb19f41bd098e41dd6ac33376f6454a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-06-04 21:09:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5c4d2125519b37dc171bc51ce1920b7d671ad437","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-05 08:14:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ab4b012dcaccf83087ac0fca6261e6bc96371397","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 08:23:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/10/check/openstack-tox-docs/e5e310e/html/ : SUCCESS in 7m 37s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/10/check/openstack-tox-pep8/fe8e0a9/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":10},{"id":"b253bee8166888c21ec1f7a43a75965b6392e48a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-05 08:23:20.000000000","message":"Patch Set 10:\n\n(13 comments)\n\nThanks Matt, melwitt.\nUnshelve a shelved offloaded server just only have the availability_zone parameter in ``request_spec`` db table, and in ``spec`` field records the availability_zone be an availability_zone and nothing more, so there is no need to make unshelve more complicated and support the specified AZ.","accounts_in_message":[],"_revision_number":10},{"id":"02ed43c45698c1b7d7fb14ec9a554b22e4caeb59","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-06 20:48:43.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8270a3cde030c0f2333e19c5efd55bf1f32641b4","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-06-06 21:27:41.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"892335c545d41aa8975f8e195fe74a867335a150","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-06 23:54:40.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"2ed9ef28838a603804f03cb6e1bfe2348dcb27b8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-06 23:54:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9061b9f270f5d8d4f2d2d8f9bef0146a748b6ef9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-06 23:55:05.000000000","message":"Patch Set 11: Code-Review+2\n\nI\u0027m OK with this now.","accounts_in_message":[],"_revision_number":11},{"id":"2afe8f3c9862a0609eb2a782f4cdfc1731a3b728","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:07:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/11/check/openstack-tox-docs/c64039f/html/ : SUCCESS in 7m 41s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/11/check/openstack-tox-pep8/01d5877/ : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":11},{"id":"37fde440f02d523a13eda785ed33f4e4ecdf0669","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-07 00:30:38.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"739739b03f4b3f01c8d140b62cffaafe73ec901f","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-06-10 05:52:52.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThe spec LGTM, but only one thing should be clarify, and that can be done a follow-up I think","accounts_in_message":[],"_revision_number":11},{"id":"65e7080127c8835e0165e1f5a5ac3c4b1c290ca5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 05:53:04.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"35cf5e12a769a55077a9ba9665f332e6da965706","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 06:01:22.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/89/624689/11/gate/openstack-tox-docs/5452877/html/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/89/624689/11/gate/openstack-tox-pep8/680c88a/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":11},{"id":"f5f433fd6dc53651de5c296bdcda296141499c5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 06:01:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"e8c7dc464594f37c386aed153bfb43d6480a5ced","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-10 06:32:54.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"33a980cd9220740fffc2f7a23164fc894a79b050","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-10 15:11:56.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ca9420bff68d4e36e9f138f094a00e70d784bc67","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-06-11 00:36:49.000000000","message":"Patch Set 11:\n\n\u003e (1 comment)\n\nThanks. Yeah, these are looks like a 400, I will consider of this when deal with the code.","accounts_in_message":[],"_revision_number":11},{"id":"e22e4070d4378e10e541fc7aa1b7f1c5b60ff6f1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-05 04:48:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"dfd1031208ae1710b002ff25bdffb637af585f09","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-05 13:52:19.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"014f3cfe7996567c101c5f74015e184adb2e0bb3","revisions":{"27a5f61f182e4ddc759b3c2f05a2f7deac5180ca":{"kind":"REWORK","_number":1,"created":"2018-12-12 12:56:22.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/1"}}},"commit":{"parents":[{"commit":"a1e34cb2752523c3ec5a44852232ec5838f4f278","subject":"Merge \"Per aggregate scheduling weight (spec)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a1e34cb2752523c3ec5a44852232ec5838f4f278"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:56:11.000000000","tz":-300},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\nBlueprint: support-specifying-az-when-restore-shelved-server\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/27a5f61f182e4ddc759b3c2f05a2f7deac5180ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/27a5f61f182e4ddc759b3c2f05a2f7deac5180ca"}]},"branch":"refs/heads/master"},"8ba585f3c464166757c1aefd26a5180bf508a944":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-12-12 12:58:28.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/2"}}},"commit":{"parents":[{"commit":"a1e34cb2752523c3ec5a44852232ec5838f4f278","subject":"Merge \"Per aggregate scheduling weight (spec)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a1e34cb2752523c3ec5a44852232ec5838f4f278"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:57:59.000000000","tz":-300},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8ba585f3c464166757c1aefd26a5180bf508a944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8ba585f3c464166757c1aefd26a5180bf508a944"}]},"branch":"refs/heads/master"},"3928a45604de81b806101e73e83ceafb6c47eac0":{"kind":"REWORK","_number":3,"created":"2019-01-04 09:45:26.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/3"}}},"commit":{"parents":[{"commit":"9109cf2e7464c433ec6e416f7296a4db97e15d61","subject":"Merge \"Remove the duplicate words\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9109cf2e7464c433ec6e416f7296a4db97e15d61"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-01-04 09:45:02.000000000","tz":-300},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3928a45604de81b806101e73e83ceafb6c47eac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3928a45604de81b806101e73e83ceafb6c47eac0"}]},"branch":"refs/heads/master"},"3257315f067bd5ab98d81f0443032c3afb3192d8":{"kind":"REWORK","_number":4,"created":"2019-01-07 03:06:31.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/4"}}},"commit":{"parents":[{"commit":"9109cf2e7464c433ec6e416f7296a4db97e15d61","subject":"Merge \"Remove the duplicate words\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9109cf2e7464c433ec6e416f7296a4db97e15d61"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-01-07 03:06:10.000000000","tz":-300},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3257315f067bd5ab98d81f0443032c3afb3192d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3257315f067bd5ab98d81f0443032c3afb3192d8"}]},"branch":"refs/heads/master"},"fce122fe6f0d617e848a701848e99695d4c93caa":{"kind":"REWORK","_number":5,"created":"2019-04-02 02:13:07.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/5"}}},"commit":{"parents":[{"commit":"32cf20135bbc9c3e3df0da21a12a4f79404c73a0","subject":"docs: Resolve issues with Sphinx 2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/32cf20135bbc9c3e3df0da21a12a4f79404c73a0"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-04-02 02:13:35.000000000","tz":480},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fce122fe6f0d617e848a701848e99695d4c93caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fce122fe6f0d617e848a701848e99695d4c93caa"}]},"branch":"refs/heads/master"},"5c7005443e64233758afd66fe54ced53d1c7bf90":{"kind":"REWORK","_number":6,"created":"2019-04-02 05:59:53.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/6"}}},"commit":{"parents":[{"commit":"32cf20135bbc9c3e3df0da21a12a4f79404c73a0","subject":"docs: Resolve issues with Sphinx 2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/32cf20135bbc9c3e3df0da21a12a4f79404c73a0"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-04-02 06:00:07.000000000","tz":480},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5c7005443e64233758afd66fe54ced53d1c7bf90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5c7005443e64233758afd66fe54ced53d1c7bf90"}]},"branch":"refs/heads/master"},"7022bd6ac63f3d7c9fc61254bb48fa46570198ca":{"kind":"REWORK","_number":7,"created":"2019-04-19 09:38:57.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/7"}}},"commit":{"parents":[{"commit":"eb3afea28eba5fb74cf137e556abd2f47981d1c7","subject":"Add host and hypervisor_hostname flag to create server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eb3afea28eba5fb74cf137e556abd2f47981d1c7"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-04-19 09:38:51.000000000","tz":480},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7022bd6ac63f3d7c9fc61254bb48fa46570198ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7022bd6ac63f3d7c9fc61254bb48fa46570198ca"}]},"branch":"refs/heads/master"},"2360bfb424bd6c6c5037e9c54cf9ac1411f48709":{"kind":"REWORK","_number":8,"created":"2019-04-30 00:56:47.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/8"}}},"commit":{"parents":[{"commit":"eb3afea28eba5fb74cf137e556abd2f47981d1c7","subject":"Add host and hypervisor_hostname flag to create server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eb3afea28eba5fb74cf137e556abd2f47981d1c7"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-04-30 00:56:37.000000000","tz":480},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2360bfb424bd6c6c5037e9c54cf9ac1411f48709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2360bfb424bd6c6c5037e9c54cf9ac1411f48709"}]},"branch":"refs/heads/master"},"a81b2b53cd7aa0ae8f56f5cb1188f299700d36d5":{"kind":"REWORK","_number":9,"created":"2019-04-30 01:58:09.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/9"}}},"commit":{"parents":[{"commit":"eb3afea28eba5fb74cf137e556abd2f47981d1c7","subject":"Add host and hypervisor_hostname flag to create server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/eb3afea28eba5fb74cf137e556abd2f47981d1c7"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-04-30 01:57:42.000000000","tz":480},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a81b2b53cd7aa0ae8f56f5cb1188f299700d36d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a81b2b53cd7aa0ae8f56f5cb1188f299700d36d5"}]},"branch":"refs/heads/master"},"d4ad4a581e9a04e35d19ea653b295ad499918c36":{"kind":"REWORK","_number":10,"created":"2019-06-05 08:14:33.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/89/624689/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/10"}}},"commit":{"parents":[{"commit":"37e4368601df54e4b01372b9919b65c807b3038c","subject":"Merge \"Amend count-quota-usage-from-placement to reflect implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/37e4368601df54e4b01372b9919b65c807b3038c"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-06-05 08:10:47.000000000","tz":480},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d4ad4a581e9a04e35d19ea653b295ad499918c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d4ad4a581e9a04e35d19ea653b295ad499918c36"}]},"branch":"refs/heads/master"},"014f3cfe7996567c101c5f74015e184adb2e0bb3":{"kind":"REWORK","_number":11,"created":"2019-06-06 23:54:40.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/89/624689/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/89/624689/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/89/624689/11"}}},"commit":{"parents":[{"commit":"37e4368601df54e4b01372b9919b65c807b3038c","subject":"Merge \"Amend count-quota-usage-from-placement to reflect implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/37e4368601df54e4b01372b9919b65c807b3038c"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2018-12-12 12:53:39.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-06 23:54:37.000000000","tz":0},"subject":"Specifying az when restore shelved server","message":"Specifying az when restore shelved server\n\nThis spec proposes support admin/user to specify\n``availability_zone`` to unshelve a shelved server.\n\nAPIImpact\n\nBlueprint: support-specifying-az-when-restore-shelved-server\nChange-Id: I0d4855c100a43548019cd0330ec325ba8b8c686d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/014f3cfe7996567c101c5f74015e184adb2e0bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/014f3cfe7996567c101c5f74015e184adb2e0bb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
