)]}'
{"id":"openstack%2Fnova-specs~638734","triplet_id":"openstack%2Fnova-specs~master~I5912498e7f5f449c44884d93585362e1a04185a9","project":"openstack/nova-specs","branch":"master","topic":"bp/flavor-extra-spec-image-property-validation-extended","hashtags":[],"change_id":"I5912498e7f5f449c44884d93585362e1a04185a9","subject":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec","status":"MERGED","created":"2019-02-22 17:54:36.000000000","updated":"2019-07-03 15:55:37.000000000","submitted":"2019-07-03 15:55:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":123,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"638734-1562169337310-d5a15b68","meta_rev_id":"118083a32c986e074ec1daef151ea0f888c4c39d","_number":638734,"virtual_id_number":638734,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2019-07-03 15:55:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-07-03 15:33:18.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-07-03 15:33:18.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-09 13:25:37.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-04-09 22:16:51.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-07-03 13:41:15.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-07-03 13:57:28.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-07-03 14:42:23.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-07-03 15:27:26.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-07-03 15:33:18.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-03 15:55:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b5a58be52ea6b77b2de1872fcedabdf0298456d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-02-22 17:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"912d15d61ee59dd6c76ebfbad6b3256b22136ac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 18:05:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/34/638734/1/check/openstack-tox-docs/912a570/html/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/34/638734/1/check/openstack-tox-pep8/8ef580e/ : FAILURE in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b709d68a43c35d38203abd2f9c716ae2d1a3f49e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-02-25 16:40:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e4802376392b3f502208c8316fc5eef2c66521f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-25 16:51:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/34/638734/2/check/openstack-tox-docs/0235a13/html/ : SUCCESS in 8m 09s\n- openstack-tox-pep8 http://logs.openstack.org/34/638734/2/check/openstack-tox-pep8/580742f/ : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"619863d135e704c8f81c97ebb47ec2ca582d8733","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-03-01 05:55:56.000000000","message":"Patch Set 2:\n\nnice~","accounts_in_message":[],"_revision_number":2},{"id":"94dc5b9c7f1f4f2f8b4ca36d7e4f00e19c47b586","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-25 23:20:05.000000000","message":"Patch Set 2:\n\n(10 comments)\n\nThis is very ambitious if you truly intend to make the schema exhaustive on the first pass. If you want to make it forgiving (anything not in the schema is assumed allowable, but unvalidated) then it might be containable to one release.","accounts_in_message":[],"_revision_number":2},{"id":"ee91cd83844aaae6713835d15b0f230a869dd2c1","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-09 13:08:50.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"033483afad5570c006a090a8c4befd1ed64e79c5","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-09 13:25:37.000000000","message":"Patch Set 2: Code-Review-1\n\nAlso it sounds like we can just done that by the json-schema https://github.com/openstack/nova/blob/master/nova/api/openstack/compute/schemas/flavors_extraspecs.py#L27\n\nSo it would be good to clarify that why we need new yaml format instead of json-schema.","accounts_in_message":[],"_revision_number":2},{"id":"e43980f765289b0cc0a46a55adbd6d5443d4c920","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-04-09 13:56:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(12 comments)\n\nI think we need to discuss about how to use the YAML file.","accounts_in_message":[],"_revision_number":2},{"id":"26d886b4d633602a64d01bef0932f8dde01773c1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-09 14:37:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3dbf1fc5206ab260aacb0627ace9b12e96bda57","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-09 14:37:22.000000000","message":"Patch Set 2:\n\n(17 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6f06390eadc9d233ea34e2babbb7eab575ecc7cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 14:44:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/34/638734/3/check/openstack-tox-docs/7fe5b0e/ : FAILURE in 5m 41s\n- openstack-tox-pep8 http://logs.openstack.org/34/638734/3/check/openstack-tox-pep8/c3003e8/ : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":3},{"id":"556d75c0b7208aee8312357cbf385d1ac75263c0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-09 15:56:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"050b3c91d1a0c8f7cd8314b79b1597263fec1c7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 16:08:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/34/638734/4/check/openstack-tox-docs/446d9ce/html/ : SUCCESS in 10m 37s\n- openstack-tox-pep8 http://logs.openstack.org/34/638734/4/check/openstack-tox-pep8/c89e1ef/ : FAILURE in 7m 14s","accounts_in_message":[],"_revision_number":4},{"id":"b780c7cbdddf472aac0c19ba343c0e6780a8a2d3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-09 16:12:55.000000000","message":"Patch Set 4:\n\nI don\u0027t know how to make the pep8 job happy [1] since it\u0027s a link and the link can\u0027t get shorter. Would welcome input\n\n[1] http://logs.openstack.org/34/638734/4/check/openstack-tox-pep8/c89e1ef/testr_results.html.gz","accounts_in_message":[],"_revision_number":4},{"id":"cfe2a889d034e860338b859e3006ebbb4a50e778","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-04-09 22:16:51.000000000","message":"Patch Set 4:\n\n\u003e I don\u0027t know how to make the pep8 job happy [1] since it\u0027s a link\n \u003e and the link can\u0027t get shorter. Would welcome input\n \u003e \n \u003e [1] http://logs.openstack.org/34/638734/4/check/openstack-tox-pep8/c89e1ef/testr_results.html.gz\n\nI think maybe if you made the link a reference, it would work. For example:\n\nNot all extra specs can be known ahead of time and for dynamic extra specs, such\nas those proposed in `Support filtering by forbidden aggregate membership`_,\nwe will provide an API that allows us to register additional extra spec\nspecifications on startup.\n\n.. _Support filtering by forbidden aggregate membership: http://specs.openstack.org/openstack/nova-specs/specs/stein/approved/negative-aggregate-membership.html","accounts_in_message":[],"_revision_number":4},{"id":"f73e91a2fc97c759a862d7046f5e6b6a7cb4b950","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-10 09:33:36.000000000","message":"Patch Set 4: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9e120ac4a118cc04b3e1df0730f703367cbd688a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-10 12:14:01.000000000","message":"Patch Set 4: Code-Review-1\n\n(21 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1e8ba41b4453d1997238369d74607a1f51d51e25","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-10 16:06:27.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b863b7c3279b018ac7ceec96fd31120ac6fec373","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-14 16:36:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"77a47e2d8b0e8132eede223b91eae99a3ad59e73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 16:46:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/34/638734/5/check/openstack-tox-docs/1ed5865/html/ : SUCCESS in 6m 25s\n- openstack-tox-pep8 http://logs.openstack.org/34/638734/5/check/openstack-tox-pep8/759aaa5/ : FAILURE in 4m 22s","accounts_in_message":[],"_revision_number":5},{"id":"8cfd5ddac581652514c7c6c2d7415b2528b33470","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-05-14 17:52:04.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)\n\ni do like this one much more then the last iteration but i have an alternitive suggestion inline that i think would be nicer long term\nand it will still allow your yaml format to work too.","accounts_in_message":[],"_revision_number":5},{"id":"23e4aa5d68c59f75766bd64b8f2ea98001729fc7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-05-14 18:34:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8e11d01afbe545ab7779d82bb3c426400b6a9691","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-05-16 17:16:39.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f51989ac8fd7b32043a48f4acbb4895bef7241af","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-20 10:07:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e7f392fe3971508664def6cee45f42e4b8eff8ae","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-20 10:12:11.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e2f53ca1a457132c60fcba32a47696b21336576c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-20 10:26:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b07b804faae866030345d38d93922c4b576f0479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:40:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/34/638734/7/check/openstack-tox-docs/b26c57e/html/ : SUCCESS in 7m 18s\n- openstack-tox-pep8 http://logs.openstack.org/34/638734/7/check/openstack-tox-pep8/3859210/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":7},{"id":"a70699eae92e0cff0365fd7ef59b97c9eab0617c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-25 22:55:57.000000000","message":"Patch Set 7: Code-Review+2\n\n(14 comments)\n\nThis is pretty high-level, but it describes the direction well. The implementation is going to reveal lots of scary dark corners though.\n\nAt the PTG we agreed to do non-exclusive [1] validation of values only in Train. I don\u0027t see that reflected in this doc. I guess it\u0027s described as staged, but it would be nice to be more specific.\n\n[1] meaning we don\u0027t do anything with keys we don\u0027t recognize","accounts_in_message":[],"_revision_number":7},{"id":"0be15912cb386939884d8d2d3b7972c6d5498572","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-07-02 11:49:55.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0c5f9b4dc96cbb7e772662f35e59793f96ccd390","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-02 18:09:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"55c9be409ea5a15aadd29542c92b230b1470a31f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-02 18:09:15.000000000","message":"Patch Set 7:\n\n(17 comments)\n\n\u003e (14 comments)\n \u003e \n \u003e This is pretty high-level, but it describes the direction well. The\n \u003e implementation is going to reveal lots of scary dark corners\n \u003e though.\n \u003e \n \u003e At the PTG we agreed to do non-exclusive [1] validation of values\n \u003e only in Train. I don\u0027t see that reflected in this doc. I guess it\u0027s\n \u003e described as staged, but it would be nice to be more specific.\n \u003e \n \u003e [1] meaning we don\u0027t do anything with keys we don\u0027t recognize\n\nThink I\u0027ve addressed this now","accounts_in_message":[],"_revision_number":7},{"id":"d48835bb8af8ea34321f90dea95960af26b76006","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 18:20:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/34/638734/8/check/openstack-tox-docs/6b22116/html/ : SUCCESS in 5m 52s\n- openstack-tox-pep8 http://logs.openstack.org/34/638734/8/check/openstack-tox-pep8/e5cf07c/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":8},{"id":"4132eb7aecadca2185e98ad85de906dfda2c8215","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-02 23:03:03.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)\n\nThanks for the updates. Still on board.","accounts_in_message":[],"_revision_number":8},{"id":"af4bf7d67765d2c9253aaa47075e84a71ab8ea16","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-07-03 13:41:15.000000000","message":"Patch Set 8: Code-Review+2\n\nLooks good","accounts_in_message":[],"_revision_number":8},{"id":"b5315c0e9359e85c8d40e22ad5751fee623b4bfe","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-07-03 13:57:28.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)\n\nso yes im happy with the current version\nthanks stephen","accounts_in_message":[],"_revision_number":8},{"id":"68b5d35ccb2b04c8434ebf6e049265007853bb8b","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-07-03 14:42:23.000000000","message":"Patch Set 8: Code-Review+2\n\nBeen wanting to see this for years, and this seems like a sound way forward to me.","accounts_in_message":[],"_revision_number":8},{"id":"8eb2ee0567c313290b7e3c8abd097985bf1b5490","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-07-03 15:27:26.000000000","message":"Patch Set 8: Code-Review+2\n\n(3 comments)\n\nYeah, that\u0027s a good move forward.","accounts_in_message":[],"_revision_number":8},{"id":"fe5447ca78124aff8cf69b5763af760bf94057d0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-03 15:31:11.000000000","message":"Patch Set 8: Workflow+1\n\nWith +9 at this point, I\u0027m going to go ahead and land this. Alex, if you wouldn\u0027t mind another look when you get a chance, please confirm that your issues from PS2 were addressed.","accounts_in_message":[],"_revision_number":8},{"id":"b38ccc3d4cf054f832f98ac688f06a3584b29b8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:31:23.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a0a608c9bd1fda9fb3128abc7a4ae1876b6ca51d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-03 15:32:08.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"730a11fc1b725fddb272d65939b35d903ff0b4be","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-03 15:32:12.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"be832b56a587496b87f4b34a7e6ce16c7f501132","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-03 15:33:13.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"ba65f8c0369b2daf6fde99b6c228a6136a014858","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-03 15:33:18.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"fab5678801fc5cc6822ac5413eea434d0bf773a9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-03 15:34:12.000000000","message":"Topic changed from bp/flavor-extra-spec-image-property-validation to bp/flavor-extra-spec-image-property-validation-extended","accounts_in_message":[],"_revision_number":10},{"id":"631ec8d78b045dd87dc0d90b098b9bdd4d57b485","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:44:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/34/638734/10/check/openstack-tox-docs/eb7a9cc/html/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/34/638734/10/check/openstack-tox-pep8/8ad8851/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":10},{"id":"e34d2e22026f0353a92c859260d1395413808b2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:44:24.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"076d3a923dc42424152e8c98ed5fe19bf1a40a04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:55:37.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/34/638734/10/gate/openstack-tox-docs/6c6c444/html/ : SUCCESS in 6m 13s\n- openstack-tox-pep8 http://logs.openstack.org/34/638734/10/gate/openstack-tox-pep8/09a3d9d/ : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":10},{"id":"118083a32c986e074ec1daef151ea0f888c4c39d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:55:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"9e4114fb4994aade7859f13308a94ffee599030a","revisions":{"09c1279c49f7dac754c9715d5af7e2317f5eb7d5":{"kind":"REWORK","_number":1,"created":"2019-02-22 17:54:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/638734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/1"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:54:15.000000000","tz":0},"subject":"Add \u0027flavor-extra-spec-image-property-validation\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation\u0027 spec\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/09c1279c49f7dac754c9715d5af7e2317f5eb7d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/09c1279c49f7dac754c9715d5af7e2317f5eb7d5"}]},"branch":"refs/heads/master"},"d2c759b6e000e6c9adf9814534275cd390d44aee":{"kind":"REWORK","_number":2,"created":"2019-02-25 16:40:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/638734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/2"}}},"commit":{"parents":[{"commit":"d796b33774cc4e21c663885a57141e93052e5546","subject":"Merge \"Amend the detach-boot-volume design\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d796b33774cc4e21c663885a57141e93052e5546"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-25 16:38:55.000000000","tz":0},"subject":"Add \u0027flavor-extra-spec-image-property-validation\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation\u0027 spec\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2c759b6e000e6c9adf9814534275cd390d44aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d2c759b6e000e6c9adf9814534275cd390d44aee"}]},"branch":"refs/heads/master"},"83b66fa2a0c9ebe17e0d3add469f431529bf5b2e":{"kind":"REWORK","_number":3,"created":"2019-04-09 14:37:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/638734/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/3"}}},"commit":{"parents":[{"commit":"78afc9d49cc1fad705092eb9b54a80af900be875","subject":"Merge \"Re-propose volume backed server rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/78afc9d49cc1fad705092eb9b54a80af900be875"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-04-09 14:36:45.000000000","tz":60},"subject":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec\n\nIt\u0027s quite a mouthful, yes.\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/83b66fa2a0c9ebe17e0d3add469f431529bf5b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/83b66fa2a0c9ebe17e0d3add469f431529bf5b2e"}]},"branch":"refs/heads/master"},"9baa886e7a0a01a071fc39f24ef9da3a8dd83836":{"kind":"REWORK","_number":4,"created":"2019-04-09 15:56:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/638734/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/4"}}},"commit":{"parents":[{"commit":"78afc9d49cc1fad705092eb9b54a80af900be875","subject":"Merge \"Re-propose volume backed server rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/78afc9d49cc1fad705092eb9b54a80af900be875"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-04-09 15:56:31.000000000","tz":60},"subject":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec\n\nIt\u0027s quite a mouthful, yes.\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9baa886e7a0a01a071fc39f24ef9da3a8dd83836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9baa886e7a0a01a071fc39f24ef9da3a8dd83836"}]},"branch":"refs/heads/master"},"efefb374a4c5ca425e88807be01d38b138776128":{"kind":"REWORK","_number":5,"created":"2019-05-14 16:36:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/638734/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/5"}}},"commit":{"parents":[{"commit":"4c0d084993993eb51e25f3943de7f2915d424de8","subject":"Merge \"Support filtering of allocation_candidates by forbidden aggregates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4c0d084993993eb51e25f3943de7f2915d424de8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-14 16:36:35.000000000","tz":60},"subject":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec\n\nIt\u0027s quite a mouthful, yes.\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/efefb374a4c5ca425e88807be01d38b138776128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/efefb374a4c5ca425e88807be01d38b138776128"}]},"branch":"refs/heads/master"},"8f7b9ad03c51c415448a219ae35f3786887c86d1":{"kind":"REWORK","_number":6,"created":"2019-05-20 10:07:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/638734/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/6"}}},"commit":{"parents":[{"commit":"2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9","subject":"Merge \"Follow up for I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-20 10:07:34.000000000","tz":60},"subject":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec\n\nIt\u0027s quite a mouthful, yes.\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8f7b9ad03c51c415448a219ae35f3786887c86d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8f7b9ad03c51c415448a219ae35f3786887c86d1"}]},"branch":"refs/heads/master"},"e3e589d64b599c7ab3f9f4ecf304318aa0330492":{"kind":"REWORK","_number":7,"created":"2019-05-20 10:26:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/638734/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/7"}}},"commit":{"parents":[{"commit":"2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9","subject":"Merge \"Follow up for I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-20 10:16:00.000000000","tz":60},"subject":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec\n\nIt\u0027s quite a mouthful, yes.\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3e589d64b599c7ab3f9f4ecf304318aa0330492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3e589d64b599c7ab3f9f4ecf304318aa0330492"}]},"branch":"refs/heads/master"},"74cb610123f72f47424167bddbc0be84dd0debfe":{"kind":"REWORK","_number":8,"created":"2019-07-02 18:09:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/34/638734/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/8"}}},"commit":{"parents":[{"commit":"2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9","subject":"Merge \"Follow up for I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-07-02 18:08:49.000000000","tz":60},"subject":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec\n\nIt\u0027s quite a mouthful, yes.\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/74cb610123f72f47424167bddbc0be84dd0debfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/74cb610123f72f47424167bddbc0be84dd0debfe"}]},"branch":"refs/heads/master"},"0f67ae4efd4707f794eea9a9420b8f7dcd444f38":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-07-03 15:32:08.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/34/638734/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/9"}}},"commit":{"parents":[{"commit":"2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9","subject":"Merge \"Follow up for I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-03 15:32:06.000000000","tz":0},"subject":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec\n\nIt\u0027s quite a mouthful, yes.\n\nBlueprint: flavor-extra-spec-image-property-validation\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0f67ae4efd4707f794eea9a9420b8f7dcd444f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0f67ae4efd4707f794eea9a9420b8f7dcd444f38"}]},"branch":"refs/heads/master"},"9e4114fb4994aade7859f13308a94ffee599030a":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-07-03 15:33:13.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/34/638734/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/34/638734/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/34/638734/10"}}},"commit":{"parents":[{"commit":"2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9","subject":"Merge \"Follow up for I9200c2b18369da8b9194a8ccbbf6f1baf6a9613c\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b2bdfd49fa84b3c05364a09fa8dac9ec8060ce9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-02-22 17:53:35.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-03 15:33:04.000000000","tz":0},"subject":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec","message":"Add \u0027flavor-extra-spec-image-property-validation-extended\u0027 spec\n\nIt\u0027s quite a mouthful, yes.\n\nBlueprint: flavor-extra-spec-image-property-validation-extended\n\nChange-Id: I5912498e7f5f449c44884d93585362e1a04185a9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9e4114fb4994aade7859f13308a94ffee599030a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9e4114fb4994aade7859f13308a94ffee599030a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
