)]}'
{"id":"openstack%2Fnova-specs~645814","triplet_id":"openstack%2Fnova-specs~master~Ie1a90b906a9bed0295a9324956a755d577c42379","project":"openstack/nova-specs","branch":"master","topic":"bp/cpu-selection-with-hypervisor-consideration","hashtags":[],"change_id":"Ie1a90b906a9bed0295a9324956a755d577c42379","subject":"Add \"CPU selection with hypervisor consideration\" spec","status":"MERGED","created":"2019-03-22 17:33:22.000000000","updated":"2019-04-25 14:32:01.000000000","submitted":"2019-04-25 14:32:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":58,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"645814-1556202721912-a42c10a4","meta_rev_id":"14da646fbcfa85427334082dce3f03e5215b0f0a","_number":645814,"virtual_id_number":645814,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2019-04-25 14:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":1,"date":"2019-04-25 13:59:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-04-25 13:51:14.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-25 14:18:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-25 14:18:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-05 22:50:50.000000000","updated_by":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"reviewer":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"state":"REVIEWER"},{"updated":"2019-04-09 13:56:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-09 14:48:39.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-04-24 05:04:43.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-04-25 13:51:14.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-25 13:59:13.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-04-25 14:18:44.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-04-25 14:32:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21dfde3bedb9a1c581f31fa1ebf11a3063debb07","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-03-22 17:33:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77cd1cc1f72a43625fa5f278237be4bc68754463","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 17:44:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/645814/1/check/openstack-tox-docs/ce5f559/html/ : SUCCESS in 7m 35s\n- openstack-tox-pep8 http://logs.openstack.org/14/645814/1/check/openstack-tox-pep8/775190b/ : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7b3657c004a00b2bee4c5c98cbc6b169a2da77b8","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-03-22 17:48:01.000000000","message":"Patch Set 1:\n\nI\u0027m away the whole of next week; will be back on 01-April (no, not a joke ;-)) and start chugging away at the feedback, if there is any in the meantime.","accounts_in_message":[],"_revision_number":1},{"id":"d0163dc0ba32a405ce24447b335a3f6754a83599","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2019-03-22 19:50:10.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nLooks like a decent start.","accounts_in_message":[],"_revision_number":1},{"id":"c9515570b253c177291d8a63f5fe071d2134dff3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-03-28 11:32:56.000000000","message":"Patch Set 1:\n\nnice","accounts_in_message":[],"_revision_number":1},{"id":"5895270ec6e7d6110208478dbbafaa9170f37ff1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-01 10:32:45.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nJust some style nits. This is clearly still WIP but the idea makes sense to me and Chris has called out most of what I\u0027d have suggested","accounts_in_message":[],"_revision_number":1},{"id":"d95a8ab1a9176878f2c9df9366a287033a420a43","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-04 08:58:23.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ef316ce4d4bfc98eead5087daa4d0f1ee9144a9e","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-05 16:14:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4594f6d3feefd1df3784f7bb60b06137a38aeb63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 16:23:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/645814/2/check/openstack-tox-docs/0e76299/html/ : SUCCESS in 8m 02s\n- openstack-tox-pep8 http://logs.openstack.org/14/645814/2/check/openstack-tox-pep8/78ac784/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"d90f61a5cea9497cca58508623ef165ae4f9523f","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2019-04-05 22:50:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"54825b9f00828a6084a053356dc7b5b6f00a6ddd","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-08 09:45:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eee07d79f0e496389e2e383d6ea94b42a1bb6292","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 10:02:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/645814/3/check/openstack-tox-docs/edb2489/html/ : SUCCESS in 8m 16s\n- openstack-tox-pep8 http://logs.openstack.org/14/645814/3/check/openstack-tox-pep8/4e452ae/ : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":3},{"id":"b083e2a967c7c0a17ffa952908bb3077b21bbba2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-08 10:02:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd0c24d788327fd8e03ca1a407c0512e83460833","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-08 10:41:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"94b91f8aedf6e5d8f99eccfcea8e0aa736093830","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 11:25:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"20ce2a5eeae9194ef68ddfe02dd0ff5c9bdefac4","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-09 07:05:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d1577292c1cf6dbc398be0a4094e477a012b30ac","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-09 07:29:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8218abbc8fb9a500262bea522b3554d5a1bf8a52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 07:40:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/645814/4/check/openstack-tox-docs/3d8ffd2/html/ : SUCCESS in 8m 00s\n- openstack-tox-pep8 http://logs.openstack.org/14/645814/4/check/openstack-tox-pep8/57b4438/ : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":4},{"id":"58a9cf4232723d636ae5b7ff29f42507a11cfefb","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-09 12:41:12.000000000","message":"Patch Set 4: Code-Review+2\n\nI have nothing against this. It seems pretty self contained.","accounts_in_message":[],"_revision_number":4},{"id":"66e8d3e21e46c8699eadd3a99ca44fb648f2d368","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-09 13:56:09.000000000","message":"Patch Set 4: Code-Review+2\n\n(11 comments)\n\nThere\u0027s a couple of nits in there that should be cleaned up but I, like gibi, think this is sane and self-contained enough to warrant inclusion","accounts_in_message":[],"_revision_number":4},{"id":"fc9eef7be62a1b5f83763e997e687d04d0fc3d5d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-09 14:48:39.000000000","message":"Patch Set 4:\n\n(13 comments)\n\nWell, this certainly says we can do a thing, but it\u0027s not entirely clear to me what specific issues in nova this solves if we do it.","accounts_in_message":[],"_revision_number":4},{"id":"b132dc7838546074796b8c12793e25a33637cdfb","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-09 14:52:10.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3b25b348df12149ad794b45e65575b54ae4e63c4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-09 15:51:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\ni think there are enough nits that this needs a respin but\ni like the general direction.\ntwo its inline.","accounts_in_message":[],"_revision_number":4},{"id":"c172534af92b517b25e32fd620132897be31bed1","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-09 16:38:29.000000000","message":"Patch Set 4:\n\n[I answered your question on IRC, but writing it here for persistence.]\n\n \u003e (13 comments)\n \u003e \n \u003e Well, this certainly says we can do a thing, but it\u0027s not entirely\n \u003e clear to me what specific issues in nova this solves if we do it.\n\nThe following is one of the issues that will be solved.  I\u0027m going to add a variant of the below blurb in the spec.  I stole a bit of your phrasing :-):\n\n\nToday operators have to be careful about how they configure the libvirt driver with regard to `cpu_model` and `cpu_model_extra_flags`, because the wrong combination can lead to failures to launch a guest.  I.e. operators have to manually validate the CPU flags they\u0027re supplying to Nova are actually supported on the given compute node.  With this change, the libvirt driver will automatically validate if a given combination of CPU model + flags can work on a given compute host—e.g. Nova will now be able to answer: \"Is this combination of IvyBridge + \u0027pcid\u0027 \u0026 \u0027pdpe1gb\u0027 supported by KVM, QEMU and libvirt on the host?\").  So the operator can now set the CPU model plus flags, and let Nova handle the rest.","accounts_in_message":[],"_revision_number":4},{"id":"3dc53f5e1e67d3454e2855c1b1470e7a8e98c767","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-22 16:47:39.000000000","message":"Patch Set 4: Code-Review-1\n\n\u003e lead to failures to launch a guest.\n\nMeaning Nova will blithely put together the XML etc., but then the actual instance won\u0027t... spawn? Boot? Work?\n\n\u003e With this change, the libvirt driver will automatically validate if a given combination of CPU model + flags can work on a given compute host\n\n...and if they can\u0027t, nova will fail the spawn (earlier)?\n\nSo the point of this is to make instance creation fail earlier - and presumably log a better reason - if there\u0027s a bad combination of host + CPU models + flags?\n\nI\u0027ll echo Matt\u0027s comments (I think) by saying that this is very much not clear in the spec. (I suspect you\u0027re so entrenched in it that it\u0027s obvious to you.)","accounts_in_message":[],"_revision_number":4},{"id":"917896d1a5731e9af6b90277a547cf452e36c58c","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-24 05:04:19.000000000","message":"Patch Set 4:\n\nActually, you add the validation of cpu_model+extra spec at another spec https://review.opendev.org/#/c/642030/6/specs/train/approved/cpu-model-selection.rst@69 :)\n\nI guess that why people confuse at here.\n\nFor this spec, actually, it is very easy, we just use the new libvirt API instead of the old API, the new API can catch more invalid cpu_model+extra flags cases than the old one :)","accounts_in_message":[],"_revision_number":4},{"id":"5d27a2a90625600840f5c1ae2c98070124168b04","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-24 05:04:43.000000000","message":"Patch Set 4: Code-Review+2\n\nAnd actually ok with this spec :)","accounts_in_message":[],"_revision_number":4},{"id":"daf53d61e38ace55693dcd317473147ec6913009","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-24 14:42:17.000000000","message":"Patch Set 4:\n\n\u003e \u003e lead to failures to launch a guest.\n \u003e \n \u003e Meaning Nova will blithely put together the XML etc., but then the\n \u003e actual instance won\u0027t... spawn? Boot? Work?\n \nExactly, Nova will put together the XML, but it will fail to spawn.\n\n \u003e \u003e With this change, the libvirt driver will automatically validate\n \u003e if a given combination of CPU model + flags can work on a given\n \u003e compute host\n \u003e \n \u003e ...and if they can\u0027t, nova will fail the spawn (earlier)?\n\nYes, my current thinking is that we should fail to spawn.\n\n \u003e So the point of this is to make instance creation fail earlier -\n \u003e and presumably log a better reason - if there\u0027s a bad combination\n \u003e of host + CPU models + flags?\n\nYeah, that is one of the points.  A related point is that: when I say we will get \"better-informed guest CPU\", it means, the newer API, baselineHypervisorCPU(), when calculating baseline CPU it will *filter out* CPU features that are not supported by QEMU on the host. \n\n \u003e I\u0027ll echo Matt\u0027s comments (I think) by saying that this is very\n \u003e much not clear in the spec. (I suspect you\u0027re so entrenched in it\n \u003e that it\u0027s obvious to you.)\n\nI will add some clarifying text to this shortly.  Thanks for the reviews, everyone.","accounts_in_message":[],"_revision_number":4},{"id":"489d17ba7432e671ad20acd9c7a8303705f575b6","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-25 11:22:44.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1c675470388f2c0be658cc1b58bfdae3994ccc29","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-25 11:36:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"143cebbebb8aa4e28b2a8efe3ac60abf6aeaf090","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:52:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/645814/5/check/openstack-tox-docs/737feca/html/ : SUCCESS in 9m 34s\n- openstack-tox-pep8 http://logs.openstack.org/14/645814/5/check/openstack-tox-pep8/5872121/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":5},{"id":"e2094b372fda1b84fcfb4d53da77f102fe56f095","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-25 13:28:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5d0ab7f0c2865fb155ce70687d5669b10bb9d7a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 13:40:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/645814/6/check/openstack-tox-docs/01c6f85/html/ : SUCCESS in 7m 18s\n- openstack-tox-pep8 http://logs.openstack.org/14/645814/6/check/openstack-tox-pep8/980c858/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":6},{"id":"fbf1a53989032c687e6ce136f3a8e54da850d248","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-25 13:47:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8db610b517b4dde8b4ebf985594bded68497884b","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-25 13:49:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"808fcb6860944c7ef00d74ea6c3dabb1faab66bf","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 13:51:14.000000000","message":"Patch Set 8: Code-Review+2\n\nThanks for the updates, kashyap. This lgtm.\n\nYou had several +2s before, and the recent changes were just adding detail, so I\u0027ll fast approve this in the next couple of hours.","accounts_in_message":[],"_revision_number":8},{"id":"474f73baea1ce1e43627ef6c1621f8206cc3d38f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-25 13:59:13.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"56b14c11db8bb7e2fa29b0165ac84ddbb554a47e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:11:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/645814/8/check/openstack-tox-docs/9e94b9b/html/ : SUCCESS in 7m 07s\n- openstack-tox-pep8 http://logs.openstack.org/14/645814/8/check/openstack-tox-pep8/ba8dd52/ : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":8},{"id":"1456d5ba3cc9be87fef6379f5bad970fa62299d7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-25 14:18:44.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"318ff52904d47a8dccc86f992b8db878e437f3cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:19:03.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0153a3d28ad6405af52b17beecf4d1316c6c15f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:32:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"50831c8ac75bd2ec12c78eda9cdf70af7c20fa6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:32:01.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/645814/8/gate/openstack-tox-docs/be777b3/html/ : SUCCESS in 8m 34s\n- openstack-tox-pep8 http://logs.openstack.org/14/645814/8/gate/openstack-tox-pep8/8f768a9/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"70811da221035044e27723933d0239277681f38c","revisions":{"9a550f4b81867e03f648dead57c25b5c7fdcde28":{"kind":"REWORK","_number":1,"created":"2019-03-22 17:33:22.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/14/645814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/645814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/1"}}},"commit":{"parents":[{"commit":"a20fed8c3b514db69a7cd56eb9897ea51ec128d3","subject":"Re-propose count quota usage from placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a20fed8c3b514db69a7cd56eb9897ea51ec128d3"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-22 14:08:02.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-22 17:32:41.000000000","tz":60},"subject":"Add \"CPU selection with hypervisor consideration\" spec","message":"Add \"CPU selection with hypervisor consideration\" spec\n\nMake Nova\u0027s guest CPU selection approach more effective and reliable by\nintroducing two new QEMU- and libvirt-based CPU configuration APIs:\nbaselineHypervisorCPU() and compareHypervisorCPU().  These new APIs are\nmore \"hypervisor-literate\" in comparison to the existing libvirt APIs\nthat Nova uses.  As in, the new APIs take into account what the \"host\nhypervisor\" (meaning: KVM, QEMU, and what libvirt knows about the host)\nis capable of.\n\nTaking advantage of these newer APIs will allow Nova to make more\nwell-informed decisions when determining CPU models that are compatible\nacross different hosts.\n\nBlueprint: cpu-selection-with-hypervisor-consideration\n\nChange-Id: Ie1a90b906a9bed0295a9324956a755d577c42379\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9a550f4b81867e03f648dead57c25b5c7fdcde28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9a550f4b81867e03f648dead57c25b5c7fdcde28"}]},"branch":"refs/heads/master"},"670cb075f49060f706c0485394b2d34652e11871":{"kind":"REWORK","_number":2,"created":"2019-04-05 16:14:04.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/14/645814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/645814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/2"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-22 14:08:02.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-05 16:13:55.000000000","tz":120},"subject":"Add \"CPU selection with hypervisor consideration\" spec","message":"Add \"CPU selection with hypervisor consideration\" spec\n\nMake Nova\u0027s guest CPU selection approach more effective and reliable by\nintroducing two new QEMU- and libvirt-based CPU configuration APIs:\nbaselineHypervisorCPU() and compareHypervisorCPU().  These new APIs are\nmore \"hypervisor-literate\" in comparison to the existing libvirt APIs\nthat Nova uses.  As in, the new APIs take into account what the \"host\nhypervisor\" (meaning: KVM, QEMU, and what libvirt knows about the host)\nis capable of.\n\nTaking advantage of these newer APIs will allow Nova to make more\nwell-informed decisions when determining CPU models that are compatible\nacross different hosts.\n\nBlueprint: cpu-selection-with-hypervisor-consideration\n\nChange-Id: Ie1a90b906a9bed0295a9324956a755d577c42379\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/670cb075f49060f706c0485394b2d34652e11871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/670cb075f49060f706c0485394b2d34652e11871"}]},"branch":"refs/heads/master"},"fe22b2bf93754d7e80b14d1c6df0a1bb1559b104":{"kind":"REWORK","_number":3,"created":"2019-04-08 09:45:32.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/14/645814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/645814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/3"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-22 14:08:02.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-08 09:45:10.000000000","tz":120},"subject":"Add \"CPU selection with hypervisor consideration\" spec","message":"Add \"CPU selection with hypervisor consideration\" spec\n\nMake Nova\u0027s guest CPU selection approach more effective and reliable by\nintroducing two new QEMU- and libvirt-based CPU configuration APIs:\nbaselineHypervisorCPU() and compareHypervisorCPU().  These new APIs are\nmore \"hypervisor-literate\" compared to the existing libvirt APIs that\nNova uses.  As in, the new APIs take into account what the \"host\nhypervisor\" (meaning: KVM, QEMU, and what libvirt knows about the host)\nis capable of.\n\nTaking advantage of these newer APIs will allow Nova to make more\nwell-informed decisions when determining CPU models that are compatible\nacross different hosts.\n\nBlueprint: cpu-selection-with-hypervisor-consideration\n\nChange-Id: Ie1a90b906a9bed0295a9324956a755d577c42379\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fe22b2bf93754d7e80b14d1c6df0a1bb1559b104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fe22b2bf93754d7e80b14d1c6df0a1bb1559b104"}]},"branch":"refs/heads/master"},"fad9c8c35d3f382428c10d6b9912ea3ab8d0caa0":{"kind":"REWORK","_number":4,"created":"2019-04-09 07:29:40.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/14/645814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/645814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/4"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-22 14:08:02.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-09 07:06:14.000000000","tz":120},"subject":"Add \"CPU selection with hypervisor consideration\" spec","message":"Add \"CPU selection with hypervisor consideration\" spec\n\nMake Nova\u0027s guest CPU selection approach more effective and reliable by\nintroducing two new QEMU- and libvirt-based CPU configuration APIs:\nbaselineHypervisorCPU() and compareHypervisorCPU().  These new APIs are\nmore \"hypervisor-literate\" compared to the existing libvirt APIs that\nNova uses.  As in, the new APIs take into account what the \"host\nhypervisor\" (meaning: KVM, QEMU, and what libvirt knows about the host)\nis capable of.\n\nTaking advantage of these newer APIs will allow Nova to make more\nwell-informed decisions when determining CPU models that are compatible\nacross different hosts.\n\nBlueprint: cpu-selection-with-hypervisor-consideration\n\nChange-Id: Ie1a90b906a9bed0295a9324956a755d577c42379\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fad9c8c35d3f382428c10d6b9912ea3ab8d0caa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fad9c8c35d3f382428c10d6b9912ea3ab8d0caa0"}]},"branch":"refs/heads/master"},"0b0b7eeb27394f89f3c77ab37aee03185d5c233b":{"kind":"REWORK","_number":5,"created":"2019-04-25 11:36:37.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/14/645814/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/645814/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/5"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-22 14:08:02.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-25 11:27:24.000000000","tz":120},"subject":"Add \"CPU selection with hypervisor consideration\" spec","message":"Add \"CPU selection with hypervisor consideration\" spec\n\nMake Nova\u0027s guest CPU selection approach more effective and reliable by\nintroducing two new QEMU- and libvirt-based CPU configuration APIs:\nbaselineHypervisorCPU() and compareHypervisorCPU().  These new APIs are\nmore \"hypervisor-literate\" compared to the existing libvirt APIs that\nNova uses.  As in, the new APIs take into account what the \"host\nhypervisor\" (meaning: KVM, QEMU, and what libvirt knows about the host)\nis capable of.\n\nTaking advantage of these newer APIs will allow Nova to make more\nwell-informed decisions when determining CPU models that are compatible\nacross different hosts.\n\nBlueprint: cpu-selection-with-hypervisor-consideration\n\nChange-Id: Ie1a90b906a9bed0295a9324956a755d577c42379\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0b0b7eeb27394f89f3c77ab37aee03185d5c233b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0b0b7eeb27394f89f3c77ab37aee03185d5c233b"}]},"branch":"refs/heads/master"},"da852b690175d6fe1182d79c99680bbafe2d5c10":{"kind":"REWORK","_number":6,"created":"2019-04-25 13:28:05.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/14/645814/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/645814/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/6"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-22 14:08:02.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-25 13:27:51.000000000","tz":120},"subject":"Add \"CPU selection with hypervisor consideration\" spec","message":"Add \"CPU selection with hypervisor consideration\" spec\n\nMake Nova\u0027s guest CPU selection approach more effective and reliable by\nintroducing two new QEMU- and libvirt-based CPU configuration APIs:\nbaselineHypervisorCPU() and compareHypervisorCPU().  These new APIs are\nmore \"hypervisor-literate\" compared to the existing libvirt APIs that\nNova uses.  As in, the new APIs take into account what the \"host\nhypervisor\" (meaning: KVM, QEMU, and what libvirt knows about the host)\nis capable of.\n\nTaking advantage of these newer APIs will allow Nova to make more\nwell-informed decisions when determining CPU models that are compatible\nacross different hosts.\n\nBlueprint: cpu-selection-with-hypervisor-consideration\n\nChange-Id: Ie1a90b906a9bed0295a9324956a755d577c42379\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da852b690175d6fe1182d79c99680bbafe2d5c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/da852b690175d6fe1182d79c99680bbafe2d5c10"}]},"branch":"refs/heads/master"},"d4f8afd74527b06a725d3f0ab124ad04b5db439a":{"kind":"REWORK","_number":7,"created":"2019-04-25 13:47:33.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/14/645814/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/645814/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/7"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-22 14:08:02.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-25 13:40:21.000000000","tz":120},"subject":"Add \"CPU selection with hypervisor consideration\" spec","message":"Add \"CPU selection with hypervisor consideration\" spec\n\nMake Nova\u0027s guest CPU selection approach more effective and reliable by\nintroducing two new QEMU- and libvirt-based CPU configuration APIs:\nbaselineHypervisorCPU() and compareHypervisorCPU().  These new APIs are\nmore \"hypervisor-literate\" compared to the existing libvirt APIs that\nNova uses.  As in, the new APIs take into account what the \"host\nhypervisor\" (meaning: KVM, QEMU, and what libvirt knows about the host)\nis capable of.\n\nTaking advantage of these newer APIs will allow Nova to make more\nwell-informed decisions when determining CPU models that are compatible\nacross different hosts.\n\nBlueprint: cpu-selection-with-hypervisor-consideration\n\nChange-Id: Ie1a90b906a9bed0295a9324956a755d577c42379\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d4f8afd74527b06a725d3f0ab124ad04b5db439a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d4f8afd74527b06a725d3f0ab124ad04b5db439a"}]},"branch":"refs/heads/master"},"70811da221035044e27723933d0239277681f38c":{"kind":"REWORK","_number":8,"created":"2019-04-25 13:49:30.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/14/645814/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/14/645814/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/14/645814/8"}}},"commit":{"parents":[{"commit":"a2a3d7203489b85cf519259d9a4eca5ed439aa71","subject":"Re-propose emulated virtual TPM spec to train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2a3d7203489b85cf519259d9a4eca5ed439aa71"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-22 14:08:02.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-25 13:49:13.000000000","tz":120},"subject":"Add \"CPU selection with hypervisor consideration\" spec","message":"Add \"CPU selection with hypervisor consideration\" spec\n\nMake Nova\u0027s guest CPU selection approach more effective and reliable by\nintroducing two new QEMU- and libvirt-based CPU configuration APIs:\nbaselineHypervisorCPU() and compareHypervisorCPU().  These new APIs are\nmore \"hypervisor-literate\" compared to the existing libvirt APIs that\nNova uses.  As in, the new APIs take into account what the \"host\nhypervisor\" (meaning: KVM, QEMU, and what libvirt knows about the host)\nis capable of.\n\nTaking advantage of these newer APIs will allow Nova to make more\nwell-informed decisions when determining CPU models that are compatible\nacross different hosts.\n\nBlueprint: cpu-selection-with-hypervisor-consideration\n\nChange-Id: Ie1a90b906a9bed0295a9324956a755d577c42379\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/70811da221035044e27723933d0239277681f38c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/70811da221035044e27723933d0239277681f38c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
