)]}'
{"id":"openstack%2Fnova-specs~647581","triplet_id":"openstack%2Fnova-specs~master~I5b876fcdc7cf616594ef46d703d2fd6c02765797","project":"openstack/nova-specs","branch":"master","topic":"bp/numa-aware-live-migration","hashtags":[],"change_id":"I5b876fcdc7cf616594ef46d703d2fd6c02765797","subject":"re add numa aware live migration for train","status":"MERGED","created":"2019-03-25 19:10:33.000000000","updated":"2019-03-26 08:30:53.000000000","submitted":"2019-03-26 08:30:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"647581-1553589053817-9906e730","meta_rev_id":"e87118e20c3f383a282de01f61795fa74750efe4","_number":647581,"virtual_id_number":647581,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-03-26 08:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-03-26 08:10:24.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-03-26 08:10:24.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-26 08:10:24.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-03-26 08:30:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0be4476bd0a3ab2e43ec73c718be7183ed0146e0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-03-25 19:10:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c02d7df4445dd071ee58f6e20346d4e625434b0b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-25 19:27:08.000000000","message":"Patch Set 1: Code-Review-1\n\nSuitable for fast approval per [1].\n\n @@ -453,4 +453,6 @@ History\n     * - Stein\n       - Re-proposed with modifications pertaining to claims and the exchange of\n         information between destination and source.\n +   * - Train\n +     - Re-proposed with no modifications.\n\nAs discussed, please update the commit title to something relevant to the bp, and add the Previously-approved tag to the commit message.\n\n[1] https://specs.openstack.org/openstack/nova-specs/readme.html#previously-approved-specifications","accounts_in_message":[],"_revision_number":1},{"id":"76b89663ddc1d1296b0e2e63c337fb07c0476006","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 19:38:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/647581/1/check/openstack-tox-docs/de2afaa/html/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/81/647581/1/check/openstack-tox-pep8/e4efecc/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4a8f9ecf06afd143305da9c250343e646fc4afe0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-03-26 05:41:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a5beaf655a63aaf7ea20c497a9a681f6e40b499f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 05:48:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/647581/2/check/openstack-tox-docs/df87cc2/html/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/81/647581/2/check/openstack-tox-pep8/dedf982/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"fe06492bbe8fc3926bba53636916ab485e4f634a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-26 08:10:03.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"740efaeb52b3f77efd5165e464dd637dec9372cd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-26 08:10:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"827e57d4b5b2fe653bf16c45ea23afed4c4283d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 08:20:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/647581/3/check/openstack-tox-docs/48a3e75/html/ : SUCCESS in 7m 57s\n- openstack-tox-pep8 http://logs.openstack.org/81/647581/3/check/openstack-tox-pep8/3d18849/ : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":3},{"id":"9aba43e0bbcf9dbe9f6f4c0d654eaa9c20e933d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 08:20:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a902ca7d520b93e683612428157f9f7a4dabd8c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 08:30:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bb3a012bbb0308940ab299031a6331d220ae2dac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 08:30:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/647581/3/gate/openstack-tox-docs/79a20de/html/ : SUCCESS in 5m 17s\n- openstack-tox-pep8 http://logs.openstack.org/81/647581/3/gate/openstack-tox-pep8/cd05b1d/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"797b9a0a77a68347757d88ea73713529a1492639","revisions":{"5bb2ab89a1c59d78dd3515712aa55dd8a0cd5441":{"kind":"REWORK","_number":1,"created":"2019-03-25 19:10:33.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/81/647581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/647581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/1"}}},"commit":{"parents":[{"commit":"ee6f94fecce729b5b5355d20f9d446035228dc53","subject":"Merge \"Fix warnings in the document generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee6f94fecce729b5b5355d20f9d446035228dc53"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-03-25 18:56:49.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-03-25 18:56:49.000000000","tz":0},"subject":"re add for train","message":"re add for train\n\nthis spec miss ff in stien.\nthis simply forward ports the spec to train\nand updates the spec history footer to reflect this.\n\nChange-Id: I5b876fcdc7cf616594ef46d703d2fd6c02765797\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5bb2ab89a1c59d78dd3515712aa55dd8a0cd5441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5bb2ab89a1c59d78dd3515712aa55dd8a0cd5441"}]},"branch":"refs/heads/master"},"b979e250f04d959785536493ea368715caa05ce1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-03-26 05:41:36.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/81/647581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/647581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/2"}}},"commit":{"parents":[{"commit":"ee6f94fecce729b5b5355d20f9d446035228dc53","subject":"Merge \"Fix warnings in the document generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee6f94fecce729b5b5355d20f9d446035228dc53"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-03-25 18:56:49.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-03-26 05:41:33.000000000","tz":0},"subject":"re add numa aware live migration for train","message":"re add numa aware live migration for train\n\nThis change simply forward ports the spec to train\nand updates the spec history footer to reflect this.\n\nPreviously-approved: Stein\n\nChange-Id: I5b876fcdc7cf616594ef46d703d2fd6c02765797\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b979e250f04d959785536493ea368715caa05ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b979e250f04d959785536493ea368715caa05ce1"}]},"branch":"refs/heads/master"},"797b9a0a77a68347757d88ea73713529a1492639":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-03-26 08:10:03.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/81/647581/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/647581/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/647581/3"}}},"commit":{"parents":[{"commit":"ee6f94fecce729b5b5355d20f9d446035228dc53","subject":"Merge \"Fix warnings in the document generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ee6f94fecce729b5b5355d20f9d446035228dc53"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-03-25 18:56:49.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-03-26 08:09:50.000000000","tz":0},"subject":"re add numa aware live migration for train","message":"re add numa aware live migration for train\n\nThis change simply forward ports the spec to train\nand updates the spec history footer to reflect this.\n\nBlueprint: numa-aware-live-migration\nPreviously-approved: Stein\n\nChange-Id: I5b876fcdc7cf616594ef46d703d2fd6c02765797\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/797b9a0a77a68347757d88ea73713529a1492639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/797b9a0a77a68347757d88ea73713529a1492639"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
