)]}'
{"id":"openstack%2Fnova-specs~649095","triplet_id":"openstack%2Fnova-specs~master~I1e5267a828133e9a6579a6790f2996c64a88fc8d","project":"openstack/nova-specs","branch":"master","topic":"scrub-backlog-specs","hashtags":[],"change_id":"I1e5267a828133e9a6579a6790f2996c64a88fc8d","subject":"Abandon flavor-classes backlog spec","status":"MERGED","created":"2019-04-01 16:18:48.000000000","updated":"2019-05-02 15:20:59.000000000","submitted":"2019-05-02 15:20:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649095-1556810459240-54e3a629","meta_rev_id":"fa06bb9232cbd455d765fd866dbe85c0e43c989e","_number":649095,"virtual_id_number":649095,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-05-01 23:01:28.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":10263,"name":"Vilobh Meshram","email":"vilobhmm@yahoo-inc.com","username":"vilobhmm"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":2,"date":"2019-05-02 15:20:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":10263,"name":"Vilobh Meshram","email":"vilobhmm@yahoo-inc.com","username":"vilobhmm"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-02 15:12:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":10263,"name":"Vilobh Meshram","email":"vilobhmm@yahoo-inc.com","username":"vilobhmm"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-02 15:12:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":10263,"name":"Vilobh Meshram","email":"vilobhmm@yahoo-inc.com","username":"vilobhmm"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":10263,"name":"Vilobh Meshram","email":"vilobhmm@yahoo-inc.com","username":"vilobhmm"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-01 16:42:18.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2019-04-01 16:42:30.000000000","updated_by":{"_account_id":10263,"name":"Vilobh Meshram","email":"vilobhmm@yahoo-inc.com","username":"vilobhmm"},"reviewer":{"_account_id":10263,"name":"Vilobh Meshram","email":"vilobhmm@yahoo-inc.com","username":"vilobhmm"},"state":"REVIEWER"},{"updated":"2019-04-01 16:42:47.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2019-04-09 11:45:51.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2019-04-09 13:56:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-19 05:30:56.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-05-02 15:12:05.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-05-02 15:20:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e465c4444aa864094a7c0b7971b2518a1ce01d4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-01 16:18:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a495779736c11357b7ca478656ec479d302bef1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 16:29:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/649095/1/check/openstack-tox-docs/5a2dda0/html/ : SUCCESS in 7m 40s\n- openstack-tox-pep8 http://logs.openstack.org/95/649095/1/check/openstack-tox-pep8/a15227c/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ad87d32081f041740eeb2b77411812af1b64c588","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-01 16:36:38.000000000","message":"Topic changed from abandon_flavor-class to scrub-backlog-specs","accounts_in_message":[],"_revision_number":1},{"id":"235a29dab229e2566f7501c78a3ca23f23626db7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 23:14:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d69c1c1d836dbb5aaca64e5dd3ab2d3dfdb71ef3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 23:19:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b9ea22b28877eb7fdb4abc42ab10217dbe20237","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 00:07:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/649095/3/check/openstack-tox-docs/a526d78/html/ : SUCCESS in 9m 10s\n- openstack-tox-pep8 http://logs.openstack.org/95/649095/3/check/openstack-tox-pep8/c18404a/ : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"351a305501f7d6489dd9b58d8a8ded4ef54bf151","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-09 11:45:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e510ec9c0cdbbc603bbf58e9cf04609016ee6a65","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-09 13:56:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a205a154601e9207885ed0a29a49327fa823c59d","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-04-19 05:30:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"15b3c69d5f8b5c288d30125acce5bbd9de6e0c96","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-26 16:55:34.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c476abccf658644e21112377c15136e5e614af24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 17:23:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/649095/4/check/openstack-tox-docs/9c12a9e/html/ : SUCCESS in 7m 26s\n- openstack-tox-pep8 http://logs.openstack.org/95/649095/4/check/openstack-tox-pep8/d6ed71f/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":4},{"id":"8601913fb372a2ac5a502bfc45f3e48eb88a9ffe","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-28 15:55:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9197948212606c0b2cd32c6fcdfc87b61641774c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 16:07:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/649095/5/check/openstack-tox-docs/8a95329/html/ : SUCCESS in 6m 56s\n- openstack-tox-pep8 http://logs.openstack.org/95/649095/5/check/openstack-tox-pep8/a69e9e7/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"06da45f2cea4df5896b3395bb7644c42737d75df","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-05-01 21:52:39.000000000","message":"Patch Set 5: Code-Review+2\n\nCool, as we keep it.","accounts_in_message":[],"_revision_number":5},{"id":"dc65a3e9d034f9280c58280dd1a0def81b8f639a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-01 22:26:50.000000000","message":"Patch Set 5: Workflow+1\n\nCarrying forward prior approvals","accounts_in_message":[],"_revision_number":5},{"id":"b66769aa9f54e173465535be234be2bdbc579347","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 22:39:17.000000000","message":"Patch Set 5: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"633a033873b451829c3944afba29fe4f67f06eb3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-01 22:44:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"756ee8a1497eb12be43d3c1f112df15c80a52265","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-01 22:45:07.000000000","message":"Patch Set 6:\n\nTrivial manual rebase","accounts_in_message":[],"_revision_number":6},{"id":"f5a01ed8f1241b1a37bd39e0512830744281f9a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 22:50:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/649095/6/check/openstack-tox-docs/dc8f7b5/ : FAILURE in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/95/649095/6/check/openstack-tox-pep8/5dc4b7e/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":6},{"id":"759e31e262b59add13e7aac5962a5e745c7335a8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-01 23:00:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f03a10e1d0a0e1b7659a93cc1a3ce52975037bb8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-01 23:01:28.000000000","message":"Patch Set 7:\n\nHad to take approved/* out of the backlog/index because there\u0027s nothing there anymore.","accounts_in_message":[],"_revision_number":7},{"id":"2784718b56397e6b82718931cbdd39d36dcad06b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 23:18:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/649095/7/check/openstack-tox-docs/5d38285/html/ : SUCCESS in 10m 17s\n- openstack-tox-pep8 http://logs.openstack.org/95/649095/7/check/openstack-tox-pep8/c8fc68e/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":7},{"id":"d15ade3462aaad2edc08e2e7402f69e2f999dd62","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-05-02 15:12:00.000000000","message":"Patch Set 7: Code-Review+2\n\nCool. \nJust fast-approving, we already got a long list of +2s.","accounts_in_message":[],"_revision_number":7},{"id":"60fae997a2e0f0c08fe388fed188a467a39e730a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-05-02 15:12:05.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"77b0b187ab53193007b4077655527a14ff363e25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 15:12:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3ad07e88c0682e10f862b70d0e5a9c615d1096fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 15:20:58.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/649095/7/gate/openstack-tox-docs/61eaeb0/html/ : SUCCESS in 7m 03s\n- openstack-tox-pep8 http://logs.openstack.org/95/649095/7/gate/openstack-tox-pep8/196c3c5/ : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":7},{"id":"fa06bb9232cbd455d765fd866dbe85c0e43c989e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 15:20:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"14f84bf8632728ae63d0bc7d24e37bb2cf4c6272","revisions":{"f03846d74f143992dd493bf2f2026fddb2b4fc67":{"kind":"REWORK","_number":1,"created":"2019-04-01 16:18:48.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/95/649095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/95/649095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/1"}}},"commit":{"parents":[{"commit":"e4934df640fce8ff604c5d78bc772a735ece4ec4","subject":"Tools \u0026 docs for backlog \u0026 abandoned spec process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e4934df640fce8ff604c5d78bc772a735ece4ec4"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:17:06.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:17:06.000000000","tz":-300},"subject":"Abandon flavor-classes backlog spec","message":"Abandon flavor-classes backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nIt is probably related to [1], but I haven\u0027t dug into that yet.  The\nauthors of either spec should feel free to modify this commit to\nclarify.\n\n[1] https://review.openstack.org/#/c/635408/\n\nChange-Id: I1e5267a828133e9a6579a6790f2996c64a88fc8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f03846d74f143992dd493bf2f2026fddb2b4fc67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/f03846d74f143992dd493bf2f2026fddb2b4fc67"}]},"branch":"refs/heads/master"},"06cb5abfd5b2fd9f383d80b419d7ce8389a2cf29":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-08 23:14:28.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/95/649095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/95/649095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/2"}}},"commit":{"parents":[{"commit":"5f2f1818688b9b29c51e4d183ac715c79e37525e","subject":"Tools \u0026 docs for backlog \u0026 abandoned spec process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f2f1818688b9b29c51e4d183ac715c79e37525e"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:17:06.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-08 23:14:28.000000000","tz":0},"subject":"Abandon flavor-classes backlog spec","message":"Abandon flavor-classes backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nIt is probably related to [1], but I haven\u0027t dug into that yet.  The\nauthors of either spec should feel free to modify this commit to\nclarify.\n\n[1] https://review.openstack.org/#/c/635408/\n\nChange-Id: I1e5267a828133e9a6579a6790f2996c64a88fc8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/06cb5abfd5b2fd9f383d80b419d7ce8389a2cf29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/06cb5abfd5b2fd9f383d80b419d7ce8389a2cf29"}]},"branch":"refs/heads/master"},"ff1d91028b90fba0ac4f50ccdf1fd6b0eb06d234":{"kind":"REWORK","_number":3,"created":"2019-04-08 23:19:36.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/95/649095/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/95/649095/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/3"}}},"commit":{"parents":[{"commit":"5f2f1818688b9b29c51e4d183ac715c79e37525e","subject":"Tools \u0026 docs for backlog \u0026 abandoned spec process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f2f1818688b9b29c51e4d183ac715c79e37525e"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:17:06.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-08 23:19:32.000000000","tz":-300},"subject":"Abandon flavor-classes backlog spec","message":"Abandon flavor-classes backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nIt is probably related to [1], but I haven\u0027t dug into that yet.  The\nauthors of either spec should feel free to modify this commit to\nclarify.\n\n[1] https://review.openstack.org/#/c/635408/\n\nChange-Id: I1e5267a828133e9a6579a6790f2996c64a88fc8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ff1d91028b90fba0ac4f50ccdf1fd6b0eb06d234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ff1d91028b90fba0ac4f50ccdf1fd6b0eb06d234"}]},"branch":"refs/heads/master"},"6e18faf058c0fb06ca45ab2a773d06785507ba1c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-04-26 16:55:34.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/95/649095/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/95/649095/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/4"}}},"commit":{"parents":[{"commit":"4659b6da5d73bd01b7d0173af60d08e1254e7375","subject":"Merge \"Add \"CPU selection with hypervisor consideration\" spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4659b6da5d73bd01b7d0173af60d08e1254e7375"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:17:06.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-26 16:55:34.000000000","tz":0},"subject":"Abandon flavor-classes backlog spec","message":"Abandon flavor-classes backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nIt is probably related to [1], but I haven\u0027t dug into that yet.  The\nauthors of either spec should feel free to modify this commit to\nclarify.\n\n[1] https://review.openstack.org/#/c/635408/\n\nChange-Id: I1e5267a828133e9a6579a6790f2996c64a88fc8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6e18faf058c0fb06ca45ab2a773d06785507ba1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/6e18faf058c0fb06ca45ab2a773d06785507ba1c"}]},"branch":"refs/heads/master"},"2155221f7126c23ec37f4dabf6eeb03930325292":{"kind":"REWORK","_number":5,"created":"2019-04-28 15:55:47.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/95/649095/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/95/649095/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/5"}}},"commit":{"parents":[{"commit":"fe5ae212972679cf6327778fe087304a0098f4d0","subject":"Abandon parallel-scheduler backlog spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fe5ae212972679cf6327778fe087304a0098f4d0"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:17:06.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-28 15:55:38.000000000","tz":-300},"subject":"Abandon flavor-classes backlog spec","message":"Abandon flavor-classes backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nIt is probably related to [1], but I haven\u0027t dug into that yet.  The\nauthors of either spec should feel free to modify this commit to\nclarify.\n\n[1] https://review.openstack.org/#/c/635408/\n\nChange-Id: I1e5267a828133e9a6579a6790f2996c64a88fc8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2155221f7126c23ec37f4dabf6eeb03930325292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2155221f7126c23ec37f4dabf6eeb03930325292"}]},"branch":"refs/heads/master"},"8ca4dccbf77eff028bb97f24f5e34ae868f79a74":{"kind":"REWORK","_number":6,"created":"2019-05-01 22:44:43.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/95/649095/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/95/649095/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/6"}}},"commit":{"parents":[{"commit":"7ad084795808d0ce6cfacea2018d43d74f18bfe6","subject":"Merge \"Abandon instance-tasks backlog spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ad084795808d0ce6cfacea2018d43d74f18bfe6"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:17:06.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-05-01 22:44:40.000000000","tz":-300},"subject":"Abandon flavor-classes backlog spec","message":"Abandon flavor-classes backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nIt is probably related to [1], but I haven\u0027t dug into that yet.  The\nauthors of either spec should feel free to modify this commit to\nclarify.\n\n[1] https://review.openstack.org/#/c/635408/\n\nChange-Id: I1e5267a828133e9a6579a6790f2996c64a88fc8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8ca4dccbf77eff028bb97f24f5e34ae868f79a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8ca4dccbf77eff028bb97f24f5e34ae868f79a74"}]},"branch":"refs/heads/master"},"14f84bf8632728ae63d0bc7d24e37bb2cf4c6272":{"kind":"REWORK","_number":7,"created":"2019-05-01 23:00:56.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/95/649095/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/95/649095/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/95/649095/7"}}},"commit":{"parents":[{"commit":"7ad084795808d0ce6cfacea2018d43d74f18bfe6","subject":"Merge \"Abandon instance-tasks backlog spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7ad084795808d0ce6cfacea2018d43d74f18bfe6"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:17:06.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-05-01 23:00:54.000000000","tz":-300},"subject":"Abandon flavor-classes backlog spec","message":"Abandon flavor-classes backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nIt is probably related to [1], but I haven\u0027t dug into that yet.  The\nauthors of either spec should feel free to modify this commit to\nclarify.\n\n[1] https://review.openstack.org/#/c/635408/\n\nChange-Id: I1e5267a828133e9a6579a6790f2996c64a88fc8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14f84bf8632728ae63d0bc7d24e37bb2cf4c6272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/14f84bf8632728ae63d0bc7d24e37bb2cf4c6272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
