)]}'
{"id":"openstack%2Fnova-specs~649100","triplet_id":"openstack%2Fnova-specs~master~Idec4df4b3fe689d73da4786993aaf8aa13c2dbf0","project":"openstack/nova-specs","branch":"master","topic":"scrub-backlog-specs","hashtags":[],"change_id":"Idec4df4b3fe689d73da4786993aaf8aa13c2dbf0","subject":"Abandon instance-tasks backlog spec","status":"MERGED","created":"2019-04-01 16:30:50.000000000","updated":"2019-05-01 22:39:16.000000000","submitted":"2019-05-01 22:39:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649100-1556750356945-cdd0ea5d","meta_rev_id":"3a9b8449601cffb349363c9f28c5468f021bacda","_number":649100,"virtual_id_number":649100,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-05-01 22:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-01 21:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-05-01 22:26:31.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-01 17:51:57.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2019-04-09 11:46:06.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2019-04-09 11:54:42.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-04-09 13:56:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-09 15:51:46.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-04-19 05:31:41.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-04-28 12:41:22.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-05-01 21:53:07.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-05-01 22:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"567ac15dbfb41e925b866cc4e4d74ebb1c06931a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-01 16:30:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"511adf2d52d2ac0b5c159b218014ef56096f5138","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-01 16:37:13.000000000","message":"Topic changed from abandon_instance-tasks to scrub-backlog-specs","accounts_in_message":[],"_revision_number":1},{"id":"a29e715edd147b973c5d57086f41ab8e137fd32e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 16:46:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/649100/1/check/openstack-tox-docs/5620074/html/ : SUCCESS in 9m 02s\n- openstack-tox-pep8 http://logs.openstack.org/00/649100/1/check/openstack-tox-pep8/2a48bba/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b737cc00ae31bc70a0c4435e3edae46699d7b768","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 23:14:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"484426c27c1fda22604ebb68db5fa532a86e0cbb","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 23:18:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8294ee46fd4ff86f076fc8093b3e5f31d8bdb9f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 23:56:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/649100/3/check/openstack-tox-docs/05f9d18/html/ : SUCCESS in 9m 06s\n- openstack-tox-pep8 http://logs.openstack.org/00/649100/3/check/openstack-tox-pep8/d97dade/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":3},{"id":"35e2100d883715a84a44b7af69ea615b3d3ff437","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-09 11:46:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c639e3e5220c4304709ee4cc4532dce57aeef3e9","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-04-09 11:54:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"def99992af81ae0b4e6364ce6cd86bddc959ee07","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-09 13:56:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6d8defee1f4f4a3d27eb242c79e8359165fccaf6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-09 15:51:46.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nWith the current pace of development I don\u0027t see something this big happening in nova now so agree it should be abandoned.","accounts_in_message":[],"_revision_number":3},{"id":"689b99f106e55adbabfae19f89e5dd3e1c6c04e9","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-04-19 05:31:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"79e0d4b54627a2c6c455711b3e7db3a764169288","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-26 16:56:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4cebeb3b56290e7e1576f4ca824332b8e9777453","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 17:38:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/649100/4/check/openstack-tox-docs/9614a78/html/ : SUCCESS in 7m 45s\n- openstack-tox-pep8 http://logs.openstack.org/00/649100/4/check/openstack-tox-pep8/f6298b3/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":4},{"id":"5bfd9ef8e83d396d2dce839fdbab2571a4c4bc8e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-04-28 12:41:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nYeah, as much as I still like the idea, its not likely now.","accounts_in_message":[],"_revision_number":4},{"id":"4010ea23592bab43917fb48e24e71d684d4f4290","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 12:47:21.000000000","message":"Patch Set 4: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"a25e55768970652b3cb0ea2c84cdeafda907cb99","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-28 15:53:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"95b2a82e95891b73427f17b2e6bb16d1830419b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 16:01:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/649100/5/check/openstack-tox-docs/8a09b98/html/ : SUCCESS in 6m 33s\n- openstack-tox-pep8 http://logs.openstack.org/00/649100/5/check/openstack-tox-pep8/0d8e0e3/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":5},{"id":"55ef4c345f60d1d3da344fe6bbea39a217f4d475","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-05-01 21:53:07.000000000","message":"Patch Set 5: Code-Review+2\n\nCool, as we keep the reference.","accounts_in_message":[],"_revision_number":5},{"id":"aa2bf8f843cbdee73930e8896ca252945329a999","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-01 22:26:31.000000000","message":"Patch Set 5: Workflow+1\n\nCarrying forward prior approvals","accounts_in_message":[],"_revision_number":5},{"id":"bdd397abc59e8fe670316c294ce3615e49dea40c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 22:26:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6739c2b0e50a478db124cbd78d38eda877e0d8ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 22:39:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3a9b8449601cffb349363c9f28c5468f021bacda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 22:39:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/649100/5/gate/openstack-tox-docs/aab3657/html/ : SUCCESS in 8m 01s\n- openstack-tox-pep8 http://logs.openstack.org/00/649100/5/gate/openstack-tox-pep8/e0942dd/ : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"942a12fb09baa79d35dcd5b97d169009fe1e7a60","revisions":{"20bc1172bf8b5d88c9c2ee676a78499868e990d8":{"kind":"REWORK","_number":1,"created":"2019-04-01 16:30:50.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/00/649100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/649100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/1"}}},"commit":{"parents":[{"commit":"e4934df640fce8ff604c5d78bc772a735ece4ec4","subject":"Tools \u0026 docs for backlog \u0026 abandoned spec process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e4934df640fce8ff604c5d78bc772a735ece4ec4"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:28:15.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:28:15.000000000","tz":-300},"subject":"Abandon instance-tasks backlog spec","message":"Abandon instance-tasks backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nThe original authors should feel free to modify or downvote this commit\nif the spec should continue to be considered for implementation now or\nin the future.\n\nChange-Id: Idec4df4b3fe689d73da4786993aaf8aa13c2dbf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/20bc1172bf8b5d88c9c2ee676a78499868e990d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/20bc1172bf8b5d88c9c2ee676a78499868e990d8"}]},"branch":"refs/heads/master"},"9a4170ea156e03a544b4d59e91854335488fb742":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-08 23:14:52.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/00/649100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/649100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/2"}}},"commit":{"parents":[{"commit":"5f2f1818688b9b29c51e4d183ac715c79e37525e","subject":"Tools \u0026 docs for backlog \u0026 abandoned spec process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f2f1818688b9b29c51e4d183ac715c79e37525e"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:28:15.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-08 23:14:52.000000000","tz":0},"subject":"Abandon instance-tasks backlog spec","message":"Abandon instance-tasks backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nThe original authors should feel free to modify or downvote this commit\nif the spec should continue to be considered for implementation now or\nin the future.\n\nChange-Id: Idec4df4b3fe689d73da4786993aaf8aa13c2dbf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9a4170ea156e03a544b4d59e91854335488fb742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9a4170ea156e03a544b4d59e91854335488fb742"}]},"branch":"refs/heads/master"},"8bd76541e5e92b04e52e28d0d162f6020d896d45":{"kind":"REWORK","_number":3,"created":"2019-04-08 23:18:20.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/00/649100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/649100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/3"}}},"commit":{"parents":[{"commit":"5f2f1818688b9b29c51e4d183ac715c79e37525e","subject":"Tools \u0026 docs for backlog \u0026 abandoned spec process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5f2f1818688b9b29c51e4d183ac715c79e37525e"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:28:15.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-08 23:17:36.000000000","tz":-300},"subject":"Abandon instance-tasks backlog spec","message":"Abandon instance-tasks backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nThe original authors should feel free to modify or downvote this commit\nif the spec should continue to be considered for implementation now or\nin the future.\n\nChange-Id: Idec4df4b3fe689d73da4786993aaf8aa13c2dbf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8bd76541e5e92b04e52e28d0d162f6020d896d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/8bd76541e5e92b04e52e28d0d162f6020d896d45"}]},"branch":"refs/heads/master"},"0a66b174fa5d6656a16bcb03c08c1b76c1b78471":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-04-26 16:56:00.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/00/649100/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/649100/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/4"}}},"commit":{"parents":[{"commit":"4659b6da5d73bd01b7d0173af60d08e1254e7375","subject":"Merge \"Add \"CPU selection with hypervisor consideration\" spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/4659b6da5d73bd01b7d0173af60d08e1254e7375"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:28:15.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-26 16:56:00.000000000","tz":0},"subject":"Abandon instance-tasks backlog spec","message":"Abandon instance-tasks backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nThe original authors should feel free to modify or downvote this commit\nif the spec should continue to be considered for implementation now or\nin the future.\n\nChange-Id: Idec4df4b3fe689d73da4786993aaf8aa13c2dbf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0a66b174fa5d6656a16bcb03c08c1b76c1b78471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0a66b174fa5d6656a16bcb03c08c1b76c1b78471"}]},"branch":"refs/heads/master"},"942a12fb09baa79d35dcd5b97d169009fe1e7a60":{"kind":"REWORK","_number":5,"created":"2019-04-28 15:53:50.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/00/649100/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/00/649100/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/00/649100/5"}}},"commit":{"parents":[{"commit":"fe5ae212972679cf6327778fe087304a0098f4d0","subject":"Abandon parallel-scheduler backlog spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/fe5ae212972679cf6327778fe087304a0098f4d0"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-01 16:28:15.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-28 15:53:47.000000000","tz":-300},"subject":"Abandon instance-tasks backlog spec","message":"Abandon instance-tasks backlog spec\n\nScrubbing the backlog directory.\n\nThis spec has been here since 2015.\n\nThe original authors should feel free to modify or downvote this commit\nif the spec should continue to be considered for implementation now or\nin the future.\n\nChange-Id: Idec4df4b3fe689d73da4786993aaf8aa13c2dbf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/942a12fb09baa79d35dcd5b97d169009fe1e7a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/942a12fb09baa79d35dcd5b97d169009fe1e7a60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
