)]}'
{"id":"openstack%2Fnova-specs~650580","triplet_id":"openstack%2Fnova-specs~master~Icccf2c5f028def28b27c6248c294039040206f16","project":"openstack/nova-specs","branch":"master","topic":"bp/detach-boot-volume","hashtags":[],"change_id":"Icccf2c5f028def28b27c6248c294039040206f16","subject":"Re-propose detach root volume spec","status":"MERGED","created":"2019-04-07 14:07:16.000000000","updated":"2019-06-05 14:13:30.000000000","submitted":"2019-04-08 11:20:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650580-1554722448555-5ed2890e","meta_rev_id":"0872b38ba82e335f0aedd0bf4d63ed8aea881936","_number":650580,"virtual_id_number":650580,"owner":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":2,"date":"2019-04-08 11:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-04-08 11:04:55.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-04-08 11:04:55.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-06-05 14:13:30.000000000","post_submit":true,"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-08 11:04:55.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-08 11:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-05 14:13:30.000000000","updated_by":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"reviewer":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"state":"REVIEWER"}],"messages":[{"id":"b49ec94842e33dd59610be99cc815b28c0537072","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2019-04-07 14:07:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c34fc3f394a38fb39a5f91c051599bb304c152d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-07 14:16:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/650580/1/check/openstack-tox-docs/8914d1b/html/ : SUCCESS in 7m 02s\n- openstack-tox-pep8 http://logs.openstack.org/80/650580/1/check/openstack-tox-pep8/a7d8273/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"707d1e6e0b1d0c6b07c03aa5f186ab6c27ab31b1","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 02:48:43.000000000","message":"Patch Set 1:\n\nThis can be fast approved. Just add the Previously-approved tag to the commit message per the process at https://github.com/openstack/nova-specs/blob/master/README.rst#previously-approved-specifications","accounts_in_message":[],"_revision_number":1},{"id":"9c269d31d607db7645778910ee9f7ce61a4de0af","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 02:49:23.000000000","message":"Topic changed from bp/was to bp/detach-boot-volume","accounts_in_message":[],"_revision_number":1},{"id":"81dbd0211f59186156c1b159787e266f99b5919d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 11:03:52.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8abaf0be5f66770c554ef480a5c0f099157812e9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 11:04:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI added the Previously-approved tag.\n\nDiff from the Stein version:\n\n 260c260\n \u003c   Zhenyu Zheng\n ---\n \u003e   Kevin Zheng\n 352a353,354\n \u003e      - Re-proposed.\n \u003e    * - Train\n\nFast merging as previously approved.","accounts_in_message":[],"_revision_number":2},{"id":"92ba2c3ad68948be60bf7363209ce6a0bf5c8747","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 11:12:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/650580/2/check/openstack-tox-docs/a3d0bee/html/ : SUCCESS in 7m 21s\n- openstack-tox-pep8 http://logs.openstack.org/80/650580/2/check/openstack-tox-pep8/f802fa5/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"174fa485d47977c89b041cd23c659ca3ad898060","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 11:12:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4c4d1d6e84a01d9e24fc0211bdd439bc84ea153","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 11:20:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/650580/2/gate/openstack-tox-docs/38f4571/html/ : SUCCESS in 7m 38s\n- openstack-tox-pep8 http://logs.openstack.org/80/650580/2/gate/openstack-tox-pep8/df57b2f/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"3e5466bba0a08201bcc7f09f578d8e20af7674d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 11:20:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d7e169e29ef47514e150b7703c1da5038142c385","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-08 15:27:23.000000000","message":"Patch Set 2:\n\nThis probably shouldn\u0027t have been fast approved given the amount of discussion on the design in the ML late in Stein about this feature and how to handle things like multiattach/tag checks and such. Kevin should have probably put details about that in the spec when re-proposing. See \"Corner case issues with root volume detach/attach\" in the Train PTG etherpad:\n\nhttps://etherpad.openstack.org/p/nova-ptg-train","accounts_in_message":[],"_revision_number":2},{"id":"173906536bb8748f54e0b3f0eeae64c2c9798e81","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-08 16:16:11.000000000","message":"Patch Set 2:\n\nCool; I\u0027ll unapprove the blueprint and we can follow up with deltas to the spec as necessary.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb3eab7a5a514ca85e7b4dac87eecc91bb7d48ad","revisions":{"2b81be5d93d817738a6fc7b5639c352bfbc5c380":{"kind":"REWORK","_number":1,"created":"2019-04-07 14:07:16.000000000","uploader":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"ref":"refs/changes/80/650580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/80/650580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/650580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/650580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/650580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/80/650580/1"}}},"commit":{"parents":[{"commit":"06b5446e53c728bf7266696e6bedda2b92b498c0","subject":"Merge \"Add a script for counting blueprints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/06b5446e53c728bf7266696e6bedda2b92b498c0"}]}],"author":{"name":"Kevin_Zheng","email":"zhengzhenyu@huawei.com","date":"2019-04-07 14:02:59.000000000","tz":480},"committer":{"name":"Kevin_Zheng","email":"zhengzhenyu@huawei.com","date":"2019-04-07 14:06:47.000000000","tz":480},"subject":"Re-propose detach root volume spec","message":"Re-propose detach root volume spec\n\nThe detach root volume blueprint was accepted in Stein,\nbut some of the work did not finish, so repropose it\nfor Train.\n\nCo-Authored-By: Andrea Rosa \u003candrea.rosa@hpe.com\u003e\n\nblueprint detach-boot-volume\n\nChange-Id: Icccf2c5f028def28b27c6248c294039040206f16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b81be5d93d817738a6fc7b5639c352bfbc5c380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2b81be5d93d817738a6fc7b5639c352bfbc5c380"}]},"branch":"refs/heads/master"},"bb3eab7a5a514ca85e7b4dac87eecc91bb7d48ad":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-08 11:03:52.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/80/650580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/80/650580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/650580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/650580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/80/650580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/80/650580/2"}}},"commit":{"parents":[{"commit":"06b5446e53c728bf7266696e6bedda2b92b498c0","subject":"Merge \"Add a script for counting blueprints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/06b5446e53c728bf7266696e6bedda2b92b498c0"}]}],"author":{"name":"Kevin_Zheng","email":"zhengzhenyu@huawei.com","date":"2019-04-07 14:02:59.000000000","tz":480},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-08 11:03:48.000000000","tz":0},"subject":"Re-propose detach root volume spec","message":"Re-propose detach root volume spec\n\nThe detach root volume blueprint was accepted in Stein,\nbut some of the work did not finish, so repropose it\nfor Train.\n\nCo-Authored-By: Andrea Rosa \u003candrea.rosa@hpe.com\u003e\n\nblueprint detach-boot-volume\n\nPreviously-approved: Stein\n\nChange-Id: Icccf2c5f028def28b27c6248c294039040206f16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bb3eab7a5a514ca85e7b4dac87eecc91bb7d48ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/bb3eab7a5a514ca85e7b4dac87eecc91bb7d48ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
