)]}'
{"id":"openstack%2Fnova-specs~651681","triplet_id":"openstack%2Fnova-specs~master~I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea","project":"openstack/nova-specs","branch":"master","topic":"bp/expose-auto-converge-post-copy","hashtags":[],"change_id":"I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea","subject":"Add spec for expose-auto-converge-post-copy","status":"MERGED","created":"2019-04-11 01:39:41.000000000","updated":"2019-07-23 13:45:45.000000000","submitted":"2019-07-23 13:45:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":224,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651681-1563889545375-358cd872","meta_rev_id":"4b7b04da0b24f6cf7273da4fb81a92d20b7851e6","_number":651681,"virtual_id_number":651681,"owner":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2019-07-23 13:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14672,"name":"Dr. David Alan Gilbert","email":"dgilbert@redhat.com"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":19124,"name":"Gerry Kopec","email":"gerry.kopec@windriver.com","username":"gkopec"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14672,"name":"Dr. David Alan Gilbert","email":"dgilbert@redhat.com"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":19124,"name":"Gerry Kopec","email":"gerry.kopec@windriver.com","username":"gkopec"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-07-23 12:58:03.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2019-07-23 13:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14672,"name":"Dr. David Alan Gilbert","email":"dgilbert@redhat.com"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":19124,"name":"Gerry Kopec","email":"gerry.kopec@windriver.com","username":"gkopec"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2019-07-23 13:30:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14672,"name":"Dr. David Alan Gilbert","email":"dgilbert@redhat.com"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":19124,"name":"Gerry Kopec","email":"gerry.kopec@windriver.com","username":"gkopec"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14672,"name":"Dr. David Alan Gilbert","email":"dgilbert@redhat.com"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":19124,"name":"Gerry Kopec","email":"gerry.kopec@windriver.com","username":"gkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-11 01:53:03.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-11 03:15:36.000000000","updated_by":{"_account_id":19124,"name":"Gerry Kopec","email":"gerry.kopec@windriver.com","username":"gkopec"},"reviewer":{"_account_id":19124,"name":"Gerry Kopec","email":"gerry.kopec@windriver.com","username":"gkopec"},"state":"REVIEWER"},{"updated":"2019-04-26 02:30:20.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-04-26 02:30:26.000000000","updated_by":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"reviewer":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"state":"REVIEWER"},{"updated":"2019-05-21 08:22:28.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2019-05-31 11:11:41.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-06-03 20:18:17.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-06-11 15:07:27.000000000","updated_by":{"_account_id":14672,"name":"Dr. David Alan Gilbert","email":"dgilbert@redhat.com"},"reviewer":{"_account_id":14672,"name":"Dr. David Alan Gilbert","email":"dgilbert@redhat.com"},"state":"REVIEWER"},{"updated":"2019-06-13 07:57:20.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-06-27 06:40:25.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-07-22 14:22:07.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-07-23 12:58:03.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-23 13:30:32.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-07-23 13:45:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da904b7b0a8d7168875908b70c8e25eefbca323c","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-11 01:39:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abd62f51607786ef82f2c3161126aeb6bd1f3dcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 01:49:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/1/check/openstack-tox-docs/9629af7/html/ : SUCCESS in 8m 38s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/1/check/openstack-tox-pep8/862b7e8/ : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"25b65cda5756ee474d0611fd58128c924a4eee0d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-12 02:00:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI think we need to figure out what happens if the host doesn\u0027t support the requested spec, and how we could do this better with traits so we can do filtering in the scheduler.","accounts_in_message":[],"_revision_number":1},{"id":"dddf11f5f061df7a4644acdd42c07d349b409896","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-12 09:29:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"043aff06c1e71d7827f699787e456ae079f5241d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 09:41:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/2/check/openstack-tox-docs/0260554/html/ : SUCCESS in 8m 45s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/2/check/openstack-tox-pep8/fe1cf56/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d006eb6d3ba501ac28da1f626899eb911c1658bf","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-04-16 09:03:14.000000000","message":"Patch Set 2:\n\nCan we abandon https://review.openstack.org/#/c/648619/ ?","accounts_in_message":[],"_revision_number":2},{"id":"437582e8ffbae7cd792dab1a8c4aa38595fb4987","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-04-16 09:26:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ac72999ecb5b7e700c7456251c9cc2098f2ad929","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-18 03:26:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ee4cffa15b810cdf837256d7d6215eb5c870161","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-18 03:27:27.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2e41eb950a96f65e274f06a731e999d1a8bec0dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 03:43:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/3/check/openstack-tox-docs/f10a38d/html/ : SUCCESS in 8m 21s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/3/check/openstack-tox-pep8/061ac84/ : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d9c83440770a581f24642dd6ed50376ecea11f0c","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-18 05:53:09.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c2f9e5af42818dec30e2c8e00634d47b633c9bfb","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-18 06:22:10.000000000","message":"Patch Set 3:\n\nI have talk with WangYa. There is case we need to solve. There must a case the user doesn\u0027t want use \u0027auto_converge\u0027 and \u0027post_copy\u0027, so if we ask the user to write a forbidden trait for COMPUTE_AUTO_CONVERGE or COMPUTE_POST_COPY. That will force the request going to a host without the auto_converge and post_copy support. But actually, we required qemu \u003e 2.5, that means all the hosts support those two features. That leads to no way to disable this two features for the instance.\n\nI suggested to introduce two new extra specs \u0027auto_converge\u003dtrue/false\u0027 and \u0027post_copy\u003dtrue/false\u0027 instead of asking the user write the required/forbidden trait directly.\n\nSo before calling the placement, when the post_copy/auto_converge\u003dtrue, we add required trait for this two features to the placement request. When the post_copy/auto_converge\u003dfalse, we just add nothing to the placement request. Then we still can schedule an instance on a host with those two feature but we disable this two features for that instance.","accounts_in_message":[],"_revision_number":3},{"id":"d437684b201cbca1b5a81d12021dcebcaac7a01b","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-18 13:25:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"384233269c6a467b533fa397311114b72ceee886","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 13:43:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/4/check/openstack-tox-docs/f1852b8/html/ : SUCCESS in 9m 45s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/4/check/openstack-tox-pep8/652b8a9/ : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":4},{"id":"63c3ab896bdc3a17fa0ff5ec389e101087f59fd7","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-18 14:45:05.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4cc430b39833fe4b12451dd6f615d072c4cd590f","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-19 06:01:24.000000000","message":"Patch Set 4: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":4},{"id":"aaf7d4ab0aad44ae112a43f3e0b1b6848fe3c0fb","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-19 06:56:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cf5b8fcb23ff2335edca64300b68ac289f923845","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-19 07:02:17.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bc1f6dd37e946df555cdd5062c498824432db807","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 07:11:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/5/check/openstack-tox-docs/d8b096a/html/ : SUCCESS in 8m 47s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/5/check/openstack-tox-pep8/2634c74/ : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":5},{"id":"c00c4247d8228aef9929d9a03f42046a453240b6","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-19 08:10:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ef49bb9036c23365cc38b0b5a80ee0090c965bff","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-19 09:47:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"14a2d906c229309b48da11d99a96a133b054e7c6","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-19 09:50:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"573ab458874af42c8942f74d6083672db0c4a8a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 11:59:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/6/check/openstack-tox-docs/1604e88/html/ : SUCCESS in 10m 45s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/6/check/openstack-tox-pep8/0708ea7/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":6},{"id":"ab4cc33d794202a8aa084c511d29634810358966","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-22 15:48:21.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"10d5fde8ff5dccdd018d69fb0ff7ba71399ba05f","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-23 01:14:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5db7e1c608d5e802eb07cbd24f42cd2524cb26af","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-23 01:15:32.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"967d0ee19a0e0480fda96a48781bdeee1d3bd98d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:28:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/7/check/openstack-tox-docs/f76cc4f/html/ : SUCCESS in 6m 57s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/7/check/openstack-tox-pep8/cca0170/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":7},{"id":"2cc8d384583a0c39603cdb01d48a1e30879001ff","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-24 23:18:14.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)\n\nThanks. Typo inline, but it\u0027s obvious what you meant.\n\nThis seems straightforward to me.","accounts_in_message":[],"_revision_number":7},{"id":"4edd2e8bebc4bcea4d6ce209534e82d017038db4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-26 02:29:54.000000000","message":"Patch Set 7:\n\n(18 comments)\n\nI\u0027ve got several questions inline, mostly for trying to clarify the proposed change and scenarios. I\u0027m potentially getting confused by the over-use of the term \"operator\" in here since operator to me is the admin of the cloud, not the end user. The operator provides the service, the user consumes it.","accounts_in_message":[],"_revision_number":7},{"id":"336b02bbfb6d1b79392f94ecf5fbfcb3b821b2f1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-26 02:30:08.000000000","message":"Patch Set 7:\n\nI think I\u0027d also like Kashyap to take a look at this.","accounts_in_message":[],"_revision_number":7},{"id":"41872ecfc0c9ce77b84b92ec7078c6faebe88024","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-26 08:47:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"11a57cb3cd19175a03adde065cb19c6f0d494c66","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-04-26 08:48:33.000000000","message":"Patch Set 7:\n\n(18 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3334c6e8717609d192779d566d6abc18cca67edf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 09:00:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/8/check/openstack-tox-docs/26f61c8/html/ : SUCCESS in 6m 55s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/8/check/openstack-tox-pep8/353d55f/ : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":8},{"id":"dce4166fdfb669c5a3df62559194d00c23c388a2","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-30 18:10:18.000000000","message":"Patch Set 8: Code-Review+2\n\n(5 comments)\n\nStill good with this.","accounts_in_message":[],"_revision_number":8},{"id":"5245bdcd3ff4bd9f348ad7f3c501ad6b386dccb0","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-30 20:11:34.000000000","message":"Patch Set 8: Code-Review+1\n\n(4 comments)\n\nLGTM, but Matt and Kashyap may want to take a look at","accounts_in_message":[],"_revision_number":8},{"id":"5ba02e21bdb719599cea842017fe8a95b651a924","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-05-05 03:22:18.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"509fd4b7b3b372ea7fb0a47a45339dd54e9025b3","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-05-05 03:22:29.000000000","message":"Patch Set 8:\n\n(8 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9e1d6e96267ebefb4dd228c610b5768f8f076083","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 03:38:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/9/check/openstack-tox-docs/d3c92d4/html/ : SUCCESS in 8m 53s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/9/check/openstack-tox-pep8/3ed00d8/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":9},{"id":"ce985ade92946239dc2452b0908a12afa211c413","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-05-20 12:29:11.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM except the prefix for the new extra specs, but I\u0027m looking for other people give suggestion on that","accounts_in_message":[],"_revision_number":9},{"id":"07013efece8e23b9ceb79d3e6d29bdcb77eab0be","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-20 19:06:50.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)\n\nStill good with this. A couple of (probably implementation) notes inline.\n\nProcedurally, at least Matt and Sylvain should have another look at this before we approve it. And Kashyap, on Matt\u0027s recommendation.","accounts_in_message":[],"_revision_number":9},{"id":"e876d7861e8abd7073043870e9a6463d0f2f7580","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-21 08:22:28.000000000","message":"Patch Set 9: Code-Review+1\n\n(21 comments)\n\nThe idea to allow exposing auto-converge / post-copy on a per-instance basis is reasonable.\n\nI have some nits on phrasing; not going to -1 them, as they can be addressed in a follow-up (or in this one, if there is a strong need to respin this.)","accounts_in_message":[],"_revision_number":9},{"id":"88fef1947fd8002919d0483b97800ce714557b00","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-05-22 10:15:49.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)\n\nSorry, but I\u0027m not OK with having image properties to be superseding the conf options, hence this -1.","accounts_in_message":[],"_revision_number":9},{"id":"598b9f9857f34e53ca50d32a6c0f8e7db81371f9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-22 20:54:10.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d4a316295e32d3a8d4d7bb95e96f21fafd765535","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-05-27 01:41:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"846aabe0d306ba0855c6a40eb9fe590d7d67e2b5","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-05-27 01:42:11.000000000","message":"Patch Set 9:\n\n(15 comments)","accounts_in_message":[],"_revision_number":9},{"id":"93efb47ca1d32bfa31683f9ca130d5da4d18b5e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 01:52:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/10/check/openstack-tox-docs/59f0ffe/html/ : SUCCESS in 7m 21s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/10/check/openstack-tox-pep8/4b4992e/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":10},{"id":"5131bdf85751ad543a0aeb2057466c21a08aa9ce","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-05-27 11:35:16.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c677059043ae6319e53906e7e4730d104c9ad657","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-05-28 07:25:28.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5187d7cfcc3ecb32a44267ecc638cab1e53efcb6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-30 16:22:00.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"e4c907df066c5d044c4506090965819cf125731a","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-05-31 03:20:35.000000000","message":"Patch Set 10:\n\n\u003e (1 comment)\n\nYes, we have policy control on update image properties API.\n\nReference:\n  https://docs.openstack.org/glance/latest/admin/policies.html\n  https://developer.openstack.org/api-ref/image/v2/?expanded\u003d#update-image","accounts_in_message":[],"_revision_number":10},{"id":"632ce91f0bdbbd7a1aefcdbb9e6f70e2a38d0dd2","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-05-31 03:21:40.000000000","message":"Patch Set 10:\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e Yes, we have policy control on update image properties API.\n \u003e \n \u003e Reference:\n \u003e https://docs.openstack.org/glance/latest/admin/policies.html\n \u003e https://developer.openstack.org/api-ref/image/v2/?expanded\u003d#update-image\n\nBy default, administrator and the image owner both have write access to image properties.","accounts_in_message":[],"_revision_number":10},{"id":"df5e411ff5b6ba9c78a8b0e6c4ae1d68fed944b2","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-31 11:11:41.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)\n\nI don\u0027t like us exposing \"post copy\" this way.\nIn the past we said this was too hypervisor specific.\n\nWe came up with the term \"force complete\" instead:\nhttps://specs.openstack.org/openstack/nova-specs/specs/pike/approved/live-migration-force-after-timeout.html\n\nI would much rather we use the \"force complete\" term here.","accounts_in_message":[],"_revision_number":10},{"id":"49a049f9c75228235d990d9ec568f588666caa3e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-31 11:32:50.000000000","message":"Patch Set 10:\n\n(2 comments)\n\nI think we can do better than the options listed here, and abstract away form the implementation details.\n\nIn fact I think we should learn from the \"maintenance-policy\" that google\u0027s cloud offers.","accounts_in_message":[],"_revision_number":10},{"id":"5fb9f2e191bacc30df06893fea0671ad8d92168d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-31 17:05:06.000000000","message":"Patch Set 10: -Code-Review","accounts_in_message":[],"_revision_number":10},{"id":"787bd7ffe82694da4d306737940182ef087a404d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-03 20:18:17.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"96eae9f39c78578b7075dd8a7c81ecb39596778f","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-06-04 09:47:56.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0818832d9cae9cbed336891234b1d0eeb797dd00","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-04 10:42:54.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"26ddeb1f1c62f229aa0d196536b576f87c866d7d","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-04 10:56:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"729ca5e9454955da41219d6ca7812d10aaccce04","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-06-04 11:30:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c0022ec5698cc43cfa79c493520fc388d2c91f75","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-04 12:05:42.000000000","message":"Patch Set 10:\n\n(2 comments)\n\nI think I understand your use case much better now, thank you!\n\nI have an alternative proposal in the comments.","accounts_in_message":[],"_revision_number":10},{"id":"1fb866f3545490ed63ec3f37e51e021734b33884","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-06-04 13:40:49.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1ddb11825f90b97b4b8f86012f00870a20a18815","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-04 14:19:11.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0e40fe3b6dd4060db43c421e2ae20b7236e61006","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-06-04 16:09:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"dba1174f5ae1341c520b4ee26d708aa0e191031c","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-06-05 01:52:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a475e14c805709816c878793bdaddd30193ac370","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:07:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/11/check/openstack-tox-docs/e6cda45/html/ : SUCCESS in 7m 34s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/11/check/openstack-tox-pep8/198ad61/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":11},{"id":"71a2cb6937cf36d385ebb856c8be0d0a0b0408fb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-11 13:15:05.000000000","message":"Patch Set 11: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9b5931280958ac7056c8f84c17eb776e93001b76","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-06-11 14:06:56.000000000","message":"Patch Set 11:\n\n(9 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7c68bb84a4cde69e90d5607015af0ca64f821dfd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-11 14:36:16.000000000","message":"Patch Set 11:\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"aad59342adbd58b04e7f1ce49a72dd9945ab1003","author":{"_account_id":14672,"name":"Dr. David Alan Gilbert","email":"dgilbert@redhat.com"},"date":"2019-06-11 15:07:27.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"26f8912b9b816f199947c90943d68b27f486d79d","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-06-13 03:52:27.000000000","message":"Patch Set 11:\n\n(8 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2c1b195fd88af01fe6716a72fc146c54f7d3068e","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-06-13 05:53:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"508a94b8ad5e12de25d02e0bedff5c7f07e04707","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 06:15:13.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/12/check/openstack-tox-docs/d7605e4/html/ : SUCCESS in 8m 06s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/12/check/openstack-tox-pep8/0c5b76e/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":12},{"id":"3af0b8f3e3e92c543ae8b85055b0c93eb3e813c7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-25 21:39:24.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"12f57bde1b34986cf6cc254abedd0bd6516fba16","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-25 21:49:50.000000000","message":"Patch Set 13: Code-Review+1\n\n(5 comments)\n\nThis seems close enough to go, but there was enough recent discussion that I\u0027m not sure the overall approach has support anymore. Need previous reviewers to weigh in again, please.","accounts_in_message":[],"_revision_number":13},{"id":"5d0c028c5527cde8b49d03bde032568d4fec573e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 21:54:05.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/13/check/openstack-tox-docs/0707fbe/html/ : SUCCESS in 6m 30s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/13/check/openstack-tox-pep8/c7e5515/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":13},{"id":"8f3ae4b505594e2fd3770b27bdea87cb42bb5630","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-06-27 06:40:25.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":13},{"id":"e23345fe51942cbe0e552233108918a9c771ff29","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-07-22 13:02:19.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"e176da8388cdca7e1d6ac151c615d50026e55eb4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-07-22 14:22:07.000000000","message":"Patch Set 13: Code-Review+1\n\n(3 comments)\n\ndirectionally i think this is fine hence the +1\nit would be nice to adress the issue raised by erric inline but i would be ok with that being in a follow up patch as they are clarification rather then change in intent. that said i also would be fine with respinning this to adress those clarificaitons.","accounts_in_message":[],"_revision_number":13},{"id":"ad81f13ba7cc44f0ae7c6650698aca0f18007819","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-07-23 06:08:34.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"236721025f82f7789f6d23d70c7748822b6e5903","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 06:22:33.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/14/check/openstack-tox-docs/73ea8fc/html/ : SUCCESS in 7m 55s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/14/check/openstack-tox-pep8/0f5171c/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":14},{"id":"baa9fbdbea138e0821a51c5e4ddab393764031ce","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-07-23 06:24:00.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ed7c8679530b47fcfaa7b63ba49ed29b8a4860c7","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-07-23 06:24:27.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"31662d178b9a56774bc82c7f600ebb1f11d565c2","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-07-23 06:30:50.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3b85ac425da0c4a4ec8ee94b129449e3c4e602e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 06:43:41.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/16/check/openstack-tox-docs/b147863/html/ : SUCCESS in 7m 20s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/16/check/openstack-tox-pep8/59584b8/ : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":16},{"id":"24ca4bddc64136daacac6342933e43b09ae98065","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-23 12:58:03.000000000","message":"Patch Set 16: Code-Review+2\n\n++ change back to \"post_copy\", other edits are sane.\n\nNeed a non-Intel core to merge this.","accounts_in_message":[],"_revision_number":16},{"id":"cdf1e67002ac06ed625f6a1c8250f454811574bf","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-07-23 13:30:32.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(4 comments)\n\nOk I have few thoughts but they are just nits. +W.","accounts_in_message":[],"_revision_number":16},{"id":"9e26febc7390e5285fca97e6a1510e1cc6739fc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:30:44.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"00b7520b6fcc24345e61033ee05cffbd901e731e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:45:44.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/651681/16/gate/openstack-tox-docs/9c39467/html/ : SUCCESS in 8m 10s\n- openstack-tox-pep8 http://logs.openstack.org/81/651681/16/gate/openstack-tox-pep8/f041afe/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":16},{"id":"7a3455a1f6c0caa05edddc1870365000dd83be75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:45:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"9037c2c83494a63b3dd4abd3598f01e51ac294ad","revisions":{"07540e3e4089c44c21e6a0ac331a6507604e071f":{"kind":"REWORK","_number":1,"created":"2019-04-11 01:39:41.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/1"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:37:06.000000000","tz":480},"subject":"Expose auto converge and post copy","message":"Expose auto converge and post copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/07540e3e4089c44c21e6a0ac331a6507604e071f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/07540e3e4089c44c21e6a0ac331a6507604e071f"}]},"branch":"refs/heads/master"},"631b54d6d3bc8479eb603d07bdc55b062531fc92":{"kind":"REWORK","_number":2,"created":"2019-04-12 09:29:25.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/2"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-12 09:29:09.000000000","tz":480},"subject":"Expose auto converge and post copy","message":"Expose auto converge and post copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/631b54d6d3bc8479eb603d07bdc55b062531fc92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/631b54d6d3bc8479eb603d07bdc55b062531fc92"}]},"branch":"refs/heads/master"},"d3f460eb9c2a6c5928cc79ea87edf9222979ee92":{"kind":"REWORK","_number":3,"created":"2019-04-18 03:26:25.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/3"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-18 03:26:12.000000000","tz":480},"subject":"Expose auto converge and post copy","message":"Expose auto converge and post copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d3f460eb9c2a6c5928cc79ea87edf9222979ee92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d3f460eb9c2a6c5928cc79ea87edf9222979ee92"}]},"branch":"refs/heads/master"},"886aca070aff3bed30aabfd7f407ebda0461a51e":{"kind":"REWORK","_number":4,"created":"2019-04-18 13:25:10.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/4"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-18 13:23:54.000000000","tz":480},"subject":"Expose auto converge and post copy","message":"Expose auto converge and post copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/886aca070aff3bed30aabfd7f407ebda0461a51e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/886aca070aff3bed30aabfd7f407ebda0461a51e"}]},"branch":"refs/heads/master"},"085837386bb3005b520e3acfc47f2d72063cc68d":{"kind":"REWORK","_number":5,"created":"2019-04-19 06:56:35.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/5"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-19 06:55:43.000000000","tz":480},"subject":"Expose auto converge and post copy","message":"Expose auto converge and post copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/085837386bb3005b520e3acfc47f2d72063cc68d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/085837386bb3005b520e3acfc47f2d72063cc68d"}]},"branch":"refs/heads/master"},"ea5911ef4de2bf2d10d9c967053fd8ac81fb6684":{"kind":"REWORK","_number":6,"created":"2019-04-19 09:47:51.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/6"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-19 09:47:32.000000000","tz":480},"subject":"Expose auto converge and post copy","message":"Expose auto converge and post copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ea5911ef4de2bf2d10d9c967053fd8ac81fb6684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/ea5911ef4de2bf2d10d9c967053fd8ac81fb6684"}]},"branch":"refs/heads/master"},"5333578d261e0a33f7540e698809557ff00e1aaf":{"kind":"REWORK","_number":7,"created":"2019-04-23 01:14:13.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/7"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-23 01:13:59.000000000","tz":480},"subject":"Expose auto converge and post copy","message":"Expose auto converge and post copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5333578d261e0a33f7540e698809557ff00e1aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/5333578d261e0a33f7540e698809557ff00e1aaf"}]},"branch":"refs/heads/master"},"461832e1d951c92977f89f4727cdf539ac65399e":{"kind":"REWORK","_number":8,"created":"2019-04-26 08:47:22.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/8"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-26 08:47:00.000000000","tz":480},"subject":"Add spec for expose-auto-converge-post-copy","message":"Add spec for expose-auto-converge-post-copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/461832e1d951c92977f89f4727cdf539ac65399e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/461832e1d951c92977f89f4727cdf539ac65399e"}]},"branch":"refs/heads/master"},"0d8ba04de99d4739e4c42b3845589cac771b7e7d":{"kind":"REWORK","_number":9,"created":"2019-05-05 03:22:18.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/9"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-05-05 03:22:01.000000000","tz":480},"subject":"Add spec for expose-auto-converge-post-copy","message":"Add spec for expose-auto-converge-post-copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d8ba04de99d4739e4c42b3845589cac771b7e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0d8ba04de99d4739e4c42b3845589cac771b7e7d"}]},"branch":"refs/heads/master"},"254d52503ffdb38dde860ac8b607bcce6bce4de4":{"kind":"REWORK","_number":10,"created":"2019-05-27 01:41:45.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/10"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-05-27 01:41:29.000000000","tz":480},"subject":"Add spec for expose-auto-converge-post-copy","message":"Add spec for expose-auto-converge-post-copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/254d52503ffdb38dde860ac8b607bcce6bce4de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/254d52503ffdb38dde860ac8b607bcce6bce4de4"}]},"branch":"refs/heads/master"},"e3d4937f1939a6a2bd9117863d0eb22ae167e2b9":{"kind":"REWORK","_number":11,"created":"2019-06-05 01:52:57.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/11"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-06-05 01:52:38.000000000","tz":480},"subject":"Add spec for expose-auto-converge-post-copy","message":"Add spec for expose-auto-converge-post-copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3d4937f1939a6a2bd9117863d0eb22ae167e2b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e3d4937f1939a6a2bd9117863d0eb22ae167e2b9"}]},"branch":"refs/heads/master"},"7f49087c7647650e6d9aadad6ea207d5ea59f248":{"kind":"REWORK","_number":12,"created":"2019-06-13 05:53:19.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/12"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-06-13 05:52:59.000000000","tz":480},"subject":"Add spec for expose-auto-converge-post-copy","message":"Add spec for expose-auto-converge-post-copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7f49087c7647650e6d9aadad6ea207d5ea59f248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/7f49087c7647650e6d9aadad6ea207d5ea59f248"}]},"branch":"refs/heads/master"},"0265b46af092db55bc2f1aeb7e209716a46fc001":{"kind":"REWORK","_number":13,"created":"2019-06-25 21:39:24.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/81/651681/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/13"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-25 21:39:18.000000000","tz":-300},"subject":"Add spec for expose-auto-converge-post-copy","message":"Add spec for expose-auto-converge-post-copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0265b46af092db55bc2f1aeb7e209716a46fc001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/0265b46af092db55bc2f1aeb7e209716a46fc001"}]},"branch":"refs/heads/master"},"e5d24e8d356c070e7b7b951f64d429fe46f05560":{"kind":"REWORK","_number":14,"created":"2019-07-23 06:08:34.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/14"}}},"commit":{"parents":[{"commit":"d11692d1b540fca035fb21358409655f0181f9a1","subject":"Merge \"Support adding the reason behind a server lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/d11692d1b540fca035fb21358409655f0181f9a1"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-07-23 06:08:16.000000000","tz":480},"subject":"Add spec for expose-auto-converge-post-copy","message":"Add spec for expose-auto-converge-post-copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e5d24e8d356c070e7b7b951f64d429fe46f05560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/e5d24e8d356c070e7b7b951f64d429fe46f05560"}]},"branch":"refs/heads/master"},"de96fbae5ee6087c4f0e95b251c377bfd98ce558":{"kind":"REWORK","_number":15,"created":"2019-07-23 06:24:00.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/15"}}},"commit":{"parents":[{"commit":"c50c3e6412e4d6c0bea02737ad6dfc131b58907f","subject":"Spec: Provider config YAML file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c50c3e6412e4d6c0bea02737ad6dfc131b58907f"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-07-23 06:23:23.000000000","tz":480},"subject":"Add spec for expose-auto-converge-post-copy","message":"Add spec for expose-auto-converge-post-copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/de96fbae5ee6087c4f0e95b251c377bfd98ce558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/de96fbae5ee6087c4f0e95b251c377bfd98ce558"}]},"branch":"refs/heads/master"},"9037c2c83494a63b3dd4abd3598f01e51ac294ad":{"kind":"REWORK","_number":16,"created":"2019-07-23 06:30:50.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/81/651681/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/81/651681/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/81/651681/16"}}},"commit":{"parents":[{"commit":"c50c3e6412e4d6c0bea02737ad6dfc131b58907f","subject":"Spec: Provider config YAML file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c50c3e6412e4d6c0bea02737ad6dfc131b58907f"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-04-11 01:30:10.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-07-23 06:30:38.000000000","tz":480},"subject":"Add spec for expose-auto-converge-post-copy","message":"Add spec for expose-auto-converge-post-copy\n\nChange-Id: I08ab3fb105fd83a0b3095ea30a0f0518cfaa24ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9037c2c83494a63b3dd4abd3598f01e51ac294ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/9037c2c83494a63b3dd4abd3598f01e51ac294ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
