)]}'
{"id":"openstack%2Fnova-specs~652101","triplet_id":"openstack%2Fnova-specs~master~I6e40b81939e76e9a9a8a8d007c7225302ff570a6","project":"openstack/nova-specs","branch":"master","topic":"bp/image-metadata-prefiltering","hashtags":[],"change_id":"I6e40b81939e76e9a9a8a8d007c7225302ff570a6","subject":"Updates to image-metadata-prefiltering spec","status":"MERGED","created":"2019-04-12 15:53:56.000000000","updated":"2019-05-30 17:00:51.000000000","submitted":"2019-05-30 17:00:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652101-1559235651687-65a39237","meta_rev_id":"036cf9a4779815cd1d32d499ee9cb5adbe364b2e","_number":652101,"virtual_id_number":652101,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-04-12 16:37:41.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-05-30 17:00:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"recommended":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"all":[{"value":1,"date":"2019-04-12 17:15:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-04-12 18:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-04-12 17:33:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-05-30 16:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-04-12 22:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":1,"date":"2019-04-13 11:51:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-05-30 16:47:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-12 16:07:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-12 17:15:47.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-04-12 17:33:21.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-04-12 18:14:02.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-04-12 22:56:37.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-04-13 11:51:22.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-05-30 16:47:51.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-05-30 17:00:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"141f30764f06f8b1bae86ab5909a84d8f095a77b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-12 15:53:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ffc254c8c6f9727f33b04141cf740e77c409854","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-12 15:55:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0bb0234fc8017c5285b74d4d961df8debe17156b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-12 15:56:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"944c78c9d20061afcc85991e3eff89840fce8486","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 16:04:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/01/652101/2/check/openstack-tox-docs/35ec561/html/ : SUCCESS in 7m 48s\n- openstack-tox-pep8 http://logs.openstack.org/01/652101/2/check/openstack-tox-pep8/c9adea5/ : FAILURE in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"2c96dac925f7ae31f3641db29d6748688fff7884","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-12 16:07:44.000000000","message":"Patch Set 2: Code-Review+2\n\nThere\u0027s a trivial pep8 fix needed here. Feel free to carry my +2 if you resolve that today","accounts_in_message":[],"_revision_number":2},{"id":"ef251b6e738117729ac0dd8a286f19774d57b27f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-12 16:17:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ea6baa7626c87a01aa45e6ccea4de5ddf4cdc03","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-12 16:19:25.000000000","message":"Patch Set 3:\n\n\u003e There\u0027s a trivial pep8 fix needed here. Feel free to carry my +2 if you resolve that today\n\nDone.","accounts_in_message":[],"_revision_number":3},{"id":"07a830560102b3cd8b0c9ade96a17b4eb12b02e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 16:27:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/01/652101/3/check/openstack-tox-docs/5c3afcd/html/ : SUCCESS in 8m 24s\n- openstack-tox-pep8 http://logs.openstack.org/01/652101/3/check/openstack-tox-pep8/781dbd6/ : FAILURE in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"a91827820e5009245b2e9485d14fc2d985432839","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-12 16:37:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fde3e4086f16088639ae24cf021fb0115929338c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-12 16:37:41.000000000","message":"Patch Set 4:\n\ngeez, I don\u0027t remember pep line width being this strict on specs","accounts_in_message":[],"_revision_number":4},{"id":"1db9b5d4826fbe99b894e71f33f0751c779abd7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 16:53:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/01/652101/4/check/openstack-tox-docs/68550e1/html/ : SUCCESS in 7m 40s\n- openstack-tox-pep8 http://logs.openstack.org/01/652101/4/check/openstack-tox-pep8/4a92d02/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":4},{"id":"05edba26d665e56cebe7e757f79290f82d6a4a74","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-12 17:15:47.000000000","message":"Patch Set 4: Code-Review+1\n\nyep these are all good impovements to the spec.","accounts_in_message":[],"_revision_number":4},{"id":"8a5f4b5fb35a7160cab280355c5bd96a6bae0460","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-04-12 17:33:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7df5e542740c0bbd1e0e49873fd76797df4319a8","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-04-12 18:14:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"120ee9a711cf30e219da3916b360fd58c8211dc4","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-04-12 22:56:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6482663c1d3a6e2955d369a91dfc96b0f7dc241d","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-04-13 11:51:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"45b76e62af82246830a3ed97ad308c21084e2275","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-30 16:47:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLooks OK to me.","accounts_in_message":[],"_revision_number":4},{"id":"799a8fe95ee9cde4b46616e77e2a8e46fe9d7782","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 16:48:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6c9ae1009a4da78ce8075d1e96724c7c034a48ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 17:00:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3dc13fd1497b9d635b5c2290432601b0f7c3d3d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 17:00:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/01/652101/4/gate/openstack-tox-docs/f33b1ae/html/ : SUCCESS in 7m 21s\n- openstack-tox-pep8 http://logs.openstack.org/01/652101/4/gate/openstack-tox-pep8/aa2538a/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2fbe267e2680e9302d181d3370c440ec168026fd","revisions":{"2efe7be705331f21d1826e0d62b3ebae477f2398":{"kind":"REWORK","_number":1,"created":"2019-04-12 15:53:56.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/01/652101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/652101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/1"}}},"commit":{"parents":[{"commit":"b99846c68a5916f89ba77acaafd198e79497d4e1","subject":"Merge \"add spec for image metadata prefiltering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b99846c68a5916f89ba77acaafd198e79497d4e1"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-12 15:51:49.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-12 15:51:49.000000000","tz":-300},"subject":"Updates to image-metadata-prefiltering spec","message":"Updates to image-metadata-prefiltering spec\n\nAdds reference links to glance metadata definitions and the companion\nos-traits StoryBoard story.\n\nAlso fixes some typos and spruces up the formatting in a couple of\nplaces.\n\nChange-Id: I6e40b81939e76e9a9a8a8d007c7225302ff570a6\nBlueprint: image-metadata-prefiltering\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2efe7be705331f21d1826e0d62b3ebae477f2398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2efe7be705331f21d1826e0d62b3ebae477f2398"}]},"branch":"refs/heads/master"},"a2ab088ef68e778694213e74b2f53bd8c9ca39e1":{"kind":"REWORK","_number":2,"created":"2019-04-12 15:55:47.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/01/652101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/652101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/2"}}},"commit":{"parents":[{"commit":"b99846c68a5916f89ba77acaafd198e79497d4e1","subject":"Merge \"add spec for image metadata prefiltering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b99846c68a5916f89ba77acaafd198e79497d4e1"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-12 15:51:49.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-12 15:55:45.000000000","tz":-300},"subject":"Updates to image-metadata-prefiltering spec","message":"Updates to image-metadata-prefiltering spec\n\nAdds reference links to glance metadata definitions and the companion\nos-traits StoryBoard story.\n\nAlso fixes some typos and spruces up the formatting in a couple of\nplaces.\n\nChange-Id: I6e40b81939e76e9a9a8a8d007c7225302ff570a6\nBlueprint: image-metadata-prefiltering\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2ab088ef68e778694213e74b2f53bd8c9ca39e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a2ab088ef68e778694213e74b2f53bd8c9ca39e1"}]},"branch":"refs/heads/master"},"660c544e194d996033e09cb8f89593dece8ca39b":{"kind":"REWORK","_number":3,"created":"2019-04-12 16:17:57.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/01/652101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/652101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/3"}}},"commit":{"parents":[{"commit":"b99846c68a5916f89ba77acaafd198e79497d4e1","subject":"Merge \"add spec for image metadata prefiltering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b99846c68a5916f89ba77acaafd198e79497d4e1"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-12 15:51:49.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-12 16:17:55.000000000","tz":-300},"subject":"Updates to image-metadata-prefiltering spec","message":"Updates to image-metadata-prefiltering spec\n\nAdds reference links to glance metadata definitions and the companion\nos-traits StoryBoard story.\n\nAlso fixes some typos and spruces up the formatting in a couple of\nplaces.\n\nChange-Id: I6e40b81939e76e9a9a8a8d007c7225302ff570a6\nBlueprint: image-metadata-prefiltering\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/660c544e194d996033e09cb8f89593dece8ca39b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/660c544e194d996033e09cb8f89593dece8ca39b"}]},"branch":"refs/heads/master"},"2fbe267e2680e9302d181d3370c440ec168026fd":{"kind":"REWORK","_number":4,"created":"2019-04-12 16:37:09.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/01/652101/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/01/652101/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/01/652101/4"}}},"commit":{"parents":[{"commit":"b99846c68a5916f89ba77acaafd198e79497d4e1","subject":"Merge \"add spec for image metadata prefiltering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b99846c68a5916f89ba77acaafd198e79497d4e1"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-12 15:51:49.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-12 16:37:07.000000000","tz":-300},"subject":"Updates to image-metadata-prefiltering spec","message":"Updates to image-metadata-prefiltering spec\n\nAdds reference links to glance metadata definitions and the companion\nos-traits StoryBoard story.\n\nAlso fixes some typos and spruces up the formatting in a couple of\nplaces.\n\nChange-Id: I6e40b81939e76e9a9a8a8d007c7225302ff570a6\nBlueprint: image-metadata-prefiltering\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2fbe267e2680e9302d181d3370c440ec168026fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/2fbe267e2680e9302d181d3370c440ec168026fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
