)]}'
{"id":"openstack%2Fnova-specs~655717","triplet_id":"openstack%2Fnova-specs~master~I615aa63a4bfe367dc93bf6805ea0038591eebd8b","project":"openstack/nova-specs","branch":"master","topic":"bp/amd-sev-libvirt-support","hashtags":[],"change_id":"I615aa63a4bfe367dc93bf6805ea0038591eebd8b","subject":"Update SEV work item to new approach based on MEM_ENCRYPTION_CONTEXT","status":"MERGED","created":"2019-04-25 15:00:13.000000000","updated":"2019-05-28 17:58:33.000000000","submitted":"2019-05-28 17:58:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655717-1559066313559-f81bca3b","meta_rev_id":"029898ce28c366ace7c6db222d4eb7daa53962d5","_number":655717,"virtual_id_number":655717,"owner":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"date":"2019-05-28 12:04:26.000000000","_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},{"value":2,"date":"2019-05-28 17:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-04-25 15:27:18.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":2,"date":"2019-05-28 12:28:29.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"date":"2019-05-01 06:08:03.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-05-28 17:47:41.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 15:01:09.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"REVIEWER"},{"updated":"2019-04-25 15:01:19.000000000","updated_by":{"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},"reviewer":{"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},"state":"REVIEWER"},{"updated":"2019-05-01 06:08:03.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-05-28 12:28:29.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2019-05-28 17:47:41.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-28 17:58:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98e2890ef9ffd3187c2608b9c659d6a58d764c0e","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 15:00:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"429533b3356cec316c8ab0231d7cdc5f2a7e26b6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 15:03:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"685ba9499a65c86b9978700240438728e8135b45","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 15:21:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d685b38afb911f721f6ab8bab6e23c2de8083142","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:21:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/17/655717/1/check/openstack-tox-docs/21528a6/html/ : SUCCESS in 7m 36s\n- openstack-tox-pep8 http://logs.openstack.org/17/655717/1/check/openstack-tox-pep8/9641aa8/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6251b2bf1e4df4716b8c681ff9bc2da2219f0e2a","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 15:22:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd085c2edb6d282aa0735d88f34fa3e1ecf33b56","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 15:27:18.000000000","message":"Patch Set 2: Code-Review+2\n\n++","accounts_in_message":[],"_revision_number":2},{"id":"066a44c28a74cc8b737dd576443645674c623fd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:38:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/17/655717/2/check/openstack-tox-docs/7cfe6d1/html/ : SUCCESS in 7m 40s\n- openstack-tox-pep8 http://logs.openstack.org/17/655717/2/check/openstack-tox-pep8/ad41b69/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f053e825649722680331515b02fdcb97f7e226e5","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-05-28 12:04:26.000000000","message":"Patch Set 2:\n\n@Jay, you got a second to review this tiny tweak?","accounts_in_message":[],"_revision_number":2},{"id":"6d5e0d1fd6cae364b965473f0d2ba61d29957bcc","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-05-28 12:28:29.000000000","message":"Patch Set 2: Code-Review+2\n\nSure, lgtm","accounts_in_message":[],"_revision_number":2},{"id":"4f9bb51dbf23c28c47e8547a55c662b8dde658ff","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-28 17:47:41.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"001d4bd91320863ce95d2280386e403e95efbaab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 17:47:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"446a1d0c3bf7c5ca6d2ffffb70aac7329cc1702c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 17:58:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/17/655717/2/gate/openstack-tox-docs/cf87fb7/html/ : SUCCESS in 7m 38s\n- openstack-tox-pep8 http://logs.openstack.org/17/655717/2/gate/openstack-tox-pep8/bc2491e/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b773181593477f6407b81b798207ac8181471f80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 17:58:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8d5531fa7f44774aa3778215afef98f6f87905e","revisions":{"381277185645990a9123740809dd1717006e79f0":{"kind":"REWORK","_number":1,"created":"2019-04-25 15:00:13.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/17/655717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/17/655717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/17/655717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/17/655717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/17/655717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/17/655717/1"}}},"commit":{"parents":[{"commit":"a4a3d991700630573fc8b0f5cea066814a8da120","subject":"Merge \"Re-approve AMD SEV support for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a4a3d991700630573fc8b0f5cea066814a8da120"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 11:33:22.000000000","tz":60},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 14:58:55.000000000","tz":60},"subject":"Update SEV work item to new approach based on MEM_ENCRYPTION_CONTEXT","message":"Update SEV work item to new approach based on MEM_ENCRYPTION_CONTEXT\n\nBetween the Stein and Train versions of the AMD SEV spec, the design\npivoted from using a trait to using a resource class.  However one\nwork item was missed from the switch, so fix it.\n\nThe significant point here is that the original RequestSpec doesn\u0027t\nnecessarily contain the SEV trait, e.g. if the user or operator has no\npreference whether to boot the VM using SEV or some competing\ntechnology.  In this case the placement service will ensure that\nlaunch happens on a host which is capable of *some* kind of memory\nencryption, and then later at spawn-time the virt driver can figure\nout which technology can be used to deliver it.  This work item covers\nthe latter stage.\n\nChange-Id: I615aa63a4bfe367dc93bf6805ea0038591eebd8b\nblueprint: amd-sev-libvirt-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/381277185645990a9123740809dd1717006e79f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/381277185645990a9123740809dd1717006e79f0"}]},"branch":"refs/heads/master"},"b8d5531fa7f44774aa3778215afef98f6f87905e":{"kind":"REWORK","_number":2,"created":"2019-04-25 15:21:19.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/17/655717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/17/655717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/17/655717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/17/655717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/17/655717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/17/655717/2"}}},"commit":{"parents":[{"commit":"a4a3d991700630573fc8b0f5cea066814a8da120","subject":"Merge \"Re-approve AMD SEV support for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/a4a3d991700630573fc8b0f5cea066814a8da120"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 11:33:22.000000000","tz":60},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 15:16:31.000000000","tz":60},"subject":"Update SEV work item to new approach based on MEM_ENCRYPTION_CONTEXT","message":"Update SEV work item to new approach based on MEM_ENCRYPTION_CONTEXT\n\nBetween the Stein and Train versions of the AMD SEV spec, the design\npivoted from using a trait to using a resource class.  However one\nwork item was missed from the switch, so fix it.\n\nThe significant point here is that the original RequestSpec doesn\u0027t\nnecessarily contain the SEV trait, e.g. if the user or operator has no\npreference whether to boot the VM using SEV or some competing\ntechnology.  In this case the placement service will ensure that\nlaunch happens on a host which is capable of *some* kind of memory\nencryption, and then later at spawn-time the virt driver can figure\nout which technology can be used to deliver it.  This work item covers\nthe latter stage.\n\nChange-Id: I615aa63a4bfe367dc93bf6805ea0038591eebd8b\nblueprint: amd-sev-libvirt-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b8d5531fa7f44774aa3778215afef98f6f87905e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/b8d5531fa7f44774aa3778215afef98f6f87905e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
