)]}'
{"id":"openstack%2Fnova-specs~660560","triplet_id":"openstack%2Fnova-specs~master~Ib06dab6480a89667a7b483c6b4b15824b760e441","project":"openstack/nova-specs","branch":"master","topic":"bp/add-host-and-hypervisor-hostname-flag-to-create-server","hashtags":[],"change_id":"Ib06dab6480a89667a7b483c6b4b15824b760e441","subject":"Update \"add host and hypervisor_hostname flag to create server\" spec","status":"ABANDONED","created":"2019-05-22 05:21:02.000000000","updated":"2019-05-24 02:14:18.000000000","total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b097d4f7f5ab37711ba561a356368b8e2ba086c","_number":660560,"virtual_id_number":660560,"owner":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-22 14:47:45.000000000","_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":1,"date":"2019-05-22 15:12:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":-1,"date":"2019-05-22 20:49:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 05:42:17.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-22 09:12:59.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-05-22 15:12:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-22 20:49:55.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"}],"messages":[{"id":"94f49e4297146174ba9ef44b6e63bce9671fcd53","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2019-05-22 05:21:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"389f052733c153e5f7dbda56dcdce6b8934f147e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2019-05-22 05:21:40.000000000","message":"Topic set to bp/add-host-and-hypervisor-hostname-flag-to-create-server","accounts_in_message":[],"_revision_number":1},{"id":"8843595f3fd1a5cf51bc7580741566d938f9ee92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 05:30:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/60/660560/1/check/openstack-tox-docs/ebc8f27/html/ : SUCCESS in 7m 10s\n- openstack-tox-pep8 http://logs.openstack.org/60/660560/1/check/openstack-tox-pep8/5e6891c/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b41bf9fae96de5d44191442a044e9c728330937c","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2019-05-22 05:40:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ea36c4a87ae27a6be3e0bd6541c9ccb2d1bd96e0","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2019-05-22 08:52:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff09b49c9707d3dbe2ee9f6c366e17dc15a132d2","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-05-22 09:12:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"65bff41e8904d1a0575ef4f2586d35b78b5f4f9e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2019-05-22 09:50:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3fc787ab2e0089b74c72619f5dd5cfa97c45c35","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2019-05-22 09:50:45.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"99e8221db654e26b49634a796de8118f364e83c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:00:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/60/660560/2/check/openstack-tox-docs/62e603e/html/ : SUCCESS in 6m 42s\n- openstack-tox-pep8 http://logs.openstack.org/60/660560/2/check/openstack-tox-pep8/6bd50a7/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b97f8330bc6f224e5750e9789c2f7089b4aca6db","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-05-22 13:02:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c452d85ef9c1144b232b34f0ecf872c6a74c8c7e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-22 13:58:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0b76b0bc904b84e762df879cd8e35b374ba315f0","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2019-05-22 14:47:27.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a7aa7b5b196450f008986826905cd022f4f76269","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2019-05-22 14:47:45.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"46f8a19e0f5a3e1c8c4c2269941c9d0ee60e9c23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 15:12:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/60/660560/3/check/openstack-tox-docs/36ebea0/html/ : SUCCESS in 7m 16s\n- openstack-tox-pep8 http://logs.openstack.org/60/660560/3/check/openstack-tox-pep8/4122a55/ : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":3},{"id":"c60f9641a75ce408fb223460b58ac3ddad41cfbd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-22 20:49:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2b3b4ef6e8d362529cfeede5660705c5a9cff07","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-22 20:49:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"42684b95e0f2ad9bf4bfef255e00d46347571a8a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-22 20:51:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"27bacdf67e02dd96072a480bd690ab530628f50b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-22 21:33:01.000000000","message":"Patch Set 3:\n\nI\u0027ve added this to the nova meeting agenda:\n\nhttps://wiki.openstack.org/wiki/Meetings/Nova#Agenda_for_next_meeting","accounts_in_message":[],"_revision_number":3},{"id":"25a681d97a51b2779605804a097b5bc1d9bd4a0f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-22 22:14:50.000000000","message":"Patch Set 3:\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-May/006518.html","accounts_in_message":[],"_revision_number":3},{"id":"42811efffbd73b9c2b9c91ca9aae2d5ead6c10f4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-23 22:12:02.000000000","message":"Patch Set 3:\n\nI think we can drop this now given the decision in the ML thread, right?\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-May/006584.html","accounts_in_message":[],"_revision_number":3},{"id":"0b097d4f7f5ab37711ba561a356368b8e2ba086c","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2019-05-24 02:14:18.000000000","message":"Abandoned\n\nWill still validate host/node in API layer.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"41ba31baf39bd6c549447b9b4f436b760d3a50d3","revisions":{"3644dc5097cdfe5e985476afe75c4ad5d04749a4":{"kind":"REWORK","_number":1,"created":"2019-05-22 05:21:02.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/60/660560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/60/660560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/1"}}},"commit":{"parents":[{"commit":"c4d28f549d469ecfc4ae2d51d38f4fc6eb84235b","subject":"Merge \"Support server power state update through external event\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4d28f549d469ecfc4ae2d51d38f4fc6eb84235b"}]}],"author":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2019-05-22 04:02:16.000000000","tz":480},"committer":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2019-05-22 05:20:19.000000000","tz":480},"subject":"Update \"add host and hypervisor_hostname flag to create server\" spec","message":"Update \"add host and hypervisor_hostname flag to create server\" spec\n\nThe spec is updated based on that something has been implemented.\n\nChange-Id: Ib06dab6480a89667a7b483c6b4b15824b760e441\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3644dc5097cdfe5e985476afe75c4ad5d04749a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/3644dc5097cdfe5e985476afe75c4ad5d04749a4"}]},"branch":"refs/heads/master"},"cdb3a95d99b5179b4652a7253a502747fb846bba":{"kind":"REWORK","_number":2,"created":"2019-05-22 09:50:11.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/60/660560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/60/660560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/2"}}},"commit":{"parents":[{"commit":"c4d28f549d469ecfc4ae2d51d38f4fc6eb84235b","subject":"Merge \"Support server power state update through external event\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4d28f549d469ecfc4ae2d51d38f4fc6eb84235b"}]}],"author":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2019-05-22 04:02:16.000000000","tz":480},"committer":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2019-05-22 09:47:15.000000000","tz":480},"subject":"Update \"add host and hypervisor_hostname flag to create server\" spec","message":"Update \"add host and hypervisor_hostname flag to create server\" spec\n\nThe spec is updated based on that something has been implemented.\n\nChange-Id: Ib06dab6480a89667a7b483c6b4b15824b760e441\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cdb3a95d99b5179b4652a7253a502747fb846bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/cdb3a95d99b5179b4652a7253a502747fb846bba"}]},"branch":"refs/heads/master"},"41ba31baf39bd6c549447b9b4f436b760d3a50d3":{"kind":"REWORK","_number":3,"created":"2019-05-22 14:47:27.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/60/660560/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova-specs","ref":"refs/changes/60/660560/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova-specs refs/changes/60/660560/3"}}},"commit":{"parents":[{"commit":"c4d28f549d469ecfc4ae2d51d38f4fc6eb84235b","subject":"Merge \"Support server power state update through external event\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/c4d28f549d469ecfc4ae2d51d38f4fc6eb84235b"}]}],"author":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2019-05-22 04:02:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2019-05-22 14:46:23.000000000","tz":0},"subject":"Update \"add host and hypervisor_hostname flag to create server\" spec","message":"Update \"add host and hypervisor_hostname flag to create server\" spec\n\nThe spec is updated based on that something has been implemented.\n\nChange-Id: Ib06dab6480a89667a7b483c6b4b15824b760e441\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/41ba31baf39bd6c549447b9b4f436b760d3a50d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova-specs/commit/41ba31baf39bd6c549447b9b4f436b760d3a50d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
